)]}'
{"id":"openstack%2Fmanila~763387","triplet_id":"openstack%2Fmanila~master~Ib46402414e8ed9b63f024313500aef85f0c47a41","project":"openstack/manila","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ib46402414e8ed9b63f024313500aef85f0c47a41","subject":"Introduce base personas for secure RBAC","status":"MERGED","created":"2020-11-19 14:50:25.000000000","updated":"2020-11-24 13:22:53.000000000","submitted":"2020-11-23 22:44:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763387-secure-rbac","meta_rev_id":"63ab401f04094cf77c927472e80b48c97cae91ca","_number":763387,"virtual_id_number":763387,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-20 04:55:23.000000000","_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-11-20 05:29:41.000000000","_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-23 22:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2020-11-19 22:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":2,"date":"2020-11-23 21:44:04.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"value":0,"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"value":1,"date":"2020-11-23 21:44:04.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 14:50:25.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-19 18:26:13.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 20:38:12.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-19 22:04:18.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2020-11-20 03:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 04:07:26.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 04:55:23.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 05:29:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 21:44:04.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"7b0a6ced3c20cfb5715e5389e207f3e5083b8055","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 14:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41d2620def1b51b6a73d3c6d98d06a8a25a0ba6f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 14:50:47.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"93844e6f0c63ff2f4531f09f910a0fa18ef9f1c1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 16:37:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/87/763387/1/check/DellEMC_Unity_Manila_dhss_true/ddb6447/DellEMC_Unity_Manila_dhss_true/47 : FAILURE in 51m 47s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/87/763387/1/check/DellEMC_Unity_Manila_dhss_false/5871d9c/DellEMC_Unity_Manila_dhss_false/46 : SUCCESS in 55m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f53f1a77f1e5f5ef10747eb02696988dcd52724f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 17:25:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4c335c61e844a28d9364cb97ce60f3 : SUCCESS in 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66a807a07060453f8a68a3d7dee837b0 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f0209bdb67947a6961608af14f39e95 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6ee06cd23a453ca947938d44d2afd8 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3da02f0bc5954f3bb2020b15abca5621 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a48a5e61151493cbed3c5641b8233a8 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7e358b5606c4b8e80f61fc2cd55b5c7 : SUCCESS in 8m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2be508d48d8e4f608f93e150abb6800c : SUCCESS in 4m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1a4020e4000449bfbb4af314cd5bda39 : SUCCESS in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f826f968e7d34eda9fe845d94be3700b : FAILURE in 36m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/996334ffebe74ab6bd12d7ce9629e4dc : SUCCESS in 49m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7026d9c1e4194573b157dd8a34343e97 : SUCCESS in 1h 16m 21s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2bcc1bbbb5cb490c93fe4a000273c09a : SUCCESS in 1h 19m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a90b4d42c01b4aaf90f529feb1d05597 : FAILURE in 1h 13m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8b6ca797c01d4eb2a818bf82a331dfbd : SUCCESS in 46m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fb2faaead6bb4ad092da1758194a31b0 : FAILURE in 2h 28m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/34aed156a15b4d488588327860c759d4 : FAILURE in 32m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5be0c31bd7694ec3bd78c82737ca6aa4 : SUCCESS in 24m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7d6f788b1f93457099e6ba5741dbddb3 : SUCCESS in 1h 04m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/919346dac6c14b329e49fdb7f49e2151 : SUCCESS in 55m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e48ba585890443a68210e8950db40046 : FAILURE in 35m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e1dc45fbdc754f4c832054b06d6ae461 : FAILURE in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5618a182b364bb835deb8dbfe7bf2e6d9aa82048","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 17:33:34.000000000","message":"Patch Set 1:\n\nrecheck\n\nknown sporadic CI failure","accounts_in_message":[],"_revision_number":1},{"id":"3c0ba590da4ad69812cb1fbe67dfc2e0ad54083c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 17:45:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(7 comments)\n\nThanks for submitting this change, Lance - i\u0027ve some doc comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"e522763a3894d85d143ddbb4b5b25c1753480da9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 18:26:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/87/763387/1/check/DellEMC_PowerMAX_Manila/1252526/DellEMC_PowerMAX_Manila/55 : SUCCESS in 1h 51m 08s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"112f08735093b4a2d073baa837f9a746cbd97a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 19:39:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9892314d8a4938bc83b5d9bca53997 : SUCCESS in 12m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abd7890c38fd4973ba0962ae0c3287d9 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae855621b474a60bedd82abba2ea684 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e70ec414aee41c79f4da2af7a772b02 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e9b1ca2ca154114812919ad27d7704f : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dbab79ca4ce4fbe91586dc846c5114b : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce6e1ac85c040eda0156825d2ac55e9 : SUCCESS in 9m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5580c39db26f41d5bc41d0416d0a42e4 : SUCCESS in 5m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98311b437e854e4490add9a6e451dc82 : SUCCESS in 4m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/982b8a02f1204c4caf2fa660bbc42226 : SUCCESS in 36m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0be46861f0ad460fbd2f5d07ff788474 : SUCCESS in 45m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/912a279910c24096a300b57a86c8d24a : SUCCESS in 1h 22m 01s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/198bee4185eb4bd8b956e5119eb74f98 : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0a202bbace0e48b2b4a5b8381d965f96 : FAILURE in 1h 00m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3bca0fbb08c04eecbadf854a0a6c6615 : SUCCESS in 1h 05m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5c6e9c41b0944a0087690f5c4d0ab662 : FAILURE in 1h 39m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0616e8649109458480a2788ae46f7e1f : FAILURE in 23m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c2d3eec50bd44c379b5d71fd03f9aa4d : SUCCESS in 26m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/077db5116f5048548563a28d614fa83d : SUCCESS in 1h 12m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cc69af26798c4e74b962931c9432e5e8 : SUCCESS in 57m 35s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/73399c7be1cc4633843dd56bafc7fc01 : FAILURE in 34m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2db62f6d5fff4582910852579e8ec5ce : FAILURE in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"171435ea4bf6c16ed76f1e246a376bb787006223","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-19 20:38:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/87/763387/1/check/DellEMC_Unity_Manila_dhss_true/d3332a1/DellEMC_Unity_Manila_dhss_true/49 : FAILURE in 56m 13s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/87/763387/1/check/DellEMC_Unity_Manila_dhss_false/a9ec328/DellEMC_Unity_Manila_dhss_false/48 : SUCCESS in 58m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec6bf9994797ab6ed3ec9eb9496ecb26a28be8bc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-19 20:58:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/87/763387/1/check/DellEMC_VNX_Manila/e87f35a/DellEMC_VNX_Manila/58 : SUCCESS in 1h 18m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"304ed8de6fb553449b7ac6a9e98048fdbae70bae","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-19 21:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/87/763387/1/upstream-check/manila-cDOT-ss/edb7abd/ : SUCCESS in 1h 49m 52s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/87/763387/1/upstream-check/manila-cDOT-no-ss/66ae988/ : SUCCESS in 1h 41m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b1c098ba79b9a5fbef94cf76b81b721621eea868","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:44:10.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"653a1e4ccc6f6e405989d831f283a10a64820213","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:44:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94f252586c5d66db5f9f27e7b783282b5530eb24","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-19 22:04:18.000000000","message":"Patch Set 2: Code-Review+2\n\nThis looks good to me - and doesn\u0027t affect anything yet, so lets merge this and work through the rest of the changes :)","accounts_in_message":[],"_revision_number":2},{"id":"26746e394af137e6ae7221b0d70b882d08b9368d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 03:35:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919d350a896b4bd4818f7edb3a2e6859 : SUCCESS in 10m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82879d72197c4c41810bc52e0614e479 : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7ca8d8e54494a77ad0a7a7a5d316ab1 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bdb205a05fa4b51913fad0ce19e154b : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9964593e9eaf4a7d9506105442d27206 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/922ff603a9b84da999ef85efa85f22f5 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e03774711efd4685a8e9356d8d557960 : SUCCESS in 6m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/717f28abe4de41d695b23d9a2e691264 : SUCCESS in 5m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dae86b6ab6994a2db123a9d9d542505d : SUCCESS in 5m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd0909b399ac4ed696e2dcc1914a9994 : SUCCESS in 31m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4033a97638c4b09bf3aa30c2ed332b4 : SUCCESS in 40m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f9c35a2eb25d4200aa493099e3da545b : SUCCESS in 1h 05m 23s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7a1da208e463412cb5b78b0f81a4c119 : SUCCESS in 1h 09m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8acac5e171e748a4a0093b5319febcb1 : FAILURE in 57m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f947fc60b5714563ae2c522bc1709048 : SUCCESS in 42m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9e547c3b95c4ec2ae9561b34867fa8e : FAILURE in 2h 01m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/32e4c750298b49b594f64919ced54ce9 : FAILURE in 34m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8c68a694b92b4e7c9a295333a9fde263 : SUCCESS in 35m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/efe2bf89788c473ca58df68b9cf497c3 : SUCCESS in 1h 08m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e6fd28730493489db245078d298a0645 : SUCCESS in 53m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1f39430fc27647dfa72c515689d9b3c5 : FAILURE in 41m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/db8bfddfd6894296b9e4557b6832ce5f : FAILURE in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab9879e501edb73a25fabcb22c8d80a263878436","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-20 04:07:26.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-87-763387-1/ns5_manila : SUCCESS in 5h 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"cf289229568c6f0cfc6e36ac9d44608578c76cd3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-20 04:55:23.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/87/763387/2/check/DellEMC_VNX_Manila/a7b3dd9/DellEMC_VNX_Manila/61 : SUCCESS in 1h 19m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"deeca74c845d831111d68c03dc1c7a6b96665934","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-11-20 05:29:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/87/763387/2/upstream-check/manila-cDOT-ss/d20f574/ : POST_FAILURE in 1h 49m 13s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/87/763387/2/upstream-check/manila-cDOT-no-ss/eae6d25/ : POST_FAILURE in 1h 53m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"72fbfbc7e8eb1ec76b86a36bb705024c548c5825","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 09:31:50.000000000","message":"Patch Set 2:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-87-763387-2/ns5_manila : FAILURE in 56h 43m 53s","accounts_in_message":[],"_revision_number":2},{"id":"1b3a057f3dd275ff57e6b7c09ec414a885564c3d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-11-23 21:44:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good, thanks Lance!","accounts_in_message":[],"_revision_number":2},{"id":"4a8f79688d4a07e8c452b04fd29a2751b8f9faec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 21:44:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a36776009a1c32b0c6c47ab56c418e6d43115129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 22:44:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98540cd1466b4a0fb3eeed5a94991792 : SUCCESS in 9m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cc06d7953af4581b205c0005267683c : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487ff3fc47ee423fa7b1d41f99e92f1d : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0744bd40717843f69b0b7fd6a0440bdd : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f672c9197e94e59ae58c02221bcccde : SUCCESS in 13m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5ff63a272dd74ed099acf2bee44bcdf7 : SUCCESS in 29m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/27ab6f952bd648c485bc856043418f2c : SUCCESS in 35m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6ab99c1387b34b1da9b62e269063674a : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":2},{"id":"fe0328ba1dcdf7fefdd40dab63bb44c364fb8a29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 22:44:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ff347418fcfbd169336e598354125af34b77b178","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 22:46:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f975e84275409bb9886c78f8538756 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2},{"id":"63ab401f04094cf77c927472e80b48c97cae91ca","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-24 13:22:53.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-87-763387-2/ns5_manila : SUCCESS in 2h 5m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a8aaf43adf6fc4bec5095346624c0844ef59855","revisions":{"00cb3e6891f07a0510c40b7fd5fd910daa5d7c54":{"kind":"REWORK","_number":1,"created":"2020-11-19 14:50:25.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/87/763387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/763387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/763387/1"}}},"commit":{"parents":[{"commit":"a10927c6bef0786f28f861fd561d550b23751fab","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a10927c6bef0786f28f861fd561d550b23751fab"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 14:46:20.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 14:46:20.000000000","tz":0},"subject":"Introduce base personas for secure RBAC","message":"Introduce base personas for secure RBAC\n\nThis commit adds some really basic check strings that we can re-use in\nthe existing policies. These checks strings implement a few of the most\ncommon and useful personas. We\u0027re also trying to implement them across\nOpenStack consistently, which will improve user experience and security.\n\nChange-Id: Ib46402414e8ed9b63f024313500aef85f0c47a41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00cb3e6891f07a0510c40b7fd5fd910daa5d7c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00cb3e6891f07a0510c40b7fd5fd910daa5d7c54"}]},"branch":"refs/heads/master"},"2a8aaf43adf6fc4bec5095346624c0844ef59855":{"kind":"REWORK","_number":2,"created":"2020-11-19 21:44:22.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/87/763387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/763387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/763387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/763387/2"}}},"commit":{"parents":[{"commit":"a10927c6bef0786f28f861fd561d550b23751fab","subject":"Bump oslo.log version to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a10927c6bef0786f28f861fd561d550b23751fab"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 14:46:20.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 21:41:35.000000000","tz":0},"subject":"Introduce base personas for secure RBAC","message":"Introduce base personas for secure RBAC\n\nThis commit adds some really basic check strings that we can re-use in\nthe existing policies. These checks strings implement a few of the most\ncommon and useful personas. We\u0027re also trying to implement them across\nOpenStack consistently, which will improve user experience and security.\n\nChange-Id: Ib46402414e8ed9b63f024313500aef85f0c47a41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a8aaf43adf6fc4bec5095346624c0844ef59855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a8aaf43adf6fc4bec5095346624c0844ef59855"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
