)]}'
{"id":"openstack%2Fmanila~763447","triplet_id":"openstack%2Fmanila~master~I340f63874af5783099ed6b353be61a2909829343","project":"openstack/manila","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I340f63874af5783099ed6b353be61a2909829343","subject":"Implement secure RBAC for services","status":"MERGED","created":"2020-11-19 21:17:02.000000000","updated":"2021-03-09 17:46:05.000000000","submitted":"2021-03-08 23:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763447-secure-rbac","meta_rev_id":"26b8dc6ce4b741fc8dd94d07f8646c58ed9471a9","_number":763447,"virtual_id_number":763447,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-08 23:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-03-07 13:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-03-06 20:58:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2021-03-08 17:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":1,"date":"2021-03-08 17:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 21:17:02.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-19 23:36:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 03:16:07.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-22 09:31:02.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-12 07:10:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-03 04:50:45.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 05:31:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-06 20:58:10.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-03-07 13:06:47.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-03-08 19:30:53.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"CC"},{"updated":"2021-03-09 17:46:05.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"bf04645b272a480d23fb7db283baca387c969dd8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58d64e9fd3b918c2f9590a1380b1967427c0a180","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:19:08.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"6e994ce2f9dafb669517d6ec5c84e0d0b8d84f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 23:36:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e30167c5714c83ae5eac43f857df0b : SUCCESS in 12m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67183615f1d2422dae01037c0b057aed : FAILURE in 12m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f28cb9b2b054607b6d0d38b11cafdc3 : FAILURE in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9b818f51f744168fdcfdb74562f62c : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fece07a4bbc4d9fb91ce00a37075302 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/615bf83f1c8541e4997c6cf69e6e4a06 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46e2b470cc004b80b7d231de6c1f26a0 : SUCCESS in 7m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0748c54207df47728bc44bf7754fca03 : SUCCESS in 5m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0d3b7c0ceb9046bb90efe2693a48a4e8 : SUCCESS in 5m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6c5d73f9cc0f4ca6bf7f9664c91ded79 : SUCCESS in 34m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4b5c9a4fe8c04426a342691a78a6881d : SUCCESS in 41m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/03ee65e1ef354a038fd8debf6a459ff3 : SUCCESS in 1h 07m 38s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b062776b00c04831a3fdc3ab6e422208 : SUCCESS in 58m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f9b8553395224e8eb1af3407b6b57428 : FAILURE in 1h 01m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2092ddfd49ae4014b78e1941951aaddd : SUCCESS in 47m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c33f82d216042bf87f4f25280746117 : FAILURE in 1h 50m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/157e6e06238346e3a5c56c794e09fbe6 : FAILURE in 27m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/064beffda63c4622a6d9de9a55d234af : FAILURE in 24m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ef7ba931f1fc4b89a2e4cf8e49d2331c : SUCCESS in 1h 21m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/aa4f46f883cb412f8b9a4a5edcfcfefb : SUCCESS in 50m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f5256afb99be4312a942da72ab250c6e : FAILURE in 48m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ac3bce95f1ec4532855bbaacb51d3994 : FAILURE in 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a69b297d5971af1add28eed399947ad796c3d250","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-20 03:16:07.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/763447/1/check/DellEMC_Unity_Manila_dhss_true/36a2e34/DellEMC_Unity_Manila_dhss_true/57 : FAILURE in 57m 13s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/763447/1/check/DellEMC_Unity_Manila_dhss_false/32b4408/DellEMC_Unity_Manila_dhss_false/56 : SUCCESS in 53m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2eaf96f7fc582523832916956a681a883107d07","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 09:31:02.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-47-763447-1/ns5_manila : FAILURE in 56h 44m 27s","accounts_in_message":[],"_revision_number":1},{"id":"61d988051c92e62df7c1e3b74c24ebf363959f84","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-24 09:21:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-47-763447-1/ns5_manila : SUCCESS in 2h 46m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f66c6e06e4da0a4cb72aaed22fa8a41c59bdfd92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-12 07:10:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"666acb0c90e33b394b5e2a2739694981f2ac1202","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:35:37.000000000","message":"Topic change-763469-1 removed","accounts_in_message":[],"_revision_number":2},{"id":"4dc98308da6e99b7874221359cc5786869c502b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:35:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"ced4d8384a1e27bacca756298bc515d6bb8113ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 04:15:41.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-47-763447-2/ns5_manila : FAILURE in 1h 51m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9d323ec2be4164ad7ad5fedab27a3ed856e2abf9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-19 06:45:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9cdd89a8ac24500d1524fc716b61ffe8185fc047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-03 01:44:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce2c076b4d38513b1b5a861559b63dd7d7d4ea34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 03:40:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0d1d3c876740afbc62e78d63e849bb : SUCCESS in 10m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c51bf4a58e749278b5f606613771d44 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6ef8e681c334443bdfe15b972068f20 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b2b416255c4293af46ec2f1a0246a3 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4766d7c406714aec93657f0bd32258a6 : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85cf9520a6424f14ae1f21c5047b2810 : SUCCESS in 9m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b08ddc46dcb6450ebebb8c4b67b9647c : SUCCESS in 9m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4bb0dae3620e432abd51b74c54fdfff5 : SUCCESS in 6m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4e79944fd24e454488e203aef1292a1d : SUCCESS in 5m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a154fbfce1b441daa8fd88831f39c8a1 : SUCCESS in 27m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/280ee46189ca4cecbd3954f9cc379d7e : SUCCESS in 32m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/44020b87aa76418ab9db3532a6443ca0 : SUCCESS in 1h 03m 54s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5d2e64f22d714e5d8ff99a425cf0a672 : SUCCESS in 1h 06m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c080d8a5976e469d9796a9b6014445d4 : SUCCESS in 1h 09m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78853adbe20241bcb51486877b9fcf8c : SUCCESS in 43m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a9b335f6007f41269c1c48b5990f27c9 : FAILURE in 1h 54m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/63415022f57646c893e62af393af745b : FAILURE in 22m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ad0ac64c87e2458188d25b528eeedf51 : SUCCESS in 23m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/611b6fa5d40a42abacaf52c6ab7fed08 : SUCCESS in 1h 10m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6fcba9b25b2d485dae5589de40501f9d : SUCCESS in 55m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6b3fa8ec1cdc4778b015bd1d00a10329 : SUCCESS in 37m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/31ce3fa4088b4925ac2e6fc929f5cbe2 : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8efb1df946f75c08f8458ba9c62616d0faebdd0a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-03 04:10:56.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-47-763447-3/ns5_manila : SUCCESS in 2h 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"53ed36854aba8ae911e2aff670755119074f82d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-03 04:25:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/47/763447/3/check/DellEMC_Unity_Manila_dhss_true/d1ad175/DellEMC_Unity_Manila_dhss_true/422 : FAILURE in 1h 26m 01s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/47/763447/3/check/DellEMC_Unity_Manila_dhss_false/c57f6ea/DellEMC_Unity_Manila_dhss_false/406 : SUCCESS in 56m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3ba827e4b49ab8726fe1d50822f90f00b6442312","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-03 04:50:45.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/47/763447/3/check/DellEMC_VNX_Manila/ee72197/DellEMC_VNX_Manila/233 : SUCCESS in 1h 09m 51s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3772bd35c644e5eebcab1bf52dc6f622a134cbbd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-03 05:31:03.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/47/763447/3/upstream-check/manila-cDOT-ss/23f7ee7/ : FAILURE in 1h 46m 28s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/47/763447/3/upstream-check/manila-cDOT-no-ss/4461576/ : SUCCESS in 1h 50m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"27cf9f1e6bb81029ec09f4e8b33f26db970cb665","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-06 20:58:10.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":3},{"id":"737ac6df039c43d1782a5267a1c32279131a67a6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-03-07 13:06:47.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"89cf871a2ce0955a8317f7004685e9348cf3309c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-08 17:33:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\ncaveat is the system reader role needs a workaround currently, see: https://bugs.launchpad.net/manila/+bug/1917504","accounts_in_message":[],"_revision_number":3},{"id":"9347143c9b278e882d5a138b1f1c610e6d1fe55c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:33:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"283b76a1c03b9b96b090b98ebce5422847d7b9d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 19:21:28.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edde585b18604a608970f413e1b30a27 : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff8d5f2ae6134683b6237de6e3646c9f : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3890006bc4a406990c45113968c8055 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e21237d5bb3444e8750f8a6ae677d6b : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3144537b096c47a59e05b6882caeb582 : SUCCESS in 8m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc0600884e934f7d9513e5fc99735adf : FAILURE in 8m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1dc21d689b774fadb28d2a6996303302 : SUCCESS in 37m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4963b18311f248aa8b862596473d7e11 : SUCCESS in 1h 25m 22s","accounts_in_message":[],"_revision_number":3},{"id":"90aeec7a032e173be50961510d64b493b46a7783","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-08 19:30:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b7ef2ac8c5eef579f99c0d7054bdb84a2b22e1b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:09:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216146931c3d424592d1f32e22c6e081 : SUCCESS in 12m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dadb3682985b42c08289e2c35293ac26 : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ac32b1c8cf14823b32d062fc5617c91 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd383bef17654c8f84ee031baa47e051 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dde351c3ed13478eabb876457f82a78b : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f17116d6b8c849b6ba96ff986d6a0dc4 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f225eb4e3b6f4cb6b5d27cff568442b4 : SUCCESS in 10m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8afa95d106634b119f5cad1e2e21dd45 : SUCCESS in 6m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea5825dc0b9a4ef8906ffccdaef5f2dd : SUCCESS in 5m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0336be1bf4b34751b592457f7b69e91e : SUCCESS in 34m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/444c45284eef4396bbf4cf95082778f7 : SUCCESS in 33m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bf929a79cd074d8aa32661a5aa7a77b8 : SUCCESS in 1h 20m 18s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/caacc093a68244d7892574081e8c9756 : SUCCESS in 1h 15m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/83c03082a8674f9b95fac3199c8f4827 : FAILURE in 1h 10m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/06a26817cc25487f9d3cf7b1c99dd83d : SUCCESS in 45m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/30190d3ab1804c30a32e5bd5d7b1f53d : FAILURE in 2h 14m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/767fab90170b4bc2a275067b5e789ff5 : FAILURE in 24m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b909563491f84fc1ae30fba515fac7b0 : SUCCESS in 21m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/666e79dc4a194a90825928cd5dc77d99 : SUCCESS in 1h 18m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/45732ef5d84e4661b6b9098c6b29b542 : SUCCESS in 1h 05m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e2b0aaf6241145c8aaa4f013adae3597 : SUCCESS in 49m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/35c0fc3fe7da4cd39292fba9b59e0398 : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"947098f1efdb68ff8c225f9c4ed3a4f6a3d4bee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:10:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04f4f20ff19f171870c04b0c16483d74d88ac749","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:27:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd8464ca1b84f9782bbb9a371e82c8c : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26e38c82d1bf4ebaa7b8436df93dfc7f : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ede134ef264d288e7490443b03f520 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9da430a96d324217860bb7b6eeae2a76 : SUCCESS in 13m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3820af9ebccd4cfb860951fcf096c4b0 : SUCCESS in 9m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a6793333f2ff48d499b20c79f50d36c4 : SUCCESS in 34m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9a35f5e1ce214b9087708a9f6286bb98 : SUCCESS in 39m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4498c2c585a543cab27926c2d333ef55 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":3},{"id":"6e3be1ea471eb643efea9b681e682e4875f575fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"583bb788a4a36b1c61b3e5ce49c350b10e35f3ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:28:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40dca0561fef491193b62d68681f3456 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3},{"id":"ce23cbbab8f63efa1194923d85df48e74bfa1837","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-09 01:14:47.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/47/763447/3/upstream-check/manila-cDOT-ss/a3d4eda/ : FAILURE in 53m 29s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/47/763447/3/upstream-check/manila-cDOT-no-ss/a0b388f/ : FAILURE in 1h 43m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"df3782b0692f63ca80bec0a5f91e73ea2fddc3eb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 10:10:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/47/763447/3/check/DellEMC_VNX_Manila/d6afc1f/DellEMC_VNX_Manila/265 : FAILURE in 1h 26m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"26b8dc6ce4b741fc8dd94d07f8646c58ed9471a9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 17:46:05.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/47/763447/3/check/DellEMC_PowerMAX_Manila/34f95fe/DellEMC_PowerMAX_Manila/305 : SUCCESS in 1h 58m 24s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8d0b6acfcd7d679b1051adf03fee0216170f940d","revisions":{"00b817ce7e792b5bee4f45bdc23afb7f8ebce361":{"kind":"REWORK","_number":1,"created":"2020-11-19 21:17:02.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/47/763447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/763447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/763447/1"}}},"commit":{"parents":[{"commit":"3bcb67c201a371bc56a9e1bf2d3682cf82412e24","subject":"Implement secure RBAC for security services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3bcb67c201a371bc56a9e1bf2d3682cf82412e24"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 15:26:22.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 15:26:22.000000000","tz":0},"subject":"Implement secure RBAC for services","message":"Implement secure RBAC for services\n\nThis commit updates the policies for services to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I340f63874af5783099ed6b353be61a2909829343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00b817ce7e792b5bee4f45bdc23afb7f8ebce361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/00b817ce7e792b5bee4f45bdc23afb7f8ebce361"}]},"branch":"refs/heads/master"},"72790c92531d9c86646bb696c0673afdb5f0c045":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-12 07:10:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/47/763447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/763447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/763447/2"}}},"commit":{"parents":[{"commit":"d8f57eff3d215de88904aae428fb5ace842a0763","subject":"Implement secure RBAC for security services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8f57eff3d215de88904aae428fb5ace842a0763"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 15:26:22.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-01-12 07:10:16.000000000","tz":-480},"subject":"Implement secure RBAC for services","message":"Implement secure RBAC for services\n\nThis commit updates the policies for services to understand scope checking and\naccount for a read-only role. This is part of a broader series of changes\nacross OpenStack to provide a consistent RBAC experience and improve security.\n\nChange-Id: I340f63874af5783099ed6b353be61a2909829343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72790c92531d9c86646bb696c0673afdb5f0c045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72790c92531d9c86646bb696c0673afdb5f0c045"}]},"branch":"refs/heads/master"},"8d0b6acfcd7d679b1051adf03fee0216170f940d":{"kind":"REWORK","_number":3,"created":"2021-03-03 01:44:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/47/763447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/763447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/763447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/763447/3"}}},"commit":{"parents":[{"commit":"5aa61a283065bda14c79b77fe748041ebd2cc964","subject":"Merge \"Implement secure RBAC for quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5aa61a283065bda14c79b77fe748041ebd2cc964"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 15:26:22.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-03 01:43:10.000000000","tz":-480},"subject":"Implement secure RBAC for services","message":"Implement secure RBAC for services\n\nThis commit updates the policies for services to\nunderstand scope checking and account for a\nread-only role. This is part of a broader series\nof changes across OpenStack to provide a\nconsistent RBAC experience and improve security.\n\nChange-Id: I340f63874af5783099ed6b353be61a2909829343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d0b6acfcd7d679b1051adf03fee0216170f940d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8d0b6acfcd7d679b1051adf03fee0216170f940d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
