)]}'
{"id":"openstack%2Fmanila~763453","triplet_id":"openstack%2Fmanila~master~I8b95e1e70f74052e5bd4af1ba29842420bafd0b2","project":"openstack/manila","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I8b95e1e70f74052e5bd4af1ba29842420bafd0b2","subject":"Implement secure RBAC for share group types","status":"MERGED","created":"2020-11-19 21:17:02.000000000","updated":"2021-03-10 01:56:52.000000000","submitted":"2021-03-09 17:44:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"763453-secure-rbac","meta_rev_id":"4b36d88408938b81fe5c6a98adbb93ff615f8fc5","_number":763453,"virtual_id_number":763453,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 17:44:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-03-09 01:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2021-03-09 16:15:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":1,"date":"2021-03-09 16:15:25.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 21:17:02.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-20 00:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 08:59:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-22 09:30:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-12 07:10:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-08 15:54:10.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-09 01:36:23.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-03-09 05:53:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 16:14:04.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 01:56:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c572cb6773259c3c541b109f0a3ce3ad8b7e2033","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df64f449a49ae81cb8bf5765f3e714074309336e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:19:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"8e8feb43cb906bd97232fd4e8615d24ef26dc398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:46:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15861588f9054008876701385e6bb4d6 : SUCCESS in 12m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad6b12de68c44af196bf8cfa8e96fcbb : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a9a8510d744423d85925c64f18859b8 : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcaf331c94754ead94ad6dd726f57a45 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f7d96baf3a64f7e87ddb0be1d53f23f : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c69f9df503a41d0a8c45d44efa2dd24 : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3667663af878496e8550ff1b9edcd896 : SUCCESS in 8m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2a68a5d1b62d42daabe69f094b7cfffc : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6db0a8ae332d457a94218e9cb772a45f : SUCCESS in 5m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/70d9d55e0f5b4f8783faf6cc7e8ad39f : SUCCESS in 28m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ce66d989f5141e3b719bc2db75385a4 : SUCCESS in 40m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b707e0bf659743ac8cb2697404c0022c : SUCCESS in 57m 44s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5a635f3e98a24790874c24ade28814ea : SUCCESS in 1h 28m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f4d57afc1259430481ae5cc2723a844d : FAILURE in 52m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/62f951441a0f4088b59daac382518b62 : SUCCESS in 1h 01m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4401cc4de42a4b2c97cffcbe7e2ff9c2 : FAILURE in 2h 21m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d75581a65ec244f2baac864d7d8a19c4 : FAILURE in 26m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/260e3848ea424fe3b2631d90ac32a7a8 : SUCCESS in 31m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/501e17e4ed8c4c1d9f184c54db43ad7b : SUCCESS in 1h 14m 49s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/88a8c06de7e143b59895a6f21e2290d0 : SUCCESS in 1h 00m 10s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/91d40a7fecb34e96a1727772102f7183 : FAILURE in 47m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/acbb7b9eda0347d68fa1ba44076ffe11 : FAILURE in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae0564fff8a44c97a392338ba563e3d34d2fa6d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-20 08:59:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/53/763453/1/check/DellEMC_Unity_Manila_dhss_true/58eaffc/DellEMC_Unity_Manila_dhss_true/63 : FAILURE in 54m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/53/763453/1/check/DellEMC_Unity_Manila_dhss_false/ab0b2ac/DellEMC_Unity_Manila_dhss_false/62 : SUCCESS in 56m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d543998649bb6b70a7c51d64f16fb416a821d58e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 09:30:17.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-1/ns5_manila : FAILURE in 56h 45m 2s","accounts_in_message":[],"_revision_number":1},{"id":"23c6e67a1ca53846e1ff256706268a93aacbc76e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-23 22:57:52.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-1/ns5_manila : SUCCESS in 1h 58m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5053c6c0a6d08e667eb635bbb3dedd67c8aa07c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-23 23:13:06.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-1/ns5_manila : SUCCESS in 1h 58m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"109be30af99344c13ce01fa0047656f5de284407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-12 07:10:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"47d48bf19f74b77a8c105f670b7fffde70f97475","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:36:56.000000000","message":"Topic change-763469-1 removed","accounts_in_message":[],"_revision_number":2},{"id":"56036b4a56cc1dbe2b37a82f88b1b5dbb645c3b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:37:00.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"065b5128e6e69746bc8b63927662e814b27d537e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 03:16:30.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-2/ns5_manila : FAILURE in 1h 52m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"3097e48f20057cfbe2964bebb160b08e4e189e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-08 15:21:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2ae281c9d53e0acab0bb30e391f94cf19ac3e92","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-08 15:54:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nPls. see inline for a couple issues to resolve before merging this one.","accounts_in_message":[],"_revision_number":3},{"id":"4b61fbb9fa4f7a629a4f5b07ca543dfcc8725a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:48:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6b2808bef24661a6b7612e959d7929 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/982881249fd1467db6a858d3c9576cc4 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e881416586744be81e0328289ee1649 : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8f2e586ee74d129b0f8a7ad0afe3b8 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebc3dc2650684b4fbfc9004a4ce02a1f : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c4adc271a94549a243894b95ec7f5d : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a047f30bf5b4777be88dbedde22d19d : SUCCESS in 10m 27s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f8b14844380344b7b78cb7bee115fdf5 : SUCCESS in 6m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fe18a108a8d41399445760920096db5 : SUCCESS in 6m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/89e60e88d5fc4a2f8035d2b8095abf6d : SUCCESS in 42m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/df0a6ef1988a430cbbd82a1419c83cfe : SUCCESS in 40m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57ce858a58f2467381385651baffdda7 : SUCCESS in 1h 36m 10s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/904602c483dc41fa9746a3ab613d8c2b : SUCCESS in 1h 08m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2370cc1a57b743339e2f41a54b7f9def : SUCCESS in 1h 29m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97a48c822e9448d8bdf2608dd8c76ab0 : SUCCESS in 48m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f4d7b1403e348f89195e19403b26490 : FAILURE in 2h 13m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5f42fd94942e4a118a46ab6c7c50ca5b : FAILURE in 25m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6ae67cc943454beb91b9b0b0200f4924 : FAILURE in 23m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/63d2b0c3dc2148b486ef3ffbb8615e8d : FAILURE in 1h 23m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d0c3d77e440a4514a7175a04e715429d : SUCCESS in 1h 07m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/321ae6ee25f146bebc55c475b161a9b4 : SUCCESS in 42m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8ec7631b199b48859d57755b4d48633c : SUCCESS in 52m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88643b8fca561bd27ff8a6a8ac971f2031de5de7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-08 21:12:05.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-3/ns5_manila : SUCCESS in 2h 13m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"3f9bef6e40149c1538cc2a209712fe78bb0329e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-09 00:19:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"93ad57d5d98c6df0bd893c2d3c26bca5cb9c7d6a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-09 00:21:23.000000000","message":"Patch Set 4:\n\n(3 comments)\n\nThanks Tom, fixed the code per your comments.","accounts_in_message":[],"_revision_number":4},{"id":"22b0ba15737e658f664b07fb6f065cf76814c899","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-09 01:36:23.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":4},{"id":"21e3988373a5e02f264b262e0cbc110d32f5a8da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:49:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5b82fd51a64d5fb998873e73beaf81 : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7270cbe72a7d4c36acaf6f11b575a2a8 : SUCCESS in 15m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d15fcaead914c3a98fbafd2270eaca8 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c15e415c47448ebc4a27870eb92cdb : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f16812dfd614e8b9637c1d35989ecda : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb391194ff5440d499a1f4aeb299ae1b : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9837986e3344ec1a7366fbe8a2e8e07 : SUCCESS in 8m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d6b9bef76c23400fa6ca73855dc21660 : SUCCESS in 5m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5057c0a28b2341aba6f9be1aa2a4b3be : SUCCESS in 4m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/375edcffe13c4dff91c907deb29f64a1 : SUCCESS in 36m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9fb7891e44834740ac54844cf40bb84c : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c87eb44220be41f18e6a036656f7767e : SUCCESS in 1h 07m 08s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d2be86ae774145a08ec9eb07f38d01eb : SUCCESS in 1h 15m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ebce005695314ef8a5ca623eff793a99 : SUCCESS in 1h 19m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/234118df4faa470693455964b2b5fb93 : SUCCESS in 59m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0b34555197354bd2b4adf722ba7fa1bd : FAILURE in 2h 25m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/720a30ff5b134e3998abeb3e697323f3 : FAILURE in 24m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/80acd5c08c3d4e338e261c829b47ac6b : SUCCESS in 29m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdecb35800234712a5d2aed6a84d10a3 : FAILURE in 1h 16m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/72aa588ae0574d2ab7bf59b071696301 : SUCCESS in 59m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/144ea12e6286425d86850645fab72a33 : SUCCESS in 40m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/77157efd1088463dbaeb7da9edf8f727 : SUCCESS in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c611da033cab87de23a2f38dcae79f7967a63b5","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-09 05:53:53.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/53/763453/4/upstream-check/manila-cDOT-ss/90bb26d/ : SUCCESS in 1h 45m 51s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/53/763453/4/upstream-check/manila-cDOT-no-ss/f30b3ec/ : SUCCESS in 1h 49m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"63c71fdaddf2c22b615e1a23421c030e1809e604","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-09 08:12:56.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-53-763453-4/ns5_manila : SUCCESS in 2h 18m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"d35a05c7c1af1f78f5a5e3b1b6aba2d4709634f2","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 16:14:04.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/53/763453/4/check/DellEMC_VNX_Manila/19b5820/DellEMC_VNX_Manila/269 : SUCCESS in 1h 30m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"dad1c34a5d0b26bdb9c9c3fbc328a15a83dbc652","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-09 16:15:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"d5a0df2a25a0d8336d1e3f4e8d26f197cb922dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 16:16:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a06f366729f9fdfd9d76b619c75e0296b93b6be5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:44:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c500faa943a04239ac0941b00597a382 : SUCCESS in 11m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c22efc8edfa346b9bd20c87ab75e61e7 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20dad7ac18464b7cb71c9f5299248428 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c78ae0d7538a4d1185cdb39997773f98 : SUCCESS in 15m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d8ddec7d0264716a695197d9063fd6f : SUCCESS in 11m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2a4a0d213f044e7db6b0a0154dc6f4dd : SUCCESS in 31m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e555d8b99dcd45bf8cb5a163f0e47a79 : SUCCESS in 38m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/823a1d32a6d049bcbedca249ddab3000 : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":4},{"id":"1bb517e4f9b3d902aad3d6f3aaf76bac49d6fa16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:44:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7b5fd0a7a34486d2d2c7bf8e59daff8916af0bdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:51:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae64e64d3554424854a7ec583346dbb : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":4},{"id":"4b36d88408938b81fe5c6a98adbb93ff615f8fc5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-10 01:56:52.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/53/763453/4/check/DellEMC_PowerMAX_Manila/51b9e53/DellEMC_PowerMAX_Manila/308 : SUCCESS in 2h 25m 09s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a7eb82572149d1ee12fa66d54e527c642e388d4d","revisions":{"615886b069f571f96d95a1553cbd67da4d0bcb7e":{"kind":"REWORK","_number":1,"created":"2020-11-19 21:17:02.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/53/763453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/53/763453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/53/763453/1"}}},"commit":{"parents":[{"commit":"e42badd0474a46a5482f51d387f91985855affe3","subject":"Implement secure RBAC for group snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e42badd0474a46a5482f51d387f91985855affe3"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:14:04.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:14:04.000000000","tz":0},"subject":"Implement secure RBAC for share group types","message":"Implement secure RBAC for share group types\n\nThis commit updates the policies for share group types to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: I8b95e1e70f74052e5bd4af1ba29842420bafd0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/615886b069f571f96d95a1553cbd67da4d0bcb7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/615886b069f571f96d95a1553cbd67da4d0bcb7e"}]},"branch":"refs/heads/master"},"578f587f5fcedf70c37b97f9bd4d16923bca14d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-12 07:10:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/53/763453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/53/763453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/53/763453/2"}}},"commit":{"parents":[{"commit":"30befdcbba0e0dcaa1249771be3a6c9e5447ff9f","subject":"Implement secure RBAC for group snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/30befdcbba0e0dcaa1249771be3a6c9e5447ff9f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:14:04.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-01-12 07:10:17.000000000","tz":-480},"subject":"Implement secure RBAC for share group types","message":"Implement secure RBAC for share group types\n\nThis commit updates the policies for share group types to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: I8b95e1e70f74052e5bd4af1ba29842420bafd0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/578f587f5fcedf70c37b97f9bd4d16923bca14d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/578f587f5fcedf70c37b97f9bd4d16923bca14d3"}]},"branch":"refs/heads/master"},"0b627d95e0486c71c61bb7f27f96ae4b90cdeaf9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-08 15:21:52.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/53/763453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/53/763453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/53/763453/3"}}},"commit":{"parents":[{"commit":"f6bf1e43e587ad5082cdd4b740379615442a86bc","subject":"Implement secure RBAC for group snapshots","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6bf1e43e587ad5082cdd4b740379615442a86bc"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:14:04.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-08 15:21:52.000000000","tz":0},"subject":"Implement secure RBAC for share group types","message":"Implement secure RBAC for share group types\n\nThis commit updates the policies for share group types to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: I8b95e1e70f74052e5bd4af1ba29842420bafd0b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0b627d95e0486c71c61bb7f27f96ae4b90cdeaf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0b627d95e0486c71c61bb7f27f96ae4b90cdeaf9"}]},"branch":"refs/heads/master","description":"Rebase"},"a7eb82572149d1ee12fa66d54e527c642e388d4d":{"kind":"REWORK","_number":4,"created":"2021-03-09 00:19:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/53/763453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/53/763453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/53/763453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/53/763453/4"}}},"commit":{"parents":[{"commit":"aa298c9a8c8984c684557440cd187a63f28d9560","subject":"Merge \"Implement secure RBAC for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aa298c9a8c8984c684557440cd187a63f28d9560"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:14:04.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-09 00:18:55.000000000","tz":-480},"subject":"Implement secure RBAC for share group types","message":"Implement secure RBAC for share group types\n\nThis commit updates the policies for\nshare group types to understand scope\nchecking and account for a read-only role.\nThis is part of a broader series of\nchanges across OpenStack to provide a\nconsistent RBAC experience and improve\nsecurity.\n\nAlso fix the HTTP method in the policy\ndoc for this API.\n\nChange-Id: I8b95e1e70f74052e5bd4af1ba29842420bafd0b2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7eb82572149d1ee12fa66d54e527c642e388d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7eb82572149d1ee12fa66d54e527c642e388d4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
