)]}'
{"id":"openstack%2Fmanila~763454","triplet_id":"openstack%2Fmanila~master~Ie1b72459ae060693badb6fe864454836a4ff1300","project":"openstack/manila","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ie1b72459ae060693badb6fe864454836a4ff1300","subject":"Implement secure RBAC for share group type specs","status":"MERGED","created":"2020-11-19 21:17:02.000000000","updated":"2021-03-10 05:42:09.000000000","submitted":"2021-03-10 02:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763454-secure-rbac","meta_rev_id":"2e1f23b393187f82d43cc3562c74964fb329ac2e","_number":763454,"virtual_id_number":763454,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 02:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2021-03-10 05:42:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-03-10 05:42:09.000000000","post_submit":true,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-03-09 16:16:01.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 21:17:02.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2020-11-20 00:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 09:56:20.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"REVIEWER"},{"updated":"2020-11-22 09:31:42.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-12 07:10:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-03-08 15:59:30.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-09 01:38:44.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-03-09 01:48:15.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 09:40:27.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-10 05:42:09.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"122f0115c1451a227244927615e2bb8236909fae","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:17:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16393a00d9863a9e0a15986b8e42cd8275e529de","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-11-19 21:19:48.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"674357bc2af5a2aea35d7773bd8ad945197918c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 00:58:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9e7bebbfa94205b2068a8e5f2773d7 : SUCCESS in 14m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd51339effe44dce9f3d6d87ed62302e : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d209dbdffe543dd8f66c99ba087d6cb : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655e2870e6c04ef3b553040404c7ebe8 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb63622e1e748e4bc5815c6d5e21fde : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04fdfc7d81f54514b13a4c12e9cec166 : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/014d4df8a8e84d3fbfdf1b10a3e793ff : SUCCESS in 11m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b195a65d00a6456bafe6e733d5c03eda : SUCCESS in 5m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a11a9934b6c14f4687f23a19c13cb388 : SUCCESS in 4m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/57b2ace9411c4ea883e33a01f2199584 : SUCCESS in 40m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/de4396f4b1dc46eca08dfae2bd56fdc1 : SUCCESS in 36m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d2e5b7460ac54cf7b416d8af8984aeb5 : SUCCESS in 1h 16m 18s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/564ba01462c546e281a5ab131626918a : SUCCESS in 53m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/798f3c30dc0a4e4884ed697ff1acd82f : FAILURE in 1h 05m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cd93e9547d3e420a950809b3d3d5bbe7 : SUCCESS in 45m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/323dfdb41bb84abba48ca67d5447fe32 : FAILURE in 2h 08m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/236d1819fc2244adb35399cb436892b8 : FAILURE in 22m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/87ec201b39d44a54ac4c79d7555419ba : SUCCESS in 28m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1a7583407e641989fc750fe60b576e4 : FAILURE in 1h 24m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c4937f95e511448d9f90206a04a8d581 : SUCCESS in 54m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/74e9e14b8d3c4ede8d77f9d8dc112253 : FAILURE in 40m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/80630235a8af4baf993a9b562577768f : FAILURE in 49m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7369c162741437ed6de5a703a682abd0ce20c9c8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-11-20 09:56:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/763454/1/check/DellEMC_Unity_Manila_dhss_true/9d983ed/DellEMC_Unity_Manila_dhss_true/64 : FAILURE in 54m 06s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/763454/1/check/DellEMC_Unity_Manila_dhss_false/7f2abd2/DellEMC_Unity_Manila_dhss_false/63 : SUCCESS in 56m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe2ba37eda6e27d28557a8653b3048f65f43dc6b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-22 09:31:42.000000000","message":"Patch Set 1:\n\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-763454-1/ns5_manila : FAILURE in 56h 43m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b31515d6c26ffd28e294ba4fcb0fac0bc5c9ee55","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-24 09:22:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-763454-1/ns5_manila : SUCCESS in 2h 46m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"77a425850f565110d4018b2996bcf66f351e44c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-12 07:10:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d7be0f94e7b50f4d883e923af3c9c89ed1a18b22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:37:08.000000000","message":"Topic change-763469-1 removed","accounts_in_message":[],"_revision_number":2},{"id":"9b95a1f4480a83e75aaf58e538308530bb537cf9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-12 14:37:11.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"cc9b36a43b9d9be1071cdf849f7af9afd86d341c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 03:03:11.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-763454-2/ns5_manila : FAILURE in 1h 53m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"da852ffefc29aff4589ffbe03175ae35ffe6f5eb","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-08 15:59:30.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"80c4b97fc215d26ff9e785708e4f21f3a9bac36e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-08 16:00:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e9433e720f41f6e36157b72850b4d9ea0c7ac1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:42:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5a836205ea46779a95177e03e8d6a7 : SUCCESS in 13m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5881982618f43959f46e2c0ce9e5e33 : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a50185c6f51d47709e5f55e41f98d018 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbdcb72e4db42e9aee53c9eab5b5400 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9f631c9f39471da12a2f6220ae56ad : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921488d205ba4b93b8eb8dd8ad2698c8 : SUCCESS in 9m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3e698fc28ae494d9f07ecf5b9dc1f2c : SUCCESS in 10m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8d99feb7588a4dab9efcede739a6e5d2 : SUCCESS in 6m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf98fe09dfb943ad921430bc4482fe75 : SUCCESS in 5m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fad5dd03e9424397a254c5b32467c711 : SUCCESS in 40m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21827d4cf706458e97cfcd40adc303b4 : SUCCESS in 34m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bfa41de2d7774535ac2f9c7980647c21 : SUCCESS in 1h 28m 06s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6b9b22d05a0247d8b2dbc0a47dc45a5b : SUCCESS in 1h 32m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/15be6b40c96c4bbea071e20ceb72a173 : SUCCESS in 1h 17m 36s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ab0117eb8924a68901ff58e254bbd15 : SUCCESS in 55m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f93c28e2fd714ddcbda9c509b185bbea : POST_FAILURE in 1h 21m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f4bed83e505b42e69db79bcfb89192e3 : FAILURE in 23m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/01cfa1dec4bc4943bffbd05f73e25e85 : SUCCESS in 28m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/192241ddbd114eb1ba69c42709b45489 : SUCCESS in 1h 28m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0a1e801b0eb44103a6b09377ebece9bc : FAILURE in 1h 01m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4cb69010b6de444a94c21be3768c2fd5 : SUCCESS in 35m 19s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bbd248f4110543c6a3197d0c15acd0f7 : SUCCESS in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4fb3a36fcb874ba18f76ec4c8f3d248cf7326831","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-08 23:17:43.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-763454-3/ns5_manila : SUCCESS in 2h 13m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"52302a013a7c99018cd9ab35753c68d8cfc7bf0d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-09 01:38:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/54/763454/3/check/DellEMC_Unity_Manila_dhss_true/012542b/DellEMC_Unity_Manila_dhss_true/441 : FAILURE in 1h 18m 14s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/54/763454/3/check/DellEMC_Unity_Manila_dhss_false/90a189f/DellEMC_Unity_Manila_dhss_false/425 : SUCCESS in 59m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc5c9c0f9464c71a91c3f4c4f9245e09a765689d","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-09 01:38:44.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":3},{"id":"f202789e33a4d6a2a27e5d6b15cba787a0eb3f72","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 01:48:15.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/54/763454/3/check/DellEMC_VNX_Manila/021bcf2/DellEMC_VNX_Manila/259 : SUCCESS in 1h 23m 14s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3071dd869a2858faf50e1f5db531420655e437c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-09 03:36:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"31bb5f556b69d79d9dbe0a502d97bc9ce7db77c8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-09 03:36:35.000000000","message":"Patch Set 4:\n\nJust rebased to break the relationship chain with a patch that was changing underneath this one.","accounts_in_message":[],"_revision_number":4},{"id":"5d0ca35cdc0bce08809a60a200a1dd46d759d993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 05:45:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011cc50197f245d29e956b3b47820281 : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29d677442ff84fa4bdbbaa9f38f7d96f : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49ccef002d24797b5a9e2f9e3f41a3f : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f818f2ff2e409db694e0e3437cc577 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a95163647cc4e26bce2c4ccb8763231 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1170913ad7241b18967d12c660439b4 : SUCCESS in 10m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a17017ed2fa44d73bf2b807f7df76dfa : SUCCESS in 10m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ff208cca0cb84cc18a77b7e5348df8fc : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/09244e619af7448885d02f5e156f55c3 : SUCCESS in 6m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9ee9fdabe6e046d4899b3f6c8af9bc8f : SUCCESS in 27m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1175b67dec3e46b4a26b19f8129afa91 : SUCCESS in 42m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/241c53d5ffde45e5bd496c20c8e86d77 : SUCCESS in 1h 05m 04s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/de1cdfdb05c347a5a577f680934c20f8 : FAILURE in 1h 20m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7a38d88cd2c141d48785179c5f1694b7 : SUCCESS in 1h 31m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7a887472182f4044b8790c0200b83b32 : SUCCESS in 35m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/527195d362844c1192098a0c2bd5111a : FAILURE in 2h 04m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0a55f6bc82214ffc97085d27ed708842 : FAILURE in 26m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9c61e02ee5994808a4fc89346b4b78e9 : SUCCESS in 21m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ccac9400727545ae96da32586d9f12b5 : FAILURE in 1h 29m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/66dd5f6a2e0d4bbc9fc4b5ba569ad86d : SUCCESS in 44m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0a3ca5ef6c0747df887c7080823e69b5 : SUCCESS in 54m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ab6cda2d0b874e63a24e6104d628ace6 : SUCCESS in 57m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0354d50d949a28c71ea5b517f081c7fa298752ad","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-09 09:37:37.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-763454-4/ns5_manila : SUCCESS in 2h 6m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"ee41fa28dba2d192c6e99aeda2a48d7e04d70800","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-09 09:40:27.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/54/763454/4/upstream-check/manila-cDOT-ss/2dce1fb/ : SUCCESS in 1h 51m 28s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/54/763454/4/upstream-check/manila-cDOT-no-ss/2e77395/ : SUCCESS in 1h 41m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"aefc03716d7a239d51969179d87dced10456e7e4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-09 16:16:01.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"370d9d7ab6af692900cf9acf20a99492db626a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 16:16:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"09c910d728d56609035869711ef541b4e212ff47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 17:59:28.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6d38da7c224b3197f54c151eedd958 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9eb47154a964368981376cb4e411fa8 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b68d366e1254a27af9e1d9b82ef3f96 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298a7bb0968f4606add85983c0b472d2 : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edc9ec2d992d4c46a32744460e3773c6 : SUCCESS in 9m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd5282c771e24e159b30f57d40e27d4e : SUCCESS in 34m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d1f1fbcd6ed4b01b70bd61896563a0d : SUCCESS in 35m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8ec4362eade4255972a344c90f3af9f : FAILURE in 1h 34m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e9509b52fa0a42a5345699bec95f97736c7fbc0a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-09 18:08:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2e22c477171e037cf0e42e723d90c414a062578e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-09 18:11:30.000000000","message":"Patch Set 4:\n\nrecheck\n\nunrelated lvm nfs mount failure","accounts_in_message":[],"_revision_number":4},{"id":"4ddbbbc4ad9ac0eab5e8e392ca4043092d836643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:27:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f49a1548a5f43508d43e952bad213a9 : SUCCESS in 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96214cc1ae1c4f0c8c7fae1254d4cfd3 : SUCCESS in 14m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09eab0f2c4144bff9a798d42ec1fcb62 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a1774a716942ceb3534af087bb6cf1 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78122f48b8204feebe5a9c9d38579c4f : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7bbc5c317b34f58bae85c0b336f0dc7 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aef71b60e70544c984ccaf6453c39b4c : SUCCESS in 9m 27s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/613ae9e1aec543cca09ee64419c1897f : SUCCESS in 6m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b34d004fc6148629a352241ce9a2265 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6b307f1c0b05410d9d77bd63e61c7e43 : FAILURE in 33m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/73d1f08e96f94e299214f568fd3533da : SUCCESS in 36m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/485c485f7172434b91305fe358f76eeb : SUCCESS in 1h 17m 44s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8d162699cdb94bcca3c392c3a8374c45 : FAILURE in 1h 18m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1a2035ef097245dc9317f4e19b1a7b0f : SUCCESS in 1h 32m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/93b42478a03a4a5387d2a0b8a0476c03 : SUCCESS in 46m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5ccf565f2258404890b2c190ac50190e : FAILURE in 1h 55m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/082de910631b4e6a8a1ff5f309fbdb05 : FAILURE in 29m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/570f5716d531437c89e9362abd8be323 : SUCCESS in 30m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b97276554c2244fa9ef959e05c1c6fb4 : FAILURE in 1h 23m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b3812cf0daed4a5d8ee5ccea758c2846 : SUCCESS in 1h 16m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1bf5caf38d7a4f9aa8f3378539ec9d25 : FAILURE in 13m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7da45f7af5914c6e87cd82e937334ed7 : SUCCESS in 1h 02m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42992b0456cecdd3a60ebcb56cffe9bca4843cfd","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-09 22:29:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"11238e7a47f634fd6c0f389064ff4969f2768942","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-09 23:36:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/54/763454/4/check/DellEMC_VNX_Manila/cf0f558/DellEMC_VNX_Manila/273 : SUCCESS in 1h 53m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"5fba76190326c273cfb1f020792f6d06089e8238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 00:44:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791dcb047761423ba167a962d4521375 : SUCCESS in 9m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c46fbdb7f74b598a322f2046bdd08f : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a76dee657db4d0da4b807032c3625ab : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ac778c1d0b4ad69bfbbd4a1f5615f7 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/511dfc6a2d4a457199075524b4598543 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c82fb45e33f473aa3a55f4e30acd29a : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ba271dce9ca41a58d601d5ca2dc12ec : SUCCESS in 8m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab7a71fe06ca469799c0282ad3677aae : SUCCESS in 8m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a6c9dc9b2954ae98a50a83c4158f86a : SUCCESS in 5m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bceb24b9bd9f4f9cb9568b523da633be : SUCCESS in 34m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3a2bd934b02c4c7ab92b192706744b57 : SUCCESS in 41m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57ca1f42f1154aa8afad23a28aac2386 : SUCCESS in 1h 07m 33s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f54eda27cba64f17ad8ae235fa090096 : SUCCESS in 58m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f9a90dec4f0343d78b1902c836e7c092 : SUCCESS in 1h 15m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/66cf96e883ea4e0b8e8e471948ec953a : SUCCESS in 42m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/674ba171b93e43e4a17f5b3b1b004515 : FAILURE in 2h 10m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a124b543432c475a862595c959bfa113 : FAILURE in 25m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/295d583ffc834327a53c9a9689068570 : SUCCESS in 28m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c8d1d19873784441ae0931ff5ba3054e : FAILURE in 1h 11m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/25b17b800031490ebd1283abb92abdf6 : SUCCESS in 55m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/652e4e1ecbcc4d9680fc4275627efe15 : SUCCESS in 39m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/306a4c8a6cdb429c90788501f2f2be28 : SUCCESS in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e50056dfed165b9c548382cc4d87f0803dfef08e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 00:44:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"094b00478dc78e6065648145d168b8e37010d3e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:12:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406961752cfc41d98fe69fbcbdf2ecf6 : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6f583d1e1d4c6a98fc5323510a7e6e : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac733ff20d2f428eaf7c2e96a95cecf0 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e85e1e02d5f4037a9d98b2e4cb45662 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a06e89a7bf464df0b795cfafe8694720 : SUCCESS in 8m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/636c06cd67f046e0aaedb18c2aaa81ae : SUCCESS in 27m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d5d229a5f9984ff68a61c8a7fb63ab2b : SUCCESS in 32m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b122654b27134d1a9c4f362de6dbadee : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":4},{"id":"3f138cb23bc43401f9721078536cdb368efda00c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:12:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3485c56b764164e03cbd438805f7600e70301955","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:13:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9942baa9da44998ed9e1faf29e76ad : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4},{"id":"4b3d2c653ce1ce46df39dd5d202064edba207353","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-10 03:07:50.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/54/763454/4/upstream-check/manila-cDOT-ss/28ff836/ : SUCCESS in 1h 48m 45s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/54/763454/4/upstream-check/manila-cDOT-no-ss/094b0c1/ : SUCCESS in 1h 41m 52s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"52351ac1f06c29ab01b7e8996142be3213923e5a","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-10 05:42:09.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/54/763454/4/check/dsvm-tempest-manila-zadara-driver/9a4642c : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"81cbc2395b5bfec3f11e804d80f38e299b05d6ec","revisions":{"a78ad4715e70bc4e41b925f4104af0e2ffd4f3f1":{"kind":"REWORK","_number":1,"created":"2020-11-19 21:17:02.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/54/763454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/763454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/763454/1"}}},"commit":{"parents":[{"commit":"615886b069f571f96d95a1553cbd67da4d0bcb7e","subject":"Implement secure RBAC for share group types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/615886b069f571f96d95a1553cbd67da4d0bcb7e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:19:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:19:56.000000000","tz":0},"subject":"Implement secure RBAC for share group type specs","message":"Implement secure RBAC for share group type specs\n\nThis commit updates the policies for share group type specs to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Ie1b72459ae060693badb6fe864454836a4ff1300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a78ad4715e70bc4e41b925f4104af0e2ffd4f3f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a78ad4715e70bc4e41b925f4104af0e2ffd4f3f1"}]},"branch":"refs/heads/master"},"1d28a3a8222863ecd9257adce7156e2bb1c0b134":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-12 07:10:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/54/763454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/763454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/763454/2"}}},"commit":{"parents":[{"commit":"578f587f5fcedf70c37b97f9bd4d16923bca14d3","subject":"Implement secure RBAC for share group types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/578f587f5fcedf70c37b97f9bd4d16923bca14d3"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:19:56.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-01-12 07:10:17.000000000","tz":-480},"subject":"Implement secure RBAC for share group type specs","message":"Implement secure RBAC for share group type specs\n\nThis commit updates the policies for share group type specs to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Ie1b72459ae060693badb6fe864454836a4ff1300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d28a3a8222863ecd9257adce7156e2bb1c0b134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d28a3a8222863ecd9257adce7156e2bb1c0b134"}]},"branch":"refs/heads/master"},"0f63139b7b85b4281afff4d21d59855cfde50690":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-08 16:00:37.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/54/763454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/763454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/763454/3"}}},"commit":{"parents":[{"commit":"0b627d95e0486c71c61bb7f27f96ae4b90cdeaf9","subject":"Implement secure RBAC for share group types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0b627d95e0486c71c61bb7f27f96ae4b90cdeaf9"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:19:56.000000000","tz":0},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-08 16:00:37.000000000","tz":0},"subject":"Implement secure RBAC for share group type specs","message":"Implement secure RBAC for share group type specs\n\nThis commit updates the policies for share group type specs to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Ie1b72459ae060693badb6fe864454836a4ff1300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f63139b7b85b4281afff4d21d59855cfde50690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f63139b7b85b4281afff4d21d59855cfde50690"}]},"branch":"refs/heads/master","description":"Rebase"},"81cbc2395b5bfec3f11e804d80f38e299b05d6ec":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-09 03:36:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/54/763454/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/763454/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/763454/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/763454/4"}}},"commit":{"parents":[{"commit":"aa298c9a8c8984c684557440cd187a63f28d9560","subject":"Merge \"Implement secure RBAC for services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aa298c9a8c8984c684557440cd187a63f28d9560"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-11-19 18:19:56.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-09 03:36:11.000000000","tz":0},"subject":"Implement secure RBAC for share group type specs","message":"Implement secure RBAC for share group type specs\n\nThis commit updates the policies for share group type specs to understand scope\nchecking and account for a read-only role. This is part of a broader series of\nchanges across OpenStack to provide a consistent RBAC experience and improve\nsecurity.\n\nChange-Id: Ie1b72459ae060693badb6fe864454836a4ff1300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/81cbc2395b5bfec3f11e804d80f38e299b05d6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/81cbc2395b5bfec3f11e804d80f38e299b05d6ec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
