)]}'
{"id":"openstack%2Fmanila~764075","triplet_id":"openstack%2Fmanila~master~I96d49f84122a34701328909c929ede4d66746911","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"I96d49f84122a34701328909c929ede4d66746911","subject":"[glusterfs] don\u0027t reinit volume list on deletion","status":"MERGED","created":"2020-11-24 23:42:18.000000000","updated":"2020-12-15 00:57:17.000000000","submitted":"2020-12-15 00:51:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764075","meta_rev_id":"498afff2295432fdb3695448b6dd72c0cfdc1261","_number":764075,"virtual_id_number":764075,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-15 00:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":2,"date":"2020-12-07 22:28:03.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-14 23:44:16.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-14 23:44:16.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 01:25:51.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-25 02:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-25 03:21:18.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-01 04:03:05.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-01 05:43:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-01 17:40:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2020-12-01 17:40:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-12-01 17:40:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"state":"REVIEWER"}],"messages":[{"id":"e86ac66750bf8d1a3241f46cc8716d38dd93c1c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-11-24 23:42:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"258f2422379d8147428007e7667738dee8ccd7a8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-25 01:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764075-1/ns5_manila : SUCCESS in 1h 42m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5c767a2c547358acc1c6976b77839b912028ae1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 02:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2a0ce0d28d425e848f21b351763ad3 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42b4c74177124bbcbf0b0aa0a742c4c5 : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb638def08c440d19f816f8426f044a1 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0253c050cf438bb7f31a6406d7ec61 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6beed45c809460cba35262c4c600300 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d560a10f2324b3b90ca25db823b6a62 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f85e71a928b440dbac849ea8f75d15b : SUCCESS in 9m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6b82f47067694af5ba6ccbdbe077980a : SUCCESS in 4m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bfbd15ff074b40fc9f14129887a6f777 : SUCCESS in 4m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/af1586c06c9548eda9e551276cbbe14c : SUCCESS in 30m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/17c3ce73b77341aa8988259da863d4e3 : SUCCESS in 50m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/30e3173745ac47eda6a4484774bf7c01 : SUCCESS in 58m 21s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f71ab72145c247708655fa15628f486f : SUCCESS in 1h 29m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b0beb0fbb704bb3a134917331f59415 : FAILURE in 1h 39m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f92dc1d8c334cbbac7a3c81353b7003 : SUCCESS in 1h 28m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6b3409d998f04684adf92ed36deb66de : FAILURE in 2h 13m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0fad9b4da06c4fbfa0dcd7597c76eea7 : FAILURE in 43m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f8be6797186349c6b6610909c95bf59c : SUCCESS in 40m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c35ace538ea472496d1a740a6082db9 : SUCCESS in 1h 53m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2ebcf53202f541dcaf77b767cc7927fd : SUCCESS in 1h 28m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ffa6f3f692ed4d859043c938a8f36ae4 : FAILURE in 37m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/690f39a9b9ff41b184a50b3d23f9f96c : FAILURE in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"238930ee0d07de5ae1292c74984389dc2b47e054","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-11-25 03:21:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/75/764075/1/check/DellEMC_VNX_Manila/842757f/DellEMC_VNX_Manila/64 : SUCCESS in 1h 18m 44s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"ae6617ff8961e0fae565b09cc17cad1a742f744e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-01 03:50:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c0938a893b975414530c08fbe02eeb096fde376e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-01 04:03:05.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/764075/2/check/DellEMC_Unity_Manila_dhss_true/38582fc/DellEMC_Unity_Manila_dhss_true/84 : FAILURE in 12m 28s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/764075/2/check/DellEMC_Unity_Manila_dhss_false/4147806/DellEMC_Unity_Manila_dhss_false/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dfcf68c61027ba410f11d5b8f68b5696803344f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 05:32:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3f3171836634898b8a2648609544e2e : SUCCESS in 9m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44648a8d074747ed9a6b29d9e7c5ffb0 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd161488958f4de2a5395190f76ab9ce : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0783079dc0154149bb45815117664d84 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a2a01c7b82a408699fcd611cb25d2de : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13512f0b5de9455f9faaa0b4011078e8 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc842e486fd74056852dc7ba75f6734b : SUCCESS in 7m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/868ae43250b94e7abf335e3119403fc7 : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ff651155627a41f0892037ca8691c8df : SUCCESS in 4m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/58def6ea7bb54f1f9c32da71aa768a61 : SUCCESS in 28m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c2180fb209de4711ad05f81a12822efc : SUCCESS in 31m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/abcede7bec1448ab8eae9c1cba0d4a11 : SUCCESS in 1h 00m 25s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e420800fd8ff48b280b3afad7411dee4 : SUCCESS in 50m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a7263d78169e42ca9917ef6b2df325e6 : FAILURE in 12m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5978e69495d946caa0da1348f21a84d4 : SUCCESS in 42m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6ec1cd01fe1241bcbe67dbe093a324fa : FAILURE in 1h 39m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/be070b466bbf432b89189140cc16c322 : FAILURE in 22m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1076b76c6b514196a1a5d7632129baee : FAILURE in 20m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/70e731e3a3b643f9bff4caa48f4ffed2 : SUCCESS in 54m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e2370eba68314df3bd43aa9a00a115b7 : FAILURE in 12m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c20808c4aadc492f90b9cf07959abb62 : FAILURE in 37m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f0d72e7d69674aecab4635a6b33a23e3 : FAILURE in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6dde7ea747da0179253aa34655dc0309cad3d6a9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-01 05:43:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/75/764075/2/upstream-check/manila-cDOT-ss/42e7906/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/75/764075/2/upstream-check/manila-cDOT-no-ss/bb5e39c/ : FAILURE in 10m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"ffdf376d7faa0ae66829e3901332ac65e163edaa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-01 05:43:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/764075/2/check/DellEMC_Unity_Manila_dhss_true/bec75c5/DellEMC_Unity_Manila_dhss_true/85 : FAILURE in 10m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/764075/2/check/DellEMC_Unity_Manila_dhss_false/55a66a1/DellEMC_Unity_Manila_dhss_false/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"85e844929b1a82dfa3deeeac0f071f8ada5b0d7f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-01 05:45:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/75/764075/2/check/DellEMC_VNX_Manila/ab50b0c/DellEMC_VNX_Manila/69 : FAILURE in 12m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"4df74a8ab86fb40590129af2c3918054600cce1e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-01 05:52:28.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-764075-2/ns5_manila : SUCCESS in 1h 51m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"72e1033b9f0a3a3644620ce36a17a84b1733d384","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-12-02 03:14:43.000000000","message":"Patch Set 2:\n\nWhen there are three share driver backends and a volume is created on node1, the \"self.gluster_used_vols\" of node1 has the parameter\u0027volume1\u0027,\nDelete the volume on node2, update \"self.gluster_used_vols\", delete the\u0027volume1\u0027 parameter,\nBut the \"self.gluster_used_vols\" on node1 has the\u0027volumeX\u0027 parameter, and there will be more and more...\nThere is no good way to delete them accurately, because the parameters exist on the node that created the volume","accounts_in_message":[],"_revision_number":2},{"id":"5090ba4ac0289e69009145a452f4b72ef3b05fb4","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2020-12-07 22:28:03.000000000","message":"Patch Set 2: Code-Review+2\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":2},{"id":"8a30ea6437b99f69de4684523910da6373a0f34d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-08 05:48:35.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e When there are three share driver backends and a volume is created on node1, the \"self.gluster_used_vols\" of node1 has the parameter\u0027volume1\u0027,\n\u003e Delete the volume on node2, update \"self.gluster_used_vols\", delete the\u0027volume1\u0027 parameter,\n\u003e But the \"self.gluster_used_vols\" on node1 has the\u0027volumeX\u0027 parameter, and there will be more and more...\n\u003e There is no good way to delete them accurately, because the parameters exist on the node that created the volume\nHi Lin,\n\nI agree - don\u0027t you think we need to update this driver to be smarter? We could use the manila database to store information instead of persisting it in in-memory objects - we have the ability thanks to \"driver private data\". It\u0027s really simple to read and write to the database - here\u0027s an example: \n\nWriting: https://opendev.org/openstack/manila/src/commit/6d8f47d3c367aa0247dca713146825cfa66b6848/manila/share/drivers/zfsonlinux/driver.py#L486-L494\n\nReading: \nhttps://opendev.org/openstack/manila/src/commit/6d8f47d3c367aa0247dca713146825cfa66b6848/manila/share/drivers/zfsonlinux/driver.py#L1284-L1285\n\nI\u0027d be willing to help review the code if you wanted to make this change.","accounts_in_message":[],"_revision_number":2},{"id":"03cee5b80e60fd996c1be234add5f069675c16aa","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-12-14 06:54:56.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e When there are three share driver backends and a volume is created on node1, the \"self.gluster_used_vols\" of node1 has the parameter\u0027volume1\u0027,\n\u003e \u003e Delete the volume on node2, update \"self.gluster_used_vols\", delete the\u0027volume1\u0027 parameter,\n\u003e \u003e But the \"self.gluster_used_vols\" on node1 has the\u0027volumeX\u0027 parameter, and there will be more and more...\n\u003e \u003e There is no good way to delete them accurately, because the parameters exist on the node that created the volume\n\u003e Hi Lin,\n\u003e \n\u003e I agree - don\u0027t you think we need to update this driver to be smarter? We could use the manila database to store information instead of persisting it in in-memory objects - we have the ability thanks to \"driver private data\". It\u0027s really simple to read and write to the database - here\u0027s an example: \n\u003e \n\u003e Writing: https://opendev.org/openstack/manila/src/commit/6d8f47d3c367aa0247dca713146825cfa66b6848/manila/share/drivers/zfsonlinux/driver.py#L486-L494\n\u003e \n\u003e Reading: \n\u003e https://opendev.org/openstack/manila/src/commit/6d8f47d3c367aa0247dca713146825cfa66b6848/manila/share/drivers/zfsonlinux/driver.py#L1284-L1285\n\u003e \n\u003e I\u0027d be willing to help review the code if you wanted to make this change.\n\nI just found your comment now. Let me fix this question, thanks, Goutham Pacha Ravi","accounts_in_message":[],"_revision_number":2},{"id":"65b0f71ba3ccb89dd8a72e33d318bb882f406e89","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2020-12-14 23:44:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good, thanks Goutham and Lin!","accounts_in_message":[],"_revision_number":2},{"id":"8d84909240a0664dc0d279d2228d325e715d0da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 23:44:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b570a2a2ad99b753b715cb93902dbfe75bca78f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:51:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8639a27f196a4cd4b5d34047672a88b6 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e45e14984577438fbb5a10a2bc065c0a : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c270bea3d93497898370db1ebf9315a : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0ac9aa529948419c2e52e64cfd1fc0 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b220c53fa90d481b9749702c40ba523e : SUCCESS in 11m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/00d743e5e5fc475b9ca18504bb21644b : SUCCESS in 35m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f53881c8fb4b4afaa199dd229a4147b9 : SUCCESS in 41m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7c9bec94606c4e0fb9f1e47d36b85815 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f7ef9d6df14e33b5c74dcb1bd4bc719f1c22353a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:51:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"498afff2295432fdb3695448b6dd72c0cfdc1261","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:57:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cf657c42934fd697c9f8897c8468e6 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"914d87377479ab3951641d658fc461ad2f338ba0","revisions":{"493d4378e3f15d3501b0f8d12d90831fc886a934":{"kind":"REWORK","_number":1,"created":"2020-11-24 23:42:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/75/764075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/764075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/764075/1"}}},"commit":{"parents":[{"commit":"d7a647e164b34867fa7348841cf8708849f2e514","subject":"Merge \"[NetApp] Fix hard-coded CA cert path for SSL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d7a647e164b34867fa7348841cf8708849f2e514"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-24 23:32:11.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-24 23:38:22.000000000","tz":-480},"subject":"wip: don\u0027t reinit glusterfs-native volume list","message":"wip: don\u0027t reinit glusterfs-native volume list\n\nChange-Id: I96d49f84122a34701328909c929ede4d66746911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/493d4378e3f15d3501b0f8d12d90831fc886a934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/493d4378e3f15d3501b0f8d12d90831fc886a934"}]},"branch":"refs/heads/master"},"914d87377479ab3951641d658fc461ad2f338ba0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-01 03:50:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/75/764075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/764075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/764075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/764075/2"}}},"commit":{"parents":[{"commit":"d7a647e164b34867fa7348841cf8708849f2e514","subject":"Merge \"[NetApp] Fix hard-coded CA cert path for SSL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d7a647e164b34867fa7348841cf8708849f2e514"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-24 23:32:11.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-12-01 03:47:43.000000000","tz":-480},"subject":"[glusterfs] don\u0027t reinit volume list on deletion","message":"[glusterfs] don\u0027t reinit volume list on deletion\n\nWe don\u0027t need to re-initialize the volumes list\non deletion, it still makes sense to add a missing\nvolume to the list, going by the reasoning defined\nin I14835f6c54376737b41cbf78c94908ea1befde15\n\nRelated-Bug: #1894362\nChange-Id: I96d49f84122a34701328909c929ede4d66746911\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/914d87377479ab3951641d658fc461ad2f338ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/914d87377479ab3951641d658fc461ad2f338ba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
