)]}'
{"id":"openstack%2Fmanila~764489","triplet_id":"openstack%2Fmanila~master~I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8","project":"openstack/manila","branch":"master","topic":"bug/1900755","hashtags":[],"change_id":"I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8","subject":"Add exception for insufficient privileges when using security services","status":"MERGED","created":"2020-11-27 18:32:05.000000000","updated":"2021-03-26 16:55:07.000000000","submitted":"2021-03-26 01:25:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"764489-bug/1900755","meta_rev_id":"19d993e7583e5bab33f48e14da36fc956f8af669","_number":764489,"virtual_id_number":764489,"owner":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 01:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-03-25 15:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":2,"date":"2021-03-25 14:49:01.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-03-25 16:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-27 20:11:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-27 20:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 14:24:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-01 17:33:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-01 17:52:30.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 01:57:24.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"CC"},{"updated":"2020-12-16 12:48:52.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"CC"},{"updated":"2020-12-24 01:44:47.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-02-17 02:17:06.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 03:21:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-18 15:44:04.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-03-18 15:47:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-03-23 14:33:22.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2021-03-25 16:32:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"e030af0bd838d5e20d486336e42e2133c22e6d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-11-27 18:32:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95c77f2ca3bbfc646d83bfc01890ac796f747f38","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-11-27 19:14:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9abd0131e33446e5685e8a771f3ad6cbad2e6543","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-11-27 20:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-1/ns5_manila : SUCCESS in 1h 39m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"942e82eb5505a040fb2037bdd4094a33971d3424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 20:33:09.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e609a29160fe45c48b1788e50f3feadc : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b3b59e8eb2442ba97084c65e970d01d : FAILURE in 20m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3d57ffa9fdf43b1b19d9e2943e92956 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea5542a7ed244c0280761806631ec7a1 : FAILURE in 10m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/657c6db6fd8940029c154f33122c72c5 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57d65f52474444f19854785c970e0df4 : SUCCESS in 12m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c830ad9a80646a5910f424d4c1e0394 : SUCCESS in 14m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e749d6e080e1485080a4b3aa7b9fafd4 : SUCCESS in 9m 23s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c7626ed3511946ba96a68b2adf484954 : SUCCESS in 10m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a0efc19066d74a5dbbc16b43b43c13b4 : SUCCESS in 10m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1277d6efcbac4149b3cb67f9374934db : SUCCESS in 33m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d1c77e44b33447f4b396c15ec77c4c51 : SUCCESS in 45m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cd2e2c25c0d34b6e925707e587768eab : SUCCESS in 1h 44m 41s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/41792404b70546d7a9e1626f5ccd10de : SUCCESS in 1h 20m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a94a13e239c9453e84964e2461207ece : FAILURE in 1h 04m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5b31bea2360440299dd98f25d1620b53 : SUCCESS in 51m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f738cc9a79dc4553a6a29906e076c2d3 : FAILURE in 1h 56m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9222e1374600417bb2cff02cd0dcd942 : FAILURE in 25m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8945776149394203ab9a8805469d2a66 : SUCCESS in 26m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0482db901ea64784a9af3529642014af : SUCCESS in 1h 09m 59s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f367fe25f85640b9a5fd83664e3236d0 : SUCCESS in 58m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3b110162022a43abb8c291e6489d3254 : FAILURE in 31m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0aa7e34730404e119a909dfabb731f19 : FAILURE in 50m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b147214a68650b243886d9923f83cc40a0ae5fd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-12-01 13:30:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"675c237c8bfd15aaebc7dbc0a9bd3bfe37100877","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-01 14:02:53.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-2/ns5_manila : FAILURE in 28m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"96088fd0d9a60f6110956f18a4be778fc9453790","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-01 14:24:24.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_true/7ba0a18/DellEMC_Unity_Manila_dhss_true/87 : FAILURE in 53m 11s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_false/a8ca351/DellEMC_Unity_Manila_dhss_false/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e75af17658b450458e03e3632811a726687150ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 15:36:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c890f1756f7b4d80ab29e5a73d373886 : SUCCESS in 9m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84a69eb2bb84fd3ad6a9e7cf8ec2ab1 : FAILURE in 14m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd430c832eca4312a5ba5b54cbe4f778 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b94803de124906b1d6ccaed6dd20c8 : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467e866129064027b2ca48360bde92db : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d35a0b2c99814ae2b9b6d23a2c6df333 : SUCCESS in 13m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de21bfaa67db4fc09e6e7b76c3cc8ec9 : SUCCESS in 14m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a7720f67ba245e5b78f403641c912e8 : SUCCESS in 8m 04s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/28463f74cfa94cfc9bd4defe38946b63 : SUCCESS in 10m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ad1b1760f884436b8a31087d5840d5f1 : SUCCESS in 10m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6ece1fd1cef74cb4bafed057b99de145 : SUCCESS in 38m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/681a2f0bb71d4d0e92b927dd8e4eefca : SUCCESS in 39m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d72aa037c21f417189c6d10908b25ef9 : SUCCESS in 1h 16m 22s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0cfeb13c07764156ab750b6f307835e9 : SUCCESS in 1h 00m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6678292b5fda4838853e1a45b47160c7 : FAILURE in 1h 07m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/65b1cc80d5794b4ca460c7cc22c2a5cf : SUCCESS in 48m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/03c0a77592554cc8a38bcc83e468ab60 : FAILURE in 1h 56m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/64e08201670449cfb41ad12109ac38ae : FAILURE in 20m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c4479437597f48b2a5891eca2a63b592 : SUCCESS in 21m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec384473a5f142b1b90ca9e299b65430 : SUCCESS in 1h 02m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6c72d4a7ce1c484cb9c610a5967ae756 : SUCCESS in 1h 04m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1164c6826e9a40e5a91586a517ff0f8a : FAILURE in 59m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/10e8c53edeca46d1a3e83825f489a7b7 : FAILURE in 1h 07m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1997611d6e7d9256dfd661f81132e5d631982f8","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-12-01 16:18:05.000000000","message":"Patch Set 2:\n\nrun-nexentaci","accounts_in_message":[],"_revision_number":2},{"id":"05cad1f5b03c0d0ebcf5e4745909b0b0c533457b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-01 16:46:11.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_true/6cd474f/DellEMC_Unity_Manila_dhss_true/89 : FAILURE in 51m 27s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_false/fa90319/DellEMC_Unity_Manila_dhss_false/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7dd7510908db52d30a3c4ff3af05467c26ed4b45","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-01 17:33:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/89/764489/2/upstream-check/manila-cDOT-ss/6632c82/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/89/764489/2/upstream-check/manila-cDOT-no-ss/da2f36b/ : FAILURE in 1h 56m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"152ba503bfc28e57138408b911c6cd21bc185673","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-01 17:52:30.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/89/764489/2/check/DellEMC_VNX_Manila/c8b8f2b/DellEMC_VNX_Manila/71 : SUCCESS in 1h 23m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"4d74adfd14ed21234817b7f24ce3cab5d8d80adc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-01 23:10:32.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-2/ns5_manila : SUCCESS in 2h 18m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f275c11c17732bc1e16f6bb97811539cdbb164c8","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-12-02 13:49:10.000000000","message":"Patch Set 2:\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":2},{"id":"ea00a67f1409c5dac1f39477f2e0c951c01c1d58","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-12-04 01:57:24.000000000","message":"Patch Set 2:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":2},{"id":"7d45eb580f3f2bae68b80edff6016484f70daae9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-04 03:11:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_true/1f1708b/DellEMC_Unity_Manila_dhss_true/94 : ABORTED in 1h 14m 14s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/2/check/DellEMC_Unity_Manila_dhss_false/ce3787a/DellEMC_Unity_Manila_dhss_false/86 : FAILURE in 10m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"033f08d971c7598f545134c047617021c6a1ba20","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2020-12-16 12:48:52.000000000","message":"Patch Set 2:\n\nHi Eduardo, thanks for proposing this fix. \nAccording with your approach, we will need to provide different exceptions and user messages for each issue that happens in the storage driver.\nHave you think on a way of providing a generic way to solve this issue? One that we can easily scale to provide more user messages to help mitigate these kind of problems?\nWe could use the exception class to hold specific errors, set by the driver, that can be converted to user messages.\nWe can discuss more about that in the irc channel or in the weekly meertings.\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"a1ab29a815becdd25efdf91648b02b9cc317a31e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-24 01:44:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7507f519f0119e6813e35b3d1bb5ef92bbb9d265","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2020-12-30 16:18:07.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Hi Eduardo, thanks for proposing this fix. \n\u003e According with your approach, we will need to provide different exceptions and user messages for each issue that happens in the storage driver.\n\u003e Have you think on a way of providing a generic way to solve this issue? One that we can easily scale to provide more user messages to help mitigate these kind of problems?\n\u003e We could use the exception class to hold specific errors, set by the driver, that can be converted to user messages.\n\u003e We can discuss more about that in the irc channel or in the weekly meertings.\n\u003e \n\u003e Thanks\n\nThanks for your comments Douglas, and sorry for the late reply.\nThe SecurityServiceFailedAuth exception in this patch currently needs to be triggered by the driver. It is my understanding that it\u0027s the driver\u0027s job to validate the security service credentials.\nFrom my perspective, I see two options:\n1. Keep validating SS credentials in the driver, and the vendors would need to add the exception in this patch to their drivers when something goes wrong with the SS authentication (e.g., wrong password)\n2. Modify the manager so it validates the SS credentials itself. The vendors wouldn\u0027t necessarily need to modify their drivers, but if they don\u0027t, the SS credentials will be validated two times: one in the manager, one in the drivers.\nThe second option it\u0027s as generic as I can think of a solution, but it would alter the way we validate SS credentials.\nPlease correct me if there\u0027s something wrong with my line of thought. I also think discussing it in the IRC channel is a good idea.","accounts_in_message":[],"_revision_number":2},{"id":"cf5f860ad882e2ea4e6fd0686bf0a488f46083a0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 07:15:49.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-2/ns5_manila : FAILURE in 1h 51m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"0e09f2384e7bfdd79f3515dbee0fdaa9d94cd866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-02-16 18:10:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a9dcda92a84f9c6fc5b6044b0f3f8687254c497a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-16 20:03:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/3/check/DellEMC_Unity_Manila_dhss_true/8a48d00/DellEMC_Unity_Manila_dhss_true/341 : FAILURE in 57m 42s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/3/check/DellEMC_Unity_Manila_dhss_false/ce72574/DellEMC_Unity_Manila_dhss_false/325 : SUCCESS in 1h 03m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"cec146c419e09ec5674b1e3c3b53060f7d686dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 20:23:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3bf6549f3cd4783946a89f1677c66c3 : SUCCESS in 9m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c25350e723c481990bf2af2c0ddbdeb : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/765ff46ad00841afa055e2906d8675a2 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18c188959234ce89963f44c5100675c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/063f74435f964eb3a6de4e0a82063745 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d41367f471814ebe855fcca2759c2ce1 : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be6f263cc9ec4f0680beb9c56fdb0263 : SUCCESS in 9m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ad2eb1441d44ddc85531c5c90704aa9 : SUCCESS in 6m 02s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f7b076bfcd204ade9633c984cf331799 : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d6aba8a04ef34246b91b8709bd3017de : SUCCESS in 9m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2605971f6e7b4975a0aeef333577883e : SUCCESS in 41m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/689477db31554d72b4c8990c6495f580 : SUCCESS in 36m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4efd611068a64e78a700f5713a063671 : SUCCESS in 1h 20m 56s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/94d5a4a3f9054d72a6571564b61cc6fc : SUCCESS in 1h 12m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/521cebdfe29848a5b4f67299a22d8a49 : SUCCESS in 1h 20m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27b6fb1def3748d4a0d9aa6f9508cf13 : SUCCESS in 50m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b13eaaeeabd44f8d94f45ee9284721f2 : FAILURE in 2h 11m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/35e69f0172d94e4c9b9bb0e1abe61c75 : FAILURE in 26m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cfccb8365ed9454c91d01397f0904179 : SUCCESS in 23m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d94fc7aea3834f19a4038cdf9f16fdfd : FAILURE in 1h 30m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d0dfa2dbfe554f5cbbfeac5fa8e1ee24 : SUCCESS in 58m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8d6119ad4b5c4616a9a040476f8c298d : SUCCESS in 43m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5216b8f4c7c34640be101daa759a4fb4 : SUCCESS in 45m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3fffae0b296542848b5b887a46f217fba810e80","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-16 21:06:42.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-3/ns5_manila : SUCCESS in 2h 30m 12s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"32fe2569ec922c78825523a9b1355ec60468de1e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-16 21:27:51.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/3/check/DellEMC_Unity_Manila_dhss_true/8ce846e/DellEMC_Unity_Manila_dhss_true/342 : FAILURE in 1h 04m 01s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/3/check/DellEMC_Unity_Manila_dhss_false/a7a9939/DellEMC_Unity_Manila_dhss_false/326 : SUCCESS in 59m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"940bbe073502274a8597ab5d56793c6c337e3352","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 21:39:28.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/89/764489/3/check/DellEMC_VNX_Manila/06d2749/DellEMC_VNX_Manila/194 : SUCCESS in 1h 15m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"14d3d0254618629f4c42a7d05036d2eac26d4cfd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-16 22:13:52.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/89/764489/3/upstream-check/manila-cDOT-ss/8ecfa60/ : SUCCESS in 1h 49m 58s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/89/764489/3/upstream-check/manila-cDOT-no-ss/8855cd6/ : SUCCESS in 1h 48m 08s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"cfcce5b729fb1168e2c5d5561b384f352f78e99c","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-17 02:17:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/89/764489/3/check/dsvm-tempest-manila-zadara-driver/28af334 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e38a23139a8a75aeb91ba3b816f038639b12ec2b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-17 03:21:52.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/3/check/DellEMC_PowerMAX_Manila/ca08e20/DellEMC_PowerMAX_Manila/234 : FAILURE in 3h 49m 26s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"721b3cae0d6e6126caa7886aca3047f8c4230775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-03-17 14:41:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"47e87c18563a9a3ccff3439a1b233f002084e4a8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-17 14:43:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/4/check/DellEMC_PowerMAX_Manila/e40f3ef/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"79dc30e811bc46960a6e7d5c50f769dfb710acbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-17 15:38:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/4/check/DellEMC_Unity_Manila_dhss_true/d08a5bb/DellEMC_Unity_Manila_dhss_true/471 : FAILURE in 55m 41s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/4/check/DellEMC_Unity_Manila_dhss_false/dee1a21/DellEMC_Unity_Manila_dhss_false/464 : SUCCESS in 51m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e1f5daa40836f1dcb7c600e8ac86c9f8c76ef600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:37:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5f867032cb4aa7aeeb566204c62833 : FAILURE in 9m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9acbcdbeb94a4e8c8b0ae40b0078a89e : FAILURE in 21m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d34be1d987e24173b654d5cd4f496f3c : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3997634d912b4f08b36994dd1857d80e : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab1ab2f3b20646e4b55dfb3b53ef2bab : SUCCESS in 8m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ceaf941b0249aeb406107b67e0a850 : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d287117b01364691ab47ab83317a2344 : SUCCESS in 11m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7270d99580fe464a85d9332394eac445 : SUCCESS in 6m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/927cdc29321d4c51b30619e52895d32a : SUCCESS in 6m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab5067af251143f5901469ef1c9218bd : SUCCESS in 6m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd996c85ba3c47ecb35ea90474e78d42 : SUCCESS in 31m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e6c73d5a38514d94b77cbd2fab98ca9d : SUCCESS in 42m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e769aae1b45947aea26b8414528deb03 : SUCCESS in 1h 04m 07s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/662fe5c23a834d1180de26f91f1e9915 : SUCCESS in 1h 38m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/970df003f6e24f67aaeded0edf3859eb : SUCCESS in 1h 31m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/24a50c05d02e42f3867e8782f2d6ed26 : SUCCESS in 47m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1ffe98a0eb5442c0aab379823aea0da5 : FAILURE in 1h 52m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b6172ad255064756b93bba6d76a1fc5c : FAILURE in 24m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d4f801ec7ca541deb4df395eac62eefb : SUCCESS in 25m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a7d643ede8c47b8925e9a4b915107c9 : SUCCESS in 1h 32m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd1d5264b05f49ca85a00baf019c619e : SUCCESS in 1h 14m 43s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/297cc5b008a044fabf793441486d50d8 : SUCCESS in 39m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6457ae2426bd4dcab4fbe7806c021631 : SUCCESS in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"95ece0dc671a7b1e9baf68b42a18120acc9fe061","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-17 16:37:20.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/4/check/DellEMC_PowerMAX_Manila/811f8a8/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"306637a17a8a3400aff86ba4806dd0bf3b003809","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-17 17:44:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/4/check/DellEMC_Unity_Manila_dhss_true/2b189dc/DellEMC_Unity_Manila_dhss_true/472 : FAILURE in 1h 06m 55s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/4/check/DellEMC_Unity_Manila_dhss_false/543f354/DellEMC_Unity_Manila_dhss_false/465 : SUCCESS in 51m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bb070614fbf7d640a40b79df84ec8e62b92f38cc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-17 18:02:39.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/89/764489/4/check/DellEMC_VNX_Manila/d8fce5f/DellEMC_VNX_Manila/317 : FAILURE in 1h 25m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"d5cce0aa72ca57bd0b7daac7625724c3e2adb991","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-17 18:36:45.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/89/764489/4/upstream-check/manila-cDOT-ss/4f06ef0/ : SUCCESS in 1h 59m 17s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/89/764489/4/upstream-check/manila-cDOT-no-ss/022d93c/ : SUCCESS in 1h 46m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"95b752640e7e7fe940ee1cc414f5a7e10ef11316","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-18 15:44:04.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":4},{"id":"733e241767b61c326e0e1f5fd6aec684a6e9a9f6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-20 08:58:53.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-4/ns5_manila : SUCCESS in 1h 59m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"b87160a0123c6e0436d355f432991e065af3466f","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-03-23 14:33:22.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThanks for working on this, I just found a small typo.","accounts_in_message":[],"_revision_number":4},{"id":"8bd60ee0ff28ff47779012e70f5ad33939ec7800","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-03-23 14:34:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bf3dd0dfd084f0d7566c56e49c71ae25ffd863e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-03-23 15:33:08.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5892769f4b1dcc633a6d048bf596ae995f628653","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-03-23 15:34:44.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks carthaca!","accounts_in_message":[],"_revision_number":5},{"id":"4f6594f73109f71701c86418dabdb6039867e752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 17:12:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b198edec7db3429c81e13877bdc5abbf : FAILURE in 11m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b1b1cc6829e404aafe055f9bd457f42 : FAILURE in 17m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71e1bb460c8e410e9186b1140b2e012b : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566dc8f32ee4473b877b8782e2cf72ab : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a7ff823eb574d828e0f139008a642d3 : SUCCESS in 9m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40c31d60f38b451face704f02218c561 : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d235403b4d06468b9b2af9cdccf3483a : SUCCESS in 9m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5285a5b94384943ba66af6adc50b59b : SUCCESS in 6m 04s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe79df034fbb4f19af7996cea73d89b1 : SUCCESS in 7m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/be17ce0f0ff244648531ff3f75b95a9e : SUCCESS in 5m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0bc6fce4716346bf8f8f6ea9c392c81f : SUCCESS in 29m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c3df24447ca748f7ae50458ca48dcee4 : SUCCESS in 57m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b67fb327d854480bb2b587f9f704bbb2 : SUCCESS in 1h 07m 39s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ebdb24ee9da945979ffa49d42af919db : SUCCESS in 1h 13m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/72d9156d005d4de29d539f772f3d5101 : SUCCESS in 1h 12m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f9d754d759d141f59232cc5715df1bc6 : SUCCESS in 1h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/929d5a45eff94839893ae9e4e8ed3b91 : POST_FAILURE in 1h 32m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5a59ec13ccbe4d259abb7c43e507f356 : FAILURE in 46m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ceb73633403a420f82c8f480b6934d32 : FAILURE in 24m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4fa5522085ea42d89dc5d2d99b0b5c34 : SUCCESS in 1h 16m 17s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4d7626c0ff8543a3bb1ba3ba03216845 : SUCCESS in 1h 22m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e01269a4a41a45eb87eeeac9bb9d95a3 : SUCCESS in 39m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ad70d744ee4d4cfe951c5b1d028bde52 : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1caefef28be00e474d2e2d82c47344bfc4c72373","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 18:33:31.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/89/764489/5/check/DellEMC_VNX_Manila/6be93ef/DellEMC_VNX_Manila/337 : SUCCESS in 1h 20m 23s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d210083acdd781bdd7f179eb538b0436d8b8a5c4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-23 18:58:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/89/764489/5/upstream-check/manila-cDOT-ss/e50e0b5/ : SUCCESS in 1h 45m 09s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/89/764489/5/upstream-check/manila-cDOT-no-ss/c881fdb/ : SUCCESS in 1h 45m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"c53f334508db21889ffd0dc8dac6bdba761e9832","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-23 21:05:59.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/5/check/DellEMC_PowerMAX_Manila/3aef540/DellEMC_PowerMAX_Manila/389 : SUCCESS in 1h 49m 03s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"d1946d262022ca5322db44e434d6da8521b3bc85","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-23 23:10:52.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-5/ns5_manila : SUCCESS in 2h 11m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"4595c5057a57afabd485db6681918ea618747402","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-03-24 07:15:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/89/764489/5/check/dsvm-tempest-manila-zadara-driver/6ea6a39 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":5},{"id":"921390dab98be23177fafacfe36ffb4312875d75","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-03-25 11:03:55.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM. Thanks, Eduardo!","accounts_in_message":[],"_revision_number":5},{"id":"1697aa797ee2e74d903f74d76765fc0863c18552","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 11:42:19.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nNice work, but please see inline.  I think we should fix up the release notes a bit before merging this.","accounts_in_message":[],"_revision_number":5},{"id":"a9e3b511edb86c7bc5de3f50289913140c5bbc87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-03-25 14:47:01.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"8ea5173323fdc6ff9a02f46272fd197ee8916f09","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2021-03-25 14:47:11.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"39874c838cb6702fbac11160ca7be4598a08448e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-25 14:49:01.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks, Eduardo - LGTM!","accounts_in_message":[],"_revision_number":6},{"id":"6d7b34b43b76bee9a19d7e3306ceeabde9a4a802","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-25 15:53:49.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/6/check/DellEMC_Unity_Manila_dhss_true/c7c8f72/DellEMC_Unity_Manila_dhss_true/543 : FAILURE in 1h 03m 18s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/6/check/DellEMC_Unity_Manila_dhss_false/c45d170/DellEMC_Unity_Manila_dhss_false/539 : SUCCESS in 58m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"bdb86025ff477f6b6f9ca3e48f18f4a24c433df0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-03-25 15:53:57.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM. Thanks, Eduardo!","accounts_in_message":[],"_revision_number":6},{"id":"a57f0b33bdae82d70a10c7cb4a6f8068bb82423f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-25 16:32:17.000000000","message":"Patch Set 6: Workflow+1\n\nLooks good, and since it was just the release note that was updated, it should pass CI. thanks Eduardo!","accounts_in_message":[],"_revision_number":6},{"id":"a40e7a5be986a1a70a249fca680f3d327122f9bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:51:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafcd22914e54aad8c164ac6194b7230 : FAILURE in 11m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/971a8ca3a38146f1a30eabb0c6b07265 : FAILURE in 17m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1461a1e1ab348c2bab1a6f66724c95e : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbab4e6e8e3642438c8bf9da6fa2c430 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90325a6b7baf4e41b5558521c50d194f : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c6f2a434049400db4788aeb7132decc : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea8c1aedb996427296219045f8d9b283 : SUCCESS in 14m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e816cf8cca4e2c913c8e78eb1bfe71 : SUCCESS in 7m 07s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3599145bc71a4dda95bb9b0aee605bc8 : SUCCESS in 5m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83914461cbe3481ca145ef2222a0e9c0 : SUCCESS in 5m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d1b5d4f90a63463b8da0d990785b6869 : SUCCESS in 32m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aa78d600fe374b03884ed22ec5037665 : SUCCESS in 36m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b2efbf203d3b488281edbce5e3172585 : FAILURE in 1h 14m 26s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d88aecdd83f1477ca5da47031cf49c0d : SUCCESS in 1h 02m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b003f19c49374376a5f69717da2e80ec : SUCCESS in 1h 28m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f93900e61c884329973cf0557ef0c86f : SUCCESS in 44m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/09eda1569c354d80b6b945ff7c3f3603 : FAILURE in 2h 41m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/486a01004d71437390aa78089efb4be5 : FAILURE in 24m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/46d097dfdd5743c288106650f6b5fae3 : SUCCESS in 27m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b4baa870090148a0ac329f59d92292d8 : SUCCESS in 1h 17m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/839104cba7774853a071996cf4305f1d : SUCCESS in 1h 00m 42s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/856567e5d25141c7bcd5e30e0dd03e9a : SUCCESS in 36m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a9dd43398ee449d9ae11c6260cb17dad : SUCCESS in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ac3dab5c4a20ad27ef616bde2b6fc6e8f9e8cba","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-25 21:28:22.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/6/check/DellEMC_PowerMAX_Manila/2da092f/DellEMC_PowerMAX_Manila/415 : FAILURE in 3h 52m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"e8c71f119b2d3e536775a558fb3659dbf7bf2289","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 21:53:08.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-89-764489-6/ns5_manila : SUCCESS in 2h 31m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"30a1601e696237432bdaa276da367beea3f14cba","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-25 21:53:45.000000000","message":"Patch Set 6:\n\nrecheck\n\ntimeout error on a scenario test for lvm driver","accounts_in_message":[],"_revision_number":6},{"id":"cc5dbc6877752ea410d4e35e4f9b0ede13742bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 00:08:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca8968fca22a4ce681267b5481ff0ac1 : FAILURE in 11m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee2ac5d35924755ac10f41d319aea38 : FAILURE in 19m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5124146b057842a9bd3ea5118b33017d : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e7428162b241f99c3f5dee0b043af9 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/862a1a5815f94bcd9546c661c34ae444 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f5a269276541a09c99817ee8c33827 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/327d2c1524a546568d1aedd0fa5a354f : SUCCESS in 12m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27801db2557a44dc846aceaaa017d97b : SUCCESS in 5m 59s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6ad81c5ca0d74381a9d058e9ba4fb958 : SUCCESS in 5m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0678aac6f713431c9fb8732257754c40 : SUCCESS in 5m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c1de1b36c97d4a5eaafb4dde2800e58f : SUCCESS in 39m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d6b8582499e34504af494f9aebdad20a : SUCCESS in 48m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a3e94e4b2c714b1b83b6c79cf8d6416f : SUCCESS in 1h 09m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3add6c9d05454b4b8bae494833d4b9a0 : FAILURE in 1h 09m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3ce3c28b84234649aba6ad1ba7fc498e : SUCCESS in 2h 07m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f86ed3cdcbd446faca150b58c4923b1 : FAILURE in 1h 04m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/44ef3b7d78b3440683041480b2efa412 : FAILURE in 2h 06m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a9626e78859342668184f4f13362b6aa : FAILURE in 29m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0d88f97906c147529d1a56fa7ac80f0b : SUCCESS in 30m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1825523767004e28a8ce00772a24be0f : SUCCESS in 1h 14m 13s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3a504945a6bf4eeaba4ef2e103e0c143 : SUCCESS in 1h 08m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7b43ceac8b164e8ea21749cc1f1c15d1 : SUCCESS in 45m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5a1931423a11453ead0b36c5e02184d9 : SUCCESS in 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bc28b7744ea027ee939bab83389e51631fa12fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 00:11:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4cdd4f6b5b3f8ef7f3eee8efc66747b1642dfac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 01:25:18.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de762a7235440919427b36497107604 : FAILURE in 10m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c919b7c0d65647e4a12b712913cee33b : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7345e0aa15444f528601aa66ab5856a7 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1bf269cb4d443cb07731c0452a1d43 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73dd0dd1cbaf4b54bdb7fd8918c42534 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef1dbe1b0959499191f9f68305c36f07 : SUCCESS in 6m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1f2b90df979f4cefab2edc0f80dbc403 : SUCCESS in 33m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/346e772c84544ba9b9a4c0c44dc44739 : SUCCESS in 39m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46410a0cee6444d9af20a14577269ea1 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":6},{"id":"6a52009be2c96f599e58e8674edbc207aa2edaeb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 01:25:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a8e6dca5edde7f502a4e5908d4eff0ed682ce3e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 01:27:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a0319f1b2149f0aee4d3c89928f063 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3dcb4492b4146e3951b58601e6fcaaf : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":6},{"id":"e073932de07c7ec62cc57d6256ebbfb31f43f922","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-26 02:32:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/89/764489/6/check/DellEMC_Unity_Manila_dhss_true/a9ab02c/DellEMC_Unity_Manila_dhss_true/552 : FAILURE in 1h 03m 58s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/89/764489/6/check/DellEMC_Unity_Manila_dhss_false/e72f6a3/DellEMC_Unity_Manila_dhss_false/548 : SUCCESS in 55m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c2691514f86333cdcbb43059ea8cede0df6021ea","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 03:34:32.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/89/764489/6/check/DellEMC_VNX_Manila/c8dcf16/DellEMC_VNX_Manila/363 : SUCCESS in 1h 26m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":6},{"id":"19d993e7583e5bab33f48e14da36fc956f8af669","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 16:55:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/89/764489/6/check/DellEMC_PowerMAX_Manila/bbbccdb/DellEMC_PowerMAX_Manila/421 : FAILURE in 3h 22m 11s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3d65f46d3e54b85cab5ecf38bae20670da563d0b","revisions":{"7ab11c0c4a92dcf9cb3f10da919c27f610d56b1e":{"kind":"REWORK","_number":1,"created":"2020-11-27 18:32:05.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/1"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"subject":"Add exception for insufficient priveleges when using security services","message":"Add exception for insufficient priveleges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ab11c0c4a92dcf9cb3f10da919c27f610d56b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ab11c0c4a92dcf9cb3f10da919c27f610d56b1e"}]},"branch":"refs/heads/master"},"bcfd5a330603d7d6768e1c4a7ad46e2823385694":{"kind":"REWORK","_number":2,"created":"2020-12-01 13:30:40.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/2"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-12-01 13:22:04.000000000","tz":0},"subject":"Add exception for insufficient priveleges when using security services","message":"Add exception for insufficient priveleges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bcfd5a330603d7d6768e1c4a7ad46e2823385694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bcfd5a330603d7d6768e1c4a7ad46e2823385694"}]},"branch":"refs/heads/master"},"4fb1fe3c84254312395b8dd0d24a4f5c7eac95d4":{"kind":"REWORK","_number":3,"created":"2021-02-16 18:10:08.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/3"}}},"commit":{"parents":[{"commit":"0d8f415e868f6faff82f973f862278d09b75efbb","subject":"tox miniversion update to 3.1.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0d8f415e868f6faff82f973f862278d09b75efbb"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2021-02-16 18:10:05.000000000","tz":-180},"subject":"Add exception for insufficient priveleges when using security services","message":"Add exception for insufficient priveleges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fb1fe3c84254312395b8dd0d24a4f5c7eac95d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fb1fe3c84254312395b8dd0d24a4f5c7eac95d4"}]},"branch":"refs/heads/master"},"b6a7ef591415042198c08e28d336d61c5c0f6ecf":{"kind":"REWORK","_number":4,"created":"2021-03-17 14:41:24.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/4"}}},"commit":{"parents":[{"commit":"9b835f03d597b687645cf15524ff88fed6af5233","subject":"Merge \"[NetApp] Implement security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9b835f03d597b687645cf15524ff88fed6af5233"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"ecsantos","email":"eduardo.experimental@gmail.com","date":"2021-03-17 14:41:22.000000000","tz":0},"subject":"Add exception for insufficient priveleges when using security services","message":"Add exception for insufficient priveleges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b6a7ef591415042198c08e28d336d61c5c0f6ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b6a7ef591415042198c08e28d336d61c5c0f6ecf"}]},"branch":"refs/heads/master"},"3b8435ee3f3931e77fe5a0176f8854de9ac6e190":{"kind":"REWORK","_number":5,"created":"2021-03-23 15:33:08.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/5"}}},"commit":{"parents":[{"commit":"9b835f03d597b687645cf15524ff88fed6af5233","subject":"Merge \"[NetApp] Implement security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9b835f03d597b687645cf15524ff88fed6af5233"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2021-03-23 15:32:10.000000000","tz":0},"subject":"Add exception for insufficient privileges when using security services","message":"Add exception for insufficient privileges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b8435ee3f3931e77fe5a0176f8854de9ac6e190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3b8435ee3f3931e77fe5a0176f8854de9ac6e190"}]},"branch":"refs/heads/master"},"3d65f46d3e54b85cab5ecf38bae20670da563d0b":{"kind":"REWORK","_number":6,"created":"2021-03-25 14:47:01.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/89/764489/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/89/764489/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/89/764489/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/89/764489/6"}}},"commit":{"parents":[{"commit":"9b835f03d597b687645cf15524ff88fed6af5233","subject":"Merge \"[NetApp] Implement security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9b835f03d597b687645cf15524ff88fed6af5233"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2020-11-27 18:19:02.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2021-03-25 14:45:53.000000000","tz":0},"subject":"Add exception for insufficient privileges when using security services","message":"Add exception for insufficient privileges when using security services\n\nWhen trying to create a share using a security\nservice while having insufficient privileges or\nproviding wrong credentials, manila raises a\ngeneric user message that provides little\ninformation to the end user on what exactly went\nwrong.\nAdded a driver-agnostic exception for when this\nhappens, accompanied with a user message to\nprovide more useful information to end users.\n\nPartial-Bug: #1900755\nChange-Id: I2b0bc5c0715c225cd5b38b55fb3967ff7fc86fa8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d65f46d3e54b85cab5ecf38bae20670da563d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d65f46d3e54b85cab5ecf38bae20670da563d0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
