)]}'
{"id":"openstack%2Fmanila~765039","triplet_id":"openstack%2Fmanila~master~If95a0443c651c1a0a2ae646879bfb829b6130f01","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"If95a0443c651c1a0a2ae646879bfb829b6130f01","subject":"Remove six","status":"ABANDONED","created":"2020-12-02 03:29:47.000000000","updated":"2020-12-02 12:08:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1000d0c56366a7b3138a07cad711d633f7c5eff9","_number":765039,"virtual_id_number":765039,"owner":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 05:09:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 08:19:44.000000000","updated_by":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"reviewer":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"state":"CC"},{"updated":"2020-12-02 12:08:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"f46d7bd9d7a889f6b5d3d81f397a71c9c41aeec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-12-02 03:29:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"708b4de7bc98f94b9f44037b547e92ed11e3f543","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2020-12-02 03:41:33.000000000","message":"Abandoned\n\nDuplicate of \nhttps://review.opendev.org/c/openstack/manila/+/756450\nhttps://review.opendev.org/c/openstack/manila/+/756981\nhttps://review.opendev.org/c/openstack/manila/+/757071\nhttps://review.opendev.org/c/openstack/manila/+/756991","accounts_in_message":[],"_revision_number":1},{"id":"734c5fc813eca8d5476eb67b6344a1c537f8633e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-02 05:09:23.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-39-765039-1/ns5_manila : FAILURE in 1h 18m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"74dac89c4ef0c43c52e462896c6247b19208dc75","author":{"_account_id":14624,"name":"Sam Wan","email":"sam.wan@emc.com","username":"samwan"},"date":"2020-12-02 08:19:44.000000000","message":"Patch Set 1:\n\nrun-DellEMC Unity CI","accounts_in_message":[],"_revision_number":1},{"id":"1000d0c56366a7b3138a07cad711d633f7c5eff9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-02 12:08:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/39/765039/1/check/DellEMC_Unity_Manila_dhss_true/564ea2b/DellEMC_Unity_Manila_dhss_true/92 : FAILURE in 3h 48m 49s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/39/765039/1/check/DellEMC_Unity_Manila_dhss_false/de21fe9/DellEMC_Unity_Manila_dhss_false/84 : FAILURE in 3h 48m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40216af37504198ee66cb9dd24c360138c4493ea","revisions":{"40216af37504198ee66cb9dd24c360138c4493ea":{"kind":"REWORK","_number":1,"created":"2020-12-02 03:29:47.000000000","uploader":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"ref":"refs/changes/39/765039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/765039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/765039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/765039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/765039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/765039/1"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-12-02 03:19:00.000000000","tz":480},"committer":{"name":"wangzihao","email":"wangzihao@yovole.com","date":"2020-12-02 03:29:39.000000000","tz":480},"subject":"Remove six","message":"Remove six\n\nRemove six Replace the following items with Python 3 style code.\n- six.string_types\n- six.text_type\n- six.add_metaclass\n- six.b\n- six.moves\n- six.wraps\n- six.BytesIO\n- six.StringIO\n\nChange-Id: If95a0443c651c1a0a2ae646879bfb829b6130f01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40216af37504198ee66cb9dd24c360138c4493ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/40216af37504198ee66cb9dd24c360138c4493ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
