)]}'
{"id":"openstack%2Fmanila~765975","triplet_id":"openstack%2Fmanila~master~I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295","project":"openstack/manila","branch":"master","topic":"bp/share-size-limited-by-share-type","hashtags":[],"change_id":"I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295","subject":"Adds support min/max share size limited by share_type","status":"MERGED","created":"2020-12-08 11:40:49.000000000","updated":"2021-03-07 23:00:52.000000000","submitted":"2021-03-07 22:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":12,"has_review_started":true,"submission_id":"765975-bp/share-size-limited-by-share-type","meta_rev_id":"ac517e09f9d30c3e67ef4fe812bdb6740c99e64c","_number":765975,"virtual_id_number":765975,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-07 22:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-03-07 14:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-03-06 21:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-03-07 21:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-03-07 21:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 13:01:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-08 13:40:03.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-08 13:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 08:07:29.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2020-12-12 00:33:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2020-12-12 04:28:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-12-12 04:28:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-12-12 04:28:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-12-12 04:28:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-12-12 08:24:08.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-12 08:33:51.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 10:28:41.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2021-01-04 10:28:41.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-01-28 17:41:32.000000000","updated_by":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"reviewer":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"state":"REVIEWER"},{"updated":"2021-02-01 07:44:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-02 01:17:22.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-04 15:53:10.000000000","updated_by":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"reviewer":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"state":"REMOVED"},{"updated":"2021-03-02 09:39:46.000000000","updated_by":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"reviewer":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"state":"REVIEWER"},{"updated":"2021-03-03 08:49:28.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"ad448aacc78e3c151e67bec553618af74e860b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-08 11:40:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe609d41abdea17b4757c31870af823779352680","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-08 11:41:45.000000000","message":"Topic set to bp/share-size-limited-by-share-type","accounts_in_message":[],"_revision_number":1},{"id":"5af07bfa9c8563eef54b10a3aa7280df784598d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-08 11:44:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3dec425f03a2fc7af72f743debacacf3e01693dd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-08 13:01:34.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-2/ns5_manila : FAILURE in 1h 5m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"6b68da79a717dd96cf6fc16a8b49e2a03012c95f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-08 13:40:03.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/2/check/DellEMC_Unity_Manila_dhss_true/cc5a475/DellEMC_Unity_Manila_dhss_true/97 : SUCCESS in 1h 55m 12s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/2/check/DellEMC_Unity_Manila_dhss_false/f91d254/DellEMC_Unity_Manila_dhss_false/89 : SUCCESS in 58m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6108d5ca6d0a2be58ffd8a6b0fdf4784bf33ad20","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-08 13:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-1/ns5_manila : SUCCESS in 1h 59m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1dcadfca063fbf2bce364f4bdd8a73c71d9a5a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 13:59:36.000000000","message":"Patch Set 2: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bce98d227142eabbd2c9daf101f1bf : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ccb055f47874829b6c20b4f8f28d5ac : FAILURE in 18m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce6d68b638f24c9b822e68a447f63dbc : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f98291ee9a649958ae4591b4b87164b : FAILURE in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d209fa29463e4b72ba0fae9dc46dd860 : FAILURE in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34fd4a78e0744efcb5cbf7dfa9d983b2 : FAILURE in 10m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edceb59a422645dc9cff29f14a047482 : FAILURE in 8m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f7e864e88be6484489db7117ca24ba99 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5abd598c9e1f44358f2a0b6c3a778009 : SUCCESS in 5m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2c583c126b1947c392aac836cbef5688 : SUCCESS in 30m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3ea4e36686f34212b7e6d08eaa6f93f1 : SUCCESS in 37m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/69668f40d5b14ddb8605bfad18ce27a9 : FAILURE in 1h 29m 26s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f6a2bfce68144aa6bcb6b1adb2f1ba48 : SUCCESS in 1h 12m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5dc3d3e072ce4e19855ebe79736d0a09 : FAILURE in 58m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5692f48df72747b2a9001c7860fdeb58 : FAILURE in 25m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e76161ef3ee442888bc21fd0ee97368 : FAILURE in 2h 06m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a63d56d88dce41dc8ede782521b4ddfa : FAILURE in 25m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dad421546f9b45e68a56ec777099e05b : SUCCESS in 22m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a3df272de7184657879d3e19ad3df0a2 : FAILURE in 1h 04m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6dcdf6e31469440a9283544aff28af33 : FAILURE in 49m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2ceaef5cd38943f9b3b1fbd548b7d245 : FAILURE in 33m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ffeca890c63d40f09145a2c45716b9ad : FAILURE in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7b19c3cd624d3a74aff8771fff70837ed0c0fd2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-08 15:36:19.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-2/ns5_manila : SUCCESS in 1h 51m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f99873a75cf8bcf29be2ed3b121597a5b34da1a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-09 08:16:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2642b4b4e486f76d2fe62e2463ad2644d7e3d5eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-09 09:32:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/3/check/DellEMC_Unity_Manila_dhss_true/e8e56fc/DellEMC_Unity_Manila_dhss_true/103 : FAILURE in 50m 38s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/3/check/DellEMC_Unity_Manila_dhss_false/d1d35ce/DellEMC_Unity_Manila_dhss_false/95 : SUCCESS in 49m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"64d2d55dc4b8310a4d7f6bcb943c23405e00449e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-09 10:05:30.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-3/ns5_manila : SUCCESS in 1h 48m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"d4091c3b34d0aebc0184da17f8b5eb3f6830d35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 10:11:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1001c8828247444e8dc7c6d78752b9ab : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c91926aa2ca44f25be532e1cf6090249 : FAILURE in 18m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48367823414447d8a40182d26234c8c5 : FAILURE in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5bec6ed620442b94366179d6131c51 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/646e0f488a24466ea72a3167d929541c : FAILURE in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f54043b3ba5c449b93ed203cc6524f48 : FAILURE in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4da89a58bab24649aa88c3881ab7b225 : FAILURE in 8m 49s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc9dc4e30b584331ba18396e5691c7b7 : SUCCESS in 9m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38d0ae3414a947e98aed15f8868f1f41 : SUCCESS in 7m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/70384b7be2834f6b850aa2b0a26ceded : SUCCESS in 35m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5627479a9d1a452987a3be01f4735c29 : SUCCESS in 34m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dae185394e2f44ff8a82712215a2c63c : FAILURE in 1h 13m 10s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8fdbc9fea80a47c186948689c57d54db : SUCCESS in 1h 21m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/215ef024b0b84970b03647524670b3c8 : FAILURE in 1h 03m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8b3624b4eb9e40cbad362929ad3b2aee : SUCCESS in 47m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a9e8b06e7cc24ba2bbd2420a4b8376a5 : POST_FAILURE in 1h 32m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d50d6808341b465fb63972ea044cc422 : FAILURE in 25m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5f808ebc402e4787a5aee2db3f795814 : SUCCESS in 28m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/27313237de5b49dd900ecea4f76a245a : SUCCESS in 1h 12m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ec58049c630648649fe7f8c90154f2fd : SUCCESS in 1h 06m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ac47ac18bc3944f390068d6e9ed6f6e7 : FAILURE in 40m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/234bac751d3343d381ab40aaaae4fbd2 : FAILURE in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3a77e424bfe79cf9f67e94bcb58b673ac8508610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-10 09:48:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c80b167b39a78cb63d9491d918990465a3f1f85e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-10 10:26:44.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/4/check/DellEMC_Unity_Manila_dhss_true/758e810/DellEMC_Unity_Manila_dhss_true/106 : FAILURE in 37m 55s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/4/check/DellEMC_Unity_Manila_dhss_false/8307ee1/DellEMC_Unity_Manila_dhss_false/98 : FAILURE in 22m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"77add13d7cd32f3f86da99f025566297beded70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:29:37.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6afc82d46b4924acb822b05705116b : SUCCESS in 9m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e85bebc30fd4427abc177d04a8f3749 : FAILURE in 12m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c91e3302f54ee3aa36490fc3ff6fff : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6850fa3cf52b41f6b3a3e328bfef870d : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68f5eec2405b4927848f9f394543a486 : FAILURE in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b80b3a5896f451d813d8cb106854e79 : FAILURE in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7b4d7b4d5e144b1bedc7de486963151 : FAILURE in 7m 30s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/77554dedcfd34c9daa7fcc33d5ef7760 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35a152e93a5543fb92ae6e78e50f2d2a : FAILURE in 6m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/79554c352965429dadc37b78368e8651 : FAILURE in 14m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e404f5438f6b40ad8b140c27bb9e4657 : FAILURE in 20m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/59649c8d58584bbf97b36412328ff4c8 : FAILURE in 34m 29s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1de997fdfbd04c8caff98385ca491342 : FAILURE in 34m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba6cf174a7a74422aef6a5ef92e06fe3 : FAILURE in 32m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7960fb1e5d094ef28651d6e969492b71 : FAILURE in 24m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/32820aad835b473e9de84818e20a0aa6 : FAILURE in 36m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4198786085c24822922c22f90e4074b1 : FAILURE in 20m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/47566f1859674a0daf19e93e3a179c23 : FAILURE in 19m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c9c35d136814cedbbb7b4f8da67a93a : FAILURE in 25m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/05f28699b9e645bfbdb03fa134db02cc : ERROR Failed to update project None in 4s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/802ec450d24343ba9bc8827a51b566ac : FAILURE in 33m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8def8cce359844b19ffce82ea5c9713d : ERROR Failed to update project None in 3s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6442ca6056136c193389574e1aa30ca7baa0c3f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-10 10:36:20.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-4/ns5_manila : FAILURE in 47m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"f1c0302aaac2b77d0fbed77a5b8d2e1103f263a8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-10 20:04:30.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-4/ns5_manila : FAILURE in 1h 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"0163bc86435d5ed92ceba46fbc2f7d88f6fbff25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-11 02:27:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a28cd631eedac377bfae0510caf6e66e92147cff","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-11 03:30:57.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/5/check/DellEMC_Unity_Manila_dhss_true/8e42bb9/DellEMC_Unity_Manila_dhss_true/109 : FAILURE in 56m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/5/check/DellEMC_Unity_Manila_dhss_false/4d94679/DellEMC_Unity_Manila_dhss_false/101 : SUCCESS in 1h 02m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"fc98f110b7fdf99973c2e09a504f665cce65a748","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-11 04:13:58.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-5/ns5_manila : SUCCESS in 1h 45m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"f7d78b88abf010f1bfa025607628af59b201c2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 04:45:07.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2819677ead0465ba921e821cdbd5fed : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a5e3a42000842dbaeec8791ab4a029e : FAILURE in 13m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c73f8ab4112847f19ef66a63f404ea41 : FAILURE in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8a392ba17540e18ef4e379b90d19c2 : FAILURE in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9aec508bb2b449fa159d082aa549f21 : FAILURE in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/993807df77a547c8ad2c8212797c050f : FAILURE in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe037f3b3154446aa7d904a708257891 : FAILURE in 23m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/85ec4c268fbe4ea1931b34f0dbed76a7 : SUCCESS in 11m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4cd1d2208d774788ac8805ea6164d5b4 : SUCCESS in 6m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8763cfb833941168d2593848ca0c3bc : SUCCESS in 24m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/43b85944accb4d2ead250e6affa60368 : SUCCESS in 55m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2dd34b35e9fe444aa4266ed668cf61d4 : SUCCESS in 1h 02m 11s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4b69f6c6333642a98986a48b713348a6 : FAILURE in 1h 39m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/49821c8257a14d899af74b5867605556 : FAILURE in 1h 16m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a2e44a3d3bef4ed5ba2161e46a490448 : SUCCESS in 1h 03m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a0e2ed96815641d9907928707d4fa707 : FAILURE in 2h 15m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/07d14418248745d2abcb68dbdb71767c : FAILURE in 22m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/70036c488c084b3ca5d5be92c990b853 : SUCCESS in 22m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/23a4b3c88f7b42daa41c2e52aee0aff6 : FAILURE in 50m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c7146d9808d04fd49b9e68ee6ec911e1 : SUCCESS in 48m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7fa6e04f03b54395bd71c8740d778e3e : FAILURE in 32m 16s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4d321f169cd1412590ece41c40fda650 : FAILURE in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9b488c0c26e303a053bb1b0a57e5481d6e31b172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-11 05:46:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4cf24f48b94b4b447be6070bbda65e50820641d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-11 06:44:11.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/6/check/DellEMC_Unity_Manila_dhss_true/5917294/DellEMC_Unity_Manila_dhss_true/111 : FAILURE in 54m 03s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/6/check/DellEMC_Unity_Manila_dhss_false/f70d90d/DellEMC_Unity_Manila_dhss_false/103 : SUCCESS in 57m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"21a900d60a635bc89bb3e42629a26937a956b948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 07:33:45.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a540d93e704d4f8399359d93424da3 : SUCCESS in 9m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e093b556a41c4496adfd0875bf4a333a : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5faf5f9f8f0d4d10a2c613b3bb4fb8bd : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fef58ba808e45649580b3c8678c81d8 : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca186e2c7823422395d1d326286cae66 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab77af3d8dc46939c01cfb962faf8ba : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ec3efb0fde1496ea5a62030fabbfeb6 : SUCCESS in 8m 52s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c8daf5dd5644c65bac8bf374c1d5712 : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f65a147b12d24bf892315b0d2839bab6 : SUCCESS in 6m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0bf6209224024e7a87280a804703d5d5 : SUCCESS in 28m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b830b1cf7c6f4777a28fb99ec2a19193 : SUCCESS in 36m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9eee0b5b8bde4454ba0b87076d62dda0 : SUCCESS in 1h 10m 00s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0cdd910d1e644ad983dd7a989f920514 : SUCCESS in 1h 05m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/af06c49c0b6a4b929fac6050d417983b : FAILURE in 51m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d3d7eecc34b144c8b793bcd312a074a5 : SUCCESS in 40m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/91d76c70744f4428a0da88ebf266ce01 : FAILURE in 1h 45m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2f9c17d76b2d46aeb6b3bb3b25705c92 : FAILURE in 21m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/85a10583a9b64b50b7f2b8ce22b92716 : SUCCESS in 24m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/847e6e953c6f4ae3af4a7c700d85fd44 : SUCCESS in 1h 14m 29s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4a8a2f4431b34433bb137545b981177d : SUCCESS in 55m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/34446c570a9d47058c1ead6e3e1a7929 : FAILURE in 47m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3b772107ac7e437595ba2ef978b85baf : FAILURE in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c47e628b4cd5fd9ef77ec3b8f455c8c177694451","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-11 07:35:46.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-6/ns5_manila : SUCCESS in 1h 49m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"9ef261f91247922a4d289160b8237ad74ba33526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-11 08:05:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ada64d0abd98880a5d77bf0d6eb5d4016ef8d9a0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-11 09:56:13.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/7/check/DellEMC_Unity_Manila_dhss_true/3ba351c/DellEMC_Unity_Manila_dhss_true/113 : FAILURE in 56m 29s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/7/check/DellEMC_Unity_Manila_dhss_false/f6d1b91/DellEMC_Unity_Manila_dhss_false/105 : SUCCESS in 52m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b8bf54c42806428aaaa431827d563be6ac7790f2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-11 10:04:36.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-7/ns5_manila : SUCCESS in 1h 44m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"daf52e9217eb60047dfbd5c8cf41e42b6a14bce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 10:33:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c92cc0b282a42c7a86cdfa366a3f6e4 : SUCCESS in 12m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/484baed44ae44a07bb569faf2960a231 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7227f0426cc44ce99a3fe99f1eb9df8c : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c425c60f034b6fbc89758f7e98b543 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10ef43ad5d574eb3ac35eb2906329cc3 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd7bdac50cd644a99b0c4396e8d518e8 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/591beaa488d84fb38858fac3b098a13e : SUCCESS in 8m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/02fd07088e1a4bf5ad9d30cb37da61ec : SUCCESS in 5m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/018245e5054a49298fac7f2c0e1626c0 : SUCCESS in 5m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/32f5a1de63524874993409196f3a1edf : SUCCESS in 28m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12e1b2c0d91f4cfba97144b6e3948cc3 : SUCCESS in 42m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2cee6ca8cc2e4e848b68cb1ff7ef8672 : SUCCESS in 1h 03m 31s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/81e30ff48d8b49c1a0dee018c9433860 : SUCCESS in 57m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/76fb8b05363a480790c355d4ef8bac67 : FAILURE in 1h 06m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6a1b358a6fc6423884eade09d6fec9cc : SUCCESS in 46m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3de66ea7f1ad4c238c6db56fd88e0961 : FAILURE in 1h 37m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8f3aff1407eb4c0a99597e0b04a2adf3 : FAILURE in 23m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6ca26c684c964b51848035c2aea4d8cb : SUCCESS in 31m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0cf8149104aa4b2e9a94428c2c9eb34b : SUCCESS in 1h 11m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/15fbc10fb80e4358a2a9bcc4d5cb57d2 : SUCCESS in 52m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/448ebec2cdd1477bbad10d732425fb97 : FAILURE in 34m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b66043ff624549b685116bc1ecc979b9 : FAILURE in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3a33ead9722f132f2ef875fb838f1fdf812169ed","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2020-12-12 00:33:20.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"90c8cad5cbb96ef27b5c080b329b0b8ddacc6c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-12 03:10:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be4c51a81334ba2a4e17ace0d1769a1 : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bd87ede6f96492d84c5b5bfe482e161 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ea385ea5cb48b8856f57037a704452 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d4f7d20e04431582aa01ebbf7b72ff : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d5441a6cd747f6b7e61204624ca4ff : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f3f14d277b34a57b34f4ae0906fe6b1 : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8bbe66601b4b7bac33f16e28e9c370 : SUCCESS in 8m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f57cd466320b432881af730b0d4eb2ab : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d427ff6410b84f9387664f56373a7de0 : SUCCESS in 5m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/186db7daf4274e05ab0e0cb957cdaeaf : SUCCESS in 33m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a932958b00f44dbb54a3df1d3525706 : SUCCESS in 37m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c1c29fd7a5a045b894545ef3691c77ad : SUCCESS in 1h 19m 01s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/854cdef7b8994824bd51e3babf205e3f : SUCCESS in 1h 09m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e94de0157f045f5a96a6c22648eaf66 : FAILURE in 1h 04m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1424a5d38a8b448d941b5622d000bb51 : SUCCESS in 50m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/689148d942fa4e00b0fbca2165b6ab3a : FAILURE in 2h 32m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/664b93c24ee44df0bbacc7180db99c72 : FAILURE in 25m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c7cc87ecb53645dda8822750096d3a36 : FAILURE in 29m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1dc7fe0578dd4e5c81420f30c8c20388 : SUCCESS in 1h 08m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd8d248b2cf349c9a2ede0873e551eef : SUCCESS in 58m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5d95076054374259ba0b1918b8ffb6ac : FAILURE in 45m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/52b441c4508446338ce80a0b013e5d2a : FAILURE in 50m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"343c4889df7f31471c7003c9661db2a08ac9af9c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-12 06:48:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/7/check/DellEMC_Unity_Manila_dhss_true/b13e139/DellEMC_Unity_Manila_dhss_true/121 : FAILURE in 50m 41s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/7/check/DellEMC_Unity_Manila_dhss_false/37dcdc4/DellEMC_Unity_Manila_dhss_false/113 : SUCCESS in 59m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"a9f0e091aa954d4682ce87d4408053aeff7381ec","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-12 08:24:08.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/75/765975/7/check/DellEMC_VNX_Manila/889bf02/DellEMC_VNX_Manila/79 : SUCCESS in 1h 22m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"e95e5bbafa1d74f1b7fc827ac5d26ca625542d35","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-12 08:33:51.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/75/765975/7/upstream-check/manila-cDOT-ss/34ffe62/ : SUCCESS in 1h 40m 40s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/75/765975/7/upstream-check/manila-cDOT-no-ss/a43d005/ : SUCCESS in 1h 49m 31s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"b7465fc1c6dcd3c6d228d02fa95d8158adca0eb8","author":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"date":"2021-01-28 17:41:32.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"73163b1c6928ea222ee9225aa2719bdea27fcb11","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-29 19:17:48.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nHi haixin, the code looks good. I just have a comment inline, about \u0027manage\u0027 operation.\nAlso, I think that would be nice to provide a release note to include this improvement in the feature section.\nThanks!","accounts_in_message":[],"_revision_number":7},{"id":"6e5a39cedd2611f12031f46891b0ae5e37d55ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-01 07:34:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e3f8e7f410b5eefd4f765238cad5922e9f553b5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-01 07:37:36.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nhi, Douglas Viroel.\nthanks for your review.","accounts_in_message":[],"_revision_number":8},{"id":"0df2f4584b4b9ff79326cd799b1abfcbfde4df3f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-01 07:43:44.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/8/check/DellEMC_Unity_Manila_dhss_true/2b94905/DellEMC_Unity_Manila_dhss_true/220 : FAILURE in 26s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/8/check/DellEMC_Unity_Manila_dhss_false/777e00c/DellEMC_Unity_Manila_dhss_false/205 : FAILURE in 8m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d55e8b28a4e62d49b14ceea4786299ff1948ec79","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-01 07:44:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/75/765975/8/check/DellEMC_PowerMAX_Manila/4936071/DellEMC_PowerMAX_Manila/175 : FAILURE in 9m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":8},{"id":"f50ff7f562900f55698b83f0905e9fa9030009cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 09:15:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f5bc5832ca458cba51c933bb8eba93 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70a3cd3f2fc24956b4d1fbd022c852e8 : FAILURE in 12m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eb9e0eb08c14d69b80ad3decbe1e6d8 : FAILURE in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fcf322fd38948a4aca08ae642169b3d : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c2534bb52df4dcdb05375ae5bd92fbf : FAILURE in 14m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f3e46bc10fe47fdba5df839e8fa43f2 : FAILURE in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b5d8898d6d54a2aaffd01a919021522 : FAILURE in 22m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/988c596a653d44f683078f881f650816 : SUCCESS in 5m 21s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/00f7cdd76a1c42e580eff26bb09c61ea : SUCCESS in 6m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f25a136c1aea4cbb9a6a963fcfb1c870 : FAILURE in 5m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/12a5844003434b298717a532d00243e1 : SUCCESS in 26m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8b423fcd25e04bf48c3bd9ff13ec7284 : SUCCESS in 38m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c6c0e7492d4b4603a77f504391e35fce : SUCCESS in 52m 03s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bb1487136daa490ab1316a44d75ddc08 : SUCCESS in 1h 04m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/adf79a39df524e5e9881bd13c142808c : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/43eeb60b20d24ad9bde97191191d56a8 : SUCCESS in 41m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e140cd9f816b47928f822b4bc57051f6 : FAILURE in 1h 38m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6fdba038ebee40248874844a30701f7d : FAILURE in 22m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/99fe50042a1b4820af69adc0d7ff6059 : SUCCESS in 17m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e23f8f08bf434ef39071a541355e424c : SUCCESS in 1h 05m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d4bb315908e34d3099d0832966d969b6 : SUCCESS in 52m 42s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8d51584e16384404840898fbc87c5d16 : FAILURE in 31m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2defd653343843b98af2b896d9bb5a4f : FAILURE in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"68d0710c5c82f538779f9d7356c9373eda9f1f3e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-01 09:17:11.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-8/ns5_manila : FAILURE in 1h 41m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"b92cc9b8da48c3b5b3796fcb725714774e3470fd","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-02 01:17:22.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/75/765975/8/check/dsvm-tempest-manila-zadara-driver/b530672 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 47m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cb910d04a500ac770582c6cc43b2d206d874ecc9","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"date":"2021-02-04 15:53:10.000000000","message":"Removed reviewer Dina Saparbaeva.","accounts_in_message":[],"_revision_number":8},{"id":"2b6ce11e012326e9984ddb8fa721718587712155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-05 08:10:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f8218536ad5283ba4af4cf85345f66445f838f7f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-05 09:55:38.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-9/ns5_manila : FAILURE in 1h 43m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"d38f012b0176cfc3750848ce0a86456523c1ecd7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-05 10:26:19.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/75/765975/9/check/dsvm-tempest-manila-zadara-driver/f3349ca : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 46m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"57ac88459c1a2e6357e5e5e32d353297ae0585d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:29:57.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/608cf5c9f7324db3b2c125f9aaa579f9 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f057806a2ad4007b5de8a1e94267a27 : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8cc8e60b7f548799256d1327e38e3e7 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321fe37fb90f4de186ee269bb52ae164 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4275dd9f9d6e4af78e30d2a952fd007b : SUCCESS in 14m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1617064c31334dd28a197bcc9b01d03c : SUCCESS in 16m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12b4ed30aa0f45248cbf45c9608a479c : SUCCESS in 16m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cb0a2d6c1c74f47bbaf0120c23cad2a : SUCCESS in 5m 54s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab62826174e4456c8fb17c53281b9ff7 : SUCCESS in 5m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5f1b2158e5b45c2afe1daa44e083cc6 : FAILURE in 5m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6f94a2f715914828b4fe2331f241fe96 : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/000e78cf576f44ffb5390aab8c6fcd82 : SUCCESS in 35m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dbae008452a64db2a1c7bbef78d5f167 : SUCCESS in 1h 07m 11s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/655146f7bd214c3fbe8878d5327e953d : SUCCESS in 56m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ec7c92ec58dd4169a25a5e878ba2ffe7 : SUCCESS in 1h 31m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0d5cecef780448a9975fe10692d8467a : SUCCESS in 41m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77391eb132414954ae81a5530069966b : FAILURE in 2h 17m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c6e82036c64c43a08cdace1691dea90f : FAILURE in 22m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/23b60ff2781b42038f9b3385547150b6 : SUCCESS in 19m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e13e916ad0a14d1e974347511f7b2e98 : FAILURE in 53m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a69fe2b04c564dd7931a6a9bed4d5ac0 : SUCCESS in 52m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3151834dca7347f48b23b1778fcf180f : SUCCESS in 33m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d484e45e6fa04a3094274fd91149e8ad : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a943bfe289fc3ee5fc8628b2e5b963a784c00a6b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-05 12:19:48.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/75/765975/9/upstream-check/manila-cDOT-ss/cc95b5d/ : SUCCESS in 1h 49m 22s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/75/765975/9/upstream-check/manila-cDOT-no-ss/2abee4c/ : SUCCESS in 1h 44m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"1889b24df757cf75b3f6766e3d4a55248f11d9cc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-05 12:46:46.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/75/765975/9/check/DellEMC_VNX_Manila/1cc775b/DellEMC_VNX_Manila/157 : SUCCESS in 1h 10m 51s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"74758de4111c5bf4f3d9222df87cf9a22189c63a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-05 12:48:37.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/75/765975/9/check/DellEMC_Unity_Manila_dhss_true/5e865c9/DellEMC_Unity_Manila_dhss_true/263 : FAILURE in 1h 44m 30s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/75/765975/9/check/DellEMC_Unity_Manila_dhss_false/7966a4a/DellEMC_Unity_Manila_dhss_false/246 : SUCCESS in 47m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"725a54a46fa9d8b985b00a4496f99e69d931d16b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-05 16:47:59.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-9/ns5_manila : SUCCESS in 1h 38m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"16a8654fe96af4292a89d8cbc34d4b3bb35c9e37","author":{"_account_id":32531,"name":"Dina Saparbaeva","email":"dinas27@outlook.com","username":"disap"},"date":"2021-03-02 09:39:46.000000000","message":"Patch Set 9: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"c7197a17265e7dd329564d11ea2710bda1c72bf5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-05 07:55:44.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)\n\nThank you for this change, Haixin - some comments inline","accounts_in_message":[],"_revision_number":9},{"id":"25fd0d0040f7a9883a8b3afe3b10183ebf092854","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-05 07:58:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1c8e0da7102ad35c95594b8d3245ae3292947580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-03-06 06:14:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0befe7a2939a558688026a8a38a7ce311070b34a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-03-06 06:31:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5dc06ed5eb0b9c21481e0b73f7de6902abca2d60","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-03-06 06:37:04.000000000","message":"Patch Set 10:\n\n(5 comments)\n\nhi, Goutham Pacha Ravi.\nthank very much for your review.\ni have added microversioni 2.61, and improved documentation.\nand add check integer when create share type or crate extra\nspecs or update extra specs.","accounts_in_message":[],"_revision_number":10},{"id":"2726ff63b0e3edf12af50fb9af30d948e656da8f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-06 07:59:27.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-10/ns5_manila : SUCCESS in 1h 44m 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"f51b4a57b8487eb59c020b2bc98559068ffb7be1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 08:11:20.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/75/765975/11/check/DellEMC_PowerMAX_Manila/90157ca/DellEMC_PowerMAX_Manila/285 : SUCCESS in 1h 39m 28s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"a0fcc3ae1f4e61593b1440fb54152572422971df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 08:26:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fe509f64274fcdb28ced2a1b0a81b5 : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b788670781433e9e98ae53fa9b233d : SUCCESS in 29m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd7c282301444bb4be0cf1dd14862106 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7f7c0249404a4d8dad8dc230320e5e : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6891118b8a154afbb62996b56a5981d3 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e3b7b6d43c04155bc8c89cce33f39e9 : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7d4df18856846e1b2a801740dfa74be : SUCCESS in 12m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe5c8fe3b1844af9862779edf08ad525 : SUCCESS in 8m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e1f9c1902bbe4cb99f861250a1070eb7 : SUCCESS in 7m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c7aea86e4978414e9c0e5d24e0e81fec : FAILURE in 8m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ef611d6f9c4342179606146f89e0fc21 : SUCCESS in 33m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b2db6a29bbff40aab9322f089583c0db : SUCCESS in 56m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9037e8e07ce340f5bada0b7a67dbc978 : SUCCESS in 1h 01m 38s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e3d00de6b2d146dcbcf2058af2a8860d : SUCCESS in 1h 09m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/da5ff5e749714482920f5891326ba963 : SUCCESS in 1h 44m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/32842de1223749848918915455134833 : SUCCESS in 31m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/75d9857b88a24d9fb4a05333eaee4b68 : FAILURE in 1h 38m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4608c712070f482a85ced4bbd336342c : FAILURE in 25m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/080787f3ef6d4cf191573f87b4896df5 : SUCCESS in 30m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ede33b34cbe440fbb00eb7b929570b71 : SUCCESS in 1h 19m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a7b98513ed7d47acadc48aded79da625 : SUCCESS in 59m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/300bc34ed08b49b4ab03230d443a8756 : SUCCESS in 39m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d1135b90013d452186568e2ff1fb51bb : SUCCESS in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6af2f246e9caa3b1207eff75ab67985a155a5e98","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-06 08:59:42.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-75-765975-11/ns5_manila : SUCCESS in 1h 44m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"c1e047943982d5833618ae08705ee8cd39830609","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 10:04:55.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/75/765975/11/check/DellEMC_VNX_Manila/60dfff0/DellEMC_VNX_Manila/247 : FAILURE in 1h 38m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"22f4f6d5581db630283a2b7b71ae793d5e488f66","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-06 10:20:58.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/75/765975/11/upstream-check/manila-cDOT-ss/1603371/ : FAILURE in 13m 14s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/75/765975/11/upstream-check/manila-cDOT-no-ss/a585c97/ : SUCCESS in 1h 54m 23s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"4daf5747e2a4a726649bb188f5b1088ad19260c8","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-06 11:36:28.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/75/765975/11/check/DellEMC_PowerMAX_Manila/afd34fe/DellEMC_PowerMAX_Manila/287 : SUCCESS in 1h 49m 58s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":11},{"id":"7c3b58cd565357d64b048fc50985e1eaeb45056a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-03-06 21:24:59.000000000","message":"Patch Set 11: Code-Review+2\n\nThanks haixin, the code looks good to me. Thanks","accounts_in_message":[],"_revision_number":11},{"id":"479ca8f57557143bb1e31cc672907f617f2bc2cf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-03-07 14:25:35.000000000","message":"Patch Set 11: Code-Review+2\n\nLGTM. Thanks, Haixin!","accounts_in_message":[],"_revision_number":11},{"id":"94da762b8c40768790ba1095c4dd9695f3cc27a8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-03-07 21:43:23.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(6 comments)\n\nSome documentation nits - please fix them up in a follow up patch, but LGTM otherwise. Thanks for making these optional extra specs tenant visible as well - and for adding the tempest tests, Haixin","accounts_in_message":[],"_revision_number":11},{"id":"9b27c9004d986593ca08d9153fc3b046f60ec909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 21:43:41.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"58558c905aca31fba25530865bd1b1294f269295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 22:59:35.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2301dcc2d8b4459dba588df1ffe4377b : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/891cb388b72d4897a9f4ffa0f2327edf : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a229f5b6645b409eb7157047bb884693 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c1396568b254bc091186f8b435532ca : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd8e57caa07f4e2aa4da0436e112e676 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded0f9f0dab64bc2b5b2eace2571be84 : SUCCESS in 5m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5fd1fc1e95e64c848ca7480c0076e5e8 : SUCCESS in 26m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1752dbb32e8f443e87053ff59ac55c42 : SUCCESS in 35m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fd185d40cf7d4b42aff68b9144de3b53 : SUCCESS in 1h 14m 18s","accounts_in_message":[],"_revision_number":11},{"id":"c078a6485b8c0db5633911ed48b97d40bb1e0687","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 22:59:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"ac517e09f9d30c3e67ef4fe812bdb6740c99e64c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 23:00:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b7757cfd7141f5ab622838279c2cd7 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/658d8a6bae7a44418abe05bdeaea0383 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"74415f6d4a7bb8b72be3e81abf79fd8597100f27","revisions":{"62c713354f79ffc7f880bcde425236afd66cea0b":{"kind":"REWORK","_number":1,"created":"2020-12-08 11:40:49.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/1"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:14.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/62c713354f79ffc7f880bcde425236afd66cea0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/62c713354f79ffc7f880bcde425236afd66cea0b"}]},"branch":"refs/heads/master"},"a2279f83719861991586ff76b39c77b169bff382":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-08 11:44:28.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/2"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:44:28.000000000","tz":0},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a2279f83719861991586ff76b39c77b169bff382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a2279f83719861991586ff76b39c77b169bff382"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2b0d059002e835aa38d5a097eefc1f19c1d219cd":{"kind":"REWORK","_number":3,"created":"2020-12-09 08:16:12.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/3"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-09 08:15:39.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2b0d059002e835aa38d5a097eefc1f19c1d219cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2b0d059002e835aa38d5a097eefc1f19c1d219cd"}]},"branch":"refs/heads/master"},"7252b2c319b41178f166171e4a99b363cda9ca2c":{"kind":"REWORK","_number":4,"created":"2020-12-10 09:48:23.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/4"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-10 09:47:38.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7252b2c319b41178f166171e4a99b363cda9ca2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7252b2c319b41178f166171e4a99b363cda9ca2c"}]},"branch":"refs/heads/master"},"0ae835f11ce3cd4d12b64093bf169279d34d3658":{"kind":"REWORK","_number":5,"created":"2020-12-11 02:27:49.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/5"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-11 02:26:58.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ae835f11ce3cd4d12b64093bf169279d34d3658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0ae835f11ce3cd4d12b64093bf169279d34d3658"}]},"branch":"refs/heads/master"},"c1ca5338bb7db23c21d14832371dafca765a28a5":{"kind":"REWORK","_number":6,"created":"2020-12-11 05:46:12.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/6"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-11 05:45:24.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1ca5338bb7db23c21d14832371dafca765a28a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1ca5338bb7db23c21d14832371dafca765a28a5"}]},"branch":"refs/heads/master"},"b3031b24215691b49fd1ba788f0a680a8ccc2c53":{"kind":"REWORK","_number":7,"created":"2020-12-11 08:05:35.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/7"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-11 08:03:38.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b3031b24215691b49fd1ba788f0a680a8ccc2c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b3031b24215691b49fd1ba788f0a680a8ccc2c53"}]},"branch":"refs/heads/master"},"814d9b27b65d63d9bb3bae10be17b0a8a6299e4f":{"kind":"REWORK","_number":8,"created":"2021-02-01 07:34:46.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/8"}}},"commit":{"parents":[{"commit":"fdfed2830db23f7226fb63a97f48bdc7c0acaf4a","subject":"Merge \"remove usage of six library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fdfed2830db23f7226fb63a97f48bdc7c0acaf4a"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-02-01 07:30:43.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start. when manage share,\ncheck it after get true size of share at manager layer.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/814d9b27b65d63d9bb3bae10be17b0a8a6299e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/814d9b27b65d63d9bb3bae10be17b0a8a6299e4f"}]},"branch":"refs/heads/master"},"de6147a8928b3edb917faed1a6645d1914c6069d":{"kind":"REWORK","_number":9,"created":"2021-02-05 08:10:20.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/9"}}},"commit":{"parents":[{"commit":"fdfed2830db23f7226fb63a97f48bdc7c0acaf4a","subject":"Merge \"remove usage of six library\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fdfed2830db23f7226fb63a97f48bdc7c0acaf4a"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-02-05 08:09:34.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start. when manage share,\ncheck it after get true size of share at manager layer.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de6147a8928b3edb917faed1a6645d1914c6069d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de6147a8928b3edb917faed1a6645d1914c6069d"}]},"branch":"refs/heads/master"},"86e48d10746e246ee5e00498f67524758099db53":{"kind":"REWORK","_number":10,"created":"2021-03-06 06:14:21.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/10"}}},"commit":{"parents":[{"commit":"067d44aab6ed67f2bfd3d340e7237961ad8c140c","subject":"Merge \"Implement secure RBAC for storage pool statistics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/067d44aab6ed67f2bfd3d340e7237961ad8c140c"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-03-06 06:13:32.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start. when manage share,\ncheck it after get true size of share at manager layer.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/86e48d10746e246ee5e00498f67524758099db53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/86e48d10746e246ee5e00498f67524758099db53"}]},"branch":"refs/heads/master"},"74415f6d4a7bb8b72be3e81abf79fd8597100f27":{"kind":"REWORK","_number":11,"created":"2021-03-06 06:31:26.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/75/765975/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/765975/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/765975/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/765975/11"}}},"commit":{"parents":[{"commit":"067d44aab6ed67f2bfd3d340e7237961ad8c140c","subject":"Merge \"Implement secure RBAC for storage pool statistics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/067d44aab6ed67f2bfd3d340e7237961ad8c140c"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:35:03.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-03-06 06:30:58.000000000","tz":480},"subject":"Adds support min/max share size limited by share_type","message":"Adds support min/max share size limited by share_type\n\nAllows set min/max share size that can be created in\nextra_specs for each share_type.the share size will\nbe checked at API level as part of share create,\nextend, shrink, migration_start. when manage share,\ncheck it after get true size of share at manager layer.\nnew extra_specs keys are supported for set min/max\nsize of share.\n\u0027provisioning:max_share_size\u0027\n\u0027provisioning:min_share_size\u0027\n\nImplements: blueprint share-size-limited-by-share-type\nChange-Id: I5ce0fabf59bfca5ebaf0be5ffe9986e2b0480295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74415f6d4a7bb8b72be3e81abf79fd8597100f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74415f6d4a7bb8b72be3e81abf79fd8597100f27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
