)]}'
{"id":"openstack%2Fmanila~765984","triplet_id":"openstack%2Fmanila~master~I4b9fc035a7903356d9136f56f0fae7991716f1e4","project":"openstack/manila","branch":"master","topic":"bug/1907244","hashtags":[],"change_id":"I4b9fc035a7903356d9136f56f0fae7991716f1e4","subject":"if share type not specified, should use default share type","status":"ABANDONED","created":"2020-12-08 11:56:25.000000000","updated":"2021-02-18 01:42:59.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"543c0d5b30b6fd3a7d9738602d4cdc45297b9a65","_number":765984,"virtual_id_number":765984,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-01-04 04:05:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":-1,"date":"2021-02-10 11:31:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":-1,"date":"2021-02-11 00:34:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 14:03:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 14:40:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-08 17:18:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2020-12-23 06:29:22.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-23 07:26:59.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-23 09:33:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-24 01:28:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2020-12-24 01:28:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2020-12-24 01:28:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-12-24 01:28:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2020-12-24 01:28:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-02 07:07:39.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a9c57215c3ca412b3b5dd3ea4aa67b417596bd59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-08 11:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19899e36fb4fac6c25900bc65534a64703200cf0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-08 12:05:23.000000000","message":"Topic set to bug/1907244","accounts_in_message":[],"_revision_number":1},{"id":"2fb728b5ab681d45056760d2a6a71c088ecb5757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 14:03:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa005dba24340638f7a404febaafc7c : SUCCESS in 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6bf036351af4cb6947a09d313da38b5 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2faba5838f4aa094bbba79e755626e : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25fa9328e0144c398eccc3ccda19d46 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/165f380017f34d11b00a09fade4e8817 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/595f7d39dbb74b8783fa5235de800e74 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f689f30967a442da9fca07f10d91a80 : SUCCESS in 7m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dff8a84f78224fbab702b3507fd45bf0 : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2dd40b36f2a0450b8245c2cfbf274f15 : SUCCESS in 4m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c876b07878ad4840ac2ca254f110c549 : SUCCESS in 32m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/192dc1b3975140fbbd4f2b333d1d46b9 : SUCCESS in 40m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/87f342812b7f4016af68eb3aad3884e8 : SUCCESS in 1h 18m 29s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2bde73408d5c46f09f4bffeb0b5a12c2 : SUCCESS in 1h 13m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2610ec9b5c494fbfbfb1270312de95e8 : FAILURE in 1h 15m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cfb28f52fcff45b08b5e42d5bd938586 : SUCCESS in 46m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7f80180f64c54c0f87a53effa0a45860 : FAILURE in 1h 59m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1c368d0f914644babb3b337728a77991 : FAILURE in 29m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e4b2626379d4e2caacb9828daeaa568 : SUCCESS in 19m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1e1b20a544b74af1b3b53bdb030d3276 : FAILURE in 1h 13m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f2101de04ef3412ebcbbaffc66eb47bb : SUCCESS in 1h 00m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c4253a25c0ef46cab95f5d2e09a5c4f8 : FAILURE in 55m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/17f8039b87f9406aa49a7481034ed43a : FAILURE in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"946d6f32849f9c317a4bca4ce4eacca8240e6723","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-08 14:40:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-765984-1/ns5_manila : SUCCESS in 1h 58m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1acfa44d23803e57085859c9354a30edf0173ded","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2020-12-08 17:18:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/84/765984/1/check/DellEMC_Unity_Manila_dhss_true/10fd878/DellEMC_Unity_Manila_dhss_true/99 : FAILURE in 1h 49m 26s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/84/765984/1/check/DellEMC_Unity_Manila_dhss_false/8f37fcf/DellEMC_Unity_Manila_dhss_false/91 : SUCCESS in 1h 03m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1ecdda3a677e5560b87a614bc7418c581d7ebd71","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-09 07:53:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a165594fef1012b675a7dc2b4dce03d0c7584f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 09:56:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07f3b7115444a1bb295698979db6313 : SUCCESS in 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc1349e60ba043a484e05f080c10339b : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7838360e950c4df3a382cc6db8f1c97f : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7988846d1cd4eeda6d1cc46dcb4b1bf : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647356e345dd49eaa686124a3c44edd6 : SUCCESS in 9m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/195f0022f99548dd9ab70c26a864da88 : SUCCESS in 10m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ba03aa8a7a34cfcbd0404938e480f36 : SUCCESS in 10m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cd11ca00c492452ca43c2f1c7500e142 : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d50667141bf4b05a331c0285942e9eb : SUCCESS in 12m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8390f7ec78a64845b7ecdc23c4b16569 : FAILURE in 28m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2667b77204574b9b8687ab52a8d3791d : SUCCESS in 33m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/959d2e6e7ec44fc89fbfc0b1ce4c5cfd : SUCCESS in 1h 09m 19s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4f9f6cb966364ca38d583bf9e78110b0 : SUCCESS in 59m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d4bb4699252e48ec941496fad9498771 : FAILURE in 57m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/816938ba0f0c4ce28ce5f574209ee7da : SUCCESS in 47m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d0a5afd227b477abdd15b2f78317b5c : FAILURE in 2h 00m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6a69ed24715a4fe28ebd093a999fec4d : FAILURE in 23m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/834686fdb05c490db68abd18b734883c : SUCCESS in 21m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b9659dd51304413a092a9ef00fb8d40 : FAILURE in 1h 19m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d95d7653c1fa48668d0a9431ccd3d203 : SUCCESS in 1h 04m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6c91a5a2122649c7a8a40b3c9683ba06 : FAILURE in 39m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ef5446826bb74955aa5ab230947c4465 : FAILURE in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49caa04e0110585b1536b1b6dda9706143b11fe5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2020-12-23 03:17:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"086b5aff7466e2357911b0eb4ff27f4782234ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 05:26:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39da47fac2c24e28a343e5107699ac13 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ef23de931b44e0a91b671ab903ba43 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2575124d3e3e4456a695a8d188be6033 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be0ebd810a647bcab93fb9677f0ba8c : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b0f7de111974ead8a7437f75d814f85 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6631877888364f49ab0613ad7d2590b3 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54c610e2a9974b4fac7eaa624c11af10 : SUCCESS in 6m 49s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1fdc08fc7be74ed9b1209246284363f9 : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d116ee98dd7496b9851c35207da89e4 : SUCCESS in 4m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/02b045dc603b4fcf98790abdd712ac94 : SUCCESS in 30m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c16da0d354d4250a0611b97a36cab6d : SUCCESS in 39m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/44448223fd7e419086d093e92c73e876 : SUCCESS in 1h 15m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f0a71f7a4f0140759d3374de2b17399c : SUCCESS in 59m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7669910f907e495bbb6ffa920ef18506 : FAILURE in 59m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/acb881915dc94b61be9a562ded587720 : SUCCESS in 45m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/453ef4dff26b47879ef75c18b4af534f : FAILURE in 2h 04m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7f814ff8d88d43ebaf8e152ebaeeab1c : FAILURE in 23m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b82005b7cf39469480b887d0430f590a : FAILURE in 25m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/489d48a1c2944ce4bdba9ef6c87b1fc1 : SUCCESS in 1h 08m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d007cebbb7aa406189cf32fde4a5eb76 : SUCCESS in 1h 01m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6b1e18f4dd74453a9fa47bf07bcb4378 : FAILURE in 38m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/df171eadc94b4924b3f76cff3b8fa018 : FAILURE in 45m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77f5fb78d7ba777e272bfa4dc02f378ba049b576","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2020-12-23 06:29:22.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/84/765984/1/check/DellEMC_VNX_Manila/cceaa82/DellEMC_VNX_Manila/88 : SUCCESS in 1h 02m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d77e40cb1917bf0d673468cc0c4cd9c7beff4e79","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2020-12-23 07:26:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/84/765984/1/check/DellEMC_PowerMAX_Manila/1ae2036/DellEMC_PowerMAX_Manila/136 : SUCCESS in 2h 00m 16s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9e56149971ce47569fe63bfbd4395d08752f3f7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2020-12-23 09:33:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/84/765984/1/upstream-check/manila-cDOT-ss/0aa5c4a/ : SUCCESS in 1h 47m 47s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/84/765984/1/upstream-check/manila-cDOT-no-ss/c822fa9/ : SUCCESS in 1h 48m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"0782d8bf262d109f4656f3e75d23d1ccbe586cd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-02 06:21:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b48584a55be199afd838a62396535224af4b41d7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-01-02 07:03:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/84/765984/2/check/DellEMC_Unity_Manila_dhss_true/90d8e73/DellEMC_Unity_Manila_dhss_true/153 : FAILURE in 41m 05s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/84/765984/2/check/DellEMC_Unity_Manila_dhss_false/07e1c82/DellEMC_Unity_Manila_dhss_false/144 : FAILURE in 32m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"18cfb0ce61c8140e58222ff83e29f0d269af925e","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-02 07:07:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/84/765984/2/check/dsvm-tempest-manila-zadara-driver/faed87e : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6e10dc9c26a6e21c57e7b8b63aed6369ec5ae70","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-02 08:05:06.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-765984-2/ns5_manila : SUCCESS in 1h 42m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e27892d797d94f782442558ef8d080ff7377007f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 08:14:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5034cadf96427497ddca356f678a05 : SUCCESS in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1edeabd488a54f9db25e042f714678f0 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7a1e051f6ea498fb030dbf5bbf2c3bc : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bdb72a77b9474bb4785b6973185757 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b575b5fdc314b3e8db5b522d299ad71 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21b76993115d4622a4a8d1199e2664e4 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0c6664aacb74f78924bc9678e9f054e : SUCCESS in 8m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ddfa34a082af49c2ba31ad87d8e6804b : SUCCESS in 6m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/81db2ad66cca4d8eb5caa2c4ed1ae686 : SUCCESS in 6m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6a6cb1a5bf84432981439f5d39234338 : FAILURE in 26m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/30fdc2cb877f4d0aa066bc819f2622ce : FAILURE in 35m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/02998351236c4780a4251164cbc40a9d : SUCCESS in 1h 06m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ed40f7ff1dc54775831490849551dd86 : SUCCESS in 50m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d64e7b8224de42369f14dd34eb06414c : FAILURE in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/34a0c642dc444c25b9edc224582ee689 : SUCCESS in 47m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4f9e8d231c0b47a88a65c2a59fa06b2c : FAILURE in 1h 46m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/78fe0fd8fd2d42ae91a11ea7d6601409 : FAILURE in 23m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9813411c1cdb4f33b0b507c44d22a1bb : SUCCESS in 26m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a78370d4b0904e76873e7cc1f6b84f69 : FAILURE in 1h 01m 33s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cd4b163e7c714afe89f28164e282bc2d : SUCCESS in 58m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/babeebd2ce2149668b59195eaf15c21c : FAILURE in 42m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4ce429b4fb204f55a0f34b270f2ed13c : FAILURE in 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"682881a55e00263b72fd64687415bc56a713cdce","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-02 08:54:20.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/84/765984/2/check/DellEMC_PowerMAX_Manila/eb78cff/DellEMC_PowerMAX_Manila/153 : FAILURE in 2h 32m 24s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ff0e0b5f1a5229a3e2495e074c8add75eebc46fb","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-01-04 01:40:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3113fd3decb2d6716b0079bf6ac799abd3efb845","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-01-04 02:28:42.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/84/765984/2/check/dsvm-tempest-manila-zadara-driver/8a48159 : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 47m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60d64e14ce14aa395cb07966cd5d2867dee72bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 04:05:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60a6a3f6a6a442698b87dd68bd608ca : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4024d022aa34b2f9ded802bd6c912f5 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d48e496243934b56bd350ecd5e2fa9e4 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9024f1e116d443c2aeccb233e6ecc88e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79321af2a76545ae8a163c0455edc8bd : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8acbd86ab5e457b8111613a383742c9 : SUCCESS in 11m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b55de211308d4668a54ab5ec35ca2d07 : SUCCESS in 7m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8b57f29fb1244e8eaf0fe452ea7aaac0 : SUCCESS in 6m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/33f3068cc54c43bab08b8951b6b8d389 : SUCCESS in 4m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ded64fc7c0d94574b976e3ab5ed4b885 : SUCCESS in 29m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0ac5d3eb1a9744fbad2fd5c42a7bfb07 : SUCCESS in 40m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/17c6fbf1ae5f464ba80c42975383e91f : SUCCESS in 1h 09m 14s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c926427b385f46849a89758b3fa1e039 : SUCCESS in 51m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/11e511c5f74244fe9e8cb4fbc390f3ff : FAILURE in 49m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8ddc089754204580a73bc134ff40c5cb : SUCCESS in 39m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/51e85d9dd8cb450fabc4809f6e48178c : FAILURE in 2h 20m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/aa2f06797c3948f4bad8c754db7dd16e : FAILURE in 19m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ffe80598b62f4d8a82df82c2ab25e90a : SUCCESS in 20m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c341b4572be9448c93e8dde01949a15d : SUCCESS in 1h 06m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/48f3b849bc244083befd9e94db85aa54 : SUCCESS in 53m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d49c9537d98348d896dcf12c81e2244c : FAILURE in 32m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/001a7e6ab5d4432a9cabe66999497e49 : FAILURE in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"acd2004edb29af76d5b5dfd0a80db18492ee53d1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 05:10:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/84/765984/2/check/DellEMC_VNX_Manila/6c8dff7/DellEMC_VNX_Manila/98 : SUCCESS in 1h 04m 33s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9a7018d75dab909e2490230aca81f57c228c290b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-04 05:51:51.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/84/765984/2/upstream-check/manila-cDOT-ss/ca4fa4e/ : SUCCESS in 1h 45m 42s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/84/765984/2/upstream-check/manila-cDOT-no-ss/17faaca/ : SUCCESS in 1h 44m 35s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"a5831b96758f63b28aa67570851c2199a8bc14c0","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-01-04 10:30:20.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/84/765984/2/check/DellEMC_PowerMAX_Manila/2cf4711/DellEMC_PowerMAX_Manila/157 : FAILURE in 3h 48m 06s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"7e35907a755d9171ed9daab31cc99c04dcd64eb1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-06 06:45:02.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-84-765984-2/ns5_manila : FAILURE in 1h 50m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"741cb78945537f9e426a4dbca1406d91afd3c11a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-02-10 11:31:24.000000000","message":"Patch Set 2: Code-Review-1\n\nHello, haixin! Thanks for working on this!\nCould you please add a release note?","accounts_in_message":[],"_revision_number":2},{"id":"20121d836a7cda363112063823f753e9594ff72c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-02-11 00:34:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nHaixin, do you have a log when this failure occurs - we already do what you\u0027re doing, please see my comment inline","accounts_in_message":[],"_revision_number":2},{"id":"2897f9c07ff55ebade9c721e7590e2a4186841ed","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-18 01:41:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"543c0d5b30b6fd3a7d9738602d4cdc45297b9a65","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-02-18 01:42:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b1cab51ab6c5cf64556401a8cc67ec4b4550886","revisions":{"766f14107b803136b4b9b64eb3a672102fb85c1b":{"kind":"REWORK","_number":1,"created":"2020-12-08 11:56:25.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/84/765984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/84/765984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/84/765984/1"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:55:01.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:55:08.000000000","tz":480},"subject":"if share type not specified, should use default share type","message":"if share type not specified, should use default share type\n\nwhen create share, if share type not specified, should\nuse default share type.\n\nChange-Id: I4b9fc035a7903356d9136f56f0fae7991716f1e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/766f14107b803136b4b9b64eb3a672102fb85c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/766f14107b803136b4b9b64eb3a672102fb85c1b"}]},"branch":"refs/heads/master"},"6b1cab51ab6c5cf64556401a8cc67ec4b4550886":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-02 06:21:36.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/84/765984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/84/765984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/84/765984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/84/765984/2"}}},"commit":{"parents":[{"commit":"d8dab2d077d47d4fa894ba05d359dde357a63fd5","subject":"Merge \"Fix logic that determines a share exists before manage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d8dab2d077d47d4fa894ba05d359dde357a63fd5"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2020-12-08 11:55:01.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2021-01-02 06:21:36.000000000","tz":0},"subject":"if share type not specified, should use default share type","message":"if share type not specified, should use default share type\n\nwhen create share, if share type not specified, should\nuse default share type.\n\nChange-Id: I4b9fc035a7903356d9136f56f0fae7991716f1e4\nCloses-Bug: #1907244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b1cab51ab6c5cf64556401a8cc67ec4b4550886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b1cab51ab6c5cf64556401a8cc67ec4b4550886"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
