)]}'
{"id":"openstack%2Fmanila~768415","triplet_id":"openstack%2Fmanila~stable%2Fvictoria~If139927730071448abc04e1ea7ebb615749e7e3d","project":"openstack/manila","branch":"stable/victoria","hashtags":[],"change_id":"If139927730071448abc04e1ea7ebb615749e7e3d","subject":"Delete unavailable py2 package","status":"MERGED","created":"2020-12-29 02:26:14.000000000","updated":"2021-01-11 21:02:10.000000000","submitted":"2021-01-11 21:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768415","cherry_pick_of_change":762273,"cherry_pick_of_patch_set":2,"meta_rev_id":"77045746552a84e43912bbfaec23aa8369d22979","_number":768415,"virtual_id_number":768415,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-11 21:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-01-11 20:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-01-11 20:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 02:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-29 04:06:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-11 20:50:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"58ae02e7270cd9dce2d485b00df1c3ced8f9ce9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-12-29 02:26:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"61925fcf812f3ea88e9575c3f07249b16b883d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 02:36:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f4f3832d44419aa0365503fa1088e0 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ae70290c1d46a0b77fa7bf0ca9f276 : SUCCESS in 4m 31s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/763c3ca660564ed59b3c1ad58209b7da : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d707d99ec7128aacd59a733681d6b45445d7d1cd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2020-12-29 04:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-15-768415-1/ns5_manila : SUCCESS in 1h 39m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"dff77e2f67d18f67109d19bcb14af673a1b75175","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-11 20:50:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"828e880077f61cb1a689f7bcb94cf27ea5dc927c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 20:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"829c453d69dfc8ac118792261daad2263d62e983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 21:00:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecedf28f5db441079ef0f682e5842068 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273bbe6c6f66400cb42159d6b4ae0245 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5e8ad477156f803ce369da6651f09ac4e22d6361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 21:00:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"77045746552a84e43912bbfaec23aa8369d22979","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 21:02:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cc10b3e1194e68bf7ca82dafda554d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d761b111b4a789f73dda41c518a380af56a3664b","revisions":{"d761b111b4a789f73dda41c518a380af56a3664b":{"kind":"REWORK","_number":1,"created":"2020-12-29 02:26:14.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/15/768415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/768415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/768415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/768415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/768415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/768415/1"}}},"commit":{"parents":[{"commit":"9e2ea96cb23d43c4d889ab7005a96b2ca9ce4c9a","subject":"Merge \"[glusterfs] don\u0027t reinit volume list on deletion\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e2ea96cb23d43c4d889ab7005a96b2ca9ce4c9a"}]}],"author":{"name":"linpeiwen","email":"root@tox-py36.novalocal","date":"2020-11-11 02:54:56.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2020-12-29 02:26:14.000000000","tz":0},"subject":"Delete unavailable py2 package","message":"Delete unavailable py2 package\n\nThe openstack Ussuri and Victoria versions no longer support the\nCentos7 and pyrhon2 environment packages. Correct the missing\nproblems in the latest document\n\nChange-Id: If139927730071448abc04e1ea7ebb615749e7e3d\n(cherry picked from commit bab899068c14d83c5a3c476167f6bdfadf23e8e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d761b111b4a789f73dda41c518a380af56a3664b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d761b111b4a789f73dda41c518a380af56a3664b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
