)]}'
{"id":"openstack%2Fmanila~770593","triplet_id":"openstack%2Fmanila~stable%2Ftrain~I96d49f84122a34701328909c929ede4d66746911","project":"openstack/manila","branch":"stable/train","topic":"bug/1894362","hashtags":[],"change_id":"I96d49f84122a34701328909c929ede4d66746911","subject":"[glusterfs] don\u0027t reinit volume list on deletion","status":"MERGED","created":"2021-01-13 18:57:06.000000000","updated":"2021-06-02 16:18:08.000000000","submitted":"2021-06-02 16:13:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770593-bug/1894362","cherry_pick_of_change":767986,"cherry_pick_of_patch_set":1,"meta_rev_id":"8fef6634d91b6b2df1df7db38e6e0488f56711c5","_number":770593,"virtual_id_number":770593,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 16:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-01-25 21:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-01-25 20:25:08.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2021-01-25 21:08:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 21:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 22:53:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 18:40:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-01-25 18:40:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-01-25 18:40:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"}],"messages":[{"id":"a69594e22b9a1e92255b9e01576f3169b3f0b3d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-13 18:57:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"f414a44d0ce56a44c69a5883a5cf8097b1b199db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 21:09:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5125a78c8c248528774a881d3e59d59 : SUCCESS in 20m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fcccfe936f04d168f8b29b528c351a0 : SUCCESS in 14m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/682dce6889734e29b1f80f793c2d71d4 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc9b1fd152b436291a5817338fbf9b6 : SUCCESS in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/582c8e660eb44a3db728a93eaa566e28 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10266600b47243dba95c643126672cdd : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3309d6e99d8f4788a5457ad0be9c6fe3 : SUCCESS in 7m 11s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/abb7a6e4688b4795a2fd0b159a88d7b5 : SUCCESS in 5m 32s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3f77cae2185d46c6a38a410250947b21 : FAILURE in 1h 24m 21s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/1ce961063e4446b78fdac2659acc3994 : SUCCESS in 46m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/330d37f2ca994ad0bbee698cf0906f3f : SUCCESS in 54m 13s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/04e7fe3e39914e8caed66fa632a33bc3 : SUCCESS in 35m 44s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/42a128b896cf4a3ab41737a29eb1954f : FAILURE in 38m 12s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/daee8ec3730647ea9c28c61ae456b85e : FAILURE in 1h 57m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/086b8390da644e8e91d2b0fce2ba9bdf : FAILURE in 21m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9d66894db51a46a4a24640ba50429558 : SUCCESS in 54m 18s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a51d3990fa1d489fa08ed98ea2d6aaf4 : SUCCESS in 45m 19s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d545f93253d44cf3b621673784b7ea09 : SUCCESS in 40m 33s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/db73567f894c4a11be059f61e131a5fa : SUCCESS in 42m 18s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/5fb9488ba9524bdf81d72a87ea209697 : SUCCESS in 1h 02m 51s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f4e113327f3c4aad80ffecb7244f1fb7 : SUCCESS in 29m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/48ba46c2812b4ff9a6337298501b34d4 : FAILURE in 35m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b765eaed23aa48cc93892301ab11e69c : FAILURE in 37m 17s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8919557e855542c9bb1a04ae93b9963a : TIMED_OUT in 1h 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7f979c0da8eae83a7cc42036eed57c4f16c3707","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-13 22:53:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/93/770593/1/upstream-check/manila-cDOT-ss/0226747/ : SUCCESS in 1h 28m 59s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/93/770593/1/upstream-check/manila-cDOT-no-ss/5ed5237/ : SUCCESS in 1h 42m 44s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f1975b6790a27e3c1b44302f8b3309afd75229f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-14 00:37:17.000000000","message":"Topic bug/1894362-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"33f21498efee75458d31192d7cbed7af91a63059","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-14 00:37:20.000000000","message":"Topic set to bug/1894362","accounts_in_message":[],"_revision_number":1},{"id":"7592afdc1f8e920101765d81a4004327d5bf3201","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-25 20:25:08.000000000","message":"Patch Set 1: Code-Review+2\n\nClean dherry-pick, backwards-compatible bug fix, driver specific.","accounts_in_message":[],"_revision_number":1},{"id":"6b76e25ef79909b1a6234aa0bd2aad435df1560a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-25 21:08:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry-pick, lgtm, thanks","accounts_in_message":[],"_revision_number":1},{"id":"c445d727589446c1b123fab8a01a4c3739050f3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c7769391a68c052673b651067c79163f2191a5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:33:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c1ec8fcb2343ef88d958f57229f429 : SUCCESS in 20m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcd4d6a17aaa477191319990f7c95ede : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fedabf53410a4a7b825355662c5813eb : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56cfb556dc5840efb4d6c2fadc685e2b : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4d613546764c79a1077097139d1410 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42b135fd886e4103b774658531fbedd9 : SUCCESS in 11m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d18f36cb2db84aeabbd18ada87f6a0c1 : SUCCESS in 4m 28s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f44e4331201d4620a3ba5f5fe961183b : FAILURE in 15m 09s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/9c19dd86dddd4270bf47d6c87d177545 : FAILURE in 8m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7dfcf2799b5e28095bc1146f72b20587d4c292f6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-28 07:09:49.000000000","message":"Patch Set 1:\n\nrecheck\n\nhttps://bugs.launchpad.net/devstack/+bug/1913283 should now be fixed\nhttps://review.opendev.org/q/topic:get-pip-py27+(status:open)","accounts_in_message":[],"_revision_number":1},{"id":"dfea942c1257f3cb00dd530746be33c3e678d1b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 09:11:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4172e2db3dab4b17a64c9cee91e27d8e : SUCCESS in 9m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6494851035364aa180388964e1c57b6b : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8e2554450e040bfb71728d22ee560a7 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c1e6f5f232449996db13671cdcb65d : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fbe624768f04b778488f3a857f9aec3 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4ddb792fcd24ad6b600222ce5d7d58f : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aee4c2968fd746debee6ca02b05d4c5a : SUCCESS in 8m 44s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9b7d760477314d50b6a1c45048b259be : SUCCESS in 6m 05s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/ef36c636069b4408a4aae2e8b1627615 : FAILURE in 1h 21m 54s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/8a211e36ef3b437996e41ad66db4e0a5 : SUCCESS in 51m 18s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/372dc0b859c84836b381f87e6acc7966 : SUCCESS in 1h 13m 39s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/da259a1eb8c64311ad28010d68ff1ec8 : SUCCESS in 1h 00m 21s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/ecd41f33584249d0804981f45f1563c1 : FAILURE in 38m 10s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a946cefd1a77473d87eedf9161a2e6f6 : FAILURE in 1h 19m 26s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a0b764cc97334ea18b7c44e524c7097f : FAILURE in 24m 34s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ee147d2dfab478986b9aa2f6c9baa2d : FAILURE in 51m 49s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1aa4ac59829448caad99f5927c506ea9 : FAILURE in 32m 20s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/20be1614be2546df863d412b395db03f : FAILURE in 54m 59s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/350632934c4e43bab36b7bb024353693 : FAILURE in 28m 53s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/16552e38b78d499db1af2384b532b6e1 : SUCCESS in 1h 22m 45s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c88440bb39b1451997f1e1a2a3503b59 : SUCCESS in 21m 35s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1ef7079de50b40c88b97719fc8a61a28 : FAILURE in 33m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ba7895919ef643aa8f042a9a959b6ace : FAILURE in 34m 08s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8edbb407a002449faa9feecbaf3e210f : TIMED_OUT in 1h 31m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf06b46f876ea08827a0ef17c61f8f025d445509","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-01-28 11:41:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"881eaed4d9da0214d378c50cc914f47a059a37ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:47:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d367cac55f54f66877c8e4149f6661f : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37698b69083c4b77a16662fe6dd20391 : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00c19cdca0264923a9c1e92b9805397d : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/586b86a86e55475da7a3439ef367db4c : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2656db718c674b11b7bd0d0c91aac02f : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972b5f3e04204e9890a1105d8814a2d0 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c7e3efb5d89459d9ac521b11b43f81d : SUCCESS in 9m 25s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/94453b9158044a34baecd76f50fe0f51 : SUCCESS in 6m 55s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/cd9a7cc019e4429f847871b82bc4ce08 : FAILURE in 1h 19m 42s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/e91c032d76cd4da7b8555fd45c71e569 : SUCCESS in 51m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec221f3a1b6c4235b0d6c7483ead1780 : SUCCESS in 1h 15m 06s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/592db8e6814b4c75a4f2e5054212944f : SUCCESS in 53m 07s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/5c201204dfd04f5d96a612a58c578aaf : FAILURE in 53m 18s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46248531945649149ade8cc2a579aaee : FAILURE in 1h 52m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/a2ead0bacd1e43edaae37e2bac589d6f : FAILURE in 31m 37s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/da1a127282b14e32896eb32aded37af1 : FAILURE in 54m 23s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0ab8bd8eed3748848ae3194c9b9968be : FAILURE in 51m 33s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/619eba51fb424a28a2fa076ed5222d9a : SUCCESS in 43m 21s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/4d00ffcc241f462c90f5e0d897067ca5 : FAILURE in 20m 49s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/07209c20b74d4b948f3af9ccc61c4e03 : SUCCESS in 1h 13m 21s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3c596333404344c0a0033ae9684a374c : SUCCESS in 26m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/735999546c58413082267ab898fb7b93 : FAILURE in 27m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c9ebfff26d2d478fa99c8203edcb6b43 : FAILURE in 33m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/893e2cf78e0b43b39ceb37ae0e6c22b0 : TIMED_OUT in 1h 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af7af15cc0bb642e3791810772d7e6329d1b4fff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-01 22:09:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d65b00da53ffbc854757768c703af6a04aa5ddc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 00:09:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bc3c0f330d64658b13ab699c5f66fae : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9bda205c0245c6a2572b91088f4593 : FAILURE in 13m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d49ce21018d416a9ebaae49f15dbde2 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9b0dc2a7534a11befd0c6b3c7c1a9f : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08588fe153fe47f9bba2752e609f2830 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/379bb0959afd497aac0e8494672cc7f0 : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/618ab473450d41308c5e92121ade9ebc : SUCCESS in 7m 42s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2b965c16f84a4bf68814c85c74da6336 : SUCCESS in 5m 06s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/3a49dad9064c4285b772ce45a4bc6e41 : FAILURE in 1h 34m 40s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/af4215c90094419b967e5bab4d0df104 : SUCCESS in 55m 01s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/288fad47e58e4de3ac8577a1232e1668 : FAILURE in 1h 36m 28s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/437c861d6698438fac1e31b2011b7b15 : FAILURE in 1h 53m 58s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/846307a450d24c478aa2b96fdb4e55a6 : FAILURE in 27m 22s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e173a28a2f3f48e89bad3463aef89582 : FAILURE in 37m 35s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6952297934e94c009bdbd2c032e3813c : FAILURE in 41m 11s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/0cda4478aac8457cb54213ffacb6ce2d : SUCCESS in 53m 16s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/17aebe57260f450f95cdb0f2c3c7a204 : FAILURE in 55m 18s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/40fdb7447fef44ff95fac25f94976697 : SUCCESS in 1h 20m 34s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c69213c243754481a0ee50880ccab02f : SUCCESS in 29m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8fecda30a0004171a7179e52d29699a4 : TIMED_OUT in 1h 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04b80087b7fbc19856c32a69a3311794f6f2369c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-02 10:43:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"105df069b482a693ea90f671ba3f4e2c91ab02eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 12:32:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030b0de1d787428b9aaa098569fe91fd : SUCCESS in 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/185b7ba43d1041d2a91b109bdefff15b : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8264691f24eb40219fbf3bbbf2517b28 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b1193d134b4f7baca8935cc6608253 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1dd25f3cbf74a6f8c21be5dd758ec38 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60234281a40c4ca88a7b08c07c577e28 : SUCCESS in 8m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d67f0325fe9143e89b0c38087cf22823 : SUCCESS in 9m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9eb963c680a745969342d63a2ab6f0ea : SUCCESS in 6m 37s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/bc7d56b1e19b46bfbfdf36aa5666ab59 : FAILURE in 1h 17m 06s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/87057957216b4427979a46d766317d46 : SUCCESS in 59m 34s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a53056c6f5f54c69b0dae6bd6451049d : FAILURE in 59m 03s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6390d3c2382241be8251ef39351ea0e6 : SUCCESS in 1h 47m 55s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/8b33916643284981a8100ba63dbf4ae6 : FAILURE in 27m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f6da1ca24ed5492b9691272f37d41c42 : FAILURE in 45m 29s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7d93ff51a5d540a0b8239b6286d8d880 : FAILURE in 29m 22s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/e7c4318df82a4763981cdd1e733a07ab : SUCCESS in 53m 03s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/5d325620092a434a82303df670e1c94d : FAILURE in 48m 05s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/3076481462cc49c1baa50ccad525d14d : SUCCESS in 1h 10m 29s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e7b0948d0b464de9b654e6d41118d01f : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36fb9eaec3fe442a8d3c85e3ef5aae16 : TIMED_OUT in 1h 30m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2c012b582845aa66e9fa8f4de6a90c54f445f6e","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-02 12:34:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"166236ef467148bc802f7b3dfb014f0bf7873a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:34:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7560b78c11184c14ab99f7bd30fc315a : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c1b7cb120bd431ab8d7a7f3c6558043 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50f5177a969040b09b61c698b8a54847 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7019186da6b4fc4a27ad3cec2da4ec2 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d92f5de77824b33a4ee69fd440f7406 : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00a2d44386db490cbea002d352b65efd : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4736200ceef4ec3aaaf41da2f811989 : SUCCESS in 13m 18s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4760058764334f5893f127fa52796241 : SUCCESS in 6m 23s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/74e994195d604b9dbfa3bfcff49ab637 : SUCCESS in 1h 24m 15s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/65eaea63e7dc42b285e771cbad0edf48 : SUCCESS in 52m 02s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/891e615a50624b2b90e1553f490a0803 : FAILURE in 1h 43m 11s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/345c62424f6645baa1f85ae1960dfb65 : FAILURE in 1h 53m 06s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/f65b706dbee542a28c58e8c925dc48df : FAILURE in 27m 52s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7079545474cf45709978b518a0754a74 : FAILURE in 39m 19s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/05d19d57a13845128c0519986454024c : FAILURE in 42m 03s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/f8cd670514144089b532b140514cbfaf : SUCCESS in 45m 51s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/c8ce72cb7114478a93b604a16b0bf37b : SUCCESS in 50m 25s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/e52516999ea54776b8a2e976d74a0925 : SUCCESS in 1h 08m 40s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/36b2d1912e6e4959afbf60ee402eb58d : SUCCESS in 29m 06s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/38a2b12800104ca8afa2adb3a1b2c668 : TIMED_OUT in 1h 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"289e8cd48879c42451b999bc5919430564f38388","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-02 14:35:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/93/770593/1/upstream-check/manila-cDOT-ss/6150ec5/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/93/770593/1/upstream-check/manila-cDOT-no-ss/07cc9aa/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bc20dc955dd8db9836a380159e0babb9f3c3f1e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 14:35:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d6f668909bc85682a292f2b4bdeae22f943d87f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:13:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b94135dd434dcf861e66d6df4e62f2 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e185f0ba586d42b1ad9e81d94d337c9b : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3325b5757efb4fe59050f5172a584b4a : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a994ac43fd64566b9a84b3b4ffc550e : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/345897592b684530af741875eb4d753e : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/695bb8088c004abeafe23fe045b9c722 : SUCCESS in 9m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e754d48f652a4ffaaebc02b665d5a83c : SUCCESS in 6m 21s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d052c566d29c4b9595395c19e6369a84 : SUCCESS in 45m 34s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/b9741801533d4876b53f2e398d3dd2d5 : SUCCESS in 1h 02m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cb08fb59949bd27ea5175b2a033bb6268bc04dfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:13:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fef6634d91b6b2df1df7db38e6e0488f56711c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:18:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ea5990ad5541d9a0fb42b766eb918b : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f5a3f4ffbe1350ebebaa46319f29e75c2375da8","revisions":{"8f5a3f4ffbe1350ebebaa46319f29e75c2375da8":{"kind":"REWORK","_number":1,"created":"2021-01-13 18:57:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/93/770593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/770593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/770593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/770593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/770593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/770593/1"}}},"commit":{"parents":[{"commit":"bff8a0d99ed73dfd494706f517d0bb1576365dd8","subject":"[stable/train] Update requirements and constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bff8a0d99ed73dfd494706f517d0bb1576365dd8"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-11-24 23:32:11.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-01-13 18:57:06.000000000","tz":0},"subject":"[glusterfs] don\u0027t reinit volume list on deletion","message":"[glusterfs] don\u0027t reinit volume list on deletion\n\nWe don\u0027t need to re-initialize the volumes list\non deletion, it still makes sense to add a missing\nvolume to the list, going by the reasoning defined\nin I14835f6c54376737b41cbf78c94908ea1befde15\n\nRelated-Bug: #1894362\nChange-Id: I96d49f84122a34701328909c929ede4d66746911\n(cherry picked from commit 914d87377479ab3951641d658fc461ad2f338ba0)\n(cherry picked from commit 2ad9bcc1b42819f7a8c7aa7200e6f8d128294205)\n(cherry picked from commit 312e4d16652383e77d3764777f531ecba4f3f0c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f5a3f4ffbe1350ebebaa46319f29e75c2375da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f5a3f4ffbe1350ebebaa46319f29e75c2375da8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
