)]}'
{"id":"openstack%2Fmanila~771413","triplet_id":"openstack%2Fmanila~stable%2Fussuri~I031a3b9775c50e78b6b86752ff8d1a4871a91c0c","project":"openstack/manila","branch":"stable/ussuri","topic":"move_filters_to_database-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I031a3b9775c50e78b6b86752ff8d1a4871a91c0c","subject":"Move shares filtering to database layer","status":"MERGED","created":"2021-01-19 20:59:52.000000000","updated":"2021-06-01 19:21:39.000000000","submitted":"2021-06-01 19:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771413-move_filters_to_database-stable/victoria-stable/ussuri","cherry_pick_of_change":770966,"cherry_pick_of_patch_set":1,"meta_rev_id":"b57ba833e8c823826f87ce87994aa50d7439ec5d","_number":771413,"virtual_id_number":771413,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 19:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-01-21 12:04:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-01-20 21:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"date":"2021-01-21 15:00:40.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-06-01 17:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 22:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 22:55:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-20 00:29:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-20 21:09:51.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-01-21 11:48:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-01-21 11:48:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-01-21 11:48:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"05ec2541e4f92bb6f68b3a9ed92f0a877dcfaf48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-01-19 20:59:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"d1de2ce0ba4fb1390c4d350ad25a4ae929e23f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 22:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261c7704c77e4e3983152f1a405dd17c : SUCCESS in 23m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb6e3a57f09a4445a81ddd9b2edd4ab1 : SUCCESS in 19m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53d441bb53134ec99c7c0bf286bdc0f9 : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdee0b96ac6a44d29698c889624cd221 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32994814bd9a4f60829c6626ed653b71 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec10b64c12004976ada6fdc1fc68106b : SUCCESS in 6m 58s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa13f58b180b4bca8975b320723a6e08 : SUCCESS in 4m 51s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/46eca800870244c2b21c30d8e4459928 : FAILURE in 1h 03m 13s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/0ae4bc003b34434d9c15a7a6a1de24fa : FAILURE in 45m 11s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1626686b697f47a98ea2f7f0dfaae07f : FAILURE in 1h 11m 04s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/bfe04a529939439185f4892f4e6b82da : SUCCESS in 48m 28s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/e6529049ce754ebab3b75b15e4c784ad : FAILURE in 45m 27s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e4eb6504bd08450e90c37501153941e6 : FAILURE in 1h 35m 01s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2d83009142a3440489e64bbcd542615d : FAILURE in 19m 00s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3776197aeb7d4fb6ab40ff3c417fcdd8 : FAILURE in 1h 04m 18s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/14773b314c8945559d5494416ee6909b : SUCCESS in 40m 22s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/6f76021e60354400958efeb1acc2e235 : FAILURE in 45m 39s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/56d09d7d0b634610b692427321f569a2 : SUCCESS in 37m 46s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/0be6d146001a4f0d9d9b4a397615d8af : SUCCESS in 1h 10m 33s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/89e0363dcd87452081f039a1f4b35c4d : FAILURE in 52m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f628c9f7058041139a6eddd59ff2be5c : SUCCESS in 31m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6b78baeebc444ebca3dc398308496c01 : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c931285eecd40b9b7ce6aba6bab46cb : FAILURE in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05bd6e30665c322695e5cd5ded391aa35e5dda42","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-01-19 22:55:56.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-771413-1/ns5_manila : FAILURE in 1h 55m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f287861a6274e687714df3341f5c1c1aed280cde","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-01-20 00:29:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/13/771413/1/upstream-check/manila-cDOT-ss/e35e0fa/ : SUCCESS in 1h 12m 52s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/13/771413/1/upstream-check/manila-cDOT-no-ss/66620ff/ : SUCCESS in 1h 49m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2b02e6cef8850fae050a9aa1e6ed76d35f203f4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-20 21:09:51.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":1},{"id":"a6f672009189c824fc6b77a84ff4c4f4afa54baf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-20 23:49:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ce86c33c3b3c2fe3d3633479a7f78ecce7467dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70731cb29bc3c68ce3ee724336605c538224d9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 01:09:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f5fc99bea348e9870750c900e9937e : SUCCESS in 11m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb8e203602141b489a9074182643920 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f816c0726a4994ae59c27b37621077 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e199e43743a4f1490599fd6ec2f4fd9 : SUCCESS in 8m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9307f6ee3be64e6aa9baabb0918a862e : SUCCESS in 8m 04s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/908edfc20ceb4a5b8a46b43a2e9bb3f8 : SUCCESS in 4m 58s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/d8da7a446e244320bd897d5b47a057aa : FAILURE in 45m 52s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/a59b27f38b7d4aa9aee6af7f009fa73d : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"eab549663d8f4963fc133132adddfc87bd8f3cfc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-01-21 01:19:02.000000000","message":"Patch Set 1: -Workflow\n\nDidn\u0027t mean to press the workflow on this one. No harm done - thanks random CI failure 😊; but we;ll wait for another review since this touches many things.","accounts_in_message":[],"_revision_number":1},{"id":"f2a57abf61fe0fb7c0d9fb4900bf3f4f728adeb5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-01-21 11:47:43.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Workflow\n\u003e \n\u003e Didn\u0027t mean to press the workflow on this one. No harm done - thanks random CI failure 😊; but we;ll wait for another review since this touches many things.\n\nThat\u0027s alright :)\nWill add more reviewers to this change","accounts_in_message":[],"_revision_number":1},{"id":"09537a29e02f8796306b0e7fb7f673028bb01e5c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-21 12:04:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry pick, LGTM, thanks Carlos.","accounts_in_message":[],"_revision_number":1},{"id":"cef3fe284a049503aeb922584973f7e65a353191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 14:07:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e6e8d45e5d44a68147c660c092d0ae : SUCCESS in 10m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/294e7d453022408586ff2d7785e295ba : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e60d75c6002c470f8468ea95f975052b : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51fb51e9165435db2c8b578ad83db51 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09e4c31641f44ec794ce75dea5676fa7 : SUCCESS in 13m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc04a6999f164191b2646b0ee0003b14 : SUCCESS in 7m 26s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2bb76f5f1e474d4b9b4dd81694c668ca : SUCCESS in 4m 41s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/dacfa46f274248818d2263eabd8bac21 : FAILURE in 1h 27m 05s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/37c139a14e79458c85a7a2a423c9ea61 : FAILURE in 47m 32s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c2f2094a4b44414186c583f5442d136e : SUCCESS in 1h 09m 31s (non-voting)\n- manila-tempest-dsvm-postgres-generic-singlebackend https://zuul.opendev.org/t/openstack/build/ad98c8165a2c4811bef03966a51e3179 : SUCCESS in 49m 39s (non-voting)\n- manila-tempest-dsvm-generic-no-share-servers https://zuul.opendev.org/t/openstack/build/68490e29b853493c8a7bd8fe5fff9c4e : FAILURE in 42m 20s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40b69460a7ca45419a8a73d6c9033d35 : FAILURE in 1h 52m 47s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/adc325908a644e3497231483f0b7cfb2 : FAILURE in 34m 08s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/af7a71ff7d8a4262a4c25d6a7fb10707 : SUCCESS in 59m 51s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0c0917f9cdb74bee976615850877c11d : SUCCESS in 42m 02s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/6551e2a5c8d84213b4b28bd5d89e3d89 : FAILURE in 48m 10s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/9fd1bd95555348c184b5159cb4f87fe9 : FAILURE in 42m 49s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/662beab51ae34caa83681bf783ea0de7 : SUCCESS in 1h 23m 58s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/552ef4a0ea3142978595cf467a24826a : FAILURE in 1h 09m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3bb3dc63c3b44e2887ed466bb9ab9743 : SUCCESS in 33m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c98cdc09e2e9439eb62bcafac2a3f47e : SUCCESS in 41m 21s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/75283929125c43c2b1f5b89d59f6f4c5 : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7e82d73ca4dceedf8c5bde6b38c0f3bdaf37545","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-01-21 15:00:40.000000000","message":"Patch Set 1: -Workflow\n\nI saw more than one error associated with manage errors, for different tests. We might need to investigate if these errors are occuring due to the new filters and if we need a fix for the tempest tests.","accounts_in_message":[],"_revision_number":1},{"id":"355e501de99156de3429fe701485698561f80a5f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-05 23:02:29.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-771413-1/ns5_manila : FAILURE in 1h 53m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4dbb58133c617c396617b46617b1902c2e48e706","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-01 15:30:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dfed3feee53f7c15e40478ea81130389f0ed97a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963408c28a484fe891deb061777716ef : SUCCESS in 11m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5052e83ac424919a87a864db2481889 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a2a58e5958a4e779d9134d76291e285 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc05a0c040684c6d80654b6262859016 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/845a413632fb405a99d0bcce125ba3cd : SUCCESS in 9m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf07fdb7649a42e5bb9b62c9897048a8 : SUCCESS in 9m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4652bac1ebef43f4a22d538e4f8c8792 : SUCCESS in 6m 21s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/13edf9b4ba6646f6a90b8985cd5dbad4 : SUCCESS in 43m 56s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/71a472d6755f441e9b7486303776a139 : FAILURE in 40m 44s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a44912923e404b3db2d8e3c837ff97a7 : SUCCESS in 1h 07m 59s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4db6fc6ade9741c1acf16da0b87a5e41 : SUCCESS in 1h 18m 30s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/86bf6204428b424caba44eccbdf6202a : FAILURE in 35m 25s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7b4ff3927a5644979dafb410db63d3b3 : FAILURE in 37m 07s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4a2f748be1e8497ba82238ec0ebd404d : FAILURE in 39m 50s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/e45349e19ef542229c1e59173cfa42de : FAILURE in 43m 17s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/dd7da87d56394216aee3ccedcf3e1e2a : SUCCESS in 41m 04s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f045f96fb34e452593322e88dbaa372c : SUCCESS in 1h 05m 59s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5614cb29e0d94aeaa5565a7eb41d1bdb : SUCCESS in 1h 08m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/18663186cf9843b48ba6ec78a78abae9 : SUCCESS in 39m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/13cff69c66a74449a9ec3db296b87ea1 : SUCCESS in 44m 14s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3df7ea25ce254b109d20da895645f5da : SUCCESS in 7m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3c7b179ef9b1114eb0e6181636ffd740c4ee4aa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-01 17:34:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/13/771413/1/upstream-check/manila-cDOT-ss/f74060d/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/13/771413/1/upstream-check/manila-cDOT-no-ss/be1a2c6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e90d754590fd2bb917ae42ac59a49c188867c75c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-01 17:48:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"673e4f023f3606ff4194ed952ec10361a2f98959","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 17:49:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1db6ed6201f4b7c56eb6e56ec74c17f6b31ebb5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 19:20:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d4da29c794d440d9dbf2836f69cb7bb : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7688420be6e4488e8e5ac0ed3cfcb755 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b38cc66d8048c0baf6503229be0da9 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90539bd7f31c4f3a81c6d74fa63deac7 : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b40d30350af45dd8493b40a755813fc : SUCCESS in 8m 28s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c3d203ce27434c3d914ff41d4427c940 : SUCCESS in 6m 16s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/968f9e94824048d984aace8574cf4c9e : SUCCESS in 52m 01s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/c1c98dfd6b8a4f689cf113e0b0738bf9 : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c407946a381b25a28af6085c0786ed7ba3a65562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 19:20:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b57ba833e8c823826f87ce87994aa50d7439ec5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 19:21:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffe9298512e4f2c93419724e546e21e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be76d7ebbaef0c061a3b19962b2f8de4a23ddb06","revisions":{"be76d7ebbaef0c061a3b19962b2f8de4a23ddb06":{"kind":"REWORK","_number":1,"created":"2021-01-19 20:59:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/771413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/771413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/771413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/771413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/771413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/771413/1"}}},"commit":{"parents":[{"commit":"969ff055f676d94bddbe5e36ca0aaea7d12887e6","subject":"Merge \"[glusterfs] don\u0027t reinit volume list on deletion\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/969ff055f676d94bddbe5e36ca0aaea7d12887e6"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2020-10-23 20:06:41.000000000","tz":0},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-01-19 20:59:52.000000000","tz":0},"subject":"Move shares filtering to database layer","message":"Move shares filtering to database layer\n\nMoves the manila shares filtering to the database in order to have\nthe queries performance improved.\n\nChange-Id: I031a3b9775c50e78b6b86752ff8d1a4871a91c0c\nCo-Authored-By: MaAoyu \u003cmaaoyu@inspur.com\u003e\n(cherry picked from commit 21c731e733fc9ebfae12e39d802eb426772608a0)\n(cherry picked from commit 1f12cb662bc65186710efb7b0d8cecb53e55255a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be76d7ebbaef0c061a3b19962b2f8de4a23ddb06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be76d7ebbaef0c061a3b19962b2f8de4a23ddb06"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
