)]}'
{"id":"openstack%2Fmanila~775451","triplet_id":"openstack%2Fmanila~master~I54bf0e685d8e436651641d154d8cd549c21164e1","project":"openstack/manila","branch":"master","topic":"bp/update-cephfs-drivers","hashtags":[],"change_id":"I54bf0e685d8e436651641d154d8cd549c21164e1","subject":"Update create snapshot and delete snapshot to use ceph-mgr client","status":"ABANDONED","created":"2021-02-12 22:12:27.000000000","updated":"2021-03-09 12:03:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ad6894c50d7897e5401edfd459319d3bb29c0e7","_number":775451,"virtual_id_number":775451,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-13 00:06:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 23:17:37.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-02-13 00:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-13 00:30:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-13 01:22:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-13 02:00:13.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-13 03:43:25.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-13 04:20:36.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9fc25ca7b6ca52bc6228df294216332ccbe73a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-02-12 22:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9653140b0b516e3653c90cac45398faa8efbfca3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-12 23:17:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/51/775451/1/check/DellEMC_Unity_Manila_dhss_true/5e2e1e4/DellEMC_Unity_Manila_dhss_true/317 : FAILURE in 59m 15s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/51/775451/1/check/DellEMC_Unity_Manila_dhss_false/23d41e4/DellEMC_Unity_Manila_dhss_false/301 : SUCCESS in 1h 04m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9c2ea96540d73bffe2586d7f513bc8745bad1dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 00:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50258667da664d0ebacd1d03a2352c79 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91cad3f1ec3c493293862503b0ea74b3 : FAILURE in 13m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83792a904b334abdbb1b20aeed5e2d1d : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd210c80057943f28f0e0f1a30c8236e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f350e623c8bb4aa392cbb2fe1d87cf93 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86a9aaa293b747f9b881cce755bfd503 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0a4fddf5234f24bd4eec4237b364c0 : SUCCESS in 7m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f7ab31943e14837a00d40223f686507 : SUCCESS in 5m 19s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0fc401cfbbc64aa0a1dc1fb472775a16 : SUCCESS in 5m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c301177404f4397a3a6a9c5da63d34a : SUCCESS in 10m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/341ab0566519402ebdb7708a99cba206 : SUCCESS in 34m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b07b1dbd375d4fae9996b518623727d9 : SUCCESS in 37m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7727ad0b83644207a1a457b449fd29d1 : SUCCESS in 1h 02m 38s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ae95bcf914064277b7689268e7d0dc32 : FAILURE in 1h 08m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca40994594e0409e86c5d08c5aeaea35 : FAILURE in 1h 17m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3364c36694da486696cd4cfa1a55a52e : SUCCESS in 44m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eabf751c3e724c5284c5611bd39b9986 : FAILURE in 1h 51m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/471b70e984664d29914f23bc418889a2 : FAILURE in 22m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fd4799bcb1ad4832a80fdb10afdd35e6 : SUCCESS in 23m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e6f1cff60be486281d6f5974c4bf93b : SUCCESS in 1h 20m 59s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/df250eb2c0e245d782fb52897e5a66c8 : SUCCESS in 45m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1184aea9ba49460488f59ee61aebbb06 : SUCCESS in 40m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f5fed2caa29a41b18ba1ac9bd33e09c0 : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f4224480f2ce840cf5da60c86c7ba6b0c8d5737","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-02-13 00:30:07.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-51-775451-1/ns5_manila : FAILURE in 2h 3m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"195ddf2288bd9257716a25f0a03c59f6c19812b6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-02-13 01:22:26.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/51/775451/1/check/DellEMC_VNX_Manila/77c75ef/DellEMC_VNX_Manila/184 : SUCCESS in 1h 15m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"59ee5bd9522ab12e7ee9bd687a76e272aeef5b02","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-02-13 01:24:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/51/775451/1/check/DellEMC_Unity_Manila_dhss_true/0d96133/DellEMC_Unity_Manila_dhss_true/319 : FAILURE in 53m 09s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/51/775451/1/check/DellEMC_Unity_Manila_dhss_false/4252489/DellEMC_Unity_Manila_dhss_false/303 : FAILURE in 58m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"67be56df18781eeb0cec41f853ce41dbd415a921","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-02-13 02:00:13.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/51/775451/1/upstream-check/manila-cDOT-ss/fc5ec73/ : SUCCESS in 1h 52m 47s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/51/775451/1/upstream-check/manila-cDOT-no-ss/2e447a1/ : SUCCESS in 1h 48m 14s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c68919c570be08f30300bf92ea1152a95879a3bc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2021-02-13 03:43:25.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/51/775451/1/check/dsvm-tempest-manila-zadara-driver/469906c : Build Failed. Use run-ZadaraStorage VPSA CI to recheck. in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8e5b08c11ece24b2574cf2e172248b5b9bbfff6","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-02-13 04:20:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/51/775451/1/check/DellEMC_PowerMAX_Manila/8e6a806/DellEMC_PowerMAX_Manila/223 : FAILURE in 3h 48m 54s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ad6894c50d7897e5401edfd459319d3bb29c0e7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-03-09 12:03:18.000000000","message":"Abandoned\n\nWill submit a patch set rados client instead of the Ceph CLI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cddb7e335c631ab04dcebc5489a645a8e9f87338","revisions":{"cddb7e335c631ab04dcebc5489a645a8e9f87338":{"kind":"REWORK","_number":1,"created":"2021-02-12 22:12:27.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/51/775451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/775451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/775451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/775451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/775451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/775451/1"}}},"commit":{"parents":[{"commit":"d7d8bb38dfeda35d4578eaede73dd048c255ce06","subject":"Update extend share and shrink share to use ceph-mgr client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d7d8bb38dfeda35d4578eaede73dd048c255ce06"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-02-12 21:09:58.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2021-02-12 21:09:58.000000000","tz":0},"subject":"Update create snapshot and delete snapshot to use ceph-mgr client","message":"Update create snapshot and delete snapshot to use ceph-mgr client\n\nUpdates the create snapshot and delete snapshot functions\nto use the ceph-mgr fs plugin\n\nThis patch set is part of the update cephfs drivers to use ceph-mgr\nblueprint\n\nPartially-implements: bp update-cephfs-drivers\nChange-Id: I54bf0e685d8e436651641d154d8cd549c21164e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cddb7e335c631ab04dcebc5489a645a8e9f87338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cddb7e335c631ab04dcebc5489a645a8e9f87338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
