)]}'
{"id":"openstack%2Fmanila~781968","triplet_id":"openstack%2Fmanila~master~Idb36bb87e480824e0dbcffa70f7cdba3f64262c2","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"Idb36bb87e480824e0dbcffa70f7cdba3f64262c2","subject":"DNM - testing proposed changes for cephfs create-share-from-snapshots patch","status":"ABANDONED","created":"2021-03-20 14:27:46.000000000","updated":"2021-03-24 02:03:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"060f430d449404a2c7ebe4c74e52e432551bba0b","_number":781968,"virtual_id_number":781968,"owner":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-21 19:39:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-20 15:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-20 18:14:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-03-20 19:41:15.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 20:13:58.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-20 22:42:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-22 06:38:36.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"8b66db53bd046d1789f7e3131df0b80417ecfa78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-20 14:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cc8390c4cdca600717c756bd0d48b17da99af4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 15:40:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b251bd8fc94004ab3bc7a59990d841 : FAILURE in 7m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8286d4f46874444a971f0afe9f2f5db : FAILURE in 11m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d590e17063746c281326ce631b5956a : FAILURE in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502fd2f3a98943b1a24171ce81dbc655 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e56b8af44e41d8a3e53972651eafc9 : FAILURE in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fb1138f60c34a4d94af6fc7b246736c : FAILURE in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66c3bf14e5394fda86cacbe1e2eced81 : FAILURE in 5m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/62daca59b453463b93cc2f9e658a1395 : FAILURE in 5m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/118fbbd1ebf64060b0065f44992b12a2 : SUCCESS in 5m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fb76c17beab0409592e27b8aa68853e8 : SUCCESS in 30m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5b7072f87da44c48bb936d2a334bbb90 : SUCCESS in 33m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/07aa62d233274e1ebc33ccfeee1f972d : FAILURE in 1h 06m 26s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/96002c7e813540d4acc04b17d762d383 : FAILURE in 31m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/23d4f0a7d25a4d54ba5c73bd4f610759 : FAILURE in 30m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/28fda3ab65c84ae092bee32410936783 : SUCCESS in 44m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5e8805fd226d40678360f4bd646b0fc0 : POST_FAILURE in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c05a5e3fae1146abad4e210904ea4759 : FAILURE in 24m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1ab4245390f2456395c3fb6a05ddef33 : SUCCESS in 25m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff711d949a674d6f9d3f98b2ab679e13 : SUCCESS in 53m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/08d474369f4548268b321962f286b1a1 : SUCCESS in 49m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f559e36b6de8411098a4c447e493d578 : SUCCESS in 39m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8a0891ccb7d34a48bfc92a28982996d3 : SUCCESS in 30m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35ebe502b1958980468287d4f16faafb14a774c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-20 16:24:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"69486a743aec1d6f19aadc67efb2f677ab0bbebf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-20 18:14:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/68/781968/2/check/DellEMC_Unity_Manila_dhss_true/4cf0a38/DellEMC_Unity_Manila_dhss_true/506 : FAILURE in 59m 38s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/68/781968/2/check/DellEMC_Unity_Manila_dhss_false/9b47dbd/DellEMC_Unity_Manila_dhss_false/499 : SUCCESS in 54m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ea10e7b2e67e2717ee3445efb391966237b78066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 18:17:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c7f033dfcd4133b2aa86a2f513b9cf : FAILURE in 10m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3915fcd79764f109da6bd7590d55043 : SUCCESS in 15m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4188b0e2ad9f48c89d931c6a81f0cccf : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f2b470927c499eb111f5191e56e22d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e379da8fc484a848715ad71b0599e44 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fefbd63a29c410783c62124cebbc519 : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5776905559e49b182dbb1e1d4a89047 : SUCCESS in 8m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1152555ed56845cca353049838e573f9 : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df8ee3919de14fb29444f8686c0527f3 : SUCCESS in 5m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/746423cf13ee40cd91b76609f7f6432b : SUCCESS in 32m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/adef90f9dd5849bdaa58db819966aab3 : SUCCESS in 50m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b59bff11946545199a1499e6c3e5d843 : SUCCESS in 1h 26m 23s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c2b95889054445c5ab25d060dde49de6 : FAILURE in 1h 16m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/13ea977d437544f3815b371dc9f7ff55 : FAILURE in 1h 36m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c847668d89ef44cb89f64905b0e65cbd : SUCCESS in 52m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f883d560c888497ab7fb6e02d04f9cd5 : FAILURE in 1h 50m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8f639268a30f4dc5b027ae59daf1df46 : FAILURE in 27m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/680a39e07b164bd8ad9084d048bdcb10 : SUCCESS in 28m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/76dad72a027c42e5a1be06aaee98c2ee : SUCCESS in 1h 18m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f6556e0af2934d7dbefb84fd9a444bc8 : SUCCESS in 1h 03m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2626a708037b4aeaaeb7e5ef2056384c : SUCCESS in 46m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c8759e0bff904be4a5569cabd062b314 : SUCCESS in 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ffcbf157f7f1650d27433de768f319bf92a6c00","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-20 19:30:25.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/68/781968/2/check/DellEMC_Unity_Manila_dhss_true/6c7baa8/DellEMC_Unity_Manila_dhss_true/507 : FAILURE in 1h 12m 02s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/68/781968/2/check/DellEMC_Unity_Manila_dhss_false/e0c25c0/DellEMC_Unity_Manila_dhss_false/500 : SUCCESS in 55m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fb265105728b695bd7e8d4f873c5670a61bcabb4","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-20 19:41:15.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/68/781968/2/check/DellEMC_VNX_Manila/af1de2c/DellEMC_VNX_Manila/329 : SUCCESS in 1h 23m 01s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"0f3412b247bcc10c5a5c370126707d980b0231db","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-20 20:13:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/68/781968/2/upstream-check/manila-cDOT-ss/5e8850a/ : SUCCESS in 1h 55m 30s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/68/781968/2/upstream-check/manila-cDOT-no-ss/03f185b/ : SUCCESS in 1h 50m 59s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cf64217abb3f1b0e3ba611cba42ed49065c05dc9","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-20 22:42:27.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/68/781968/2/check/DellEMC_PowerMAX_Manila/99fb6fc/DellEMC_PowerMAX_Manila/371 : SUCCESS in 2h 52m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"38eba57d74234981d93e92a9772fb053b42d1647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-21 17:08:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd44b01c74cb6470a16eff66ec9e1ff0e734bca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 19:39:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3e54bbafcf4612acbc3c88325c6061 : FAILURE in 10m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39de4d62fa7d4c76b70db000451ee8d4 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b60081ce5f94dd28de33fb5167b76c7 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4bb01f47db4aecb54a8fa3c15e2fb7 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cc6410b3a0e4a7ba1416ff0473cfc4a : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4a01b5150dd483e89fb33e713c12d3a : SUCCESS in 11m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f996e5c5b004fa9a1dd70f32851a169 : SUCCESS in 10m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/256168761fc44ad4ab3e0a029fe259b4 : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eca7656531864c82a4fbd6aa269c116f : SUCCESS in 6m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d581763ae59e458db318e4a770e8b891 : SUCCESS in 27m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/90002066cc8749b59ee4ee73fbd61e2f : SUCCESS in 36m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d9a0f718ab2f45c6a136ec7cecfa9f54 : SUCCESS in 59m 24s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/340c4fa13b6f4ab59bee23363b110217 : FAILURE in 54m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/466ea94f88b04beba2e005a31e0f75f0 : FAILURE in 1h 23m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bde99927b17e443eb67f18702f5afe93 : SUCCESS in 43m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8ccb9756d1b741098816f09df080d49c : FAILURE in 1h 49m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/68ca419b3a6c4f56836431611e281afc : FAILURE in 20m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0b1bf010014648dfab5d278103650844 : SUCCESS in 21m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1f0a0212ed494c139c0052f0d7ce672a : SUCCESS in 1h 02m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/27cb6a8d6dbf410ba1694d62306d9958 : SUCCESS in 1h 06m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/37ef935aed824e99921ad9d96a2f2a14 : SUCCESS in 36m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0f4f1d1070d94b7ba258079b491888cd : SUCCESS in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8716615d2a96581a3a7b40a9ce522ccec8467c03","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-03-21 20:13:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/68/781968/3/check/DellEMC_Unity_Manila_dhss_true/cd61ba1/DellEMC_Unity_Manila_dhss_true/510 : FAILURE in 1h 03m 05s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/68/781968/3/check/DellEMC_Unity_Manila_dhss_false/61e9dbd/DellEMC_Unity_Manila_dhss_false/503 : SUCCESS in 54m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"1e07f94a9f116cbdc62ceea08c3d6897469e3447","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-03-21 21:05:37.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/68/781968/3/check/DellEMC_VNX_Manila/d70090c/DellEMC_VNX_Manila/330 : SUCCESS in 1h 25m 42s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"5aaaa95bd4f2d77cf0f1fd61696ae327bfca358b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-03-21 21:34:24.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/68/781968/3/upstream-check/manila-cDOT-ss/b6239be/ : FAILURE in 2m 14s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/68/781968/3/upstream-check/manila-cDOT-no-ss/1e8f7da/ : SUCCESS in 1h 54m 22s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e2874a7c2550a4d93d08dd031fbebea5c927756b","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-03-21 22:30:32.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/68/781968/3/check/DellEMC_PowerMAX_Manila/04142f9/DellEMC_PowerMAX_Manila/374 : SUCCESS in 1h 47m 44s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"a1ff7fa4a21bd270dc37b0636bc8514e8bdb2faa","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 06:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-781968-1/ns5_manila : SUCCESS in 1h 56m 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b7bf6101a98d132874f21e289e94f4c23181e62d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 07:49:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-781968-2/ns5_manila : SUCCESS in 1h 52m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"884f15e4150f8c87b283c4bcae48750167fdc124","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-22 10:52:57.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-68-781968-3/ns5_manila : SUCCESS in 1h 55m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"060f430d449404a2c7ebe4c74e52e432551bba0b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-03-24 02:03:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4928279aaf9c2ed166a835489fb5df625acf5ae","revisions":{"2363a41f1685f8dbb050fcc0158b892a32985498":{"kind":"REWORK","_number":1,"created":"2021-03-20 14:27:46.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/68/781968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/68/781968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/68/781968/1"}}},"commit":{"parents":[{"commit":"4015e2e58cb6f188730ed682606cc68c627f95cc","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4015e2e58cb6f188730ed682606cc68c627f95cc"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-20 14:26:54.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-20 14:26:54.000000000","tz":-240},"subject":"DNM - testing retry getting clone status","message":"DNM - testing retry getting clone status\n\nChange-Id: Idb36bb87e480824e0dbcffa70f7cdba3f64262c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2363a41f1685f8dbb050fcc0158b892a32985498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2363a41f1685f8dbb050fcc0158b892a32985498"}]},"branch":"refs/heads/master"},"8c3bbb1c3fa0536f1a4d1b7f0a2207a3386db7dc":{"kind":"REWORK","_number":2,"created":"2021-03-20 16:24:56.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/68/781968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/68/781968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/68/781968/2"}}},"commit":{"parents":[{"commit":"4015e2e58cb6f188730ed682606cc68c627f95cc","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4015e2e58cb6f188730ed682606cc68c627f95cc"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-20 14:26:54.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-20 16:24:51.000000000","tz":0},"subject":"DNM - testing retry getting clone status","message":"DNM - testing retry getting clone status\n\nChange-Id: Idb36bb87e480824e0dbcffa70f7cdba3f64262c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c3bbb1c3fa0536f1a4d1b7f0a2207a3386db7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8c3bbb1c3fa0536f1a4d1b7f0a2207a3386db7dc"}]},"branch":"refs/heads/master"},"f4928279aaf9c2ed166a835489fb5df625acf5ae":{"kind":"REWORK","_number":3,"created":"2021-03-21 17:08:01.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/68/781968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/68/781968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/68/781968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/68/781968/3"}}},"commit":{"parents":[{"commit":"56a4f8fac3c29e4c4929e97dc8d28e9184b35757","subject":"Add create share from snapshot in CephFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/56a4f8fac3c29e4c4929e97dc8d28e9184b35757"}]}],"author":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-20 14:26:54.000000000","tz":-240},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-03-21 17:02:47.000000000","tz":-240},"subject":"DNM - testing proposed changes for cephfs create-share-from-snapshots patch","message":"DNM - testing proposed changes for cephfs create-share-from-snapshots patch\n\n      - added consistent LOG.debug for major methods\n      - LOG.debug for ceph_argparse.json wrapper\n      - fixed rados_commands in delete_share\n      - fixed up clone status checks\n\nChange-Id: Idb36bb87e480824e0dbcffa70f7cdba3f64262c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4928279aaf9c2ed166a835489fb5df625acf5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4928279aaf9c2ed166a835489fb5df625acf5ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
