)]}'
{"id":"openstack%2Fmanila~782854","triplet_id":"openstack%2Fmanila~master~I0489ede104840df3ce295ad55fc99e4cf6475493","project":"openstack/manila","branch":"master","topic":"bug/1908352","hashtags":[],"change_id":"I0489ede104840df3ce295ad55fc99e4cf6475493","subject":"Adds explicit error message","status":"ABANDONED","created":"2021-03-24 20:06:44.000000000","updated":"2021-03-25 01:47:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2b688f0fdb1feaf929ce7604e95711d465a4af2","_number":782854,"virtual_id_number":782854,"owner":{"_account_id":32635,"name":"Paul Ali","email":"paulali@protonmail.com","username":"paulali"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-25 01:47:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"c45c8f78c44ab7fd38d0dd43bef4669c059d6090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32635,"name":"Paul Ali","email":"paulali@protonmail.com","username":"paulali"},"date":"2021-03-24 20:06:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8033aa41ea929535989c6e051a752c5adf4a83a9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32635,"name":"Paul Ali","email":"paulali@protonmail.com","username":"paulali"},"date":"2021-03-24 20:57:15.000000000","message":"Abandoned\n\nError in uploading the release note","accounts_in_message":[],"_revision_number":1},{"id":"a2b688f0fdb1feaf929ce7604e95711d465a4af2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-03-25 01:47:56.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-54-782854-1/ns5_manila : SUCCESS in 2h 24m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8992abfc8d1bd41c516e8741f27ce7dcb849e62b","revisions":{"8992abfc8d1bd41c516e8741f27ce7dcb849e62b":{"kind":"REWORK","_number":1,"created":"2021-03-24 20:06:44.000000000","uploader":{"_account_id":32635,"name":"Paul Ali","email":"paulali@protonmail.com","username":"paulali"},"ref":"refs/changes/54/782854/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/782854/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/782854/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/782854/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/782854/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/782854/1"}}},"commit":{"parents":[{"commit":"7e9b23c7ea6f54811d4fb282775945d43efbc120","subject":"Adds explicit error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7e9b23c7ea6f54811d4fb282775945d43efbc120"}]}],"author":{"name":"paulali","email":"paulali@protonmail.com","date":"2021-03-24 19:42:03.000000000","tz":0},"committer":{"name":"paulali","email":"paulali@protonmail.com","date":"2021-03-24 19:42:03.000000000","tz":0},"subject":"Adds explicit error message","message":"Adds explicit error message\n\nWhen a user sets a quota that is greater than 2147483647, an\nexlicit error message that \"Quota limit should not exceed 2147483647\"\nis communicated. This is better with regard to the initial message\nwhich was \"ERROR: The server has either erred or is\nincapable of performing the requested operation. (HTTP 500).\"\n\nCloses-Bug: #1908352\nChange-Id: I0489ede104840df3ce295ad55fc99e4cf6475493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8992abfc8d1bd41c516e8741f27ce7dcb849e62b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8992abfc8d1bd41c516e8741f27ce7dcb849e62b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
