)]}'
{"id":"openstack%2Fmanila~784449","triplet_id":"openstack%2Fmanila~master~If28d97a9916ce56d7f7bf93274f5618eee01c209","project":"openstack/manila","branch":"master","topic":"bug/1922243","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2022-05-06 04:31:14.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2021-11-26 22:14:41.000000000","reason":"Carlos Eduardo replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-06 04:31:14.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-03-04 07:22:27.000000000","reason":"haixin replied on the change"}},"hashtags":[],"change_id":"If28d97a9916ce56d7f7bf93274f5618eee01c209","subject":"Check project permissions for share replicas","status":"MERGED","created":"2021-04-01 16:43:43.000000000","updated":"2022-05-06 04:32:41.000000000","submitted":"2022-05-06 04:31:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"784449-bug/1922243","meta_rev_id":"ec8680db24026a97c095e56337ff7341750cb2d3","_number":784449,"virtual_id_number":784449,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 04:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"recommended":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-02-21 23:38:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-03-04 07:22:27.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":1,"date":"2022-03-07 11:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-05-06 02:49:59.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"}],"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 17:56:26.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-04-01 18:33:13.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-01 18:58:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 23:08:40.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-06 17:16:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-27 15:53:08.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-06-28 11:35:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-09-10 07:40:05.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-11-04 15:23:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-02-15 23:23:48.000000000","updated_by":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"reviewer":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"state":"CC"},{"updated":"2022-02-16 01:54:06.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-16 03:04:07.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 11:04:40.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"ada122bdb9e74463a9c2b52bba22ef5fe784a268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-01 16:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd2dd399679f8e2cc5e4cf164e5e1af29230c270","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-04-01 17:56:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/49/784449/1/check/DellEMC_Unity_Manila_dhss_true/f780404/DellEMC_Unity_Manila_dhss_true/571 : FAILURE in 57m 24s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/49/784449/1/check/DellEMC_Unity_Manila_dhss_false/9daa3ea/DellEMC_Unity_Manila_dhss_false/575 : SUCCESS in 1h 12m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9009b8db39c0d4d89a1c29ee3e4d5357225d10de","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-01 18:33:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/49/784449/1/check/DellEMC_PowerMAX_Manila/7ca7926/DellEMC_PowerMAX_Manila/437 : SUCCESS in 1h 49m 13s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9a690f22f16f6cadb44ae6a70b095fdfaf96b664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:58:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5a886b9f074a44a94338fa8b610317 : FAILURE in 11m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25eebbe45a81481c8f767213cba85cc2 : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/760c310e0ab24b2fb51a175d1d2be466 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0f8375acf2407184e715e64b081c1e : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ab0e8a9ff1b40a1b2873ed242c3accd : SUCCESS in 11m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7da79805c8a459c816623ab9c563988 : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f45d1afe4a4d8c903a03bea831c285 : SUCCESS in 9m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/43c63e99a2fd4cab95f62e1468254a59 : SUCCESS in 6m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3c08ecc20821432b8c170adb1cb9ac5a : SUCCESS in 7m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6b6a5e9cf8ab43259418a52ac174dc7d : FAILURE in 30m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c0fb901338204f4b8ff1f8c9c48b1302 : FAILURE in 10m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fdd61c4ebe0248d7bbdade26f4c4d4ad : FAILURE in 11m 07s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4062928e1e8c42c3af74202b78f72471 : FAILURE in 11m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6f9582b95e7642cdb936a2b797653d55 : FAILURE in 1h 28m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7aaa9c4bd3cd476d99da249190c2ee4c : FAILURE in 45m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/79e5ede5211440c2b86eb92c94bc69dc : FAILURE in 1h 58m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/56e7cdd4aae044e297774f1564471cd2 : FAILURE in 24m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/afa098fd77a849ebb0b042ab41d0a1a7 : SUCCESS in 24m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f0c0355923fa4dbbba84bad1d5d38792 : FAILURE in 1h 15m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b5bbbbf39ced4d8987b63176d89f12a0 : SUCCESS in 58m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/020a9710ab13424c882d0d499c457aa7 : SUCCESS in 39m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/341db4b262164481872ab636ad254317 : SUCCESS in 41m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"594af11e7e6a1c524586b110791e7c17158a22ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-01 23:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-1/ns5_manila : SUCCESS in 2h 23m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b4738a35a74cc8626eec7219d7b2f0f81f13805b","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-06 16:14:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71c7b10078591fd7d8d7ac78d4325b71d155cafe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-06 17:16:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/784449/1/thirdparty-check/pure-devstack-manila-tempest-aio/8809162/ : FAILURE in 1h 00m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5fba6393d870213ff1afc8a61c2d49af700259f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 18:19:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b480c3c8bbc4aabb4c2391c6c50fea0 : FAILURE in 11m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3ba3c59035489091686d91d94e05a2 : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2e88544a78140bc8545f62d8dc02695 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd25748554942de833de4610df8811a : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e351d8c280410da7fb4fdc8139c541 : SUCCESS in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47733502366541d1a398ba026e77a1c1 : SUCCESS in 16m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11a48e0281b64d21b911f0a29582da0e : SUCCESS in 11m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a5065c35d514c34ab6eec6673b571ca : SUCCESS in 7m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c95c6ffbee84ad385a4078728ad4e1c : SUCCESS in 7m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/70c7d983a8b34ab1ac3add4242b7bf1c : FAILURE in 31m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/25cfac40e53f427596599757c3d69a92 : FAILURE in 34m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9d2e9c4696e4be9b4243ef21418e587 : SUCCESS in 1h 28m 05s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1879bcf095bc42e597aa273a91aeeaee : SUCCESS in 1h 07m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/554c22d6a3b14e5a816b0f2b2ed46837 : SUCCESS in 1h 26m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/89ae5c8a26dd4e798c373c74c92f2ba0 : FAILURE in 54m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/694f736cff5a4af5bec6acc423aa20f4 : FAILURE in 2h 04m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f1aff336459241d1871d6de747f44456 : FAILURE in 24m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/26b1d56161cd4c0f937b022f0b9d4668 : SUCCESS in 29m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/40ac3b31b261494b82596d68d93d7405 : FAILURE in 1h 04m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/da2547b50ecf476a9c8e2429386ef309 : SUCCESS in 1h 00m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0bb1f69d2316447893b7a1029c45b4d6 : SUCCESS in 39m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e02255ba7f444f73a2b479de52a0cb57 : SUCCESS in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6461327993fdd77d144c4e63cfe2c2a02d60436b","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-11 16:12:36.000000000","message":"Patch Set 1:\n\n\"tearDownClass (manila_tempest_tests.tests.api.admin.test_replication.ReplicationAdminTest) [0.000000s] ... FAILED\nDetails: {\u0027code\u0027: 409, \u0027message\u0027: \u0027Share a7a62295-5c65-42ec-9d63-272a08ceb38a has replicas. Remove the replicas before deleting the share.\u0027}\"\n\nerror in tempest cleanup could be related","accounts_in_message":[],"_revision_number":1},{"id":"d6935a890d186ba43decec8d43c1ea56d22bc0c0","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-11 16:14:33.000000000","message":"Patch Set 1:\n\n\"{2} tearDownClass (manila_tempest_tests.tests.api.test_replication_negative.ReplicationAPIOnlyNegativeTest) [0.000000s] ... FAILED\n\nFile \"/opt/stack/tempest/.tox/tempest/lib/python3.8/site-packages/manila_tempest_tests/services/share/v2/json/shares_client.py\", line 330, in delete_share\nDetails: {\u0027code\u0027: 401, \u0027title\u0027: \u0027Unauthorized\u0027, \u0027message\u0027: \u0027The request you have made requires authentication.\u0027}\"\n\nanother error in cleanup, auth related...","accounts_in_message":[],"_revision_number":1},{"id":"3818d42e9c01092249b050f36828d66e5947421d","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-26 12:06:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"069f7ff15d35c35ebd2cd540f42601b9f27ed62f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 19:05:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820530e6791b48569d3e50a271ee2c3e : FAILURE in 15m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e2d2808dd774c0797608daf04fab33d : FAILURE in 30m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a805a7518074943b7414823ecb942cc : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df7787f0b3246cab12add4e411ef677 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/797d5f0d818b421bb6ba49bab869bab2 : SUCCESS in 17m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed3e01fe56ce48868a905eb7aa8f710a : SUCCESS in 11m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/527fce45e8f0407e936a3d8efae01961 : SUCCESS in 10m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c33285fc33b24585b812900109f9e7ac : SUCCESS in 10m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d42240266b9c4620b4a4f4ed9088e9ed : SUCCESS in 6m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5572a1f5866041629022988bb3ccca44 : FAILURE in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d77062a9891b461c9fb2177d0fc6a4e9 : FAILURE in 40m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9da42414b0584262be5550649018e05f : SUCCESS in 1h 12m 39s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8f7cb7cda7d344fcb591295c848c05d7 : SUCCESS in 59m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fbc96fe1ac234dd08a9a18fbbfa2a050 : RETRY_LIMIT in 52m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5b9cc0bf963745a582a367edb66f4a57 : SUCCESS in 43m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8d6e25cefaad495ebf8f95ad778268fd : FAILURE in 1h 55m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1ad47b16babb4b1fab41f0565267fda6 : FAILURE in 22m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/193edd3e60184b8faeaf39944a02acb3 : SUCCESS in 17m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3418afefe3bd4d9b9a697a20bd249750 : FAILURE in 1h 01m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/32eb6c53785a4b0b8bacad708bbfabfd : SUCCESS in 1h 01m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bd7c0920b2e748d2900996b898d3db7a : SUCCESS in 28m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/863a86289f424f81988e2423cbd5c183 : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c4382762f966a386962c366fbf5dd52f891a736","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-24 16:06:14.000000000","message":"Patch Set 1: Code-Review-1\n\nI think this will work but it needs a bugfix reno.","accounts_in_message":[],"_revision_number":1},{"id":"be8cc558730ecdba86fd63b7a4463135dab50f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-06-28 11:35:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f397e46526394650cb62fdaed21396236ced575","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-06-28 11:37:19.000000000","message":"Patch Set 1:\n\nAdded the reno but IMO it would be nice to have unit tests for this... With UT, bugs like this could be easily avoided...\n\nAlso, the bug mentions:\n\"I suggest to introduce the same pattern like we have for shares with an option to get everything for admins (all_projects\u003dtrue) and filtering by project by default.\"\n\nIt\u0027d be nice to have this, but we could open up another issue as an RFE and work on it separately.","accounts_in_message":[],"_revision_number":1},{"id":"477575fafcaab469104b8dc1d671642303cb3800","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-28 12:22:29.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/784449/2/thirdparty-check/pure-devstack-manila-tempest-aio/ed257ff/ : FAILURE in 46m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b4c9c08e010e89515cd034298a36ae5ca3515b17","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-28 12:44:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/49/784449/2/check/DellEMC_Unity_Manila_dhss_true/74e3f33/DellEMC_Unity_Manila_dhss_true/163 : FAILURE in 1h 01m 39s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/49/784449/2/check/DellEMC_Unity_Manila_dhss_false/0b73291/DellEMC_Unity_Manila_dhss_false/161 : SUCCESS in 1h 08m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b1b825659035615a47ef36063ec0573215f6a450","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-28 13:35:16.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-2/ns5_manila : FAILURE in 1h 35m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"477fc24e29401d9a5d93e6ea97fa5982385689da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:37:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae6974335e941d0acdc42ee0f2a74a6 : SUCCESS in 13m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53e3acbc476e4b769a6f5bc6b6a4019b : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6cb07c55aff4901bf8941831d45b671 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4e533103b94081a679a486f088d9db : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aafa80df054949da9018bdedc7f9c065 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d92faa9251b4a93869b717f4450547f : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df7983ee9e23459d94778764d24d23a7 : SUCCESS in 12m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6eb4a6afca84a50873a150ea4526b0c : SUCCESS in 7m 49s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/96871765bf78493680f9d068102da796 : SUCCESS in 7m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d21de345586741d3b59db019043437ac : SUCCESS in 6m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2aa7381fb2ea4c2dadd528ee810278ef : FAILURE in 39m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5558016c65164cb4b9d91afa4e378fda : FAILURE in 43m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/505e9bbff76d49fa89be31856ec3cc17 : SUCCESS in 1h 12m 22s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1622537fcc994a6c8ed1feef977720b3 : SUCCESS in 1h 12m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6c12b2a5a4104d62b21efa0b9f7ceaf0 : FAILURE in 33m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9c8443e7209044d4b175979af53693ab : SUCCESS in 56m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/88e77cadf1fd4d00bc82c2e1901a674f : FAILURE in 2h 56m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9ab5a01524dd4c5890eb2fc4681b0417 : SUCCESS in 34m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7fabcd506af146ac925bccaacd12c599 : FAILURE in 29m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f312c3fcbdac4e049330b25350391ac0 : FAILURE in 1h 11m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/63d0430a54f0498bb2f765ee137f9b28 : SUCCESS in 1h 09m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ff2577ef9bb44476a0e61109b1365b5b : SUCCESS in 45m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0282a252f01349b5b5fe312c3a581ebe : SUCCESS in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab105d82c9a13edb2ea9ff5da66c2edacb712319","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-30 22:15:01.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-2/ns5_manila : FAILURE in 2h 11m 3s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"649033dafcb7dcb470bc9de862c22c05eb4cb5c9","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-10 07:40:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nhi, carthaca,\nit looks good to me aboute this change.\nplease see comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"c8b7a02942b8a343d1a63580760476123b0278bf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-09-16 15:34:47.000000000","message":"Patch Set 2:\n\nrecheck\nlogs are gone","accounts_in_message":[],"_revision_number":2},{"id":"221fd56b8b776a89f7eb48c178d2d8d757aaa124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 17:23:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec54aae938c4b86971693ff28b22b54 : SUCCESS in 12m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4d3832f1160419e81363e7fff9e7287 : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e621f27f638d420c9bc5e58322292559 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22339e4f7394bde95d5760c96d64437 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/526efca45a50424d92ecb1635ad2f6d8 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2a1fdd60e9040cb87cf57c5251b782b : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7195b85669b746ad8c8d56496317600e : SUCCESS in 11m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab066eaac59f4b4dbe415173e6d08365 : SUCCESS in 6m 30s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3a2e1a2195954e05bd2f52cd27a9f8ae : SUCCESS in 5m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/32e2c4b42f0843fe9cef575eb21ede3e : SUCCESS in 4m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/250cba1b368e4f058242644a1ca8ebc0 : FAILURE in 30m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5b333b4356ac4071add31f40eb7590a3 : FAILURE in 36m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a81c5f18ba31432db30d007e11d63f6c : SUCCESS in 1h 07m 11s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e44c05f7954e4ef3afa537e85679f5f2 : SUCCESS in 1h 08m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8c0b78e86ca14c0f955b829f916d0646 : FAILURE in 1h 23m 25s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f0c8efa83924841abb4ba5f387d489c : SUCCESS in 43m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62ab39434dfa41708b1b46c7e3477c29 : FAILURE in 1h 43m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8686f142ca344e6781c5d1337ed67196 : SUCCESS in 31m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/581e3261831849e4971d730dd87abb7e : FAILURE in 23m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/464fbc882b3442839379215c7d8e9aba : FAILURE in 1h 07m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0fa635a980384901af24d0ce730ebb6e : SUCCESS in 1h 16m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1fdc6361fe4e413c82475f685bcabccf : SUCCESS in 36m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/96f1efe9c5934050b7d0fd46daae6e4a : SUCCESS in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59492e13e216694a8ceb632e7909c3fded1cacae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-09-16 21:56:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"80d3f8cc860bead01f710cf8b1ed29e7cb499564","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-16 22:53:11.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/784449/3/thirdparty-check/pure-devstack-manila-tempest-aio/6d00001/ : SUCCESS in 55m 14s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/49/784449/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/5cd3a3d/ : FAILURE in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"337df7920d5152c86d085b6775dd6a489c758bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 00:14:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0864603dd8c744ee8c2307a611bade3b : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b4264c12e464ce7a0afa07b2eb232ba : SUCCESS in 19m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73b1c624ea0d4ae8a178e8a5b76eac49 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33314c325a734f059771fc05f93e92f3 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34b7866cc694fc09513090792f203fd : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd230dd4bc98421bbbf12ee1ce464d3b : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dd68e848e564bf3b7c090474f9f2124 : SUCCESS in 10m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ee828c5596745218879848a89a8d530 : SUCCESS in 5m 57s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7613520df944ac499afdac801de17c2 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49f604d7de11475ab601e73fd716cfb1 : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/34ed1e04abed456d893e934431c1bbba : FAILURE in 31m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/81801cf0a8cf415baab83ff2b16cdb28 : FAILURE in 36m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49873d7534f3447ebec4d23c0a074062 : SUCCESS in 1h 05m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/088e7bba92ea49f4bd97ffcb3241a489 : SUCCESS in 59m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/76bda064aae44e8190f90d376bcd6fff : FAILURE in 1h 26m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/af5aef37653c4d27bd987c7fad78c3a3 : SUCCESS in 43m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/35b6eee06532400a847dc3faf59e040c : FAILURE in 2h 12m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/72aebe2b11594405814dc1cc93f183e8 : SUCCESS in 31m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/582a22ea938049f28501086b839bf4c5 : SUCCESS in 24m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0b1064c4e7c4ea8bbcc35a7551b5483 : FAILURE in 1h 05m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/639c028b8fba4e22808a9677f0d1e071 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e28fc1b82c134f44a66071e06247b63f : SUCCESS in 50m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/390d16e249cf4e338423196419c86421 : SUCCESS in 52m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3cca7311ecb0b2621ecc91232181f43b49b58b4e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-17 03:35:26.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/784449/3/check/DellEMC_Unity_Manila_dhss_true/355359b/DellEMC_Unity_Manila_dhss_true/358 : FAILURE in 3m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/784449/3/check/DellEMC_Unity_Manila_dhss_false/c3da6df/DellEMC_Unity_Manila_dhss_false/354 : SUCCESS in 1h 16m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"9873dbac014bdd4bdfbd0a2b9f08e37338449978","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-17 04:07:19.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-3/ns5_manila : FAILURE in 1h 24m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"721c0618112a781314fa50099bbe2f3704b1101d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-04 15:23:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75e8f6e94ac5b779585ca2e9bc36adcd5f879965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 17:35:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9fa01ddf3b4ca59d685c61d1173e2a : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a4d29160b024d969b1dfe782f43c694 : SUCCESS in 17m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01d5c50cd67043ae80dcaa4719d2a4eb : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7017df7e7a0949258f00697219ce4d2b : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb3b1a4be5040db92a1afcb5a1a3f5f : SUCCESS in 19m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34d6af1c8f8e49449eacdc1eccd4b0b4 : SUCCESS in 12m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b279e413354c7d97ee04506878bbb0 : FAILURE in 20m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7650621d696e440d9bf8dac97f3186e1 : SUCCESS in 7m 55s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b70a283dedbe4d31aba2d8baef45a23c : SUCCESS in 5m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb1b7d0eef83450e873816923235d208 : SUCCESS in 6m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2aa69c33f4044f63b307ee40797519e8 : POST_FAILURE in 33m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6959532693294d5fb86d415139d0c793 : FAILURE in 37m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8d6b348857e4854a893fde358720ae3 : SUCCESS in 1h 22m 12s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1141ad9ee7c24055b93ce48a06480e96 : SUCCESS in 1h 33m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cc7bbaad299c46229e2a96f6bb5a8cf9 : FAILURE in 2h 03m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/93dcf968b9ba45609df33afec6497748 : SUCCESS in 49m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/45f23f66af5d41fe99efefed41b9907b : FAILURE in 2h 01m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d205503467894f56bccc1724f570ca88 : SUCCESS in 31m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/edad131ed2a6439c9bd547c6be1e5878 : SUCCESS in 28m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a20bcd55276e416bb470d12ff1f8dcd7 : FAILURE in 1h 08m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/54141414e840486aaeb9a8f4feb875d7 : SUCCESS in 58m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b762a5b0af644b1fb7949b75a30d947c : FAILURE in 30m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a981ce5f8b4440d4919cef85506385e9 : FAILURE in 40m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4f713530d51d9120e2e74cb6e386a16c30eb900","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-26 12:19:33.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-3/ns5_manila : FAILURE in 1h 3m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"bbc0c8bf3ecddeed9b1df7bbfbe0e7c11068364c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-26 22:14:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6a8cb97968b7fa901317c39e0ad837e4ac78555","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-23 19:29:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"be0f88a90a4698f537ac39ed71ae24101d2275d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 20:55:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e358c5b7c854de092193495bc44a54e : POST_FAILURE in 14m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c49bbb653c8479a94ec0bc88df00ae1 : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8d2430f8d4a44838997059e7928ec01 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21336af0ec92411584348dc8a2a281df : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13bb020da6c846469856de58eb172c45 : SUCCESS in 16m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51fb6cf7e6224f76b829b88122794ef1 : SUCCESS in 18m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7f89bad7d7c4fe8954836d92a7f27da : SUCCESS in 4m 56s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e2345ffc04a463ab03a2a3bef78b3d2 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3c44cd433e6425596a330a21c29a3e3 : SUCCESS in 5m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/86b68412db90462fbcc607a41e55f06f : FAILURE in 23m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5f00f4e5d663492bbcb34e4cb0ebffec : FAILURE in 36m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/84108e264e134ba3ae2ea11293c86469 : SUCCESS in 1h 01m 01s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4364f00dc9024c2d94a726ba720abb78 : FAILURE in 1h 03m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cb08f107166340bd8d9ce934088d77a1 : FAILURE in 1h 24m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/71ad82337b114259955f87f4cdd0b559 : SUCCESS in 44m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5ad1b940baf74358b71e14ae60105e98 : FAILURE in 1h 22m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/29da8e5c255e46db94fcf7250cae6699 : SUCCESS in 21m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e292db8b88e74b538c725c0c68dc5b5b : SUCCESS in 22m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d517aeb690634ffc8b420109422e3fa6 : FAILURE in 1h 05m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b463c04ead9b4522b4eae9dc786a61eb : SUCCESS in 44m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3a55afffbd744be8b3767826080d73ff : SUCCESS in 31m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/32a949275a7643d59234420f374d4d80 : SUCCESS in 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d33bcf91bc8b90c410802a3ccdf90d339b4b8a80","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-23 21:06:15.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-4/ns5_manila : SUCCESS in 1h 36m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"c63a0a9e93ce5117758ad8b3d2f592c7026e7151","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-23 22:43:59.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/f409d6977c884c0f9e955a1d8679c566 : RETRY_LIMIT in 1h 05m 19s\n- pure-devstack-manila-tempest-fa-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/1675fc4775e945a28ad120d6e8f4f448 : RETRY_LIMIT in 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5bacbc6462f90bc11fdcbb29796f865756fab27e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-15 22:20:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5782528e2e74c2f4153f78bf78151371183f529c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-15 23:22:05.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b7a90bcfcf320a073fab83651eb03c4707d6f953","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 23:22:07.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"ffd918d8212ac0a0cb9c1e4bf8090d53c215cdb5","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-02-15 23:23:48.000000000","message":"Patch Set 5:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":5},{"id":"750b93a2617882b210bd8701ceaf188f3ef0e590","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-16 00:07:56.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/784449/5/thirdparty-check/pure-devstack-manila-tempest-aio/af0caaf/ : FAILURE in 44m 26s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/49/784449/5/thirdparty-check/pure-devstack-manila-tempest-fa-aio/1a9aedf/ : FAILURE in 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4918d76856410328c8ee627cb529d0d6e1e402a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:57:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8996a119477d4146bef50e40e6777171 : SUCCESS in 8m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/094507a6ed524bc4ac47ee5c2dfddeb8 : SUCCESS in 28m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ae6af7fcabe42349906aeb10538e51a : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17ba82a166084f8894c65177ca78834e : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b70e9ebd0a184494ab29fb55e6bdce86 : SUCCESS in 16m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a9e56c64724b0cba749979c465dd3d : SUCCESS in 18m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d24534e378ee4d0b880025d24add8d86 : SUCCESS in 5m 22s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/839b5a9311354580ab82337f0c2dd18b : SUCCESS in 6m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac8d0d991b7a4d4f90eeec909be4edf7 : SUCCESS in 4m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2209c67b48c74a22a6f44884fbb314ab : SUCCESS in 25m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/320a7edc62dd4e2dbeead143617dc3b4 : SUCCESS in 30m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d0d7f77066741289e343c75065addb7 : SUCCESS in 1h 00m 53s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/395ea4f3a0e04dbcbd3f4828608191d2 : SUCCESS in 1h 01m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d31bff8e1dc49ce9df7a78ff67cb53a : FAILURE in 1h 21m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/95334dcda62e423399fc8d5ad04f07e8 : SUCCESS in 44m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0acca82d2b4342f69a9a5a6692b565d3 : FAILURE in 1h 33m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/75e415caac844fe09b5ee1e705d8284f : SUCCESS in 40m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/47697b8fc59d4e0abfbfb00111bb276c : SUCCESS in 21m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d6e8c0a366d045899ff91d6a9ab2012b : FAILURE in 1h 01m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cc04cbcc339e46169b9d92aac22d5985 : SUCCESS in 1h 07m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fd241afdc012490b9fd277cb2fba643b : SUCCESS in 40m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fa5341583d074d0ab473405cca204e62 : SUCCESS in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1e9f8a5af3f55f9513faeff996b597c4bd0070a6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-02-16 01:54:06.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/49/784449/5/check/DellEMC_VNX_Manila/4f86fe2/DellEMC_VNX_Manila/144 : FAILURE in 55m 57s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"0f0ab6e00f6dd0f4479b316929467d438942a913","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-16 02:07:40.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/49/784449/5/check/DellEMC_PowerMAX_Manila/53b2397/DellEMC_PowerMAX_Manila/147 : FAILURE in 1h 09m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"385865f4e070edd36285ecd19b3f1343dc082f75","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-16 02:17:13.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-784449-5/ns5_manila : SUCCESS in 1h 43m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"49f9774e75eaf865dc09ec21d37956cc6d7256f3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-16 03:04:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/784449/5/upstream-check/manila-cDOT-ss/7772d9e/ : SUCCESS in 2h 05m 35s\n- manila-cDOT-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/784449/5/upstream-check/manila-cDOT-no-ss/95fa94c/ : SUCCESS in 1h 50m 29s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b7f52ad45196cf9cdc9a0ec59a5b385ce1780e99","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2022-02-16 08:16:36.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f8fdd8e8a0c6ff50a566e309407568ff6720e4c9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-16 22:49:35.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a593ae546a9fa2346fcf4e3f030fc29a24778c39","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-21 23:38:37.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"52de3112c2b747a4b868cbed5bffb3dc189cd977","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-23 06:20:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37a9df1258227b35a6bf9f7120bd66ff8a084374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 08:06:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22b0f58df7140799fb337769ca926c1 : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea53c814b5df4dde96a17802fbbf4ab8 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3f88eb94b574b7186f42c1cafc66153 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a15494bd30c481e9ec0dd6ee43123cc : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d04f0bd1de4851b329ea60cdf55492 : SUCCESS in 24m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e584c7122a434fd9b211bf5a475e02cd : SUCCESS in 18m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/245161ae338c4c008d29ee38f321d1be : SUCCESS in 5m 58s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e77f6a4c66d44d35a2404a394d71bdfb : SUCCESS in 5m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b0db2e520a14d1d83c8bfaf84c3fc4a : SUCCESS in 5m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e9e20d1e8b204c00b696545909d4f83a : SUCCESS in 31m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e37680e806ef4780a9dbe795a85f3cdb : SUCCESS in 33m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c2496e94a6b4a3083a87a2ec61d1697 : SUCCESS in 1h 13m 50s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9137c6ba620f437cae6392acc8aa479d : SUCCESS in 1h 14m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/917896d1a7934c008ba7bd66929a239f : FAILURE in 1h 43m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/09f453ffb9184994b595d457d8a81daf : SUCCESS in 51m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b1e507898b9f4e86b958f52bb35d6c23 : FAILURE in 1h 43m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8daf8d23783448f2a07bb92606ee8fb5 : SUCCESS in 35m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cb4d5a1231cd4c7787f5f38fc99ee339 : SUCCESS in 22m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c73ada5f061147488551b3c40f82e6fe : SUCCESS in 1h 07m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c87435b08b104d17bd66f99a93c553b9 : SUCCESS in 57m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/51de2533fb3946819b90f2253380ebb1 : SUCCESS in 38m 05s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cff478f61962483587dd26a264b8d0af : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"abe1da43198c6b53235b966986c2e3821de0de28","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-23 19:46:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ae9481250ccd89df29ab273af7c1d6117333badd","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-03-04 07:22:27.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e3f2877aa718407f736d8d0fded9799b53d0dda4","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-03-07 11:04:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ce72387d5d275ba1706b491bd34d082e2390b441","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-05-06 02:49:59.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"671a5e8333a9da98cfcbb59b12380b1ffa85b12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 02:50:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3a8019c8309addba08e7796376411b03620bece2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:31:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da7672b08c34e1d9dff56ada168e15d : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b35faa712724f08bfd1b326cc548cf8 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2550bb79a14ebdb795b5bf267aa93d : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e89c75efc295494796cc0bbb06693de2 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7728326871764ef18e7ac9fe5527ca95 : SUCCESS in 5m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c47885afc862416489916d50d5304385 : SUCCESS in 24m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/03c3858b584f434db10d051876d7ea62 : SUCCESS in 28m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d8952f06d3d49bfa2c9b78736cc0b68 : SUCCESS in 1h 35m 12s","accounts_in_message":[],"_revision_number":5},{"id":"82aaf937fb09319b6597e5eeba43dd835ce15fe3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:31:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ec8680db24026a97c095e56337ff7341750cb2d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:32:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6756ba4d983e478cad68e1856008b409 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/413afe111b4a463da972fe5e83bbbaff : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6484de925119ffb3614921b59045a958a404d43d","revisions":{"10e4a9a25cc7c405a27e6a7fdcdb836b3b168162":{"kind":"REWORK","_number":1,"created":"2021-04-01 16:43:43.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/49/784449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/784449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/784449/1"}}},"commit":{"parents":[{"commit":"a3aaea91494665a25bdccebf69d9e85e8475983d","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3aaea91494665a25bdccebf69d9e85e8475983d"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:38:58.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:43:21.000000000","tz":120},"subject":"check project permissions for share replicas","message":"check project permissions for share replicas\n\nderive project from parent share, only for non-admins\n\nChange-Id: If28d97a9916ce56d7f7bf93274f5618eee01c209\nCloses-Bug: #1922243\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10e4a9a25cc7c405a27e6a7fdcdb836b3b168162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10e4a9a25cc7c405a27e6a7fdcdb836b3b168162"}]},"branch":"refs/heads/master"},"948a4b3dcaa3fd95fb3cb7d40f93584a16456ad3":{"kind":"REWORK","_number":2,"created":"2021-06-28 11:35:09.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/49/784449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/784449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/784449/2"}}},"commit":{"parents":[{"commit":"a3aaea91494665a25bdccebf69d9e85e8475983d","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3aaea91494665a25bdccebf69d9e85e8475983d"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:38:58.000000000","tz":120},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-06-28 11:35:02.000000000","tz":-180},"subject":"Check project permissions for share replicas","message":"Check project permissions for share replicas\n\nFixed the issue of returning more share replicas than we should\nbe in a project context. Derived project id from parent share in\nthe share replicas listing query in non-admin share replicas\nlisting request scenarios.\n\nChange-Id: If28d97a9916ce56d7f7bf93274f5618eee01c209\nCloses-Bug: #1922243\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/948a4b3dcaa3fd95fb3cb7d40f93584a16456ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/948a4b3dcaa3fd95fb3cb7d40f93584a16456ad3"}]},"branch":"refs/heads/master"},"ec4d1b1a1fa5525c34ff772de768afbef419001a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-16 21:56:35.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/49/784449/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/784449/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/784449/3"}}},"commit":{"parents":[{"commit":"a3aaea91494665a25bdccebf69d9e85e8475983d","subject":"Merge \"Add Python3 xena unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3aaea91494665a25bdccebf69d9e85e8475983d"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:38:58.000000000","tz":120},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-09-16 21:56:24.000000000","tz":0},"subject":"Check project permissions for share replicas","message":"Check project permissions for share replicas\n\nFixed the issue of returning more share replicas than we should\nbe in a project context. Derived project id from parent share in\nthe share replicas listing query in non-admin share replicas\nlisting request scenarios.\n\nChange-Id: If28d97a9916ce56d7f7bf93274f5618eee01c209\nCloses-Bug: #1922243\nDepends-On: I69c31a33c1aad8edae2d90ad6101da1be650be85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ec4d1b1a1fa5525c34ff772de768afbef419001a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ec4d1b1a1fa5525c34ff772de768afbef419001a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fd9eccf2b4ae912f6da9f109279091826c1c1605":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-23 19:29:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/784449/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/784449/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/784449/4"}}},"commit":{"parents":[{"commit":"03fea855d97b35d7d388d029bbc80af7200afde3","subject":"remove usage of six library in api layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03fea855d97b35d7d388d029bbc80af7200afde3"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:38:58.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-23 19:29:15.000000000","tz":0},"subject":"Check project permissions for share replicas","message":"Check project permissions for share replicas\n\nFixed the issue of returning more share replicas than we should\nbe in a project context. Derived project id from parent share in\nthe share replicas listing query in non-admin share replicas\nlisting request scenarios.\n\nChange-Id: If28d97a9916ce56d7f7bf93274f5618eee01c209\nCloses-Bug: #1922243\nDepends-On: I69c31a33c1aad8edae2d90ad6101da1be650be85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd9eccf2b4ae912f6da9f109279091826c1c1605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fd9eccf2b4ae912f6da9f109279091826c1c1605"}]},"branch":"refs/heads/master","description":"Rebase"},"6484de925119ffb3614921b59045a958a404d43d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-15 23:22:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/784449/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/784449/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/784449/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/784449/5"}}},"commit":{"parents":[{"commit":"b7f417afdfe4c4a1d18d533e00d1d5b6a18bb33c","subject":"Merge \"Fix quota-set no required parameters on update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b7f417afdfe4c4a1d18d533e00d1d5b6a18bb33c"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-01 16:38:58.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-02-15 23:22:05.000000000","tz":0},"subject":"Check project permissions for share replicas","message":"Check project permissions for share replicas\n\nFixed the issue of returning more share replicas than we should\nbe in a project context. Derived project id from parent share in\nthe share replicas listing query in non-admin share replicas\nlisting request scenarios.\n\nChange-Id: If28d97a9916ce56d7f7bf93274f5618eee01c209\nCloses-Bug: #1922243\nDepends-On: I69c31a33c1aad8edae2d90ad6101da1be650be85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6484de925119ffb3614921b59045a958a404d43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6484de925119ffb3614921b59045a958a404d43d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
