)]}'
{"id":"openstack%2Fmanila~786367","triplet_id":"openstack%2Fmanila~master~I719c670416c06f8eb9228a3537034285ae3733b4","project":"openstack/manila","branch":"master","topic":"bug/1924230","hashtags":[],"change_id":"I719c670416c06f8eb9228a3537034285ae3733b4","subject":"Don\u0027t run periodic_share_replica_update() on active replicas","status":"MERGED","created":"2021-04-15 08:09:58.000000000","updated":"2021-09-16 22:25:17.000000000","submitted":"2021-09-16 20:24:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"786367-bug/1924230","meta_rev_id":"a271111ad4809788b537bc29ac22adea1f591e76","_number":786367,"virtual_id_number":786367,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 20:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-09-16 19:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-09-10 07:48:54.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-09-16 19:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 09:58:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 10:01:45.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-15 10:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 10:21:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-16 10:41:05.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-04-16 11:23:10.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"CC"},{"updated":"2021-09-09 15:32:58.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-09-10 07:48:54.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-09-16 15:40:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-09-16 19:28:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-09-16 19:54:34.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-16 20:10:38.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"97e27d40d11b2912380dbe73b077d520f6907677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-15 08:09:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d136dad48745efb59e6fbb6cfe1bc009eea49ad3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-15 09:58:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-67-786367-1/ns5_manila : SUCCESS in 1h 47m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"9567d94607ea3884c17383246ee1e9950b162688","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-15 10:01:45.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/67/786367/1/check/DellEMC_PowerMAX_Manila/176c9c1/DellEMC_PowerMAX_Manila/480 : SUCCESS in 1h 51m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"51ef925b3a0f0f66309518fa7580840724f76542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 10:20:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393f4aeaacd64810ad361c11a111ac06 : FAILURE in 10m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41cfc000594749c191b72c8580a252f8 : SUCCESS in 19m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a11daf5dc3b490c98b48fb28209e959 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d33cf64148e4a5c90fca518046e897f : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1efbfe0fbf4799825d59afe6547dd7 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27c0132ffc9f47908407d13704c745b8 : SUCCESS in 9m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9c40efe5b70440aa7d9ad7849685fa1 : SUCCESS in 14m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1de7591876934b0da8b5142ab6af07dc : SUCCESS in 6m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/571064a2da9e41de80b481d32eda12a5 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6e3f9ab1ef9f4fa69ee92a2611270e4a : SUCCESS in 29m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/498b806a7fe74aefa34db41be611d60b : SUCCESS in 38m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3b1d0fd160f748ba9fadf489b3810a7b : SUCCESS in 1h 03m 31s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/11cafc18a50443009a93879a85fd03df : FAILURE in 59m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f96a207bd9ed41a491a7c724e46e8896 : SUCCESS in 1h 36m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bf2db34da3c043458990d6b4714d9857 : FAILURE in 44m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/779391e9ff364248b5c5fe9e22752a4f : FAILURE in 2h 08m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8690aeb8e44f4bc2ba279bf1d7b7833c : FAILURE in 17m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/73e1bf7a45ed46d68b80a75eda53dd34 : SUCCESS in 26m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/51bed1ad3f3f4f2ab2786392bae89470 : FAILURE in 1h 04m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3f17dc4c1f7648aebf754ea621171c61 : SUCCESS in 56m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/362f07a3d25347b2be5d2b34f5d39eab : SUCCESS in 35m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ef0b37c680cc4d259f3e5f61faaaacae : SUCCESS in 33m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5743240a91a544f5e8d1fbec65e96b2f8fd286bc","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-15 10:21:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/67/786367/1/upstream-check/manila-cDOT-ss/dc16ebf/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/67/786367/1/upstream-check/manila-cDOT-no-ss/fa03d47/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c032d5ca091abc3aba6d00ac9cbce9aa77de288c","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-04-15 12:08:20.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/67/786367/1/check/DellEMC_PowerMAX_Manila/afc5772/DellEMC_PowerMAX_Manila/481 : SUCCESS in 1h 47m 21s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"76af0e6230b4584dcc8a11a44387deeb3c91e682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-16 08:45:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6fbc99bbd3d403d46b2d31291f3ea87f1be8b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:30:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354b1e95bc2647b389d30ee2eecb94d4 : FAILURE in 10m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b2e23818d74ced9210c790ad44d373 : SUCCESS in 19m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa8de60502df4de79fa57ccf1d9f03a0 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e73bf8470041c09d0c0cbd6e1bb841 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467fc18bf73f461db942bec72b619387 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/485516ea6a054f2b9ae9f8227ad9940e : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0053f6a72c60429aa8feb692aa9090d5 : SUCCESS in 12m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b745f138adb742c4a3beb21cfec63fd2 : SUCCESS in 6m 14s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e1badaabd16043fb9592cf150bf0d668 : SUCCESS in 6m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f729bb1c2f72494083e6d22919b8493a : SUCCESS in 5m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/279ee7e106494eb7b915d7071e1f1c68 : SUCCESS in 25m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a21c4ab29c354a549bd73420b04a1911 : SUCCESS in 38m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6190b76220347b5873be13ef06870e8 : SUCCESS in 1h 09m 28s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7e0eb7dc919b4d9daaa453627bf3dd37 : FAILURE in 1h 13m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/11d81b9cc54e4512aa30fec2ee58dae1 : SUCCESS in 1h 29m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/93ca4655ef0c4c6cb45d4bb554ac8a01 : FAILURE in 43m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6d29831fe3b14c7c8c57fd4a0c955978 : FAILURE in 1h 43m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/21697afabbcf41e5965dbd6f68c1f184 : FAILURE in 23m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/55a67befbefb43aa9a5daeb59dadb8b0 : SUCCESS in 20m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5037db95a9d846138caffd9f0ffe85b3 : FAILURE in 1h 18m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6d85b5abccf046e2b82a3f928b76f182 : SUCCESS in 1h 00m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8e177ac940ec44c5a8257f2c87ce6f57 : SUCCESS in 34m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8b9f38836faa436383f9fdf60a94d7dd : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"226d42ed8927ff1d9652fc17f7b2ab7d0626d9c6","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-16 10:41:05.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":2},{"id":"02f363d1537ce7f1e11473b2fd2a7dab164e4fa5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-16 11:15:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-67-786367-2/ns5_manila : SUCCESS in 1h 44m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"4da10b058d5d38862259a63899e7bfa388396bf1","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-04-16 11:23:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fd35c34d4f94577b0566d2d2950cd7b4caac361f","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-16 11:59:12.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nThanks Douglas for taking a look - please see my thoughts inline","accounts_in_message":[],"_revision_number":2},{"id":"302170db8d14f69a4ff062c6a7c5ce421db15150","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-16 12:13:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/67/786367/2/upstream-check/manila-cDOT-ss/b27b8a8/ : FAILURE in 47s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/67/786367/2/upstream-check/manila-cDOT-no-ss/b72619d/ : SUCCESS in 1h 40m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d27597ad5891684beb4db78a1ecd67a83ee98c7c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-04-16 12:37:06.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHi Maurice, thank you for raising this issue and providing a fix. A few more concerns inline. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"4247cbac0bea7f032733d531639384163df18a70","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-16 13:04:29.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for the persistence, Douglas. There is still another issue, which will go to a separate fix, so this one here is only a partial","accounts_in_message":[],"_revision_number":2},{"id":"d71ef8552bd66373bd9448a5988d68f3a947b9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-04-16 13:09:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2b76d432de2af7ed514aff20d35a9c00627fa91a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-16 15:00:09.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-67-786367-3/ns5_manila : SUCCESS in 1h 50m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"bc62a0e985b35a4f60a65e824c0828e2ce980c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 15:08:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9359b405f5e544a3beba39038a0e6fc9 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75f2cc2ca5f64b7d9dd9e60b90a242f3 : FAILURE in 34m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae73b779b03a4b849d5b4ab4c0a34727 : FAILURE in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4854d60ce5254267a15f740d78bc9649 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d78563bced74baeb75714fa894e0350 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c72e7369c6134fd8a4fe3a4d7e6a9155 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3273d324c11d4005b714e03aa2b154e3 : SUCCESS in 9m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3b87ee3f0da480f8142543052bd8338 : SUCCESS in 6m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/79d11a2112464fa5ba6406b501be7062 : SUCCESS in 5m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/afa96eb5dda846d7a98594c6e1752403 : SUCCESS in 5m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9fd05b08bb5d47e78fca5bed7a1da460 : FAILURE in 34m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f5bb7d1b2d4a4a62870bef1e084c3be5 : SUCCESS in 38m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1bdfd038a3c84b24beea5cdd12a5da19 : SUCCESS in 1h 15m 04s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3382db139a12468b876f81ff2fe97f16 : FAILURE in 1h 05m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/92fb1dc176e14032a9c14930a2857a14 : SUCCESS in 1h 49m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f64cfc95005473bb130e6bd09164918 : FAILURE in 48m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e7b19df3ee84bdd95ad3f63f7b94fad : FAILURE in 1h 52m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2f32bc95817848cb9b76ab0f154de911 : FAILURE in 31m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a4e5bd90a0d341febecdeaa8a76b873a : FAILURE in 25m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/93bddd8de3854a63a4ac7542349b5754 : SUCCESS in 1h 19m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f3f76b36bbbd4bbaaf635342abca195e : SUCCESS in 1h 02m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c4961331a43042ca9b73545c433cf645 : SUCCESS in 36m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6ce60ea23fd34d55872f52122e3c0a0b : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0eb50ae36ac6d8628d6ecc483de50413d4adda6","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-26 12:04:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4321bb0df2cd849e4ad76230d96b815f80cab784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:01:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654779d8375940e8876782ad3d144173 : FAILURE in 19m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd8aa9ba14414cd9bc869db834c76cb8 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f66d6c0d3474b8db848f1e406acf77a : FAILURE in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1095ce47d1c94ee29700b63df49173c4 : SUCCESS in 16m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2b3f9802274705a418b8a7e3821a66 : SUCCESS in 21m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c14f5ba6bdb4c72b5aeff145b0e865e : SUCCESS in 9m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc9af8d297d54e3095d452e670e375e6 : SUCCESS in 20m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a78a9d93b824348b8f7142efb0af56c : SUCCESS in 6m 28s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/91140a9d2d7c4805a52fc2f1803a7b58 : SUCCESS in 6m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0a29c45826eb4a7688cb28f899a89d32 : SUCCESS in 5m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd23be468d9a4005b4e34889beb7e45c : SUCCESS in 29m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3793ce70183465694ac3dd3c993a52c : SUCCESS in 37m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/958611e7932c4a3c91e67dd76db7b486 : SUCCESS in 1h 24m 01s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eac0160552fa41fba1a2bafa53c074ef : FAILURE in 1h 23m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a1e363cffd2c4617a7414035315ed24c : FAILURE in 1h 55m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9c93c12347ae41e78df6a2a45c00480f : SUCCESS in 59m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c1479457dba248d5836eacb393b7d473 : FAILURE in 1h 43m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d32e9af6ec36429482edf421ebde5783 : FAILURE in 23m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/22a4a70f196c4546945b2c9bb51b8063 : FAILURE in 19m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/98d1f780f4154588affa5c299d87c128 : SUCCESS in 1h 08m 33s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a896464c6fc446038e9bb7de2e56512a : SUCCESS in 58m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0a2b1015f82c4aa89be665a9976c8dbb : SUCCESS in 30m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2686564948c44cfd9634b043a0265eb5 : SUCCESS in 38m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f6eb22aa2629032642b6167268c9c5c169decc4","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-26 14:09:46.000000000","message":"Patch Set 3:\n\nrecheck\n\ntimeout on downloading urllib3","accounts_in_message":[],"_revision_number":3},{"id":"97cd5eda3ff9a040eebc9ca7a039496a5c573cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 18:54:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ffa5bd7d0dd44a8997b53223fe50194 : FAILURE in 10m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c38cd1a40c3c4122a702fca5d9fe97d0 : SUCCESS in 26m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b01e7b8654c64411ad776946f184f200 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c95c1e7a83649d8831391e3d2cdadb7 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd96635054f94e249f495d078755ca76 : SUCCESS in 8m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a27e310fcbb140e1a110163e7f46be83 : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73b9653206834b6fa87350fd83e97e77 : SUCCESS in 9m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57fd13c6f5854e40a6220e209c1294f6 : SUCCESS in 5m 55s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5da446e0ac0844e695b55ffaf8ab57ea : SUCCESS in 5m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1f79c952f26944a9b86e20399146b2a5 : SUCCESS in 5m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97dab4076c5545b98d3be25c4c251c9c : SUCCESS in 32m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/360784147e6049fbbbd5d5b35597dc53 : SUCCESS in 33m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f757f2ed550a47f4b2baaf00f1800f6c : FAILURE in 1h 07m 22s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6c135076551f4cc689e1e9f4da1b763b : SUCCESS in 1h 07m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f4112523cf154d5eb3c7bfbbe9ab1b3f : SUCCESS in 1h 40m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/99fea60d790643b9bef0152e941c6b14 : SUCCESS in 48m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/560040629e8d4010b2f36ee5c1961c40 : FAILURE in 1h 21m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/90c7d609d75e49b2a688d41c1860fa46 : FAILURE in 25m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/eb22ef852a2b4fa4ae32e5bbee1c85a7 : SUCCESS in 24m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2641f9f381f14a4ca2dabace95f53632 : FAILURE in 1h 19m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/09870600fbb34ab996f34900e4d53785 : SUCCESS in 56m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2bb7949047a247ce8e22c01f2fb2ebf6 : SUCCESS in 36m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/36232b7c3e234d22a9774ace37edde3a : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ad57547f641508614b196c760167d049e3891893","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-26 19:09:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"61d7544dbcdb8f6f0eb33a96ada75b9b87b81488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 21:24:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ae143596984739a903408187b6ea78 : FAILURE in 10m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcca6a41e5ae4359b4c8b6daa63456be : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5c74d55ebc24057b7f0c6afce472013 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eaf1b2f07384a4b84cb792203114005 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2feb5ed4625a4088824b9df152bce8a0 : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/035bf780964447b9b48bb51c02d78c03 : SUCCESS in 16m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4fc9037212d4ca9a4bc14628d115ec9 : FAILURE in 19m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba4bc99d48524a6298f85572ff859ba3 : SUCCESS in 5m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f15e0f0b15741c1a9c1ebe6c9391438 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e937339b74d247bbaec3822bde74218a : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24b876da6a50417c91acca2764489368 : SUCCESS in 26m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b716f2d085934ae183307dd86be8d9d5 : SUCCESS in 36m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cf8f6840e45b4caf81e967b4d5b3a77d : FAILURE in 1h 13m 48s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c17f399d9b97436ab9006de428e50e6b : FAILURE in 1h 51m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/88cdbf25dd4245d3875126de7060be77 : FAILURE in 2h 12m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fbbc6bcc17484e0291ca3855e00ed778 : SUCCESS in 45m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/921e4d5b2c6745c0a533f0f6955946ec : FAILURE in 2h 04m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d4ebe79622fe4e8b94c3a3b7de2b90e3 : FAILURE in 26m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/06b437557d8449ec899827b377122ce7 : FAILURE in 15m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f1b3a270adbe471ab69348558c2451f9 : SUCCESS in 52m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a19020be92ac4f2490bfea9f9fa85a5d : SUCCESS in 53m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cd5726fa263848299291d8c7b47aac60 : SUCCESS in 38m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2930b4099a284b158b7d5ae8e2493cb9 : SUCCESS in 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3aebbce66d84a48a70b4b15dca5ca968ddcab0d","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-27 10:38:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cfe3a7b95788eb8eacf993e137891863f724acfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:13:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43194feecc0c4a8eb855ff35bcfbe44d : FAILURE in 11m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d545fb518c3496189d89abc293709db : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b2faddd81964c3fb4876fbef93f7181 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9053d24b0b0b4a73971a9e626fbccfdf : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e3917d01c34504ba98b70a483d054f : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/595e0ac64daa48e4a357e62143617098 : SUCCESS in 13m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0ac5377b4614bfe879c0f1b70471e94 : SUCCESS in 12m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14869d017ed34ece9321e03da4d7781e : SUCCESS in 5m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b8d46a864224daeaae705ecf52d159b : SUCCESS in 5m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f6b7bd889f334985bdb4b6927299859c : SUCCESS in 5m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/556569a10353436dbb7ec78be8a6e67b : SUCCESS in 33m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be86198e1a734012bc955ad6f605efd7 : SUCCESS in 40m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8717b1082f34dcd8b6b218c2015ed5d : SUCCESS in 1h 24m 05s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/44df06c4a1184befbf7bb8647f749691 : SUCCESS in 1h 18m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cd08285aceeb412ba03cbf3bad4d3935 : FAILURE in 1h 36m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f26488c7ce82498789560cb6aa9a72ba : SUCCESS in 59m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e1d555895ad142fe913fdf7523e63176 : FAILURE in 2h 27m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a56892e038e8470f861fdbc19da388cd : FAILURE in 23m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ec1e44eafaf54fde89116ddc578edf0b : SUCCESS in 25m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9344f6e22b194bf7bb6728a37aa91b2f : SUCCESS in 1h 04m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bf2c053611984fcb85d9ea963dde048d : SUCCESS in 58m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6b863d2a27ea43cb9c928b2c89c6474d : SUCCESS in 34m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/220365d42ec747e6b7bdbf05dcaa951d : SUCCESS in 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f1f4362a59ab4fb68443b096f46537061aaabbfd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-28 10:09:55.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/67/786367/3/upstream-check/manila-cDOT-ss/f0e32c0/ : FAILURE in 7m 26s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/67/786367/3/upstream-check/manila-cDOT-no-ss/1fd392d/ : FAILURE in 7m 32s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7b8ee613fa0c2ddbafcd1763dbc0796a11756a5d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-10 07:48:54.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM.","accounts_in_message":[],"_revision_number":3},{"id":"8170aa371fab17a7eb07b45e4f843f9d18df615d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-16 15:40:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7830e31d2aca497650dd149a004e47fa1b69d2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 17:23:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0865e3dfb5b34b18b4b5ef1d81632c8e : SUCCESS in 11m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a68fffebeac74f20a987b56c5ea4a6d3 : SUCCESS in 21m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6074f5ef3ac49cf9cfcdd9d24f0ca7e : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8219973eb661475aa72d989d785187cc : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b1c60f70d9415da3fd19afa6ccade9 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e017b014c7fc43aa8e66b7e629d31d93 : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01bab4ed1b414a4da62c88ec1a1b22a8 : SUCCESS in 11m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1523953069dc468ab5743b23502f8285 : SUCCESS in 6m 04s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a0da13e6ccb34b2a9f5455afa9252679 : SUCCESS in 5m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e72d5f0d9c1b4380b8b480df9e857a9a : SUCCESS in 7m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ebd03597e1674c4e8da0b9d5fc492f96 : SUCCESS in 27m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a77eff5f96e4417fb55af58e8557bbf0 : SUCCESS in 32m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a0f74acc10114deeac697fec1baa3b1b : SUCCESS in 1h 03m 24s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c2424b1077fb4647b80dac9b4b3efd7e : SUCCESS in 1h 10m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4f18515866644f518b47dfff64cfe2c3 : FAILURE in 1h 24m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40e5615100a54c109c63306127cf52b1 : SUCCESS in 41m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1156552754b342a4be04e2028a7fd554 : FAILURE in 1h 41m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/40bb27ca26e34e3a86085233873fbc91 : SUCCESS in 26m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0362aadc48cf434daf2964e2d2b6e45b : SUCCESS in 22m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/46c0eb4ae8a5403798e7a835f2589cca : SUCCESS in 1h 20m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3ee525268ec54130bc5dc8219325bbe9 : SUCCESS in 1h 03m 10s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0fdbf12d835447fd8fd502e374d51bdb : SUCCESS in 33m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/927c6bf9827344c0972bceccfe7448b9 : SUCCESS in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"177eb17276bd52dea151a1fa679961d8b887afb9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-16 19:24:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/67/786367/3/upstream-check/manila-cDOT-ss/31efb28/ : SUCCESS in 2h 00m 41s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/67/786367/3/upstream-check/manila-cDOT-no-ss/8e37124/ : SUCCESS in 1h 49m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0bb39613dbf95d33f488af5d2a4b68cde18c31e4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-16 19:28:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c65e4795e4af42a70d0bf5310bbcd056e578115e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 19:28:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a87e7a683a38b20bbb43b78d321a036da38037af","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-16 19:54:34.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/67/786367/3/check/DellEMC_VNX_Manila/04e2bbb/DellEMC_VNX_Manila/183 : FAILURE in 36m 31s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3ea91e78639bddea2964b377755300f336968227","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-16 20:10:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/786367/3/check/DellEMC_Unity_Manila_dhss_true/f99b458/DellEMC_Unity_Manila_dhss_true/353 : FAILURE in 3m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/786367/3/check/DellEMC_Unity_Manila_dhss_false/9647d5b/DellEMC_Unity_Manila_dhss_false/349 : SUCCESS in 1h 20m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"dd242ea54d9581631954fd8962bfb05554112fdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 20:24:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd9e682c2a5423e8ccd817572bbce7d : SUCCESS in 12m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190d1188b8d648cda79c18d73a030d37 : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f841c0de42d94dd9b806263341b78992 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3afdcfd0e08e49828922d8d42262d279 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/193e790bb097423db2f99858aa26ecfb : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945e1babcbd64c36bd6bf360c4378a49 : SUCCESS in 6m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2ee65bafdc054bd6901018965a28fc96 : SUCCESS in 27m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a51b8e866494c758e9cf7edca1da8a7 : SUCCESS in 37m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9f52f6270b0a429f96e41040bbdf8d14 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"801e3824b5bcc0e4ec35fd1597cb7b046dc446ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 20:24:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ea93db4e00d46fd911f627f38e6a7165917225da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 20:27:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def0927e5477472da17e04dcaf1910a8 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/006c83e032514a08ac32b1b5d96f4233 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3},{"id":"a271111ad4809788b537bc29ac22adea1f591e76","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-09-16 22:25:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/67/786367/3/check/DellEMC_PowerMAX_Manila/6d6f7c2/DellEMC_PowerMAX_Manila/258 : SUCCESS in 3h 09m 17s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5ba3117f20261d6a49a0ee0131c3c01c2bfb4d2d","revisions":{"a2ad167c477c71ca7b08a7efa5a763a72127b874":{"kind":"REWORK","_number":1,"created":"2021-04-15 08:09:58.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/67/786367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/786367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/786367/1"}}},"commit":{"parents":[{"commit":"ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b","subject":"Merge \"Add api reference for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-15 08:06:41.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-15 08:06:41.000000000","tz":120},"subject":"Don\u0027t run periodic_share_replica_update() on active replicas","message":"Don\u0027t run periodic_share_replica_update() on active replicas\n\nSimilar to periodic_share_replica_snapshot_update() skip active\nreplicas and do what the comment above says.\n\nChange-Id: I719c670416c06f8eb9228a3537034285ae3733b4\nCloses-Bug: #1924230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a2ad167c477c71ca7b08a7efa5a763a72127b874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a2ad167c477c71ca7b08a7efa5a763a72127b874"}]},"branch":"refs/heads/master"},"7412add382798ae2720460451fda5cad10ed3535":{"kind":"REWORK","_number":2,"created":"2021-04-16 08:45:13.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/67/786367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/786367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/786367/2"}}},"commit":{"parents":[{"commit":"ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b","subject":"Merge \"Add api reference for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-15 08:06:41.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-16 08:45:06.000000000","tz":120},"subject":"Don\u0027t run periodic_share_replica_update() on active replicas","message":"Don\u0027t run periodic_share_replica_update() on active replicas\n\nSimilarly to periodic_share_replica_snapshot_update() skip active\nreplicas and do what the comment above says.\n\nChange-Id: I719c670416c06f8eb9228a3537034285ae3733b4\nCloses-Bug: #1924230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7412add382798ae2720460451fda5cad10ed3535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7412add382798ae2720460451fda5cad10ed3535"}]},"branch":"refs/heads/master"},"5ba3117f20261d6a49a0ee0131c3c01c2bfb4d2d":{"kind":"REWORK","_number":3,"created":"2021-04-16 13:09:14.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/67/786367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/786367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/786367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/786367/3"}}},"commit":{"parents":[{"commit":"ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b","subject":"Merge \"Add api reference for share server migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad4315eb360ccc9afbd78d4bc5b5e5e0716df87b"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-15 08:06:41.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-04-16 13:08:53.000000000","tz":120},"subject":"Don\u0027t run periodic_share_replica_update() on active replicas","message":"Don\u0027t run periodic_share_replica_update() on active replicas\n\nSimilarly to periodic_share_replica_snapshot_update() skip active\nreplicas and do what the comment above says.\n\nChange-Id: I719c670416c06f8eb9228a3537034285ae3733b4\nPartial-Bug: #1924230\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5ba3117f20261d6a49a0ee0131c3c01c2bfb4d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5ba3117f20261d6a49a0ee0131c3c01c2bfb4d2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
