)]}'
{"id":"openstack%2Fmanila~788608","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~I304bf59b3f8c0d5b847078a5752bac8ac4f21690","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/1922075","hashtags":[],"change_id":"I304bf59b3f8c0d5b847078a5752bac8ac4f21690","subject":"[Glusterfs] Fix create share from snapshot failed","status":"MERGED","created":"2021-04-29 10:21:50.000000000","updated":"2021-05-19 18:35:21.000000000","submitted":"2021-05-19 18:33:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788608-bug/1922075","cherry_pick_of_change":784189,"cherry_pick_of_patch_set":4,"meta_rev_id":"d985a377d6ba2115aedc77bed6aefbe3026f2477","_number":788608,"virtual_id_number":788608,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 18:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-05-04 10:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2021-04-30 10:03:44.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-05-19 17:22:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 10:49:36.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-04-29 12:11:07.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 12:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 02:14:59.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 10:35:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-05-19 17:22:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"daa10ea42641c24bb6a4499b3263bab70958e5a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-29 10:21:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f31d43422fded7e684f135711b0fabcd96d91a5a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-04-29 12:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-08-788608-1/ns5_manila : SUCCESS in 1h 48m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a6418f679e46335479554f5780a7008e3ad396d9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-29 12:21:38.000000000","message":"Topic bug/1922075-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"6cff9e8e33cf52f045d59c990b10fc9e5f1d8e7f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-29 12:21:57.000000000","message":"Topic set to bug/1922075","accounts_in_message":[],"_revision_number":1},{"id":"d3d4eecca4cdee26cd266230a3473d49a8e8313d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68129b834c648d08b21b24ccecdc7d3 : FAILURE in 10m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c07d199a20b4572be56cb4a95289d88 : SUCCESS in 21m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ba2ea0052234ab3ac9f1ed894fd6ac6 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b28bd4d39a4cde87b30a2ec7d60838 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60655f09426245c88054e453f112e64a : SUCCESS in 9m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8fd60d44a8c4ef89f12e977280874d7 : SUCCESS in 14m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b743415b13245759316303c8400d364 : SUCCESS in 9m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e90d89c89794733b29d98b2f58e0d60 : SUCCESS in 5m 42s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5560015ea1894fd0ac362a527c5fbf92 : SUCCESS in 5m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7c50d71651c4295b1ca6b19237e0a81 : SUCCESS in 5m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bfa8ec8c018745c3942e8820f60df7a9 : SUCCESS in 26m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f97514bb7da04c63bda75e6581c0cceb : SUCCESS in 39m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/43e516f7a8bc4d25b3d439a7dde4811a : FAILURE in 1h 19m 18s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dd857b2b5beb488b946adff0b718059d : SUCCESS in 1h 16m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8e98ed50d15d4493b64edefb342adb87 : SUCCESS in 1h 03m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4858b8b98a134bc18df5a8391b4af664 : FAILURE in 47m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f03cc175b36c4da5a0412167e09665e3 : FAILURE in 2h 28m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/305dc4df190a402798be32d11f772860 : FAILURE in 26m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/12f82783fe0646149a50c974c1318aed : SUCCESS in 28m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/03f65af2f9ef463da8ad0db5a0b86071 : SUCCESS in 1h 13m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e27677ca25884484a5df971f91547bbd : SUCCESS in 1h 05m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/161635590238448caa0518dd78d607a5 : SUCCESS in 36m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/14c30ea2cc3f49ffaee6d25d4dd20ce5 : SUCCESS in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f69b443869b3e2ae631ceae71f3d9c65bd29f64","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2021-04-30 00:44:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"814a8e3c1a8326e70f94e95056be894068a2bdbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 02:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a198d960e8a841e6896dd8e65316b93f : FAILURE in 10m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14e7d449eeac4f278215469a01c9869b : FAILURE in 20m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/367b873b6bf34b1d916caaf74fd97403 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a3a3a4b76944768888ac924ea74ae5b : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/477fee02b3344717a8d831766bdb74a4 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2a68fdfb9244d4cbad29a2bfae8446f : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/179c0a2d3af84325bbab8954ee6ddd73 : SUCCESS in 10m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55dc17e8d0049a5aaa46beec59c365f : SUCCESS in 5m 42s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c1980a3fafb4235ac8e7ab24d69afbd : SUCCESS in 7m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f7396d70e7674e57b6721f818583bf0c : SUCCESS in 6m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2920851c8ac44289ab95de399760c62a : SUCCESS in 26m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/870113d21c174619aa50ed0335a9ae8e : SUCCESS in 42m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ef91858765247f4860b416711f01cd9 : SUCCESS in 58m 41s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c0a4544c5d2140fc95a35745b6062f06 : SUCCESS in 59m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e1a40e1cfa20454d90c5b671c74d925a : SUCCESS in 1h 18m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27c575d274124d449e27cd2a8de687db : FAILURE in 43m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ec3501b27e1040e29d2b61c84565eaf2 : FAILURE in 1h 27m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e1c870bfe23e4dc3bf3e8fb9d67b20c0 : FAILURE in 23m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/88b6303b27c646cba79ddc5716b0c210 : SUCCESS in 22m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/17fbcc8cc09d4239a3962fb105a90302 : FAILURE in 1h 11m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0262a58d6dec4e548e7663f78eba10e0 : SUCCESS in 53m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9ae61fc15b824b908b616f80057b216d : SUCCESS in 35m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/627be4b3053f4c59bb1b057f421afcf8 : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6c7a06fd0921b4717658ca4c9ff3ebef5bcc64e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-04-30 02:14:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/08/788608/1/upstream-check/manila-cDOT-ss/33e44ce/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/08/788608/1/upstream-check/manila-cDOT-no-ss/51995b8/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"26797e4c46994d58963e50093f20df3a61e7a8c8","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-04-30 10:03:44.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry pick, good bugfix; no config, api or db changes.  Driver specific.","accounts_in_message":[],"_revision_number":1},{"id":"310686799d804ff98cd487f5574cd06c2a68013e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-05-04 10:35:54.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry-pick. LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"7cb61df3f869b18520eeb11d1491003e88b82f94","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-19 17:22:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6fa68ca76a121e3d39f3dca03b9db8a19d70f06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b59cd07aacb5b785d8e663e4d7cc4d832e30f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:33:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f31ea857bf44145ac551e9dd6276ebe : FAILURE in 10m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c444f797620c4ce58af962457d6e2894 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465ea19c53ab4b40badf874f6ee33990 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/213827580df74b4db05cbf6d90d7b78c : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5f7e5b876c94fd49044118f0254596e : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1d8085682af47dc8b96ceaccf0ab745 : SUCCESS in 5m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d9f202029da5421e8f86f4151acc85cb : SUCCESS in 30m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1771ab332eb149e2a57befc858e66d07 : SUCCESS in 43m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f6680892eaa0480f8b93de8ebd994406 : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6100c3c325692a76794d06f37b7c6f8c4418489c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:33:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d985a377d6ba2115aedc77bed6aefbe3026f2477","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 18:35:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a11b85c10842d5951b51c54729b17c : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/743d47da77d445d8a848ae091d7634a2 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3635ade670cce08525c946eb9029940d86fcafa8","revisions":{"3635ade670cce08525c946eb9029940d86fcafa8":{"kind":"REWORK","_number":1,"created":"2021-04-29 10:21:50.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/08/788608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/08/788608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/788608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/788608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/788608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/08/788608/1"}}},"commit":{"parents":[{"commit":"8de6a20379e1bc7e3e1f4ddfb0f95dc56254849f","subject":"Merge \"Reuse _set_instances_share_data for replicas.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8de6a20379e1bc7e3e1f4ddfb0f95dc56254849f"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2021-04-01 02:56:36.000000000","tz":0},"committer":{"name":"Lin PeiWen","email":"591171850@qq.com","date":"2021-04-29 10:21:50.000000000","tz":0},"subject":"[Glusterfs] Fix create share from snapshot failed","message":"[Glusterfs] Fix create share from snapshot failed\n\n1、After performing a snapshot clone of the glusterfs vol,\n  the status of the vol is\u0027Created\u0027, and the parameter\n  \"gluster volume set nfs.rpc-auth-reject\u0027*\u0027\" is required for\n  the vol in the\u0027Started\u0027 state.\n2、The cloned volume needs to activate the snapshot,\n  if the snapshot is already activated, you need to skip the activation step\n\nCloses-Bug: #1922075\nChange-Id: I304bf59b3f8c0d5b847078a5752bac8ac4f21690\n(cherry picked from commit 63e255248b7072c92c084a0cea269a27bd47f813)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3635ade670cce08525c946eb9029940d86fcafa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3635ade670cce08525c946eb9029940d86fcafa8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
