)]}'
{"id":"openstack%2Fmanila~789502","triplet_id":"openstack%2Fmanila~master~I6cf8a08379b2f2b0fd1bcada4cd20c39adc01f47","project":"openstack/manila","branch":"master","topic":"bug/1927060","hashtags":[],"change_id":"I6cf8a08379b2f2b0fd1bcada4cd20c39adc01f47","subject":"handle replica state on migration complete","status":"MERGED","created":"2021-05-04 07:45:50.000000000","updated":"2021-08-29 18:23:33.000000000","submitted":"2021-08-29 18:21:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789502-bug/1927060","meta_rev_id":"d68245ba9eb83d773c9e580af7ee4739340d8268","_number":789502,"virtual_id_number":789502,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-29 18:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-08-29 17:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-08-13 19:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2021-08-13 08:11:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-08-29 17:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 08:08:55.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 08:59:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-05-04 09:21:21.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 09:42:47.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-05-04 10:55:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 10:58:50.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-04 11:57:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 16:24:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-06-01 21:21:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-06-24 15:21:09.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-08-13 08:11:27.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-08-13 19:31:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"607267afd55950559683cdee188c6ab6a2a5fc3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2021-05-04 07:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4791f5bce32f6334b264a5318f7d9df536ef671a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 08:08:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/789502/1/check/DellEMC_PowerMAX_Manila/99edb2e/DellEMC_PowerMAX_Manila/24 : FAILURE in 22m 47s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"ba0d7e381713e7fd186afd0b7e44d06ade66ffc7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-04 08:59:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/789502/1/check/DellEMC_Unity_Manila_dhss_true/745e64e/DellEMC_Unity_Manila_dhss_true/40 : FAILURE in 1h 13m 26s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/789502/1/check/DellEMC_Unity_Manila_dhss_false/c8cd428/DellEMC_Unity_Manila_dhss_false/41 : SUCCESS in 49m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d38e508102c03059b62043d5c91657d6113152b1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-04 09:21:21.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-02-789502-1/ns5_manila : SUCCESS in 1h 34m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6130d6ede848980a19996dff98404e50d023047b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-05-04 09:42:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/02/789502/1/thirdparty-check/pure-devstack-manila-tempest-aio/626b956/ : POST_FAILURE in 1h 53m 21s","accounts_in_message":[],"_revision_number":1},{"id":"76bf50492eca537e8e0c6962e2fde55d251925b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 10:55:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66d135b54fc46a09ac10b524c30751a : FAILURE in 11m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2670d7e6d11a45fe94767d8f8be0d6fd : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/244760c45ab4493681dbd68a76f60b20 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b058888b9224f6c98d722b079f9d5f8 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4decb6841f0c4e798f7046e16c135835 : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f12f1bfde8f34c5b8071cd6616c3f2cf : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9903af1247e14dbaab392a560c53fd9c : SUCCESS in 9m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1dfc30a8b41d4c25916de26f0e00e449 : SUCCESS in 6m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f4f95f5c127f48d0b23cd3c5fa3c42e6 : SUCCESS in 5m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aabf96697d18452283ea56ccb435174c : SUCCESS in 28m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cae7fc456ff54692bd286610619f41cc : SUCCESS in 32m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f5b9c32ec45f4436b10bfefe975a07ff : SUCCESS in 1h 11m 54s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fd283a086da644b19b9b9c9b612dc2f9 : FAILURE in 1h 21m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b576924a46134f4aa272e48e1872aca0 : TIMED_OUT in 3h 07m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b96e9d47ef76485ab1b4329fbc94f0bc : FAILURE in 41m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1d1ad48fe8fb46e292ca0a7eb2aff7bf : FAILURE in 1h 14m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9adf8ceb71e54b108f0108b1bf7e33d5 : FAILURE in 20m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4b0a1842c0a94cda872bae3969575d6e : FAILURE in 28m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5c0a86625d3a4cc1b2c6896888361330 : SUCCESS in 1h 09m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/952813bf7e6e40328c733c8b6acd9e26 : SUCCESS in 1h 02m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b629f0fd564c474f89a93cbc49aed243 : SUCCESS in 32m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8bf8a8d7c589442ba4506835b7db9bb7 : SUCCESS in 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c75ad3bc6e3c4dd36a4591fabb85d0cf1ab7487","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-04 10:58:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/02/789502/1/upstream-check/manila-cDOT-ss/6f052dd/ : FAILURE in 2m 41s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/02/789502/1/upstream-check/manila-cDOT-no-ss/0b35ecb/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7e7a3f1638ba5610de27590d207c8e23c8bcdc5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 11:18:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/789502/1/check/DellEMC_PowerMAX_Manila/1fcfea4/DellEMC_PowerMAX_Manila/25 : FAILURE in 22m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a6fcdaebdbab7a026f839a262d5eb678b81ce6de","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-05-04 11:57:09.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/02/789502/1/check/DellEMC_VNX_Manila/ff1c87b/DellEMC_VNX_Manila/22 : SUCCESS in 1h 01m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"01a2ae6604c17365664942c95547d1e40c608976","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-05-04 12:10:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/02/789502/1/check/DellEMC_Unity_Manila_dhss_true/298f895/DellEMC_Unity_Manila_dhss_true/41 : FAILURE in 1h 14m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/02/789502/1/check/DellEMC_Unity_Manila_dhss_false/d855ac4/DellEMC_Unity_Manila_dhss_false/42 : SUCCESS in 54m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"970661623f6b678798f12eb48621a01e5e9342ea","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-27 16:24:46.000000000","message":"Patch Set 1:\n\nLooks great, just a couple of missing things -\u003e unit test changes and a release note - adding these to the same patch makes for an easier back port","accounts_in_message":[],"_revision_number":1},{"id":"f8a1aad916828d26e62603f91ae95404edce0f52","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-06-01 21:21:54.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry-pick. LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"ea96c68bccb3baf1986baa01694fc88c6c521ac0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-06-10 12:31:51.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e Clean cherry-pick. LGTM!\n\nRealized now my mistake on the vote 😐\nI can submit the unit tests and release note for this fix :)","accounts_in_message":[],"_revision_number":1},{"id":"5c0a366c0d7fd79eab0a7672129943d4b32c3753","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-01 03:58:28.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-02-789502-1/ns5_manila : FAILURE in 1h 29m 13s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4dca532ce508a6eb407ca7425cd5920048cf6246","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-22 19:31:05.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+2\n\u003e \u003e \n\u003e \u003e Clean cherry-pick. LGTM!\n\u003e \n\u003e Realized now my mistake on the vote 😐\n\u003e I can submit the unit tests and release note for this fix :)\n\nCarlos - still working on this?","accounts_in_message":[],"_revision_number":1},{"id":"149ae7fdb1a1c52ce51aff272bfb108df9f6bed6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-07-27 20:22:43.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: -Code-Review\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review+2\n\u003e \u003e \u003e \n\u003e \u003e \u003e Clean cherry-pick. LGTM!\n\u003e \u003e \n\u003e \u003e Realized now my mistake on the vote 😐\n\u003e \u003e I can submit the unit tests and release note for this fix :)\n\u003e \n\u003e Carlos - still working on this?\n\nYep... Hopefully in the next days I\u0027ll have an update","accounts_in_message":[],"_revision_number":1},{"id":"ce8767f61f71778bdd91ec7cad3eaf37568860a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-12 23:03:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e007ee0a2e9a410af4d59dc3a89022a26014bcf1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-12 23:05:28.000000000","message":"Patch Set 2:\n\nGot a chance to update this change. Please check the changes :)","accounts_in_message":[],"_revision_number":2},{"id":"0c78b963a0bc8e067ef761097dc29b066354c3fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-12 23:06:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f5bcda2a94c1efffda903729219c0a508f9eae80","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-13 01:03:58.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/02/789502/3/thirdparty-check/pure-devstack-manila-tempest-aio/914a026/ : FAILURE in 1h 10m 56s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/02/789502/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/1cddb76/ : FAILURE in 1h 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70b7ac20cf3ecf0eac66ab365b6186b72b22bd9a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-13 01:50:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/02/789502/3/check/DellEMC_PowerMAX_Manila/d61c3b6/DellEMC_PowerMAX_Manila/180 : FAILURE in 2h 43m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3eb0e642785c28b700304c4dfd9beca90ba443d8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-13 05:12:35.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-02-789502-2/ns5_manila : FAILURE in 17m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5d7ed585366cea073feb10d7aebe5f696f246eb3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-13 05:57:44.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-02-789502-3/ns5_manila : FAILURE in 17m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"68f795d41eb84d694ddf3844d7b49a8d1f763279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 07:14:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92736073a298406f98837fb284f0fdcf : SUCCESS in 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdafb43e537f4fbdb6bcc3ea7d82f5a6 : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c823e88e2e8e425b8a6751e05a90f898 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505551a526444f319157b5132d13bfa0 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f05ab0a44c4d1884ca4f4ca68b498f : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29587e9af16740c5adf914554bac05c7 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8d1dc678fa1433d9b04a6786102dab8 : SUCCESS in 8m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8836252b002f41abadc5143c5d9d065c : SUCCESS in 5m 45s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/abc8d51f8fe94af5856793df46644b00 : SUCCESS in 6m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c154c569bb15488b9b35f18d2f8019ef : FAILURE in 5m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f52144e318374bda8f7765d213aea2c6 : SUCCESS in 28m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3610daf200c44849319bb36cc470e09 : SUCCESS in 36m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/06cf7b79037041b39f572ead8759bd3d : SUCCESS in 52m 55s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b6a1609e373b480081b075eec02b52d8 : SUCCESS in 1h 25m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/19bba028a58944028838dbf0094116ef : SUCCESS in 1h 55m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/678d5908add345698679b3a794eb92b7 : SUCCESS in 48m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c96f94ce47ed4a8aa301c1ab914f0435 : FAILURE in 1h 50m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7f092f81e27242c98b7d96b84339fa93 : SUCCESS in 30m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b18e25354b564b60b39c3136bada8ad4 : SUCCESS in 20m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d3a34c5431d4813932b7b1c38735428 : FAILURE in 1h 04m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4642b55b26ce4466ae2624f5b48f138b : SUCCESS in 46m 25s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c65c823fb2784473980bcbc735741f09 : SUCCESS in 40m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/07c82240d62443b7a2d7bfb23af48d68 : SUCCESS in 42m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14ebd7934c42b1f371eecd9e11baeaf1b5925b13","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-13 07:29:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/02/789502/3/upstream-check/manila-cDOT-ss/29754d9/ : FAILURE in 11m 00s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/02/789502/3/upstream-check/manila-cDOT-no-ss/9038ac2/ : FAILURE in 11m 21s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e6f6a241dbc268eddf0fed03c877dd8f708c3e5f","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-08-13 08:11:27.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"efbb423fa371ed58f647d07319994e94e5e446e3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-13 13:12:14.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-02-789502-3/ns5_manila : SUCCESS in 1h 50m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"6e7ad5fb6d1c2691d87ed31512fa25085f518dab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-13 19:31:29.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM, thanks carlos!","accounts_in_message":[],"_revision_number":3},{"id":"936d84fa8b6abf8a7d1ef3cc59a80127d96c53bf","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-29 17:11:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks for proposing this fix, lgtm.","accounts_in_message":[],"_revision_number":3},{"id":"f9167f46cccac705b7999f0fbfa097896a613e0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 17:12:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4cc0ae3aa1992cf568cc9b6f837cd33ef30bec74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:21:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2f5713542541ecbaf1e06b4aefb78e : SUCCESS in 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/184a629582d94a2394f8fbdf45912eb6 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebae0c4f070b49469e611e13c5b0aa8f : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cfcfdbc7a654e5c827fc1cb27afc005 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84bc88b7fb4e4f0f9b0db37c94b519a8 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/899f9c8716e14c36ba36a8567c7456ef : SUCCESS in 6m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6af82f07437d4ddd82959d978f0998a1 : SUCCESS in 28m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a8018ad7ba184d86aebe0ad3a0dd1654 : SUCCESS in 36m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7e3f763b617b49db9e49e0e86485cf71 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":3},{"id":"d5ed506fd0f5202b9cb3cf83cc90efb169811147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:21:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d68245ba9eb83d773c9e580af7ee4739340d8268","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:23:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3092c24417c2433681f59de90d47b5aa : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74a2715b95c143a382ad2ca86eb4d403 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"68e3783522d363bdc6f21b9940a12ea36fc79375","revisions":{"9160526111191f70732f6bbd91ab76ab366aad47":{"kind":"REWORK","_number":1,"created":"2021-05-04 07:45:50.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/02/789502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/02/789502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/02/789502/1"}}},"commit":{"parents":[{"commit":"faf44b54dc463c1af02c92893387f7ab2a535e32","subject":"Merge \"Updates CephFS drivers docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/faf44b54dc463c1af02c92893387f7ab2a535e32"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-05-04 07:41:58.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-05-04 07:45:23.000000000","tz":120},"subject":"handle replica state on migration complete","message":"handle replica state on migration complete\n\nSet share instance replica_state to \u0027active\u0027 after successful migration\nif the share is involved in a replication setup.\n\nChange-Id: I6cf8a08379b2f2b0fd1bcada4cd20c39adc01f47\nCloses-Bug: #1927060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9160526111191f70732f6bbd91ab76ab366aad47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9160526111191f70732f6bbd91ab76ab366aad47"}]},"branch":"refs/heads/master"},"c98140f5f1f4a302b5390a8377c4ef83dd6e817c":{"kind":"REWORK","_number":2,"created":"2021-08-12 23:03:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/02/789502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/02/789502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/02/789502/2"}}},"commit":{"parents":[{"commit":"18668d24652b653cb2b562d4708b87ba99794b85","subject":"Early validate for CIFS without security service.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/18668d24652b653cb2b562d4708b87ba99794b85"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-05-04 07:41:58.000000000","tz":120},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-08-12 23:02:20.000000000","tz":-180},"subject":"handle replica state on migration complete","message":"handle replica state on migration complete\n\nSet share instance replica_state to \u0027active\u0027 after successful migration\nif the share is involved in a replication setup.\n\nChange-Id: I6cf8a08379b2f2b0fd1bcada4cd20c39adc01f47\nCloses-Bug: #1927060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c98140f5f1f4a302b5390a8377c4ef83dd6e817c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c98140f5f1f4a302b5390a8377c4ef83dd6e817c"}]},"branch":"refs/heads/master"},"68e3783522d363bdc6f21b9940a12ea36fc79375":{"kind":"REWORK","_number":3,"created":"2021-08-12 23:06:28.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/02/789502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/02/789502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/02/789502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/02/789502/3"}}},"commit":{"parents":[{"commit":"18668d24652b653cb2b562d4708b87ba99794b85","subject":"Early validate for CIFS without security service.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/18668d24652b653cb2b562d4708b87ba99794b85"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2021-05-04 07:41:58.000000000","tz":120},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-08-12 23:06:22.000000000","tz":0},"subject":"handle replica state on migration complete","message":"handle replica state on migration complete\n\nSet share instance replica_state to \u0027active\u0027 after successful migration\nif the share is involved in a replication setup.\n\nChange-Id: I6cf8a08379b2f2b0fd1bcada4cd20c39adc01f47\nCloses-Bug: #1927060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/68e3783522d363bdc6f21b9940a12ea36fc79375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/68e3783522d363bdc6f21b9940a12ea36fc79375"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
