)]}'
{"id":"openstack%2Fmanila~793571","triplet_id":"openstack%2Fmanila~stable%2Fussuri~Id6094827341bf6ef8856cd4e7af11b36e9afb560","project":"openstack/manila","branch":"stable/ussuri","topic":"bug/1876820","hashtags":[],"change_id":"Id6094827341bf6ef8856cd4e7af11b36e9afb560","subject":"[ci][devstack] Install bridge-utils for Container driver","status":"MERGED","created":"2021-05-28 18:19:17.000000000","updated":"2021-06-01 16:18:16.000000000","submitted":"2021-06-01 16:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793571-bug/1876820","cherry_pick_of_change":725407,"cherry_pick_of_patch_set":1,"meta_rev_id":"b7237744924ced83793a3457a0d1061e6c19dfc4","_number":793571,"virtual_id_number":793571,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 16:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-06-01 16:18:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-06-01 16:18:16.000000000","post_submit":true,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-31 23:38:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-06-01 10:48:02.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 20:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 22:37:04.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-29 01:46:55.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-29 10:20:21.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-05-29 19:28:43.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-05-31 23:38:19.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"141f4dc8c188412d9bcbf6e8831dd41e1c390eeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 18:19:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d269d47ccc042960259d690e10c4addf9f46aa36","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 19:08:38.000000000","message":"Topic bug/1876820-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"a1a26374682215464640f27f3f43f756d6e84617","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 19:08:41.000000000","message":"Topic set to bug/1876820","accounts_in_message":[],"_revision_number":1},{"id":"59b788df44404bd2e74c9b870a6197365374534d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 20:02:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5334bd626bd44f19958f985d990f4192 : SUCCESS in 14m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da746fbd32174c22bbb2189c6d869065 : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cafc7e5fb5e0499faf4927a2667017ac : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b802a1126c54478960d883acf671bc0 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/305ede5ec742445f9eacc1e0489ed427 : SUCCESS in 10m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/faf67ef6f8f0471b81825ccc0a877a9b : SUCCESS in 8m 13s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/10494924cdb0419a9218b2cae7841bfa : SUCCESS in 5m 58s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/2ff81b794a4e4d36bf6f6d3be480f79f : SUCCESS in 51m 32s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/4add9033f7e44bffa9e1af3e9f5da3e9 : SUCCESS in 53m 50s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6d4124484df420f800ae6695db39df2 : SUCCESS in 1h 07m 46s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cacd418482ce45128c67a0281f1030d9 : SUCCESS in 1h 37m 32s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/ab5a6d7392e94847abc74fa6476a5b26 : FAILURE in 19m 49s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7201bf65634c4c8fa22d2994817f0fda : FAILURE in 39m 28s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/670ab14642134ba4a718508c6c9457c5 : FAILURE in 29m 38s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/e9fc94debaee4b94a061fddabe89c7cc : FAILURE in 42m 21s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a26cf9b8bf504fc390aeceaf0e374b9c : FAILURE in 44m 19s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/48d242c49a764ead958981001997926b : SUCCESS in 1h 19m 10s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c2b5cae0c7b84628a7e13c20c7fe112a : SUCCESS in 1h 01m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c69662da1f0a42fb85afe03aab7e7d57 : SUCCESS in 38m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1a8124d5a9d645c2a9072072911b7025 : SUCCESS in 30m 54s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8a448c60c3a14f1dbbee47751fd274d4 : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f6ebf1f4a016344eb9df04f5f5aaa39788a5218","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-05-28 20:52:45.000000000","message":"Patch Set 1:\n\nrecheck\n\nKnown sporadic test failure","accounts_in_message":[],"_revision_number":1},{"id":"ec8eb9f8834368c059ef37df789e60bf5c7a0541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 22:36:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3aebbf860b40c9b3edcba3d6747897 : SUCCESS in 10m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d6dfecbd67b419cb821dc765ebf0055 : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f637f5c7485640ba9401097930900ed4 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d6f457fc01a47a2ba9a9739c6414ea1 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7af0c735f8ad47668b9e474f81ceff28 : SUCCESS in 10m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7bc6a4a5fca40c49dfdb63661376e2c : SUCCESS in 8m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d5cf2046bc7e4e47a129c717c6f68214 : SUCCESS in 6m 38s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/88ea8fd3ebe34917b3ea66f3708a62d3 : SUCCESS in 36m 51s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/9e472dbf400b4329b6f3c91bde4b1616 : SUCCESS in 44m 12s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/15d3f6f167b54e4cbe60ee39574077af : FAILURE in 1h 16m 53s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/136dec08418443a0aed2ae31224164a6 : FAILURE in 1h 43m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/2f0dea01c0ed4013aa9442ca99e77302 : FAILURE in 21m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b3ca471e584486bb026f5e3b57c86fb : FAILURE in 28m 09s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cc6cdeb4c51e4570bee11fa296b4454b : FAILURE in 36m 12s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/bfb7411fe29c4cdba5660e30b485d1b3 : FAILURE in 35m 24s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/cce21bf65f824b6191344863fed66073 : SUCCESS in 44m 19s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/6716ba7d82694cbc82e3dcf588af77f3 : SUCCESS in 1h 01m 00s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/dfcce5fd973241358eac99ea4f48a1d9 : SUCCESS in 1h 01m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9bb58dd06cfb497e9b94dd0ee2001060 : SUCCESS in 34m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/be4fbc462acb4e5a9238a271b39da8e8 : SUCCESS in 31m 09s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02c66a6a6d8f49a9bec8e472735e769e : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86ae90a197fde4674824981855b78491b1229bf0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-28 22:37:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/71/793571/1/upstream-check/manila-cDOT-ss/67c8b14/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/71/793571/1/upstream-check/manila-cDOT-no-ss/803347f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f2dd72f58a71c71a26f445519dab89f4b0005b14","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-05-29 01:46:55.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-71-793571-1/ns5_manila : SUCCESS in 2h 14m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"4ee449e449a45bc1ebfe26778b5631ec824a8337","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-29 10:20:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Goutham.  Clean cherry pick; test-only code.","accounts_in_message":[],"_revision_number":1},{"id":"2d6accba0e62e1136d31670b7d168ec7e7a9407e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-05-29 19:28:43.000000000","message":"Patch Set 1: Code-Review+2\n\ncontainer job is green, thanks!","accounts_in_message":[],"_revision_number":1},{"id":"683479c378bea3c837d2fcd794cf070badf67bff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:23:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdc575292ce31bc9f7f4bb27e17b558a3b708ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:28:56.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"ce9b04d0a734a2b02df229be789de8034759dba0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-31 15:44:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ff1a3a5f71df205afcc46e563d4e9e9a08168138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 17:33:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86918e20866d4ab59cfd312b5d0d5f10 : SUCCESS in 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/392be4f44c7e4c6ba12a96b58fc8d2d5 : SUCCESS in 14m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/846c7c4dc0cd487bb3ebf88787359e96 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7b4719f4224c7fafc3757a8b03f61a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11cb693800af420bb5d66afae164b171 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f47847e175b482b89c979a4a3781151 : SUCCESS in 7m 18s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f0a3641397b4dd190ac3fe5e7cb37d2 : SUCCESS in 5m 35s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1a52e66e3e2f45a7a8b7f165412f16aa : SUCCESS in 46m 35s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/5875ab1670f04f568d8b32df35ae36e0 : SUCCESS in 51m 41s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/9fd89a39e4b841e780225b1e2db3c935 : FAILURE in 1h 14m 38s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a9377b19139d43d38e2724502c2403ee : FAILURE in 1h 45m 09s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/03022dd31f744af18430bad869647a7a : FAILURE in 24m 56s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c2506d2bec444c85b56d311453a64ae8 : FAILURE in 35m 53s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c3f3c7d9a1f14a43a9370bfe63c4db9f : FAILURE in 38m 17s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/8694f3bbdcd74ace8e1e01a6fb80cbf0 : FAILURE in 38m 45s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/a6dc71d1502642f19b7722f4654f0cdb : FAILURE in 45m 53s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/22d4be9cdcbc4d8eb495d9fed61a3d54 : SUCCESS in 1h 00m 41s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0a4bcf100fc846e8a9f44d25f8468b51 : SUCCESS in 1h 08m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cd18bd3d08934d98a5105614ffc88308 : SUCCESS in 37m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/099258ffebe24f539ad0dcb05598d7c0 : SUCCESS in 38m 52s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5bd2a7b7d24e445584f3ba3da7a21457 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7dbb4ed2cfe58d8f98648fcce3541fe2d610f66c","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-05-31 19:56:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"38487ab999d1ab30ac5552c1d095e7755b0fe736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 21:28:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f95d3c862c46a98ee42c738c311ddb : SUCCESS in 10m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/486541e1c93f43748c7dac6cd5ca021b : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/250b6a043dd14296900c21247787314e : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9ceea8f71944789693eafce42734b4 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b774094c87f42b589d4c8d174a4e7b0 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3de931f5efdc401bad61c3fe126266fa : SUCCESS in 8m 27s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8c862c73f33a4d6aa84ea09e43423b81 : SUCCESS in 5m 49s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/5b9785c8fb4842ae8a43dc34a3261ace : SUCCESS in 35m 30s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d3e4a649fe124028bf63ee7f9e00233d : SUCCESS in 51m 05s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1f2ae1fed6a142b39c60f386bff97d68 : FAILURE in 1h 06m 35s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b781e32321b46969b5b3ef068d33a67 : SUCCESS in 1h 29m 18s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/d7a28c27a38f4d9cba83eca3512ec574 : FAILURE in 20m 44s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4487b3506bee4bd698c3be5aee3bff6e : FAILURE in 30m 37s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0ce063cd5cb04edfa679828c5c633a1f : FAILURE in 32m 08s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/a579c367fdea4d1e9a57c6472d7f9a6c : FAILURE in 33m 59s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/2de6b67cd8674691916f279cc44a1436 : SUCCESS in 42m 45s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/1f730a4bc50e406a96638b5d88382127 : SUCCESS in 1h 17m 52s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3d7f1ae0c9be46e8918d00f36fb5a881 : SUCCESS in 1h 01m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3ed6328d0f044fb18db667c878116da6 : SUCCESS in 37m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4f33862a2d4d47688c6562f71019fb19 : SUCCESS in 39m 56s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/69a19d332d184c8eb7c6317ff0d1fa24 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cba6bfe9eaa95b3f8ceebb380ad8b5979fa998c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-05-31 21:28:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-ss/be370eb/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-no-ss/a2eea47/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"904fe5b5ef7e1b9b71cff07da23b9906e029cc6a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-05-31 23:38:19.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good and the container job is passing. Thanks Goutham!","accounts_in_message":[],"_revision_number":2},{"id":"caaafd7ee8f4037935b4b04a7e244d599911cc17","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-01 00:24:33.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-71-793571-2/ns5_manila : SUCCESS in 2h 7m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"890fd143bb113b373b6d15b663bbb1c2d86ce5e9","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-01 10:48:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1c953c50c02235bdf74126f67f2ab98a9e3b4ed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:48:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"29aba33060bc7cf2fb699051cc1a73612e466eb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 12:44:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/916293bb1a6c4d9aba0938b89607856a : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cffbf57147ae4f0191ee4e7b907b2a02 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b33067e4a7d4718ac8fcf394da30219 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cddfbebfae547028a3e8496e17fba03 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbeee2153f564f3a96ba240fdaa45ec5 : SUCCESS in 7m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8603a400108d4b938e29da30e1ec8798 : SUCCESS in 5m 22s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/b63daef770c84a7b8e08fdf4cf96daf7 : SUCCESS in 49m 33s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/c15e1fd77da7409182efbf70929484d6 : FAILURE in 1h 55m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3a0b4b2e3ef41eb86e9abcc66a5040cf1416bd78","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-06-01 13:01:38.000000000","message":"Patch Set 2:\n\nrecheck\n\nlvm-ipv6 job failed in two scenario jobs with ssh timeout","accounts_in_message":[],"_revision_number":2},{"id":"84c05974b746fb04bb20029aaf17526235dcb3d3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-06-01 13:02:06.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e recheck\n\u003e \n\u003e lvm-ipv6 job failed in two scenario jobs with ssh timeout\n\ntwo scenario tests :)","accounts_in_message":[],"_revision_number":2},{"id":"81541ab6f937396b3c91d8fffaccb142c8b0a841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:41:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44dad5e563bd406daba2e99b5fd0fb68 : SUCCESS in 9m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12805a1295e948e4a27e073ca7952bc6 : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae6818f4336f4b559c713b777c32a1ca : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e515311b89c24096aacf43dff960c323 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf261b67b47a4eb4a157dfdeb2a46e3e : SUCCESS in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9b0099dfa9f4654aee1cf1f8bca4ca7 : SUCCESS in 7m 30s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e4b004e3dab49e498c1563394e929d1 : SUCCESS in 6m 01s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/b4d68dbdc6654806ae71133d7dedea3d : SUCCESS in 42m 32s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/ad9763901b8a42b880de18cfc43ea815 : SUCCESS in 53m 27s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/1094e107bbd94cc4a736182b2d9c74d8 : SUCCESS in 1h 33m 08s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bca95f147ac242189e7f238322e9cc2c : FAILURE in 1h 13m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/fdf98ab19bed4d43b1d18593f59caf03 : FAILURE in 25m 57s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/678cb8bec4a84adf8b6a7183f7e7af49 : FAILURE in 38m 10s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/071e2f0437af4acda9876b3ac584e55f : FAILURE in 37m 06s (non-voting)\n- manila-tempest-dsvm-glusterfs-native https://zuul.opendev.org/t/openstack/build/15fc724896e64a9ca08fc327888faa3f : FAILURE in 40m 54s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8b22f39664954f788c1e00e8372135ba : SUCCESS in 58m 27s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/fe9f656096354559bf5e301169aca509 : SUCCESS in 1h 10m 04s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f32c901332444601978fa6e8d3b7107b : SUCCESS in 1h 12m 07s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0eca0992a9224ebd8dff13b92bacc22f : SUCCESS in 37m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1fe5497e0f024ea2aae63cbb61a40f39 : SUCCESS in 39m 04s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/edc9180c8f534ea4a776b2454714a2d9 : SUCCESS in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7b57c0359005915a46fee1a6511728f1f8db5c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:42:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7d60d76be1f3d1ddced4580169472a4eacf2098","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-01 14:43:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-ss/d7ef63e/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-no-ss/d9539cd/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc3bbd2def5dcc36e1b763744ad8fb2c76b37cce","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-01 14:44:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-ss/cbd58ea/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-no-ss/975d81b/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"94245307a38830c395b9d00df0ec49548b7b9545","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-01 14:46:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-ss/e7df608/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/71/793571/2/upstream-check/manila-cDOT-no-ss/3f28d56/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4b11327df2f2b68d417681ba5866c1f3dea643ff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-06-01 15:29:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0311960977584817ee740c6d2cecf0583551ecce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:16:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a54500863a4530b60e4d01b630f482 : SUCCESS in 11m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62a90d06b3af4de894ac8d5c0989830d : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb65b9d2e60343389502b3a29b00d87f : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f757f0376ebc4ee18919be9137c2e5a2 : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97dd0d1e4a5d4b55a235de71bbff7395 : SUCCESS in 9m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1cdb0830f2e546cb8f37cee47c6e9b36 : SUCCESS in 6m 14s\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/5d224d684cc44a3f888d53d547599318 : SUCCESS in 57m 37s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/ccab73ace0144adab7ecdc611996008e : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":2},{"id":"420624358fa6678a41a0324f16d1537803a9374a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:16:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"efe82250dd55d75c653195626ac157f09aebe2fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:18:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc579f1f2904704b8553a7fb3baca49 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a853b0cc91ccad5581cc19bc851cf8e16c04a031","revisions":{"e72d0da7f7d0042ed6d546ed41e7d390d34742ec":{"kind":"REWORK","_number":1,"created":"2021-05-28 18:19:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/71/793571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/71/793571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/71/793571/1"}}},"commit":{"parents":[{"commit":"460f44704df3303e092de3e9781224fdbbe8502d","subject":"[NetApp] Fix access rules for CIFS promoted replica","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/460f44704df3303e092de3e9781224fdbbe8502d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-04 21:48:52.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-05-28 18:19:17.000000000","tz":0},"subject":"[ci][devstack] Install bridge-utils for Container driver","message":"[ci][devstack] Install bridge-utils for Container driver\n\nDevstack no longer installs this package after [1]. The driver\nneeds to replace the use of brctl to completely remove this\npackage from our test environments.\n\n[1] https://review.opendev.org/724443/\nPartial-Bug: #1876820\n\nDepends-On: I6c7db2286d544ff8156926e07efd57628cfdc9ff\nChange-Id: Id6094827341bf6ef8856cd4e7af11b36e9afb560\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 3b509399e3c4c95f0cc6cf1b0e2fc80fe315b9ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e72d0da7f7d0042ed6d546ed41e7d390d34742ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e72d0da7f7d0042ed6d546ed41e7d390d34742ec"}]},"branch":"refs/heads/stable/ussuri"},"a853b0cc91ccad5581cc19bc851cf8e16c04a031":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-31 15:44:56.000000000","uploader":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"ref":"refs/changes/71/793571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/71/793571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/793571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/71/793571/2"}}},"commit":{"parents":[{"commit":"e48602024474be0a6d9cc0a0a37d66805f957847","subject":"Merge \"Fix CI issues and adjust test jobs\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e48602024474be0a6d9cc0a0a37d66805f957847"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2020-05-04 21:48:52.000000000","tz":-420},"committer":{"name":"Tom Barron","email":"tpb@dyncloud.net","date":"2021-05-31 15:44:56.000000000","tz":0},"subject":"[ci][devstack] Install bridge-utils for Container driver","message":"[ci][devstack] Install bridge-utils for Container driver\n\nDevstack no longer installs this package after [1]. The driver\nneeds to replace the use of brctl to completely remove this\npackage from our test environments.\n\n[1] https://review.opendev.org/724443/\nPartial-Bug: #1876820\n\nDepends-On: I6c7db2286d544ff8156926e07efd57628cfdc9ff\nChange-Id: Id6094827341bf6ef8856cd4e7af11b36e9afb560\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 3b509399e3c4c95f0cc6cf1b0e2fc80fe315b9ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a853b0cc91ccad5581cc19bc851cf8e16c04a031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a853b0cc91ccad5581cc19bc851cf8e16c04a031"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
