)]}'
{"id":"openstack%2Fmanila~798197","triplet_id":"openstack%2Fmanila~master~I3e6170f8def53aabd4bc5858fa297061e2873477","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"I3e6170f8def53aabd4bc5858fa297061e2873477","subject":"TrivialFix: Fix the filter name in config helper","status":"MERGED","created":"2021-06-26 16:44:53.000000000","updated":"2021-06-30 23:37:25.000000000","submitted":"2021-06-30 14:48:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798197","meta_rev_id":"4147f1e2c2105185d5eb3db977870b00d8ea87e4","_number":798197,"virtual_id_number":798197,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 14:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":2,"date":"2021-06-30 12:40:48.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"all":[{"value":1,"date":"2021-06-30 12:40:48.000000000","_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-26 17:10:40.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 17:30:52.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-06-26 18:04:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-26 18:43:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 18:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-26 19:45:49.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 20:57:55.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 12:40:48.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"}],"messages":[{"id":"bbc9a5685ba464a1e1d5a83aef52fd23df54013b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-06-26 16:44:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fae9c68aae0f2c196c50674a1ef6a1572a76756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-06-26 16:59:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a94cb02b0a7ac13628a262025b0a73530cba110a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-26 17:10:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/97/798197/2/check/DellEMC_PowerMAX_Manila/da98e82/DellEMC_PowerMAX_Manila/106 : FAILURE in 11m 01s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"1571a9eb7ff93cc56188026bab374961b9d33402","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-26 17:30:52.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/97/798197/2/thirdparty-check/pure-devstack-manila-tempest-aio/3b95cb6/ : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":2},{"id":"34a40ee2ab4bccf535ca73636be8e02c2d31d6e4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-26 18:04:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/97/798197/2/check/DellEMC_Unity_Manila_dhss_true/0332400/DellEMC_Unity_Manila_dhss_true/159 : FAILURE in 14m 56s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/97/798197/2/check/DellEMC_Unity_Manila_dhss_false/a22d4a4/DellEMC_Unity_Manila_dhss_false/157 : SUCCESS in 1h 04m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c486f4b3a7a832c45ab134026abafa3e83f2f262","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-26 18:43:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-97-798197-1/ns5_manila : SUCCESS in 1h 57m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5dd2224b169cbadd5dc89f21abd9f63fef30b22c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 18:49:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7e51ffeecd4c49910fd49e96df27ed : SUCCESS in 11m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f26b8235a9f48678330fe485de1a04b : SUCCESS in 20m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7dcb7ec3f7548f79af0a123db6586a2 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6dc7babfdd5483da7a8348480a0ef5b : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cee75cce6a4b47b8ab9fc4763639e926 : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/947ca81f55f945cdbea5382d7b5f3178 : SUCCESS in 10m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8a0510ba7840b7ab88fcf075515534 : SUCCESS in 10m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/55564a5a3fbf4c0fa718bb02ef04e732 : SUCCESS in 6m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70505c89179e45e2a5ab9874ef526fa4 : SUCCESS in 7m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b8b8e88807ec468db5c7eb867f80e0ed : SUCCESS in 29m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ca22ca13f69f4c0c92fb9914d1f8fc13 : SUCCESS in 40m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6abda44bce614518852a037038ca0ce6 : SUCCESS in 1h 17m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/28cbae1ad86c41d287aab5eadb3eeb3d : SUCCESS in 1h 13m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c61c696e7dee41ffbe43be9468239591 : FAILURE in 23m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/93d08c59a9c24a4b96f9b251d1daed07 : SUCCESS in 38m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8fe80eb2ef0442e6bbee572595cae72d : FAILURE in 1h 48m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d8a719135bd646db863e3c1f640fab63 : SUCCESS in 24m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/edb41912cd644aaf9180fd2a0c70b7bf : SUCCESS in 25m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9a71da89d65b49459e5c88985a424d52 : FAILURE in 58m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bbca980ff0b54169b8213b0f0451d85a : SUCCESS in 59m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/31785034c18a4664b080ac11de28bf0d : SUCCESS in 37m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b2b8197fb4074ed68fc79f7075209c8e : SUCCESS in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72c3bb9b9963b43ea434597a6edbfbfe2a189b4a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-26 19:29:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-97-798197-2/ns5_manila : SUCCESS in 1h 59m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"a2f49fc8688df4d9c1b786cde9cdcbfd8b474f34","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-06-26 19:45:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/97/798197/2/check/DellEMC_VNX_Manila/905ff3c/DellEMC_VNX_Manila/95 : FAILURE in 55m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"c4b56db0689f380aaf862f7569900b31eda0948e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-26 19:47:39.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/97/798197/2/check/DellEMC_PowerMAX_Manila/6cde833/DellEMC_PowerMAX_Manila/107 : FAILURE in 57m 39s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"3cca7377ae9b68ae8c9990a0c43a8541608969f1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-26 19:53:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/97/798197/2/check/DellEMC_Unity_Manila_dhss_true/ebbb4c8/DellEMC_Unity_Manila_dhss_true/160 : FAILURE in 1h 03m 52s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/97/798197/2/check/DellEMC_Unity_Manila_dhss_false/d3059b5/DellEMC_Unity_Manila_dhss_false/158 : SUCCESS in 59m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"819d2d69d79f935a39da1b1991351071ebb8a766","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-06-26 20:57:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/97/798197/2/upstream-check/manila-cDOT-ss/4fd3e63/ : SUCCESS in 1h 38m 29s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/97/798197/2/upstream-check/manila-cDOT-no-ss/b85e514/ : SUCCESS in 1h 42m 55s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"90bff0617a408f753b8b94a5693096e0b274d348","author":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"date":"2021-06-30 12:40:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"8dbf5ab1c7f965b3f04fc070e4f851f4f0a77b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 12:41:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39f211885cf121cd3c0be5c899e0b309bd7cea4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:48:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75888ac5f234256a36422b63b0d04cf : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e91d386382847dd90bbf8e13d3d98f5 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4ee64a08ec4b719dc0db82d4142318 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c9571b5213544eba121926a554e40dc : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03cba61dde5c476dbe761b9cdd9a4264 : SUCCESS in 11m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07e832ad72534e458f1acb6b2402fe79 : SUCCESS in 31m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/62b0d1b0dbec42638de4c00f77bd57b3 : SUCCESS in 42m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dd28ffb79cc64bae85987f678f2bc49b : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":2},{"id":"24198a246576d39983b3924898e613335dfd671f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:48:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"19ffa34715b892e4bb7d3008dcce59977b5a72c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:50:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae01e05e0094263ab469ae4810f8e34 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4147f1e2c2105185d5eb3db977870b00d8ea87e4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-30 23:37:25.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-97-798197-2/ns5_manila : SUCCESS in 2h 12m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"854b1dc958db4bd903e7b3c69c242295ad1518bd","revisions":{"3efc8b061160bfb1d50a297a89f3c9f68db2d3f4":{"kind":"REWORK","_number":1,"created":"2021-06-26 16:44:53.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/97/798197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/798197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/798197/1"}}},"commit":{"parents":[{"commit":"1261576a684b05ddc5f62db7c366bcbadc2a97a8","subject":"Merge \"PowerMax Manila - version change for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1261576a684b05ddc5f62db7c366bcbadc2a97a8"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-06-26 16:37:36.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-06-26 16:37:36.000000000","tz":-180},"subject":"Fix the filter name used by scheduler config help","message":"Fix the filter name used by scheduler config help\n\nThe config `use_scheduler_creating_share_from_snapshot` has the\nhelp message with wrong name of the scheduler filter that the user must\nenable with the option. The filter name is `CreateFromSnapshotFilter`,\ninstead of `CreateShareFromSnapshot`.\n\nTrivialFix\n\nChange-Id: I3e6170f8def53aabd4bc5858fa297061e2873477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3efc8b061160bfb1d50a297a89f3c9f68db2d3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3efc8b061160bfb1d50a297a89f3c9f68db2d3f4"}]},"branch":"refs/heads/master"},"854b1dc958db4bd903e7b3c69c242295ad1518bd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-26 16:59:14.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/97/798197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/798197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/798197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/798197/2"}}},"commit":{"parents":[{"commit":"1261576a684b05ddc5f62db7c366bcbadc2a97a8","subject":"Merge \"PowerMax Manila - version change for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1261576a684b05ddc5f62db7c366bcbadc2a97a8"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-06-26 16:37:36.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-06-26 16:47:50.000000000","tz":-180},"subject":"TrivialFix: Fix the filter name in config helper","message":"TrivialFix: Fix the filter name in config helper\n\nThe config `use_scheduler_creating_share_from_snapshot` has the\nhelp message with wrong name of the scheduler filter that the user must\nenable with the option. The filter name is `CreateFromSnapshotFilter`,\ninstead of `CreateShareFromSnapshot`.\n\nChange-Id: I3e6170f8def53aabd4bc5858fa297061e2873477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/854b1dc958db4bd903e7b3c69c242295ad1518bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/854b1dc958db4bd903e7b3c69c242295ad1518bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
