)]}'
{"id":"openstack%2Fmanila~798206","triplet_id":"openstack%2Fmanila~master~I660a0f703edf8bfd91f78ed1b0c258828a1d12a2","project":"openstack/manila","branch":"master","topic":"bug/1778975","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-15 03:58:32.000000000","reason":"Goutham Pacha Ravi replied on the change"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-02-08 08:26:47.000000000","reason":"haixin replied on the change"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2022-02-15 19:20:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I660a0f703edf8bfd91f78ed1b0c258828a1d12a2","subject":"Fix quota-set no required parameters on update","status":"MERGED","created":"2021-06-26 19:29:11.000000000","updated":"2022-02-15 19:22:40.000000000","submitted":"2022-02-15 19:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"798206-bug/1778975","meta_rev_id":"ba00222688fb34991a590e8d31f5187409f2e2fe","_number":798206,"virtual_id_number":798206,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 19:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-02-15 17:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-02-15 03:58:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-02-15 17:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-26 20:04:44.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-06-26 20:49:17.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 21:00:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-06-26 21:01:44.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-26 21:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-02 05:01:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-08-29 18:10:05.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-09-06 01:10:37.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-01-20 17:53:26.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-20 18:42:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-21 02:01:39.000000000","updated_by":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"reviewer":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"state":"REVIEWER"},{"updated":"2022-02-10 04:15:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-10 14:02:16.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-15 17:56:18.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"dbbbe5b25ed7cbf81de0ee56c114ef7b07e39ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-06-26 19:29:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f03c899876583a9e85eea0d70e0eab59e55828ae","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-06-26 20:04:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/1/thirdparty-check/pure-devstack-manila-tempest-aio/405810b/ : FAILURE in 34m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3e846efaedd3d6ae9a3a6f4e3e2dc4297e99d647","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-06-26 20:49:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/06/798206/1/check/DellEMC_PowerMAX_Manila/7fb0874/DellEMC_PowerMAX_Manila/108 : FAILURE in 1h 01m 32s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e8768332ee935ebe7e39024911c931eb9eaa38a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-06-26 21:00:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/06/798206/1/check/DellEMC_Unity_Manila_dhss_true/5da286f/DellEMC_Unity_Manila_dhss_true/161 : FAILURE in 1h 01m 40s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/06/798206/1/check/DellEMC_Unity_Manila_dhss_false/0a7587c/DellEMC_Unity_Manila_dhss_false/159 : FAILURE in 1h 11m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"29407356299de598e4538551d2962e757879e6b6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-26 21:01:44.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-1/ns5_manila : FAILURE in 1h 31m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a7373cf5e514952c04fc1eba7b5719639684e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-26 21:06:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6754e5004f2420ebeab8f5e9d7fc2e6 : SUCCESS in 11m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dba1b21741f74054a1c23b6d750f5678 : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd26c197e0604731914ecee8043d95a6 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3fa36930778423a8f8ecb9668a65c46 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/719adecbe3524051b9c693e367a87868 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1049cb1d078d4d3a9d58fcbc51eca27f : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f4a1ee4cc124804989bca9305508b15 : SUCCESS in 11m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab8312263d7c40b19bb652b4404dee9b : SUCCESS in 8m 22s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9f84c6942564cab876c75ba21ba9d93 : SUCCESS in 8m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/74c93ae3611742a1934b58f5b4ffad62 : SUCCESS in 8m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7f88ce786a9c4a719fcd00987f79eda6 : FAILURE in 26m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3ee5f7f68ed450b860817a98ed425fc : FAILURE in 33m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ec45a5c28ff548709ed548bf6a2ff705 : SUCCESS in 1h 09m 49s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ade32addec1e4aa9903b08caec5e056c : FAILURE in 1h 13m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59a9908fdf7d4ab59c47e4d2fa8f0b2e : FAILURE in 23m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4e8e2464412b42d58d575197ee78107e : SUCCESS in 46m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/55078b47a3b445e5b8ecf15c661b2cf2 : FAILURE in 1h 35m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/dfe92d89d9d645c982452d94a21e714d : FAILURE in 25m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/950e43213bf248d28e9d4ad394f14056 : FAILURE in 24m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ac762767f0684e38b99c9d6db15e46f2 : SUCCESS in 54m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/263d819814da4faaa773d936fca53d22 : FAILURE in 1h 03m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b57aef678824357a7c10bb0bf899223 : SUCCESS in 42m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/039de9cfb9c04f69800c33d7dfeba1b1 : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2ec98e086b03022685012b7c8bc360a3848dc18","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-06-30 22:39:32.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-1/ns5_manila : FAILURE in 1h 57m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e910dbe25d694bf5860ae9d5d2634de12689582","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-02 05:01:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7a40db65a533aab0cb03cfcadccfe5deed2c0b2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-02 05:32:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/1/thirdparty-check/pure-devstack-manila-tempest-aio/b41e1e7/ : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"822870abc5fc942fb9df48416583f4bd410fa720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 09:54:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e982f7386eda4e17a70d7ad641c7968e : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08b1b187ab874f4ca067a694a7adfb3c : FAILURE in 20m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dda05cd75a741b9a238b662888807d8 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2a1e38958848ef8e8f0b5fed355a59 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8acea37c7ea64857a00308d37321bacb : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c65e2d62f0e2448abea3f38a526a4510 : SUCCESS in 12m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f28f4751f6e4156a830f9b15fc33442 : SUCCESS in 12m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ce32b1dcb046bb84c4415cb4a6cc1b : SUCCESS in 6m 14s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b750172459304d37a5f149e234175588 : SUCCESS in 7m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73f5bae0bcd64e8daab7f90468c7f38a : SUCCESS in 5m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/873bf02dd620419b91392d652c3a20c4 : FAILURE in 25m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/15c2ba725dbc498f9239ca76197e9772 : FAILURE in 38m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/38d0016a6fe9434da497bb10c77a51c5 : SUCCESS in 1h 02m 35s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6b40b683b60d4e9c9b1bd9c43a2569c8 : FAILURE in 1h 09m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c477d3009fcf4594a8c4d8929ebef0bb : FAILURE in 1h 33m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f36a42e2ad25475cbb1bb645c1693bfd : SUCCESS in 44m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/03c143b0b7694387bc76c34333f608f6 : FAILURE in 1h 27m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8c32099061db44c99c422ed16c296998 : FAILURE in 29m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b9d1672cca724768a6e37cd9151c2e16 : FAILURE in 25m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2cfaae390e434d1b832d1f4014033693 : SUCCESS in 1h 11m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/73cae022d99f4bd7ae8283b887c3fed3 : FAILURE in 58m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d05164acb3bf4d2b8c9815a0d7bca592 : SUCCESS in 36m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/80c90125e71d48a6bf83dd423eb73099 : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"954b5735c2462caae4b4ce18dddcfd62dc3c13e2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-29 18:10:05.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d2162b7b8988a9fdd0683cb53d45a1a09ed7ba40","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-29 19:05:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/1/thirdparty-check/pure-devstack-manila-tempest-aio/1e29f5a/ : FAILURE in 54m 39s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/06/798206/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6ef9792/ : FAILURE in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f0a975c5c6ba08e2950d8d8c4290b3f50f73523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 20:00:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97aeed4e8b7941b3854fc8294c1f2fff : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bed2cc036c11401198c324d43c0e1e35 : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/318afb570db741e4a0a3e55e3066cebf : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae3a1a8b1d8471e9bee1175004683ee : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13087864bfaa4afe839c5c442e7f6d68 : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f934eda4a78a4e0695ef948c3de31976 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e03e569590a497c803d81b86d60b9f1 : SUCCESS in 9m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b6316d186742c296272e54691ff29c : SUCCESS in 6m 06s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1cc200af07ce4572b15bacc1f6c39fd8 : SUCCESS in 5m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f959d8d573ca40fba096f6dbcfa95b27 : SUCCESS in 5m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8908bdc3f8fb4ad99029cdf1eca2dd84 : FAILURE in 36m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5982746a80f145198dc021b103b9fd37 : FAILURE in 38m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ddd0b8bc11b847eb81f9821df8bdf6c3 : SUCCESS in 1h 07m 49s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/30d453b0995b4cef917baf9357f4038a : FAILURE in 1h 17m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d43de1dee0b140cb967d3925c98eeef2 : FAILURE in 1h 44m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8eb96d85d399480f8c898ff9e407a32c : SUCCESS in 49m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d792954e7da249c7b518a5569102a184 : FAILURE in 1h 28m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2368e6289fc94112a954fa5c58676ed4 : FAILURE in 36m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a6c0c87b78f64ccd9a85f1ad7f17a061 : FAILURE in 32m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/32275564cfe848c29f739a9f474c98da : SUCCESS in 1h 06m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/431e5b0ce85d4c7f9513608cc72a4ad5 : FAILURE in 1h 03m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/87074a8ea60343e8920a33a31dba2564 : SUCCESS in 37m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/670ae9db664546368c5d80b91f7ba096 : SUCCESS in 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4039d6d3e181dd6658de91ad4e2995e654d5aa28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-09-03 00:08:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f50a63df2af54521ea47d8243dd16c62d5016a21","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-09-03 00:55:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/2/thirdparty-check/pure-devstack-manila-tempest-aio/2ba7c31/ : FAILURE in 45m 53s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/06/798206/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/62e755b/ : FAILURE in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"971fdfbe33504ff440a9939d14fd8091280ffdbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 01:52:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82df863f96dd4eedbefdb1673aa69bb7 : SUCCESS in 10m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93b19f5ee7ab43a1860d2eff2932a89f : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e11fc51f649f4744ba07540b6522dd09 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337e20cf4cfc453398b05dbdab27e044 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d866abd606f45debfccdefd9f5e2df3 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f171d439fb7749c58a706e63014819d0 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22496dcfbc87405999a511213ee4b729 : SUCCESS in 10m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63c8e35b985745519ab8b13fd07784f9 : SUCCESS in 6m 03s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4897933a26bf4a4dbb7ded2e8a95fdc3 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ec20d9930da491d8971b19711a16048 : SUCCESS in 5m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/577e4c2dcaa146ecb78982587e1ad253 : FAILURE in 36m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af880001494741a19880ae30b98ea450 : FAILURE in 45m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51063aa3d24640d2bebc2e553499b7b7 : SUCCESS in 1h 01m 19s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fad8732bf75b41129fe3b115f70de9a5 : FAILURE in 1h 27m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2ad0503960134f91a1d742781dfc51bb : FAILURE in 1h 30m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/491d977c20cc4a439ca43ff04f762d24 : SUCCESS in 49m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87a5874041ab470f9b60c81623a5700a : FAILURE in 1h 38m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/04f1ddda0950470cb2f688c7d6991551 : FAILURE in 31m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1d0b98f771a049b69d42fe4581079bd5 : FAILURE in 26m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3d9944aea54f40d1b7552a34199a959b : FAILURE in 1h 11m 50s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7726e37afa0441dfa65e399283ebc321 : FAILURE in 1h 28m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/eb45d519ffc44aafb4630bc01eec1a6d : SUCCESS in 40m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e18ae0357a2247589867c5e67ae2f7f1 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58c7b84b0c16020127e3a81b057c3db26c0d17e5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-09-03 06:30:53.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/06/798206/2/check/DellEMC_PowerMAX_Manila/c657799/DellEMC_PowerMAX_Manila/247 : FAILURE in 1h 53m 45s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"fb8839b0ba0a164b77cd428829a92bef45d39974","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-05 01:28:36.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-2/ns5_manila : FAILURE in 1h 52m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"aa2b6093e000ff78dacc74565bf6c7db1eb7d28c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-06 01:10:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nhi, Felipe,thanks for this.\nplease see comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"8a4cce175f2a3dfe500b072b02cdcbf20d8a6112","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-07 18:34:46.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-2/ns5_manila : FAILURE in 1h 41m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"82e1cb00b3936b8b6c041a3ebb4acde9d657267b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-01-20 12:23:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2362bde80cd98cea52bbf56d061d7096db42e5d3","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-01-20 13:13:50.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"13bfc64a36fd088ad74d314f657e6ef743930604","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-20 13:24:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/3/thirdparty-check/pure-devstack-manila-tempest-aio/a9184a7/ : FAILURE in 1h 00m 15s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/06/798206/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/903be6d/ : FAILURE in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"927fc753576f705885991b47adad1e35dff9eb9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-01-20 14:00:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d6b939a3533a07aa28bf8ef2de8c33433a32134","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 14:29:12.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-3/ns5_manila : FAILURE in 1h 48m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"f0e2561a4f1166c70522607b06161a63cd69482f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-20 14:57:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/4/thirdparty-check/pure-devstack-manila-tempest-aio/3673600/ : FAILURE in 55m 57s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/06/798206/4/thirdparty-check/pure-devstack-manila-tempest-fa-aio/2715195/ : FAILURE in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f72fca54fdee0ae919c3535ff4994b9b277f0eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 16:02:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa8408a7462d4cd1bab45ffcf375d47f : SUCCESS in 10m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1635b3a99748429cebaa982f918691 : SUCCESS in 31m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/394bac2f51b649f38bd4a62eccc38529 : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6489d8d065fa4399b9ec38c6913f335c : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d38e74f6434ae995e9500b28934438 : SUCCESS in 27m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0de515ec475d44b4abf5cd5df78aeb65 : SUCCESS in 26m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4f08fcb0054ceeb89fe8fcbab51936 : SUCCESS in 6m 01s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b10d0cbefc75431fa31190e68ae97ec3 : SUCCESS in 4m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c5adb25e34148198776076f83c52bc9 : SUCCESS in 5m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/05de22dff2be431b81d0ae1755a2bcf5 : SUCCESS in 30m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2050aad0689743d0b9eafce40b496790 : SUCCESS in 38m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3b296c2231a8452f86fcfb3c5d8c0592 : SUCCESS in 1h 17m 27s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b02706f7c92e4277b662b9e4a4427474 : SUCCESS in 1h 16m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8adbf5c8b5f44a51b76ea91d4a9c0145 : FAILURE in 1h 53m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1cde987f117349fa9b4fda63aebc19aa : SUCCESS in 52m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f035dd7e8aaf41b5bfc4bb7b88ff4db4 : FAILURE in 1h 42m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7416a1ef23a2403f988f07538703d025 : SUCCESS in 39m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5505d4411c8647238071ee113ec65df4 : SUCCESS in 25m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db54226e746d4cb681bdbf7c0828239e : SUCCESS in 1h 11m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/70f2a00579f0445a9697e17c10cfdc67 : SUCCESS in 1h 04m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cb446f4af3154d75b33b8687bf61c06d : SUCCESS in 38m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8f8340dbd56e45e793696728c06ca05f : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"311f10be547c3e9bafd61d956c3495a53885b78e","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-01-20 16:48:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/06/798206/4/check/DellEMC_PowerMAX_Manila/dc8ca86/DellEMC_PowerMAX_Manila/81 : FAILURE in 45m 14s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"cc9eabd5bd13e0b198c45a38fb4a8cdd3b9adde1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-20 16:58:47.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/798206/4/check/DellEMC_Unity_Manila_dhss_true/db772b2/DellEMC_Unity_Manila_dhss_true/256 : SUCCESS in 2h 57m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/798206/4/check/DellEMC_Unity_Manila_dhss_false/fe9c230/DellEMC_Unity_Manila_dhss_false/252 : SUCCESS in 1h 29m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"e4b8e6b521f43f2b28f8293428b4cb0ba122d123","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-20 17:53:26.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/06/798206/4/check/DellEMC_VNX_Manila/1c8c418/DellEMC_VNX_Manila/122 : SUCCESS in 1h 50m 21s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":4},{"id":"e52a68b9dd891810fe79708431c581ee237bd629","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-20 18:42:41.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/06/798206/4/upstream-check/manila-cDOT-ss/473acea/ : FAILURE in 2h 36m 40s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/06/798206/4/upstream-check/manila-cDOT-no-ss/6dc09c7/ : FAILURE in 2h 34m 09s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"4eed4f694e4f01312e068a98f63327e846f51877","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-20 21:48:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-4/ns5_manila : SUCCESS in 2h 19m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"feeb59b8570c95a2808e8f7ab1c6699ef6d4a46e","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2022-01-21 02:01:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"06a9cc71fe9671d2df5022dab9ef54ef813eab71","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-01-27 11:39:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74c065f627d19505491bd02658ee56f363859699","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-08 08:26:47.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"605369e3fe9bf3fb9ed695b0842b84ad41554125","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-10 04:15:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f8e818acb155cd3513934a329ab5991bfc0c47bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-02-10 13:20:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31b69667e362250649cb4be11d93bbde90c7e9ee","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-02-10 13:21:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"83592b395663cd57740dbe07af85a15b644e52a7","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-10 14:02:16.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/06/798206/5/check/dsvm-tempest-manila-zadara-driver/12019fc : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":5},{"id":"ab5e2b75bb217d45b7768dcded417d1c13239673","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-10 14:13:21.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/06/798206/5/thirdparty-check/pure-devstack-manila-tempest-aio/5279312/ : FAILURE in 51m 16s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/06/798206/5/thirdparty-check/pure-devstack-manila-tempest-fa-aio/cdf2115/ : FAILURE in 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"125ec176e91e6351a6b60ac938fc53c9b89ec097","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-10 15:08:58.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-06-798206-5/ns5_manila : SUCCESS in 1h 47m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"793f0ef8e625a4f338c6d111aaac5778f925d1ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:42:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854c5e5f5d1e4efe82266bade4d40b1d : SUCCESS in 10m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df42dbbaf79e4350b825165cae9d8510 : SUCCESS in 28m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/529b6da298c045ddaaf4dd4d4613e033 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a17b7cad4264dae8e8823502d68176f : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a22bfe55e1545a0b0b51f83c6410916 : SUCCESS in 16m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1dd703edc824d569568b05b040a688d : SUCCESS in 17m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c8c855f50cb4ad6a9cd81f64bd04b5d : SUCCESS in 5m 35s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7625ce48e4534c68b941f54e7923be5c : SUCCESS in 5m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/13a194b07aa1441c953bf56e71126f6d : SUCCESS in 6m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/87c6b4648dff4bb6a849ad8b7328d559 : SUCCESS in 27m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4b7567954d5d45ec84de5989bc9ccd59 : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/422d22961e304b159e0512968ccf4d8a : SUCCESS in 1h 05m 04s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7db838a200674c958f2da787646653bc : SUCCESS in 1h 17m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/20a9a5fe8a4543a9ba8e2a7278d5f599 : FAILURE in 1h 23m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5d4bdfe433224fec867030d4715b1158 : FAILURE in 45m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cad976afb9b8478da8d1038da7690251 : FAILURE in 2h 18m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d71303dd15eb4108b5af46a177edfeac : SUCCESS in 29m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3997488032624877b68d3bad01abbb6a : SUCCESS in 27m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b05f17ab7f146799e8e2019f36c25d3 : FAILURE in 1h 19m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c3a2045bd9ab4e06b9ca19c014017d0c : SUCCESS in 1h 00m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b11d94ccc1274b3f96d317e0545b6003 : SUCCESS in 44m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fc6ec1f0790b4217ac141b373e098e36 : SUCCESS in 41m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"158aecbfbee6fe5f1482db6cb9b44b99fd0ab399","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-10 17:37:58.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/798206/5/upstream-check/manila-cDOT-ss/e18c023/ : SUCCESS in 1h 41m 42s\n- manila-cDOT-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/798206/5/upstream-check/manila-cDOT-no-ss/f4de627/ : SUCCESS in 1h 54m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"ca12fffb52d0cbc8656642b84152f66465d0d524","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-02-10 18:01:45.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/06/798206/5/check/DellEMC_PowerMAX_Manila/e132dd3/DellEMC_PowerMAX_Manila/131 : FAILURE in 1h 13m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"df63cde23eb4ae60a64dd8b5041a4f5305e2a1e0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-15 03:58:32.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c48dc6669dea698d70610fe0c0bddffb723c81a1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-15 17:56:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe011e8903d63e5f3c41c385d35c3fc0d4a418ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 17:57:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7a9704be09da27a91ee39fe7f64ca8cbdda2de00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:20:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68db7c7933e349a8b59fb08d5ff75c04 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48ba4dfecf7243f5a917ee3d39eb01bb : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a23a67750064da3984dda15566f17df : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9861faaa31da46c798f9d6d53087270e : SUCCESS in 16m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7593059c8fb94983b16a8df822432f30 : SUCCESS in 26m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d15d81d513047468e601754082f3ee4 : SUCCESS in 5m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4c5e75a3267848ecb12c9a21a06e7ba3 : SUCCESS in 28m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d6d778933ebb45eeb5c87396e64292ea : SUCCESS in 33m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/813895e13c46445a8e7078777669e3aa : SUCCESS in 1h 19m 21s","accounts_in_message":[],"_revision_number":5},{"id":"764390f31a37dbc7ce1b2850080ec9e6ceafee2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:20:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ba00222688fb34991a590e8d31f5187409f2e2fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:22:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499186f642124541961f5a3c348489fd : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddf0e04f6bde4080bfcd6e49c199ef86 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c07f2d8ba8c97294858af1fb6388135de17b509d","revisions":{"933a048d593692cd9bdd1e568f514f7d8bedd6ce":{"kind":"REWORK","_number":1,"created":"2021-06-26 19:29:11.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/06/798206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/798206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/798206/1"}}},"commit":{"parents":[{"commit":"7b521675efd8e0be76ca6a75c30f6c46ff9c9a10","subject":"Merge \"Direct mgr commands to the mgr daemon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b521675efd8e0be76ca6a75c30f6c46ff9c9a10"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-04-01 19:50:58.000000000","tz":0},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-04-23 17:54:38.000000000","tz":-180},"subject":"Fix quota-set no required parameters on update","message":"Fix quota-set no required parameters on update\n\nFixes an issue on manila api while trying to update a quota\nwithout specifying a value.\n\nChange-Id: I660a0f703edf8bfd91f78ed1b0c258828a1d12a2\nCloses-Bug: #1778975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/933a048d593692cd9bdd1e568f514f7d8bedd6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/933a048d593692cd9bdd1e568f514f7d8bedd6ce"}]},"branch":"refs/heads/master"},"2097c7c37902f90efd4b2cdfac1d7d7f8dd37640":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-03 00:08:11.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/06/798206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/798206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/798206/2"}}},"commit":{"parents":[{"commit":"d955928947a70fa1f7c50f158ba2d35a69a83762","subject":"Merge \"[Optimise] Use ThredGroup to manage periodic tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d955928947a70fa1f7c50f158ba2d35a69a83762"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-04-01 19:50:58.000000000","tz":0},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-09-03 00:08:11.000000000","tz":0},"subject":"Fix quota-set no required parameters on update","message":"Fix quota-set no required parameters on update\n\nFixes an issue on manila api while trying to update a quota\nwithout specifying a value.\n\nChange-Id: I660a0f703edf8bfd91f78ed1b0c258828a1d12a2\nCloses-Bug: #1778975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2097c7c37902f90efd4b2cdfac1d7d7f8dd37640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2097c7c37902f90efd4b2cdfac1d7d7f8dd37640"}]},"branch":"refs/heads/master","description":"Rebase"},"68cf677aad7986820b6d5f95231d7c987a9e510e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-20 12:23:36.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/06/798206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/798206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/798206/3"}}},"commit":{"parents":[{"commit":"d1509256dac8d2e66d873e4b9d38e4e06f344b59","subject":"Merge \"OnlyHostFilter allows user to specify host during share create.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d1509256dac8d2e66d873e4b9d38e4e06f344b59"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-04-01 19:50:58.000000000","tz":0},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-01-20 12:23:36.000000000","tz":0},"subject":"Fix quota-set no required parameters on update","message":"Fix quota-set no required parameters on update\n\nFixes an issue on manila api while trying to update a quota\nwithout specifying a value.\n\nChange-Id: I660a0f703edf8bfd91f78ed1b0c258828a1d12a2\nCloses-Bug: #1778975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/68cf677aad7986820b6d5f95231d7c987a9e510e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/68cf677aad7986820b6d5f95231d7c987a9e510e"}]},"branch":"refs/heads/master","description":"Rebase"},"1f3c7dc7a9a2ad7015dea9cf812c9ba3ac3f7ee1":{"kind":"REWORK","_number":4,"created":"2022-01-20 14:00:33.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/06/798206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/798206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/798206/4"}}},"commit":{"parents":[{"commit":"d1509256dac8d2e66d873e4b9d38e4e06f344b59","subject":"Merge \"OnlyHostFilter allows user to specify host during share create.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d1509256dac8d2e66d873e4b9d38e4e06f344b59"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-04-01 19:50:58.000000000","tz":0},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-01-20 13:59:40.000000000","tz":-180},"subject":"Fix quota-set no required parameters on update","message":"Fix quota-set no required parameters on update\n\nFixes an issue on manila api while trying to update a quota\nwithout specifying a value.\n\nChange-Id: I660a0f703edf8bfd91f78ed1b0c258828a1d12a2\nCloses-Bug: #1778975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f3c7dc7a9a2ad7015dea9cf812c9ba3ac3f7ee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f3c7dc7a9a2ad7015dea9cf812c9ba3ac3f7ee1"}]},"branch":"refs/heads/master"},"c07f2d8ba8c97294858af1fb6388135de17b509d":{"kind":"REWORK","_number":5,"created":"2022-02-10 13:20:55.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/06/798206/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/798206/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/798206/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/798206/5"}}},"commit":{"parents":[{"commit":"d1509256dac8d2e66d873e4b9d38e4e06f344b59","subject":"Merge \"OnlyHostFilter allows user to specify host during share create.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d1509256dac8d2e66d873e4b9d38e4e06f344b59"}]}],"author":{"name":"Felipe Rodrigues","email":"felipe_futty@hotmail.com","date":"2020-04-01 19:50:58.000000000","tz":0},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-02-10 13:20:17.000000000","tz":-180},"subject":"Fix quota-set no required parameters on update","message":"Fix quota-set no required parameters on update\n\nFixes an issue on manila api while trying to update a quota\nwithout specifying a value.\n\nChange-Id: I660a0f703edf8bfd91f78ed1b0c258828a1d12a2\nCloses-Bug: #1778975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c07f2d8ba8c97294858af1fb6388135de17b509d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c07f2d8ba8c97294858af1fb6388135de17b509d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
