)]}'
{"id":"openstack%2Fmanila~799982","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~I2d47dbb1251dfa8d529f813724fb7fd97d7d3a0b","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/1921927","hashtags":[],"change_id":"I2d47dbb1251dfa8d529f813724fb7fd97d7d3a0b","subject":"Handle service client authorization errors","status":"MERGED","created":"2021-07-08 15:15:52.000000000","updated":"2021-08-13 12:46:54.000000000","submitted":"2021-08-13 12:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799982-bug/1921927","cherry_pick_of_change":783904,"cherry_pick_of_patch_set":2,"meta_rev_id":"86128336f6ae5c9727931563dc41031e5ff65155","_number":799982,"virtual_id_number":799982,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-13 12:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-08-02 11:27:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-08-13 11:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-08-13 08:42:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-02 14:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":1,"date":"2021-08-13 11:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-08-13 11:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 15:51:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-08 17:23:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-08 17:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 19:23:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-27 20:00:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-07-27 20:00:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-08-02 14:45:11.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-08-13 08:42:41.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-08-13 11:55:12.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"6f0e47cd4761a80aa699195b8ffde23f0ffbbcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-08 15:15:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5101e15cc2de23b3cb8efcdfd152d2bd09ea9515","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-08 15:16:17.000000000","message":"Topic bug/1921927-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"e90987bd339428094bd4af7783d95b179672f593","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-08 15:16:20.000000000","message":"Topic set to bug/1921927","accounts_in_message":[],"_revision_number":1},{"id":"89f72d7b571be3735cf026c4ffbff6193a973b65","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-08 15:51:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/82/799982/1/thirdparty-check/pure-devstack-manila-tempest-aio/da17064/ : FAILURE in 34m 45s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/82/799982/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/f616032/ : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d1fef343d740c7aceff3e2312a6f967022d9a268","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-08 17:23:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-82-799982-1/ns5_manila : SUCCESS in 2h 6m 30s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e417bb7016767ef41b52b5194c55798a447fb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 17:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6ceebdce384bfe9c15aefba579d920 : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bd00cb4cbb94b47831611419c5840d6 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21426d79d1d14dc5bc3391d54bdc5a48 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f9ec2a982e43cb9490107b4681aa93 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9c592aadca84ced89a1e15a8620577d : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a045bbecdb9241899be1eea30c8e5ecb : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c03933a31c6848adaeaa59a2f06621e0 : SUCCESS in 9m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77c27c72192245fe8f9ea94cfec96d50 : SUCCESS in 6m 08s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b8fc4e1f6dc479f8e4a50c2c0fc5f2c : SUCCESS in 6m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54bc57b0504c4835aa93f97637aaab3f : SUCCESS in 6m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f3360770fdbb485cb55ee43ca14bf0c9 : SUCCESS in 30m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a5c6e527e1454f93ade0c60d3ee9e9f8 : SUCCESS in 39m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa195075a68b45e0979e9bc3be06ce8d : SUCCESS in 1h 44m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7fcf660bab874649a67b450dc95f589d : SUCCESS in 1h 11m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2d9c2c919cf840e2897f5f172e0150db : POST_FAILURE in 1h 34m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9c44e4246fb540d7acf1814dbc0e7526 : SUCCESS in 58m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5432451768c246cfa61850c1dca559a4 : FAILURE in 2h 09m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3eb72c4119b14086b97a0c53bec2305d : SUCCESS in 40m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7558aad23a684b56a74595373e579d17 : SUCCESS in 23m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9232f7a390e423ead9b1057125c4a19 : SUCCESS in 1h 03m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c44966363cf4452cb455ab8a8e98730a : SUCCESS in 1h 05m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/56f7321f9a6f46c493d6e9e49158f5d1 : SUCCESS in 40m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dcc7281d1dd546cd95c8a31e7665d043 : SUCCESS in 42m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b628906d3fe782e13f500b200b3230ee3cb52931","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-08 19:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/82/799982/1/upstream-check/manila-cDOT-ss/5109744/ : SUCCESS in 1h 46m 16s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/82/799982/1/upstream-check/manila-cDOT-no-ss/e4e03fb/ : SUCCESS in 1h 51m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bde1458becfec6d7456877e206b8547f60c38c37","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-02 11:27:19.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry-pick. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"cbf04e7433cee44237125bcaaf713effa20452a9","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2021-08-02 14:45:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"35aeaba468291d88666d35b5336394a25447589e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-08-13 08:42:41.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b74875592229c308db7deb251a263001bf460962","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-13 11:41:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean cherry pick, LGTM, thanks","accounts_in_message":[],"_revision_number":1},{"id":"b6c9106d0164bc364aafd8a30eba4a69df9bba14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 11:41:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8de8ebafbbf13d27862597f6e716b70fb587e24a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-13 11:55:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2da939de9832e41d156815241ae86da6f907a94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 12:45:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81e235a06964baaa39b53cc2502788a : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc4832ae97ae4e25b455ef771ced3e27 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269dffca2b6342cfa84f3b212ad4a7ad : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9483246ac38448bfba2842264cfe5564 : SUCCESS in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de18086033024bd68c1788acd5476882 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cc7609dda354d71994141e0d4f495a4 : SUCCESS in 6m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d7527a095ddf461d80ceb03279a8c9f8 : SUCCESS in 30m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fb08b968053f4d64b6c49d21f92893e5 : SUCCESS in 35m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d392de074f864b9f9c2058f1f15dfd1a : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6aa63d2a7deffbde279eb828a61ab593caaebdf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 12:45:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86128336f6ae5c9727931563dc41031e5ff65155","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 12:46:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a732dd3701ee4c4ab5265cc34b3a9ce7 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45baa318f94d4bcfbd45ed696fe99795 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"113dc2878a0d72c22822bfeee933a6511c3e2dbf","revisions":{"113dc2878a0d72c22822bfeee933a6511c3e2dbf":{"kind":"REWORK","_number":1,"created":"2021-07-08 15:15:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/82/799982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/799982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/799982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/799982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/799982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/799982/1"}}},"commit":{"parents":[{"commit":"59a0c147df9e9255b3199fbb6846311f0fe8cef1","subject":"Filter shares by share type \"extra_specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/59a0c147df9e9255b3199fbb6846311f0fe8cef1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-03-30 16:13:08.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-07-08 15:15:52.000000000","tz":0},"subject":"Handle service client authorization errors","message":"Handle service client authorization errors\n\nAuthorization errors when connecting\nto nova/glance/neutron/cinder can be raised as\nBadConfigurationExceptions in the logs. We shouldn\u0027t\nbe raising exceptions derived from the service client\nlibraries.\n\nChange-Id: I2d47dbb1251dfa8d529f813724fb7fd97d7d3a0b\nCloses-Bug: #1921927\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 04e06968fd7990adac9eb92a885af557a8acc772)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/113dc2878a0d72c22822bfeee933a6511c3e2dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/113dc2878a0d72c22822bfeee933a6511c3e2dbf"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
