)]}'
{"id":"openstack%2Fmanila~800552","triplet_id":"openstack%2Fmanila~master~I21633d4ffe14983365b0b6239216ed5d0bbfaace","project":"openstack/manila","branch":"master","topic":"bug/1934889","hashtags":[],"change_id":"I21633d4ffe14983365b0b6239216ed5d0bbfaace","subject":"[NetApp] Fixed scoped account replica delete","status":"MERGED","created":"2021-07-12 19:45:17.000000000","updated":"2021-08-29 18:42:06.000000000","submitted":"2021-08-29 18:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800552-bug/1934889","meta_rev_id":"b2f50af98fb5b45db9427f1af892b5cb20fc57e8","_number":800552,"virtual_id_number":800552,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-29 18:38:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-08-28 21:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-08-29 17:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-08-28 00:19:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-08-29 17:34:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 19:45:17.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2021-07-12 19:52:55.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-07-12 20:35:48.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-12 21:15:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-12 21:36:30.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-12 22:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 22:29:50.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-12 22:38:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-07-13 01:46:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-27 23:54:42.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-28 21:31:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"7fc360cad5a04700992f9d9d973bd4af8ec6613d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-07-12 19:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93f24f5b4991b27892a4132e289c87fadca82284","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-12 20:35:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/52/800552/1/thirdparty-check/pure-devstack-manila-tempest-aio/b4d6bb2/ : FAILURE in 49m 31s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/52/800552/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/4885b56/ : FAILURE in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0bd1a76373a3b65c5594015ab242a1b96d4b7253","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-12 21:15:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/52/800552/1/check/DellEMC_Unity_Manila_dhss_true/b619b56/DellEMC_Unity_Manila_dhss_true/178 : FAILURE in 1h 29m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/52/800552/1/check/DellEMC_Unity_Manila_dhss_false/afa1e51/DellEMC_Unity_Manila_dhss_false/175 : SUCCESS in 1h 04m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"249ca3f43896d273c983548c583a897d27cdd487","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-12 21:36:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/52/800552/1/check/DellEMC_PowerMAX_Manila/aa5f1a6/DellEMC_PowerMAX_Manila/126 : FAILURE in 1h 50m 52s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"b99eea5c3c80b54e5a1433bef221a32582c508a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 22:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e4c2f70e95343c6be9edc4dd57d0973 : SUCCESS in 12m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad35314941bb4273b223e920fde7a0b3 : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b1d02d94d3a468baf047558efb1438e : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a426cae941394218b07a62e8bdc399ac : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b23a5a3c8b56475aa2fa23bb848c9508 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24a98e03f6504bd9b0ace0903fe377b0 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cc5645435cf485093d7dc4bbe4ffcd3 : SUCCESS in 9m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae176ca2712c4a0aab2508fa39a4c6a1 : SUCCESS in 6m 25s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/efb0b875ffd74c7b9437b97d34828fc2 : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f78de3a62564dfbb37512dd21b71477 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/96a30727982843b894c5a0bb9868930e : SUCCESS in 34m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cca5ff61b3924a48b0017e0296c78c35 : SUCCESS in 33m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7acb018042e1446b9a35ee4f38e10b07 : SUCCESS in 56m 57s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/da7b0e41431e43ca9e704549ec282e20 : SUCCESS in 1h 19m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dd5f1f8cce914fd18e7ec28f7b15734a : FAILURE in 1h 42m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e4fb59c251294117ab815724aaaa4b70 : SUCCESS in 41m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0b9ecdefd9bc427889928d94be24ba1e : FAILURE in 2h 41m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/bde3ed7d26144adb9d3c3c6d69f37800 : SUCCESS in 27m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/02eacb753cdc4d39aec4591f3a19460e : SUCCESS in 19m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e8707c37c3f74042b8991a240944e006 : SUCCESS in 1h 20m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3cc69aa24e69457399c9cdc6c789ec16 : SUCCESS in 53m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/59d3802d51a04e3b9ead9dbc20b0d0a9 : SUCCESS in 40m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5bcda22fc5bf4279a481586a46252293 : SUCCESS in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"376651db58b7425b564fd14f417769aabbc8318b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-12 22:29:50.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-52-800552-1/ns5_manila : SUCCESS in 2h 6m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c8af8100a41ab55b3d533dccff3872e83ca9db5b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-12 22:38:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThank you Felipe and Nahim, waiting on NetApp CI","accounts_in_message":[],"_revision_number":1},{"id":"828a692827dc77e5b83c4753e6f8bbec0931602e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-13 00:57:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/52/800552/1/check/DellEMC_Unity_Manila_dhss_true/26948aa/DellEMC_Unity_Manila_dhss_true/180 : FAILURE in 2h 10m 01s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/52/800552/1/check/DellEMC_Unity_Manila_dhss_false/24d6a90/DellEMC_Unity_Manila_dhss_false/177 : SUCCESS in 1h 06m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f1df352f292aa39f93e2b15a6ffa4c877414a677","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-13 01:22:36.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/52/800552/1/check/DellEMC_PowerMAX_Manila/e56da7f/DellEMC_PowerMAX_Manila/128 : SUCCESS in 1h 58m 15s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"c9f1d91d06457458b450b857a9fc80cc8786a2dd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-13 01:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/52/800552/1/upstream-check/manila-cDOT-ss/40e713d/ : SUCCESS in 3h 07m 07s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/52/800552/1/upstream-check/manila-cDOT-no-ss/fe6d280/ : SUCCESS in 3h 16m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b10842cfece8c42749738840f2d018cc54218f0b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-13 23:23:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46e46e3b432158231f119bfa63a9acbac66f93b2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-07-24 00:35:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nI think that you folks should fix these two minor details before merge the fix. wdyt?\nThanks Felipe and Nahim","accounts_in_message":[],"_revision_number":1},{"id":"7db54a07823d6cc1bd0c36d4e5e49f2840c52d17","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2021-08-27 18:33:48.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThank you for your review. I\u0027ll upload the changes in the next patch.","accounts_in_message":[],"_revision_number":1},{"id":"519de95591e3ad389d6b5374c5b8c36872a4e373","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2021-08-27 19:05:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfc748b122ebd677d7aa7a5af2c7d6e7ec97a1d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 20:35:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c741fbd0454140b45f9fbadd04749c : SUCCESS in 11m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6456326da95b4fc5896ef037ee3ad3cb : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95fc01129f934f0aa72f64ac0b1bda30 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2eecf85271247e5835c0a136818b684 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf47b47617341c2ae6e0dcefe8c8f5c : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/687f539e773245309047d487b9caf6fa : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1499c0df3e4e11beeabc3a7315ffa9 : SUCCESS in 10m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1369bd73226540b1b9f4db37c3712b44 : SUCCESS in 6m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/58b212399b6a4e6abd76a2ce6f7678f0 : SUCCESS in 7m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5efdae3ccdc242c4a3f6a406ef9b45cf : SUCCESS in 6m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1f55b5408b89419da4cb968156d66b9c : SUCCESS in 34m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a97a78ea94b7439c993ecc24231b285f : SUCCESS in 34m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/04ab5003c9ea4897982000424fcc3637 : SUCCESS in 59m 14s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db00cc14316c47b8869053ae691c5741 : SUCCESS in 59m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3f6d05a50b3f4cab8f19673e9084ea71 : FAILURE in 1h 13m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/071266adcc1f4881992de58f6aab7668 : SUCCESS in 38m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e0f83491f9b54c7b968dfde4a60899ed : FAILURE in 1h 28m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4cca8795b5ca41e888249f1387684f9b : SUCCESS in 23m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e2beb4b25fb44b6c90487bc1932a36f1 : SUCCESS in 24m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2a4a0d24066c4d9fb4a4a0ae97e2857d : FAILURE in 53m 06s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d9e3541766f2454fb60c7fd5461141e0 : SUCCESS in 1h 01m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/49ef1d4c48c0423eb0fcc437a98e3502 : SUCCESS in 26m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/812e1a45c93d4d52993dff95a4e666a2 : SUCCESS in 35m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39c280e81be9b485dbbd92ca5cad9055896224dc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-27 20:48:52.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/52/800552/2/check/DellEMC_PowerMAX_Manila/8815561/DellEMC_PowerMAX_Manila/214 : SUCCESS in 1h 43m 27s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"318e7f25d23e9fd1710f4f37e5a6a70782e6075a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-27 21:00:34.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/52/800552/2/check/DellEMC_Unity_Manila_dhss_true/db3db03/DellEMC_Unity_Manila_dhss_true/304 : SUCCESS in 1h 48m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/52/800552/2/check/DellEMC_Unity_Manila_dhss_false/ac1ee10/DellEMC_Unity_Manila_dhss_false/301 : SUCCESS in 1h 09m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"4a941fa61499e5744111712d51d047a69006fd75","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-27 21:15:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/52/800552/2/thirdparty-check/pure-devstack-manila-tempest-aio/27359f2/ : FAILURE in 2h 07m 59s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/52/800552/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/4e6ac37/ : FAILURE in 1h 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74b5022623bb9517d762b92b9c67887dcffc472d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-27 22:21:53.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/52/800552/2/upstream-check/manila-cDOT-ss/6dda7e9/ : SUCCESS in 1h 45m 33s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/52/800552/2/upstream-check/manila-cDOT-no-ss/00b5ad2/ : SUCCESS in 1h 44m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3d948c72d76e43811aed167935ef4ac8a712275","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-27 23:54:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/52/800552/2/check/DellEMC_VNX_Manila/6e23716/DellEMC_VNX_Manila/147 : SUCCESS in 1h 41m 49s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b24a3633e4d36fd7df16524885564a3198f75350","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-28 00:19:42.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks Nahim","accounts_in_message":[],"_revision_number":2},{"id":"b4e3af11a655d432b29e53809f791d12977ed7c9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-28 02:01:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-52-800552-2/ns5_manila : SUCCESS in 1h 58m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"15be5e31cbfcb7b88f213fecbaee3694c0d818df","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-28 21:31:54.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM. Thank you for proposing this fix, Nahim","accounts_in_message":[],"_revision_number":2},{"id":"907ac5e132f5ed38f3a4d854b966f1bca41f6044","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-29 17:34:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks for updating the fix. LGTM, thanks","accounts_in_message":[],"_revision_number":2},{"id":"7e54e19ce52a33c922efea7ff86d080536826d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 17:35:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0bf2509b948e99e1b39bf097bbf70bd88ad6dee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:38:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46e099f18e44285b22a843514a7bb38 : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ccf8dec3e9642f0bc75b915f35a2e33 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d7c67e4a0b406a91947eb4f75b0d46 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6893fbd74ed45748cac011aac6733b0 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc37281aa6540d09d63c1c717575116 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/788d32d7ddc8408db1398f934f0e14f5 : SUCCESS in 6m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8ddf556c40146cc8a643f8929c47319 : SUCCESS in 25m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ae4cf3464c34a2fadf2e60733a7c534 : SUCCESS in 31m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/58ef33a6c4604268ac321fcffb8d8eed : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4c37f383b230f598de9d8284de000e03b0d95128","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:38:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b2f50af98fb5b45db9427f1af892b5cb20fc57e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:42:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44e842357fa4b61a59fe712044e10ff : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f033b2be56394ac6bb5a70d32dc478f1 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a745e9c7c48f0bfdd830f13094f588fe085b4845","revisions":{"98e7e8010fd1cc53342f8890722e4882ba42909c":{"kind":"REWORK","_number":1,"created":"2021-07-12 19:45:17.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/52/800552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/800552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/800552/1"}}},"commit":{"parents":[{"commit":"6c329e1a80f34cfbf6b077bbafe15693cb34a919","subject":"Merge \"Replace oslo_utils.fnmatch with fnmatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6c329e1a80f34cfbf6b077bbafe15693cb34a919"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-06-24 17:13:13.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-07-12 17:38:31.000000000","tz":-180},"subject":"[NetApp] Fixed scoped account replica delete","message":"[NetApp] Fixed scoped account replica delete\n\nIn the ONTAP 9.8 and older, there is a bug in the zapi call\n`snapmirror-release-iter` when using the scoped account\nconfiguration. The operation is returning success, but the release\nis not occurring.\n\nTo avoid this issue, the Netapp driver was changed to release with\nthe `snapmirror-release` zapi call, instead of the one that has a\nbug. This new zapi call requires the field `relationship-id` to be\nspecified and it is being retrieved from ONTAP before calling it.\n\nA small fix, not related to the bug, was made in the zapi call to\n`snapmirror-release-iter` because the `relationship-info-only`\nfield was in the wrong place according to the documentation.\n\nCloses-Bug: #1934889\nChange-Id: I21633d4ffe14983365b0b6239216ed5d0bbfaace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/98e7e8010fd1cc53342f8890722e4882ba42909c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/98e7e8010fd1cc53342f8890722e4882ba42909c"}]},"branch":"refs/heads/master"},"a745e9c7c48f0bfdd830f13094f588fe085b4845":{"kind":"REWORK","_number":2,"created":"2021-08-27 19:05:10.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/52/800552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/52/800552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/52/800552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/52/800552/2"}}},"commit":{"parents":[{"commit":"de785fea87cc403f13969a9a185d9ce4bb1a5fe4","subject":"Merge \"Add missing [oslo_reports] options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de785fea87cc403f13969a9a185d9ce4bb1a5fe4"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-06-24 17:13:13.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-08-27 18:41:36.000000000","tz":0},"subject":"[NetApp] Fixed scoped account replica delete","message":"[NetApp] Fixed scoped account replica delete\n\nIn the ONTAP 9.8 and older, there is a bug in the zapi call\n`snapmirror-release-iter` when using the scoped account\nconfiguration. The operation is returning success, but the release\nis not occurring.\n\nTo avoid this issue, the Netapp driver was changed to release with\nthe `snapmirror-release` zapi call, instead of the one that has a\nbug. This new zapi call requires the field `relationship-id` to be\nspecified and it is being retrieved from ONTAP before calling it.\n\nA small fix, not related to the bug, was made in the zapi call to\n`snapmirror-release-iter` because the `relationship-info-only`\nfield was in the wrong place according to the documentation.\n\nCloses-Bug: #1934889\nChange-Id: I21633d4ffe14983365b0b6239216ed5d0bbfaace\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a745e9c7c48f0bfdd830f13094f588fe085b4845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a745e9c7c48f0bfdd830f13094f588fe085b4845"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
