)]}'
{"id":"openstack%2Fmanila~801079","triplet_id":"openstack%2Fmanila~master~Ie82a523b4f8ab3d44d0fc9d70cb1ca6c059cc48f","project":"openstack/manila","branch":"master","topic":"bug/1936667","hashtags":[],"change_id":"Ie82a523b4f8ab3d44d0fc9d70cb1ca6c059cc48f","subject":"Replace deprecated import of ABCs from collections","status":"MERGED","created":"2021-07-16 15:18:57.000000000","updated":"2021-07-22 17:09:09.000000000","submitted":"2021-07-22 17:07:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"801079-bug/1936667","meta_rev_id":"6f9e03019178bd714a8179d5a1ccf84782a5f70c","_number":801079,"virtual_id_number":801079,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 17:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-07-22 16:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-07-22 16:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-16 15:54:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-16 16:25:44.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-16 17:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-16 17:19:28.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 18:30:43.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 18:47:11.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 21:43:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-07-17 06:01:11.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d84fdf5b0cec9895594ad78fe078fcdc8b885524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:18:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7dc2e2ad1e54ae07ba470c5bd6787f09f681d24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-16 15:21:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"50662c9cf29e976d4b439c12c64117d59287a1ba","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-16 15:54:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/79/801079/2/thirdparty-check/pure-devstack-manila-tempest-aio/f8b59df/ : FAILURE in 32m 19s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/79/801079/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/0fd80f7/ : FAILURE in 32m 24s","accounts_in_message":[],"_revision_number":2},{"id":"26cb913238419540114d286f6d9f66dd87295729","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-16 16:25:44.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/79/801079/2/check/DellEMC_Unity_Manila_dhss_true/afce8b9/DellEMC_Unity_Manila_dhss_true/207 : FAILURE in 57m 25s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/79/801079/2/check/DellEMC_Unity_Manila_dhss_false/5186ee4/DellEMC_Unity_Manila_dhss_false/204 : SUCCESS in 1h 03m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"bc5cb1387614b6d6a30ec70ca9aff5ea1f8c38ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 17:17:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85e655d1391437ebcd46ae45f753073 : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30d09c8f601478c86dffa5630f0c62c : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c0ed3a58cf448aca5c41a23db00cbb9 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69dc3f50709f418cb5e74f1d53737536 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e47c5c354d47473daf647ee5a2d6b983 : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a47a2ca2fc67427ebbf919181fa30cab : SUCCESS in 10m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b06a7160b7d84c5480131268cf463072 : SUCCESS in 10m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/70381be0f6574b258d243f62b2e3da1d : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2b16606cb8244d1bda96e4f09a6bb21 : FAILURE in 6m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/488d0ba84a7d4fa08692732a2d17e623 : SUCCESS in 38m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b654dcfacbc445f697ad1b0b5145fc8a : SUCCESS in 36m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a242b258d25940d18fa678e60bd05831 : SUCCESS in 1h 24m 37s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3af0790272734a86b3b41f9d92297c3b : SUCCESS in 1h 12m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ccf1d941f35b4bc9b6282b39ddc7afe3 : SUCCESS in 1h 51m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d9d9ec1a3bba43609e07f8849f994cd9 : SUCCESS in 49m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/320860678f19425a88bdd1be7c1bf1ad : FAILURE in 1h 39m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/71029dad0261463fbb84f48016cd5811 : SUCCESS in 27m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/578464ccef8f48cdbdd23de2c884f34d : SUCCESS in 26m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cdb43c276a8645b1aced34bc6192e413 : SUCCESS in 1h 20m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/26c1421ff4ec4bb3ae59c07be3a8ccec : SUCCESS in 55m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b38236f7b8b4f93894402484540bba5 : SUCCESS in 42m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7221c465344c4b1c8a8bcde8c1b39db4 : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5e6317333a834f2416793b4800e598843f668fd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-16 17:19:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/79/801079/2/upstream-check/manila-cDOT-ss/0d9ebda/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/79/801079/2/upstream-check/manila-cDOT-no-ss/bab7a9a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a056bb4f9fdd234bb5c671842654696582eaa32","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-16 18:30:43.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-79-801079-1/ns5_manila : SUCCESS in 2h 4m 45s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4926583901532a54a2212b23373a741f6833d07","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-16 18:47:11.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/79/801079/2/check/DellEMC_PowerMAX_Manila/932e8f9/DellEMC_PowerMAX_Manila/145 : SUCCESS in 2h 24m 46s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ec4428589534dd18192ac13f1219a1a3efa46e69","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-16 18:53:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/79/801079/2/check/DellEMC_Unity_Manila_dhss_true/d8de1ef/DellEMC_Unity_Manila_dhss_true/209 : FAILURE in 1h 03m 46s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/79/801079/2/check/DellEMC_Unity_Manila_dhss_false/799bbb5/DellEMC_Unity_Manila_dhss_false/206 : SUCCESS in 1h 16m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3987526897e90d8859bcce2feda837e691119ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-16 19:15:31.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-79-801079-2/ns5_manila : FAILURE in 2h 4m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"a5f8d42fac8bbc6c7e81a601078bbd1df6531eb5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-16 21:43:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThanks for this change. Please see a comment regarding consistency in the imports across this project; it\u0027s not a biggie because neither our hacking rules nor pep8 prevents the way you\u0027re doing it, but, consistency enhances readability and maintainability in cases.","accounts_in_message":[],"_revision_number":2},{"id":"788a81b65990f30e8324264e5ff2880b071442e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-17 03:20:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0dd6bc15326a111717e8275eed6c738857ab3063","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-17 03:52:56.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/79/801079/3/thirdparty-check/pure-devstack-manila-tempest-aio/12752f8/ : FAILURE in 31m 19s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/79/801079/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/a775084/ : FAILURE in 30m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c1f6f5fe414061fe6d38080fe85a79a2d26786c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-17 04:32:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/79/801079/3/check/DellEMC_Unity_Manila_dhss_true/b84148f/DellEMC_Unity_Manila_dhss_true/210 : FAILURE in 1h 07m 49s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/79/801079/3/check/DellEMC_Unity_Manila_dhss_false/b08da41/DellEMC_Unity_Manila_dhss_false/207 : SUCCESS in 1h 11m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"17158d5905260d15ce61e4f3f9feef6629e2bee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-17 04:59:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314370f3ebe446008862eb2ad72d4142 : SUCCESS in 10m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fb4491edd52484d9e75ac2822aa1fbd : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/366ed127be034a478522078c6febeffc : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423bbbdc7c4948aeb07090ec0e9d5323 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2410179dc94c49e5b7d647fd2f81f904 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f252774b550e457bad851575e44b4d6d : SUCCESS in 10m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaee8fa3543848ac9c1160a54b8562d7 : SUCCESS in 9m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ee63c8b15c8944398401d1aa3621af64 : SUCCESS in 5m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9c88963f0c204a12acf25d35fab94a9e : FAILURE in 5m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9194fe2a9dea4b34992435baa38b8183 : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c6cd238bdee845aca9a6c00f8f2905d6 : SUCCESS in 36m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0c4114e70e144e198d4b43c8ced48359 : SUCCESS in 1h 01m 48s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c11c60f1014b4731b51085ced2c37158 : SUCCESS in 1h 07m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/55a4d0a28b5d4953a6be85dc48bc6a8e : FAILURE in 1h 20m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/13c3224ad7c145e2a94f8bc207b3a852 : SUCCESS in 46m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8969b53299a441db85337b90cb3464c0 : FAILURE in 1h 36m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/feb6bee0e2094f1f9aa26161ded2e7bb : SUCCESS in 26m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/638ef7249f1841768f1a1e759025ef3a : SUCCESS in 21m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/660d63a5be8440019f4b1e8d3a6e964d : FAILURE in 1h 00m 19s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5c49f84e86e7459b8de1863d97148e37 : SUCCESS in 56m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/02f8158777554c4bafacad9835a10e63 : SUCCESS in 33m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0ec95da04ca043e0873f067e001cd6cf : SUCCESS in 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f5f390d9f3c43ddd71f3e7e55637fda87961016","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-17 05:00:34.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/79/801079/3/upstream-check/manila-cDOT-ss/61edb99/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/79/801079/3/upstream-check/manila-cDOT-no-ss/e99b9e2/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c4d3c8d89982dbbca1d4632d1bc43737bd7dead4","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-17 05:07:08.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-79-801079-3/ns5_manila : SUCCESS in 1h 45m 57s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"85bc1923d342e1d58f2950400a98b9869a3fb7f5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 05:53:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/79/801079/3/check/DellEMC_PowerMAX_Manila/3dac71e/DellEMC_PowerMAX_Manila/146 : FAILURE in 2h 32m 47s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"0c5ea7505aa504e48b67763fd3385bd9e63e1f12","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-07-17 06:01:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/79/801079/3/check/DellEMC_VNX_Manila/f3d1a73/DellEMC_VNX_Manila/112 : SUCCESS in 1h 00m 54s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"c6cbb3e2f1cfe5ae97f55185e1541636c6f7f0c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-17 06:12:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/79/801079/3/check/DellEMC_Unity_Manila_dhss_true/670ef53/DellEMC_Unity_Manila_dhss_true/211 : FAILURE in 1h 10m 37s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/79/801079/3/check/DellEMC_Unity_Manila_dhss_false/f1e309c/DellEMC_Unity_Manila_dhss_false/208 : SUCCESS in 1h 12m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"0ce786809da115f60280ef915c2919fb04507647","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-22 16:01:35.000000000","message":"Patch Set 3: Code-Review+2\n\nThank you!","accounts_in_message":[],"_revision_number":3},{"id":"736924acaafe0da2abe8c09e47a857aa3815b938","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-22 16:01:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bce2187b60cda9d82c0158cb354c9f894ce9ece7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 16:02:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6d0fb366c1bf88deba3c0604d1945e2dff6ee355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 17:07:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec132e774b584c088c9ac63aba0b4072 : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/670ecf95d5794fb5a2490b8a13ea3c31 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247e4f4cf20643c488ac2c305f4e657b : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a81ae8554245a8b9aee45451c345cd : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/861624d6c4ed447bb507c682f42f5050 : SUCCESS in 8m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f366c025552e4a65ada8280e2855dde7 : SUCCESS in 28m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fbdf46df8c83437eb9ebef8834864ca5 : SUCCESS in 33m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74a0e355531545e6934cc8d0eb2602c9 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7f0854afd53e49fe0f56480dbce79bcdc70f9a4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 17:07:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6f9e03019178bd714a8179d5a1ccf84782a5f70c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 17:09:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b3ab5e06fc4a6c9ec6cc97b43401e2 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c03f96ffaeb64e8cd374b6ed15d24cc21fdec034","revisions":{"4863d8019f24bbe856cac1337a86707f14451615":{"kind":"REWORK","_number":1,"created":"2021-07-16 15:18:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/801079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/79/801079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/79/801079/1"}}},"commit":{"parents":[{"commit":"35af72a87520ac0f063672a0306838e383012dfb","subject":"Merge \"init_host should be called before RPC consumer is created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/35af72a87520ac0f063672a0306838e383012dfb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:16:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:16:26.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections was deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ie82a523b4f8ab3d44d0fc9d70cb1ca6c059cc48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4863d8019f24bbe856cac1337a86707f14451615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4863d8019f24bbe856cac1337a86707f14451615"}]},"branch":"refs/heads/master"},"fdb1729a19dbee17a64f82c3f8e6e0b9d901a79b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-16 15:21:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/801079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/79/801079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/79/801079/2"}}},"commit":{"parents":[{"commit":"35af72a87520ac0f063672a0306838e383012dfb","subject":"Merge \"init_host should be called before RPC consumer is created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/35af72a87520ac0f063672a0306838e383012dfb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:16:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:21:14.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ie82a523b4f8ab3d44d0fc9d70cb1ca6c059cc48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fdb1729a19dbee17a64f82c3f8e6e0b9d901a79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fdb1729a19dbee17a64f82c3f8e6e0b9d901a79b"}]},"branch":"refs/heads/master"},"c03f96ffaeb64e8cd374b6ed15d24cc21fdec034":{"kind":"REWORK","_number":3,"created":"2021-07-17 03:20:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/801079/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/79/801079/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/79/801079/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/79/801079/3"}}},"commit":{"parents":[{"commit":"35af72a87520ac0f063672a0306838e383012dfb","subject":"Merge \"init_host should be called before RPC consumer is created\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/35af72a87520ac0f063672a0306838e383012dfb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-16 15:16:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-17 03:20:17.000000000","tz":540},"subject":"Replace deprecated import of ABCs from collections","message":"Replace deprecated import of ABCs from collections\n\nABCs in collections should be imported from collections.abc and direct\nimport from collections is deprecated since Python 3.3.\n\nCloses-Bug: #1936667\nChange-Id: Ie82a523b4f8ab3d44d0fc9d70cb1ca6c059cc48f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c03f96ffaeb64e8cd374b6ed15d24cc21fdec034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c03f96ffaeb64e8cd374b6ed15d24cc21fdec034"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
