)]}'
{"id":"openstack%2Fmanila~801372","triplet_id":"openstack%2Fmanila~master~I2e7d6edafc929e1c63425f6f4be02335fc263a34","project":"openstack/manila","branch":"master","topic":"metadata","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2022-01-28 18:36:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2e7d6edafc929e1c63425f6f4be02335fc263a34","subject":"WIP: Metadata for User facing resources","status":"ABANDONED","created":"2021-07-19 18:16:19.000000000","updated":"2022-01-28 18:36:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f29f54ad501dc7fad01c9610da5dbb0a201a5f1d","_number":801372,"virtual_id_number":801372,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-06 17:15:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 18:51:17.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-19 19:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 21:53:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-27 23:15:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-07-27 23:48:05.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-18 21:31:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"189fa9c4a39f969dc8f727759d3bef942b1a78c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-19 18:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4e7be704d0b21a1f442396be2a24e32ad52c2dc","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-19 18:17:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a332c5ec9320e7ce75eb58cec43801efdc65ea67","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-19 18:51:17.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/1/thirdparty-check/pure-devstack-manila-tempest-aio/2257291/ : FAILURE in 32m 38s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/47d3074/ : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fd90a264e8e27c365ed2c58b6609502dbea882e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 19:10:27.000000000","message":"Patch Set 1: Verified-1\n\n(27 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590cd25f70044bc7b43a9be5decb4563 : FAILURE in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0ba666781504551aacd3bc6039f4a70 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a4ea59a87624b529eb375e2a874dce2 : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009ca8e525ab459cbe25e6115e0f90da : FAILURE in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/439e81c8054e4783bf8a2c38ae06ab26 : FAILURE in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c683e148a7724aa391ba2b54e29572b6 : FAILURE in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8c333f03abe43ba86ce6b30996a7153 : FAILURE in 6m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e0b28fddd76c45fdb55b781f1bdb54b3 : SUCCESS in 6m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/207c78a461834409aa6b727f6f2b36a1 : SUCCESS in 5m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4139d3c16572449fa4d300952c01c7e4 : FAILURE in 13m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8b424bd2f153410eabb27866e38ce7ad : FAILURE in 13m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d5b543ba78104be48e44525bcf7fe99d : FAILURE in 21m 54s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ccb5b8eef8b34b7d892b91baebe1bda7 : FAILURE in 22m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b626461446f94f54ab496f5e7515f044 : FAILURE in 28m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d83c53c2dc744dd1a9184c6dfb8021b1 : FAILURE in 25m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/90a5bbdd8b79460f85135f1bd4e3c5ff : FAILURE in 22m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/24d5015d58d9467e8f99dd05537c1a59 : FAILURE in 19m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e31f7b74737c4fc7a508e7f47ab48ba9 : FAILURE in 20m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7f315b6b21a64875b032e28884f6d3fb : FAILURE in 20m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c459e6bce89e4e388a042f04dec27034 : FAILURE in 48m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6af32ff71f024fa39d290eb102193986 : FAILURE in 20m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/531f7678367e443e90d4937a434ffed3 : FAILURE in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"706b9d25534c535323bae8d5c31cfa79f2708347","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-19 21:53:45.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-1/ns5_manila : FAILURE in 1h 2m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3d4e0cb4c447f29fbe488bad024f432c7e58f4e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-27 21:18:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1dbb90a542d5a40bdaad6ad8e8912d718c4638b2","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-27 21:19:37.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"b2bdc190e66d8ab207dff68bb2aeb96e6981d91e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-27 22:59:25.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-2/ns5_manila : SUCCESS in 1h 40m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"611106cc6a5d906d6ac50a152868371439ab563d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 23:07:09.000000000","message":"Patch Set 2: Verified-1\n\n(56 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceed49f914224d1cba8d9df90d0740dd : SUCCESS in 10m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19deb618327a41f88e9e180db3fbd51e : FAILURE in 19m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/728478d849b04a63afe7b538f8293510 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2f3ca1ed5d49e1893b3d7f752845f6 : FAILURE in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cb729ab2946416193da7dcea04950d8 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f8b0b5dc40449979f4828a6efeffeb1 : SUCCESS in 10m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdeeace1aef646ab90bcae2b80fa09af : SUCCESS in 9m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/83b6b4895422420da93380f2538638b6 : SUCCESS in 5m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6364ed0eabe645cd90aa8c7a03c5c816 : SUCCESS in 5m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/13c0901a72964c89994a38058be7f091 : SUCCESS in 33m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c60255061e724594aa15543fdcd1ae48 : SUCCESS in 29m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/23c6399bc9ad49c5aea68e948c8cf220 : SUCCESS in 1h 05m 51s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9564f8b1ac2041279c1c5fedfd290228 : SUCCESS in 1h 20m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4e9dbedf49b6478281b9935ad379541a : FAILURE in 1h 45m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/043d26370bbd4948bd519706a953c960 : SUCCESS in 48m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cbeb705a8f874ee8885bf658b922563b : FAILURE in 1h 20m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2a3407885eb04ca2a75d46a050957d9c : SUCCESS in 37m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f4c3e359c26a461daa5a513fc8a8e3b3 : SUCCESS in 27m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0f17e64af4f54f4a91334a47cf44acf5 : FAILURE in 1h 05m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8849246b4d374667ab068be63066a0ed : FAILURE in 48m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/87d4e347d809402b9a233ad48fc2f0ce : SUCCESS in 44m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/106a32ae2398446ebdea6533abe7b503 : SUCCESS in 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e3156ca53d82ee3f1932f94bd8075e5741c59c0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-27 23:15:43.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/2/check/DellEMC_Unity_Manila_dhss_true/c010308/DellEMC_Unity_Manila_dhss_true/216 : FAILURE in 1h 44m 33s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/2/check/DellEMC_Unity_Manila_dhss_false/add0d2e/DellEMC_Unity_Manila_dhss_false/211 : SUCCESS in 1h 02m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a84f7b4fa1173c67f9740f09e2b15fa904404d8a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-27 23:27:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/2/thirdparty-check/pure-devstack-manila-tempest-aio/0b60fc3/ : FAILURE in 53m 40s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6e374de/ : FAILURE in 2h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a133e63647109ccd1c87c66566d4e48c5f238015","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-27 23:48:05.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/2/check/DellEMC_PowerMAX_Manila/5367a82/DellEMC_PowerMAX_Manila/148 : SUCCESS in 1h 43m 07s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2a11928d70cae3b9eb3f436350eb172b1a3cc6ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-30 19:16:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"94fc2fa38c60579c65a0f32b03959d635c1bd9c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-30 19:40:22.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/3/thirdparty-check/pure-devstack-manila-tempest-aio/5c2c44a/ : FAILURE in 22m 56s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/56093f0/ : FAILURE in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9d31f6b2af9b7d6a4d22065ae386be617d627d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-30 19:51:12.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/3/check/DellEMC_Unity_Manila_dhss_true/0ce0e43/DellEMC_Unity_Manila_dhss_true/235 : FAILURE in 34m 51s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/3/check/DellEMC_Unity_Manila_dhss_false/dbfa12b/DellEMC_Unity_Manila_dhss_false/232 : FAILURE in 34m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a085ad784a8511f700c29b080c40ae85aa0d1a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 20:09:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7c8dfae95849aa9acafca1a442c04d : SUCCESS in 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/404c80cf4d2249a5943ff2ac6ab57b2b : FAILURE in 18m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e77b2a69eb50489aab12270b5b94efb2 : FAILURE in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e1255ac9d84f0ba13546d191696626 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53d43aefe5c8400785e547ce97ccf545 : FAILURE in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b316c40996e1421b8386c9b2c0b75cea : FAILURE in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8fbe84dcab46f5b6dfd874efb8ac04 : FAILURE in 10m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/668fb452c310467db63fc57a45f09c5a : SUCCESS in 5m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/769c4b07ebda4bb3bf2b915fca5ac290 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/41b82f2bad034119919211962bb83fc8 : FAILURE in 19m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3d92035159ac4fb28afa3de95a5738dc : FAILURE in 19m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1582f7ee37540fe96ca00f14f1ceb2a : FAILURE in 31m 54s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/20c60fde822e439c932b6d3ffc6656f1 : FAILURE in 34m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3ca8b92332094d48a233abfd90ff83b8 : FAILURE in 34m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a998444944bd4e929c8f4724fca38759 : FAILURE in 29m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c0e0fe53bf544410b1732c3d86fc02c5 : FAILURE in 32m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1ab132a6cc1f43198c3ef72c70c3fd30 : FAILURE in 22m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7f295d4363604b18a2c648ba95607be1 : FAILURE in 20m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fabecce1366a41258dd48e7ec18d9632 : FAILURE in 31m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0cbbb85e42c44fce86dc4a0782fe4ae4 : FAILURE in 47m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3b97f4cc1a054667ac70e6a3e72930ce : FAILURE in 34m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4d489a31deff4180bdb29480861cc2f0 : FAILURE in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9512830dab3d221ff9825ccaca3830d8aec47b5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-30 20:51:02.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-3/ns5_manila : FAILURE in 1h 25m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"6faa9153126bb12d89e9e8aaee129350f3838e1a","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-30 21:06:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/3/check/DellEMC_PowerMAX_Manila/9c2fc3b/DellEMC_PowerMAX_Manila/159 : FAILURE in 25m 53s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"2194f9c6e5768a2fa5482926b0464b4b8d9ad701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-07-30 21:47:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25411e452655290085171bdd166f22f4980e353b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-30 22:07:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/4/thirdparty-check/pure-devstack-manila-tempest-aio/5dd33e0/ : FAILURE in 18m 33s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/4/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6cb37a1/ : FAILURE in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18f51f83c7a4862fc355473e8d78252cf1326fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 22:41:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b2056671148438f9263af78570db3be : SUCCESS in 15m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0141553792544f48b1271162ca8bec89 : FAILURE in 25m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c38ba01cd44491bc9d7f8b68c5932a : FAILURE in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f40f468f7b4499d96912b76af24dba0 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f31ff53b9a241aca5e0c865517f2ab4 : FAILURE in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0691a48c83c437392456a2441e19c30 : FAILURE in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fd8e17a3e6846f5a2194b0e42eaed2f : FAILURE in 9m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/40138f739b434b53bffaa07fe40bb027 : SUCCESS in 8m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1be22c4399b6416f94b2e9646c50fb56 : SUCCESS in 5m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d833154bc0e34e5fa9c1c281d33bbaab : FAILURE in 16m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/da8e59018d544d05be23edbce8a83c27 : FAILURE in 16m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/580a61d5d2ea498a9d9cd68a4cd9915d : FAILURE in 26m 12s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/62c1695155d24ec389dad563451a7a92 : FAILURE in 28m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/724ca2bc2c5a4be5a12dcba74715d254 : FAILURE in 30m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c3e155174c1548b6bd85b4569f2fcf05 : FAILURE in 27m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ded572e99bdb463fa0db6eca5507fae7 : FAILURE in 27m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f9ecbcef275643bfa98d9f0746f5183a : FAILURE in 19m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6aa721ac86a24904bbec62c5d24e2999 : FAILURE in 20m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/12946b4f4df54de3b765e4b764858e61 : FAILURE in 25m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/25ba1506708f41aca0daac5c43c8fac7 : FAILURE in 48m 28s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/76c9e2000d5c409aa247dca77a978ffc : FAILURE in 29m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/357bfb62817a45adab468ce52e79cdc0 : FAILURE in 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ecda3213ad4316a44497cfe2da501562a022aafc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-30 22:50:57.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-4/ns5_manila : FAILURE in 1h 1m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"05bd4f60c0ffb82fa61854e85a3162854f0b0842","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-07-30 22:58:05.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/4/check/DellEMC_Unity_Manila_dhss_true/3130740/DellEMC_Unity_Manila_dhss_true/237 : FAILURE in 27m 35s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/4/check/DellEMC_Unity_Manila_dhss_false/5bfa5fb/DellEMC_Unity_Manila_dhss_false/234 : FAILURE in 25m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"d025049b4d5b2d00464ea64d60e10a21e62d1a63","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-07-30 23:20:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/4/check/DellEMC_PowerMAX_Manila/5e69449/DellEMC_PowerMAX_Manila/161 : FAILURE in 21m 30s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"a05e2f7570f6b5a66baf19aedafe75a69cb703dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-08-04 22:44:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7669db1b014e4952a9bf2d3b795557040509e356","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-04 23:41:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/5/thirdparty-check/pure-devstack-manila-tempest-aio/5f8ef67/ : FAILURE in 55m 56s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/5/thirdparty-check/pure-devstack-manila-tempest-fa-aio/33949a0/ : FAILURE in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fc396a6ec71d5b014ea049f120fa8a716b524e90","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-05 00:33:08.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-5/ns5_manila : SUCCESS in 1h 47m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"d99eb2a84bf3df46e611051a510c6093111f2741","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-05 00:42:35.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/5/check/DellEMC_PowerMAX_Manila/bf053ac/DellEMC_PowerMAX_Manila/164 : SUCCESS in 1h 57m 40s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":5},{"id":"625c225d346683bd3633254d46d7aaa5ea60ac9e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-05 01:02:25.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/5/check/DellEMC_Unity_Manila_dhss_true/9573089/DellEMC_Unity_Manila_dhss_true/240 : FAILURE in 58m 21s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/5/check/DellEMC_Unity_Manila_dhss_false/563b722/DellEMC_Unity_Manila_dhss_false/237 : FAILURE in 2h 17m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"aa314c301b1dd5c89d90ac5fa18fa5dd730228f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 03:00:26.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05fe2d727ad542a681213fadd93dbac1 : SUCCESS in 11m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d9ddbab7304167b0df4e38f8c9a396 : FAILURE in 21m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d8373a7c63842ec96b3a10bd43d7802 : FAILURE in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5895d814714d90b6a69f0e44a0649f : FAILURE in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12da29f6546c41e793081aa30091af81 : FAILURE in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e51666b5902483192f9e3cabc2c6b5e : FAILURE in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73fc288651934ce0bb1e5487a1eb6b61 : FAILURE in 12m 25s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c0e065cba8dc4bda8d9eed906b39d4cc : SUCCESS in 6m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8199171aef6746798a9854ddf00f399a : FAILURE in 6m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44c73db8b2744c37a79805e9210a9a6a : FAILURE in 25m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/02c46286203b4be7be67a5a0e83a2f58 : FAILURE in 35m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6167fc6ba5634b6292a6b5d0fc6adb4b : FAILURE in 35m 29s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/31e67a5be40d4026b0229c010013a21e : FAILURE in 1h 14m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7657989dd8b74de780bb4f96dd4851a1 : FAILURE in 1h 42m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9050c82564ac4f33a68df01ddf3ca9d9 : FAILURE in 2h 37m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f774d749ba974c14aa371de0c27fda7a : FAILURE in 1h 34m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b32073dd9158411a91fdcfcfc4eb05ab : SUCCESS in 30m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/584dfc24ba1c403fbd10a98ba96ff889 : SUCCESS in 26m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7456e27246474f709846853fbe169973 : FAILURE in 1h 44m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/06dcea2c9c98450d9da910394e3894f2 : FAILURE in 1h 02m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dc9df6226f404b4b9c8d43ed1633e30a : FAILURE in 42m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e2599ae1030d45b197846f1d7afdde35 : FAILURE in 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7252453944b07c770709a3e421f61c788408d742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-08-06 15:05:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4e22900f610498d7ef86e223bf4608b9140d5873","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-06 16:02:36.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/6/thirdparty-check/pure-devstack-manila-tempest-aio/99285fe/ : FAILURE in 56m 02s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/6/thirdparty-check/pure-devstack-manila-tempest-fa-aio/ca5b107/ : FAILURE in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d081a1e00a863bdba0084f498930d426cac740ff","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-06 16:31:23.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-6/ns5_manila : SUCCESS in 1h 25m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"d0c94c456e26b17afb9dc5618b4d5cc8011c3ceb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-06 17:35:03.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/6/check/DellEMC_PowerMAX_Manila/f21abac/DellEMC_PowerMAX_Manila/171 : FAILURE in 2h 29m 18s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"99fe7780e662492faeca977965bab14b63d186fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 18:11:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dba7c6371f743bc8e72fc10368d4a08 : SUCCESS in 10m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8266c2e61bf548c78eec305cd27d4e9f : FAILURE in 21m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8077f5630c764c4f83134a64b2c8e8d0 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88384c67497418ca0f9c8f80100d4d7 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb164c390849467aba853e8ca1266c0a : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23dc2fe7331b438da0b3ea35d06ac060 : SUCCESS in 9m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2dd95aaf9874bf7a3f2cb34690ed345 : SUCCESS in 11m 08s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d7706480f6f4f019374d5ddc86ed6ef : SUCCESS in 5m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/241447790ca341258278a426e9275755 : SUCCESS in 5m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e21a3a3649904086ba0e89a03af91727 : FAILURE in 26m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2ace059cd64c452a996ad7be351ed951 : FAILURE in 26m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/239c373d56164ffeb571b446aedbbec3 : FAILURE in 42m 08s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bc73d23f2c97453db5ad60478b182b1f : SUCCESS in 1h 27m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/63d62d03bbcb4e07b8d0050056c44ab7 : SUCCESS in 1h 57m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d2b436d32114d7c8f2ed3dc361d002f : FAILURE in 3h 01m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/60be96b80d164bc9b26337103a8f3316 : FAILURE in 1h 18m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e11e4a6fbe454009865491ce1bc7ffef : SUCCESS in 42m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9ace2ceeea6a485f9e67697bd69df347 : SUCCESS in 30m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a686a5e5d16424eadda33781701c2f9 : FAILURE in 1h 50m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0d29628e59994f2b9f658acc2ef0cc9b : FAILURE in 53m 29s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f8bf718511344d44aeca872ed0185bad : FAILURE in 43m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5bffdff21ae742f1b4e9b15d80775fc1 : FAILURE in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fa83f6b488c1cc87efa37ece630aaa836aba7ae1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-06 18:19:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/6/check/DellEMC_Unity_Manila_dhss_true/1c8d3b0/DellEMC_Unity_Manila_dhss_true/246 : FAILURE in 3h 13m 44s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/6/check/DellEMC_Unity_Manila_dhss_false/444b252/DellEMC_Unity_Manila_dhss_false/244 : SUCCESS in 2h 27m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"6bbc56b3430d99f0278858ca0dd110fde0338345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-18 21:31:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7feaefbb076d2cd3f5f57ed0f7f45ff928098370","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-18 21:36:23.000000000","message":"Patch Set 7: Workflow-1\n\nNot for Xena as we discussed; this needs some more work; but is progressing well. I\u0027ve squashed your commits ashley, and made some modifications - tests will still break, and more changes are necessary..","accounts_in_message":[],"_revision_number":7},{"id":"281c71908d4c45769f28355220cf5fe1bc18c65d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-18 23:04:05.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/7/thirdparty-check/pure-devstack-manila-tempest-aio/3d55ae5/ : FAILURE in 1h 30m 08s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/7/thirdparty-check/pure-devstack-manila-tempest-fa-aio/e12ec73/ : FAILURE in 46m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ab20f48b27ce29a46e17cda404bee93475a22acd","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-18 23:17:34.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila http://publiclogs.emc.com/72/801372/7/check/DellEMC_PowerMAX_Manila/b468d44/DellEMC_PowerMAX_Manila/185 : FAILURE in 1h 46m 04s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":7},{"id":"5081034baa89846a68f60085f0f791c9073f35ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-19 01:26:52.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/72/801372/7/check/DellEMC_Unity_Manila_dhss_true/fff5678/DellEMC_Unity_Manila_dhss_true/275 : FAILURE in 3h 55m 23s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/72/801372/7/check/DellEMC_Unity_Manila_dhss_false/559b90d/DellEMC_Unity_Manila_dhss_false/272 : FAILURE in 3h 54m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"b43fdbed884547727ea768162953a9708e12dafc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:30:57.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4042d16ef0204af38ae25b6956de5531 : SUCCESS in 11m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0c7e8427073436db960dd9d5f1f3778 : FAILURE in 20m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c24b75e36ca4ac18b5e2d9ac1658f66 : FAILURE in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2889164cf34e4f5b93ae12dcba5a07b4 : FAILURE in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7246473d0b1640a996a1c9afb5597c59 : FAILURE in 9m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/458034abfa4b4e989e5f3168027b77ae : FAILURE in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b077ded6034a0c8b0383404c7fbdfe : FAILURE in 10m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a7466f3dbe134799a1dc21493077aa7f : SUCCESS in 6m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35ea1aac218f4035b7ecf2ebb7b06989 : FAILURE in 6m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a07a8bd979c41739bde7e2ec23a73c0 : FAILURE in 21m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dea051633de44358a00540fd3334482f : FAILURE in 26m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/84bbc8925f394d68a81164bd72306065 : FAILURE in 37m 53s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f95c327375764896928e06276c317ae3 : FAILURE in 56m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e2e09b64809845788ed60fe61721e039 : FAILURE in 1h 53m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c061f9864e404960ba7226e291bd323f : FAILURE in 33m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0c7f3f030d424febb2d75304cba36751 : FAILURE in 1h 21m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/033faf63ae71474f8c8fec2a0cbbfda4 : FAILURE in 26m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c934ea4d24cc4f2e9f446f751a364dd5 : FAILURE in 20m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/34a2c440f9f04f299909d7ca838c4fa3 : FAILURE in 31m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07a0bac116c04080972b71ca888ba38b : FAILURE in 42m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c05cf5beaea14f0ca3869c9074e31565 : FAILURE in 34m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ddcba718e0bf412ba8eab72c1e477843 : FAILURE in 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ebf3f97ea95d56694882dc52edfa1dc0940feb22","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-08-19 07:21:35.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-7/ns5_manila : FAILURE in 1h 49m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"c13858a23749ff4b8935283b65007e7cd6e9e69c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-10-26 19:14:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"66e7381e0ba60b797736408a8b90635b4bf3e282","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 19:28:48.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/72/801372/8/check/DellEMC_Unity_Manila_dhss_true/d0a2b39/DellEMC_Unity_Manila_dhss_true/49 : FAILURE in 14m 01s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/72/801372/8/check/DellEMC_Unity_Manila_dhss_false/996724c/DellEMC_Unity_Manila_dhss_false/52 : FAILURE in 12m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d13afff483854a8c41ce47e795507c48fba98882","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-10-26 19:42:30.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/72/801372/8/thirdparty-check/pure-devstack-manila-tempest-aio/f305d61/ : FAILURE in 25m 57s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/72/801372/8/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6ef2777/ : FAILURE in 26m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1492d72cb0d0cd8af9e348054954cee1cf352120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 20:11:43.000000000","message":"Patch Set 8: Verified-1\n\n(33 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bf4f1cbcfd436b986e513e2acd5abd : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc6eb72c29c4d72ae0963cd9e87f49c : FAILURE in 15m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9df31e3970045a0bff0d557fe09de36 : FAILURE in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb6c000701204b9888956c0ca863c45d : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb90a0dee5dd4a0e8e7d7411a6b15614 : FAILURE in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b857edff23d349ff938f3a24c26eca64 : FAILURE in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/710ec8bf852f4390aba5bcad207a0ae6 : FAILURE in 5m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/482770c4d09d44669946f84c0c529a01 : SUCCESS in 7m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/98fb7c82bf2c4d5c9cc304557ddc1ffd : FAILURE in 6m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ebe63d6c814b463a8db3b677918630cb : FAILURE in 19m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2c6c71a1254e4377a8f55a74567e225c : FAILURE in 21m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d365531aa3c4e9c9174652447a46cdd : FAILURE in 36m 40s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/178a6aa7e2704c28b7d94f273fbd6e5a : FAILURE in 41m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/989d35be13384af5a25adade353e40ce : FAILURE in 38m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bd53d696e9644426ba6475b2b74a0c4f : FAILURE in 33m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aa204570404f4b50b58b8febe57cae27 : FAILURE in 32m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/28ac8ec36a3740208a0e3aef1d4fd31d : FAILURE in 21m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a8cf35bca9754807b661d08de5e9781a : FAILURE in 21m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/be4cdc2fcc034e9a8eace28849b75ca9 : FAILURE in 39m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a51640abe0d64ca4a6399b7677aa315d : FAILURE in 48m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d1f4eae0c6674a3cbb84fec66969a110 : FAILURE in 34m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a5d301c753d48f9b3743efedd2141af : FAILURE in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a74e13a24116cea2ec1caa4644bf4b381dd18c97","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-26 20:32:41.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-8/ns5_manila : FAILURE in 1h 5m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"fe5b000dd0bdf03e83a32d6d5cf058aa0df7e33e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-06 16:26:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bc7fca29f908fadf80fea5ef560d33d49997c498","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-06 17:08:00.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/72/801372/9/check/DellEMC_Unity_Manila_dhss_true/5148f93/DellEMC_Unity_Manila_dhss_true/198 : FAILURE in 41m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/72/801372/9/check/DellEMC_Unity_Manila_dhss_false/8c8f22f/DellEMC_Unity_Manila_dhss_false/187 : FAILURE in 9m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"811781ea7af96e420806ce19fd7029319df531de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 17:15:41.000000000","message":"Patch Set 9: Verified-1\n\n(49 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98778e23a77545e1867397f790fea493 : FAILURE in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/633b2fe795bd4f418cb43d2e404ea83b : FAILURE in 23m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8f28b4418af4b4cb2687973f12ee079 : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8387dc2873045b2aebb00893a1d1469 : FAILURE in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3941500956415b85c52617285ec3bb : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f058063dd340b6b36c2805a0e9d642 : FAILURE in 6m 53s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/77443e660ff3465a95150808fc86a1c4 : SUCCESS in 6m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/60b1c2c9209a4d02a71e9229a7367b5e : FAILURE in 7m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66da991a01e34bb788966590e4744ca5 : FAILURE in 18m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/17e31e705d944026824dcbefd32abd6f : FAILURE in 15m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e3d4b2a0a4f644188499a4f362209ebf : FAILURE in 30m 20s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/26da00207d1049dca1ee6cbe3bf28a80 : FAILURE in 30m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cd7acbdfe1c3456ca40250e4ac8429c9 : FAILURE in 37m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d57326de39834b2987ce90ce2eae026c : FAILURE in 45m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99eeb9a67a424fbfb4a5244bbb7e0645 : FAILURE in 34m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/66ed755e77dd47fea4bd6eb497fba85b : FAILURE in 17m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1e6996216749468db477c3bf8b264c4b : FAILURE in 14m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8ad111584d054764a476fdf0935035ce : FAILURE in 19m 56s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a1af495638ce4069b77d91130349fea3 : FAILURE in 47m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/912a23b33a1c4bcc911fc98584f9844f : FAILURE in 35m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1560f827fbb34bd29744de0f187681fc : FAILURE in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5fa3054a5305fc300b13eb387359ac5e2c75dc41","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-06 19:38:40.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-72-801372-9/ns5_manila : FAILURE in 1h 15m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"55a0a32198665714320c9e06e5e3d3aae9d619b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-06 20:41:21.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/973ca8366dab41aaa8cd11160ef9b6ad : RETRY_LIMIT in 35m 46s\n- pure-devstack-manila-tempest-fa-aio finger://os-ci-sf-all.cinder-ci.dev.purestorage.com:7979/dde4591e9d2743ae98f442de2c91ec7b : RETRY_LIMIT in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4614a0b18101a3af28597aa5fcf08a6ea2a644ed","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-01-06 23:28:47.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio pure-devstack-manila-tempest-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"f29f54ad501dc7fad01c9610da5dbb0a201a5f1d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-28 18:36:28.000000000","message":"Abandoned\n\nsplit patch, check tag: bp/metadata-for-share-resources","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"0930ac1f88ce95809f9bef411c4f697b1434f772","revisions":{"9b2bf0ccf65062ff3236c37623bd385cc01fe6b3":{"kind":"REWORK","_number":1,"created":"2021-07-19 18:16:19.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/1"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"subject":"Add Share Snapshot Metadata Resource","message":"Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9b2bf0ccf65062ff3236c37623bd385cc01fe6b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9b2bf0ccf65062ff3236c37623bd385cc01fe6b3"}]},"branch":"refs/heads/master"},"ebb9f8c39e432b1329555a06fa7995892b1eea1e":{"kind":"REWORK","_number":2,"created":"2021-07-27 21:18:42.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/2"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-27 21:18:33.000000000","tz":0},"subject":"Add Share Snapshot Metadata Resource","message":"Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ebb9f8c39e432b1329555a06fa7995892b1eea1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ebb9f8c39e432b1329555a06fa7995892b1eea1e"}]},"branch":"refs/heads/master"},"d879a15a7e6f5310701511ce2e814060f03ec08b":{"kind":"REWORK","_number":3,"created":"2021-07-30 19:16:03.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/3"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-30 16:29:25.000000000","tz":0},"subject":"Add Share Snapshot Metadata Resource","message":"Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\nCan be used with curl commands\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d879a15a7e6f5310701511ce2e814060f03ec08b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d879a15a7e6f5310701511ce2e814060f03ec08b"}]},"branch":"refs/heads/master"},"b97fc93f41e0139f846a4e6a2700da4e772df5bd":{"kind":"REWORK","_number":4,"created":"2021-07-30 21:47:43.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/4"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-30 21:44:30.000000000","tz":0},"subject":"Add Share Snapshot Metadata Resource","message":"Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\nCan be used with curl commands\n\nThis change adds resource metadata tables to the migration including:\nShare\nShare Instance\nShare Instance Export location\nShare Access Rules\nSnapshots\nSnapshots Instance Export location\nSnapshots Access Rules\nShare Groups\nShare Group Snapshots\nSecurity Services\nShare Networks\nShare Network Subnets\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b97fc93f41e0139f846a4e6a2700da4e772df5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b97fc93f41e0139f846a4e6a2700da4e772df5bd"}]},"branch":"refs/heads/master"},"ef0fce81af85832419fdb519e6d8c6155cb8f7e3":{"kind":"REWORK","_number":5,"created":"2021-08-04 22:44:37.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/5"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-08-04 22:39:13.000000000","tz":0},"subject":"WIP:Add Share Snapshot Metadata Resource","message":"WIP:Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\nCan be used with curl commands\n\nThis change adds resource metadata tables to the migration including:\nShare\nShare Instance\nShare Instance Export location\nShare Access Rules\nSnapshots\nSnapshots Instance Export location\nSnapshots Access Rules\nShare Groups\nShare Group Snapshots\nSecurity Services\nShare Networks\nShare Network Subnets\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ef0fce81af85832419fdb519e6d8c6155cb8f7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ef0fce81af85832419fdb519e6d8c6155cb8f7e3"}]},"branch":"refs/heads/master"},"9bfb03469611fc2b140f5440c3eed6433a94c1af":{"kind":"REWORK","_number":6,"created":"2021-08-06 15:05:27.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/6"}}},"commit":{"parents":[{"commit":"1a395050395a7af3f2b2f11f3578a90a1d14e312","subject":"Merge \"Add Pure Storage FlashBlade driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1a395050395a7af3f2b2f11f3578a90a1d14e312"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-08-06 14:59:44.000000000","tz":0},"subject":"WIP:Add Share Snapshot Metadata Resource","message":"WIP:Add Share Snapshot Metadata Resource\n\nAdds a share snapshot metadata resource including:\n\nNew share snapshot metadata database table\nAccompanying share snapshot metadata table model\nAccompanying database methods to access and manipulate metadata\nAccomapnying API methods to access and manipulate metadata\nAdditional exception\n\nThis code will be abstracted to include more user-facing resources.\nCan be used with curl commands\n\nThis change adds resource metadata tables to the migration including:\nShare\nShare Instance\nShare Instance Export location\nShare Access Rules\nSnapshots\nSnapshots Instance Export location\nSnapshots Access Rules\nShare Groups\nShare Group Snapshots\nSecurity Services\nShare Networks\nShare Network Subnets\n\nPartially-implements: bp/metadata-for-share-resources\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9bfb03469611fc2b140f5440c3eed6433a94c1af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9bfb03469611fc2b140f5440c3eed6433a94c1af"}]},"branch":"refs/heads/master"},"76c11780839385212dbd0999a00936f5740e91fa":{"kind":"REWORK","_number":7,"created":"2021-08-18 21:31:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/72/801372/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/7"}}},"commit":{"parents":[{"commit":"f56a326da8a1a1cfb4f8223da19430d8293033b6","subject":"Merge \"Add generic fuzzy matching logic to the database layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f56a326da8a1a1cfb4f8223da19430d8293033b6"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-08-18 21:31:00.000000000","tz":-420},"subject":"WIP: Metadata for User facing resources","message":"WIP: Metadata for User facing resources\n\nThis change adds resource metadata tables and\nAPIs for the following resources:\n\n- Share Instances\n- Share Instance Export locations\n- Snapshots\n- Snapshot Instance Export locations\n- Snapshots Access Rules\n- Share Groups\n- Share Group Snapshots\n- Security Services\n- Share Networks\n- Share Network Subnets\n\nDocImpact\nAPIImpact\nPartially-implements: bp/metadata-for-share-resources\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76c11780839385212dbd0999a00936f5740e91fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76c11780839385212dbd0999a00936f5740e91fa"}]},"branch":"refs/heads/master"},"5a6bc65a20b502b9f673a9e70ad61d055fd7a344":{"kind":"REWORK","_number":8,"created":"2021-10-26 19:14:30.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/8"}}},"commit":{"parents":[{"commit":"a659470852f920621e5473dc323751039a24da87","subject":"Merge \"Adapt CephFS driver to do not try to escape export ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a659470852f920621e5473dc323751039a24da87"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2021-10-26 19:13:33.000000000","tz":-240},"subject":"WIP: Metadata for User facing resources","message":"WIP: Metadata for User facing resources\n\nThis change adds resource metadata tables and\nAPIs for the following resources:\n\n- Shares\n- Share Export Locations\n- Share Replicas\n- Share Replica Export Locations\n- Share Instances\n- Share Instance Export locations\n- Snapshots\n- Share Access Rules\n- Share Groups\n- Share Group Snapshots\n- Security Services\n- Share Networks\n- Share Network Subnets\n\nDocImpact\nAPIImpact\nPartially-implements: bp/metadata-for-share-resources\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a6bc65a20b502b9f673a9e70ad61d055fd7a344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a6bc65a20b502b9f673a9e70ad61d055fd7a344"}]},"branch":"refs/heads/master"},"0930ac1f88ce95809f9bef411c4f697b1434f772":{"kind":"REWORK","_number":9,"created":"2022-01-06 16:26:06.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/72/801372/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/801372/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/801372/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/801372/9"}}},"commit":{"parents":[{"commit":"142990edc027e14839d5deaf4954dd6fc88de15e","subject":"Merge \"ratelimit should not be enabled in nolimit pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/142990edc027e14839d5deaf4954dd6fc88de15e"}]}],"author":{"name":"ashrod98","email":"ashrod98@gmail.com","date":"2021-07-19 18:08:30.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-01-06 16:22:49.000000000","tz":0},"subject":"WIP: Metadata for User facing resources","message":"WIP: Metadata for User facing resources\n\nThis change adds resource metadata tables and\nAPIs for the following resources:\n\n- Shares\n- Share Export Locations\n- Share Replicas\n- Share Replica Export Locations\n- Share Instances\n- Share Instance Export locations\n- Snapshots\n- Share Access Rules\n- Share Groups\n- Share Group Snapshots\n- Security Services\n- Share Networks\n- Share Network Subnets\n\nDocImpact\nAPIImpact\nPartially-implements: bp/metadata-for-share-resources\n\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0930ac1f88ce95809f9bef411c4f697b1434f772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0930ac1f88ce95809f9bef411c4f697b1434f772"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
