)]}'
{"id":"openstack%2Fmanila~802017","triplet_id":"openstack%2Fmanila~master~I40349cbe106f688a67043a73f8563f87f069831b","project":"openstack/manila","branch":"master","hashtags":[],"change_id":"I40349cbe106f688a67043a73f8563f87f069831b","subject":"Replace deprecated scheduler filter module name","status":"MERGED","created":"2021-07-23 12:21:20.000000000","updated":"2021-07-23 20:39:46.000000000","submitted":"2021-07-23 20:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802017","meta_rev_id":"84a6596fb6a6ad250a0567c08c56042adc2e9c2d","_number":802017,"virtual_id_number":802017,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-23 20:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2021-07-23 19:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2021-07-23 19:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 13:29:36.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-07-23 13:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 13:59:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 18:08:45.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-23 19:19:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"ca9cfde5e18cb38874f68759b2be870d293a820e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-23 12:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92d03c7b769ee53731dda7ce514b40d21fa9ddd7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-07-23 13:29:36.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/17/802017/1/thirdparty-check/pure-devstack-manila-tempest-aio/1206908/ : FAILURE in 1h 07m 21s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/17/802017/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/db4b7c3/ : FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1658a6c5b887b1f458d54250c36d40ef8eed62e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 13:59:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b66e928ddb1448a1854bcf486ca6d0e1 : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49ef72da93ab466ebd5e9c3e28150657 : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5632606c8cd747c6b295bfba8321521a : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e066fd12d3b4fbdb83f35cba59c8ef9 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65b4ef83e736406d87af564707f232af : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/544090e3cc754f1d99f06b9831c4be6f : SUCCESS in 9m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/721edef60d574b78a09552ca674c381a : SUCCESS in 11m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/60ca2bcb39264beea9d37e9641215e86 : SUCCESS in 5m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bdea156f97bb4a32888a4be321775ffd : SUCCESS in 5m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/839254549c804697accf0870c11a3bf3 : SUCCESS in 28m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b93a25cfc84e43f7a6b443ba438f06c2 : SUCCESS in 37m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/142addca94fd443d84a38fcbe0ad3b61 : SUCCESS in 1h 08m 52s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/98a994e32b86463fa2b4c8274df019a7 : FAILURE in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/646158908e0d414a9150e7b75b268556 : FAILURE in 1h 37m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3e82e2d0e4e4009bc05b6ec3b67bea9 : SUCCESS in 45m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eeb31827afaf4778957f3db2a8680e0e : FAILURE in 1h 26m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/02973fae29dc4e52accbfef0d7f42d24 : SUCCESS in 35m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b4edea4abdcf4dfaa22d0f5d5e138f5c : SUCCESS in 30m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e9556da15184af7b995275d61f4127e : SUCCESS in 1h 01m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/80dc2df8a1704ada93015f3518a0819d : FAILURE in 57m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7528fafbf3854a20acd6db140d00d3bf : SUCCESS in 41m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/da1e8b1056aa4feaa27dde17a1d84b40 : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"853c6085b52fca828d946eca5c2f8833fd6de353","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-07-23 13:59:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/17/802017/1/upstream-check/manila-cDOT-ss/a35f278/ : NOT_REGISTERED\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/17/802017/1/upstream-check/manila-cDOT-no-ss/f7bf5be/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"b96b57272e47ba40423bcd9743b2a8d1497178d8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-07-23 18:08:45.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-17-802017-1/ns5_manila : SUCCESS in 1h 58m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d01b57c88797d636e9166f237bdb118d9677a55a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-07-23 19:19:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you! you\u0027ve a keen eye!","accounts_in_message":[],"_revision_number":1},{"id":"ad63db13c0cb750c3f08d5ab047659205f5c2b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 19:20:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a773dbeb77496ed2001c08b44966ff888bcaa35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:38:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3e86f1fae449eb9eef2f078fa4671e : SUCCESS in 10m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35bb3a8e102f44abb0fe5e60bc7c30c2 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbdd522128948a2b5e93b249c3af57b : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eba8109f278f4f2d8841909acfc1fef7 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/075c1f51d80a4852a09c43a3d1424b92 : SUCCESS in 10m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7cb747678da04e2582ef10fb6e1292ad : SUCCESS in 29m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/38c59d5236b4419c8cea1b3bb74f8d65 : SUCCESS in 36m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/429ad6fe62934299a0462080eff07389 : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"3ffa9d1ff5cbc82335e36d3934314b6c87193265","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:38:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84a6596fb6a6ad250a0567c08c56042adc2e9c2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 20:39:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a48955d1e5043c28a77c4954189fd6c : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87382e141761519f8ad3cc2e2bc8300b31fccaef","revisions":{"87382e141761519f8ad3cc2e2bc8300b31fccaef":{"kind":"REWORK","_number":1,"created":"2021-07-23 12:21:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/802017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/17/802017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/17/802017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/17/802017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/17/802017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/17/802017/1"}}},"commit":{"parents":[{"commit":"82ea5d568001b20334ab3b976ccab7d7337432e7","subject":"Merge \"Replace deprecated import of ABCs from collections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/82ea5d568001b20334ab3b976ccab7d7337432e7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-23 12:20:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-23 12:20:00.000000000","tz":540},"subject":"Replace deprecated scheduler filter module name","message":"Replace deprecated scheduler filter module name\n\nmanila.scheduler.filter_scheduler.*Scheduler was deprecated long ago[1]\nin favor of new manila.scheduler.drivers.filter.*Scheduler.\n\n[1] 3ffb4979f328259793079a5ece6f3b79e23f5187\n\nChange-Id: I40349cbe106f688a67043a73f8563f87f069831b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87382e141761519f8ad3cc2e2bc8300b31fccaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/87382e141761519f8ad3cc2e2bc8300b31fccaef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
