)]}'
{"id":"openstack%2Fmanila~806764","triplet_id":"openstack%2Fmanila~master~I34020da6a929f5c7ae1c63bc0f2ef338c49dbe10","project":"openstack/manila","branch":"master","topic":"bug/1942124","hashtags":[],"change_id":"I34020da6a929f5c7ae1c63bc0f2ef338c49dbe10","subject":"[NetApp] Fix list of mandatory services for CIFS share creation","status":"MERGED","created":"2021-08-31 15:26:44.000000000","updated":"2021-09-02 05:05:32.000000000","submitted":"2021-08-31 22:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806764-bug/1942124","meta_rev_id":"55f72315e1ef4601724afe4f206f463437aae63c","_number":806764,"virtual_id_number":806764,"owner":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 22:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-08-31 17:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-08-31 21:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-08-31 21:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-08-31 21:08:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 17:03:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-08-31 17:19:46.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 17:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 17:44:15.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-08-31 18:51:29.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-31 21:08:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-08-31 21:08:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2021-08-31 21:09:29.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-08-31 21:30:01.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-02 04:20:28.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"886c4785a18749cbd27f596ff80dec4df2ccb202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2021-08-31 15:26:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1483e2bdfe49449a97212224304699cf5ee6f95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2021-08-31 15:30:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17918e770cfafc8b9fb43f3499aadacfbf992d7b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-08-31 17:03:45.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/64/806764/2/thirdparty-check/pure-devstack-manila-tempest-aio/b79105f/ : FAILURE in 1h 29m 52s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/64/806764/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6b6c7a9/ : FAILURE in 54m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a6346dedd066e82df5e29b9f99bdcbb987afd531","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-31 17:19:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/64/806764/2/check/DellEMC_PowerMAX_Manila/556e2d0/DellEMC_PowerMAX_Manila/231 : SUCCESS in 1h 48m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"3c9080b8522bd0944221cd18ddd443ed363faad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 17:39:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fba0c8858c546e784d1a6b22deef04c : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d757f05763045e48003210a74b4ce52 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8f2b13a853f4814aa4bef8bba094525 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240fa510f4404be68fa1c90c40034720 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f6cb141895b42c7a50bc89b8078b9a9 : SUCCESS in 12m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/439f9a4799854a7f9667bb658b147d7c : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/241691b659ae4d9880871dfae6fa4ec0 : SUCCESS in 10m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/57534d6059434d2585723c86aa6b44e2 : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4a11f9394ce24e7fb4cd871d1f3d8e97 : SUCCESS in 6m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ebe05981509f44a98430872d74e51daa : SUCCESS in 34m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c172c516495b49009a6d1844c8f5fc5e : SUCCESS in 36m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5da7d3b925414f84bc25af1192a75948 : SUCCESS in 1h 01m 55s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f826ec3906614e8b9c4d53ccfd2b116c : SUCCESS in 1h 17m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2c537344b5545929b348a990e4d5969 : FAILURE in 1h 20m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/952d68819d8043af97069a6e67c06b8a : SUCCESS in 38m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b89650efe2b84333bae9dcb0f8e7a0f7 : FAILURE in 2h 03m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/daa9b136800842ed85f01603096c0dde : SUCCESS in 32m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6111812714294c27a17bda14b5ffb68a : SUCCESS in 24m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e428781829c14258b5565140d53f9a16 : SUCCESS in 1h 16m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/063c1c520f8b463b8c53ab4a76996403 : SUCCESS in 1h 00m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/49bcd4d0305443ba9b922df33333a68e : SUCCESS in 36m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1ed6fd9523ae4f40b6836020021d56fb : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f875acaa01faca56ffae4a6cc3638a91a07edc5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-08-31 17:44:15.000000000","message":"Patch Set 2: Code-Review+2\n\nNice catch, Nahim! Let\u0027s wait on NetApp CI output","accounts_in_message":[],"_revision_number":2},{"id":"b76d2815560f9092548ffcf05764e123a89026e6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-31 18:51:29.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/64/806764/2/check/DellEMC_VNX_Manila/d2ff319/DellEMC_VNX_Manila/154 : SUCCESS in 1h 11m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b74c028f36dac658398572793ac20921a21430a5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-31 21:08:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0acf71b51dd44b33a5b524b4d8d0e14a71cf987d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 21:09:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4428acd2bdeab72f3a1c6a909ec615af1f0dc28","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-31 21:09:29.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":2},{"id":"5d26b526fd6eb5c094120b1602f81c70a493b5ef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-31 21:09:54.000000000","message":"Patch Set 2:\n\nThanks for the fix, I missed the point that this was a mandatory \"and\" check - i.e., all of the items in the list of security service types will be required. Makes sense to limit to \"active_directory\" for CIFS","accounts_in_message":[],"_revision_number":2},{"id":"aadc17bf1aafa32b11ed810ea52e824c1f512553","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-31 21:30:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/64/806764/2/upstream-check/manila-cDOT-ss/7c5123b/ : SUCCESS in 3h 35m 14s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/64/806764/2/upstream-check/manila-cDOT-no-ss/1436699/ : FAILURE in 3h 49m 36s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d47023f77382042525edf8cc53a16b29b95dc77f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 22:26:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f247a2203ae940bd85658ac94533b4fa : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a35b79da5b0a456b9cdd866c9036d656 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13ab8f65e2b44789e768dce02daa2eb : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddf7e19bc7c747d1ab004ffb688fc56c : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1193224248447b183f1fa485856fb0a : SUCCESS in 12m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3f4cbe8cca624ab4a37bf0e21a70d0e0 : SUCCESS in 25m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e0844208f25d43309f3960266b1dbbed : SUCCESS in 37m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6759f85e13cc4dbead492e80cbd4dec3 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e10edc1f966e440f9cba5118a59331faea6c3c8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 22:26:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f1e04052bc5553e835282e654ec1a9a8bdaf1a20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 22:28:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d56432b21c84aa69a94c584aaff1f17 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2},{"id":"eab587af55964dbd518f7ada5b5b3aa21e61f361","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-08-31 22:30:42.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/64/806764/2/check/DellEMC_PowerMAX_Manila/3388966/DellEMC_PowerMAX_Manila/234 : SUCCESS in 1h 44m 57s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"10355181ff5e5a4dc7641108b9047631c11ce69c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-02 04:20:28.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-806764-1/ns5_manila : FAILURE in 17m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"55f72315e1ef4601724afe4f206f463437aae63c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-09-02 05:05:32.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-64-806764-2/ns5_manila : FAILURE in 17m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58031e369305337475aa20efb060b3061a5681cb","revisions":{"531ad8c12ccd93eeda1edba6b4ecfc323452c177":{"kind":"REWORK","_number":1,"created":"2021-08-31 15:26:44.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/64/806764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/806764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/806764/1"}}},"commit":{"parents":[{"commit":"b50a0bb9deb6842e45b40f30d56f18b31c9fdd03","subject":"Merge \"Add config option reserved_share_from_snapshot_percentage.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b50a0bb9deb6842e45b40f30d56f18b31c9fdd03"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-08-31 14:52:11.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-08-31 15:20:34.000000000","tz":0},"subject":"[NetApp] Fix list of mandatory services for CIFS share creation","message":"[NetApp] Fix list of mandatory services for CIFS share creation\n\nTo create a CIFS share, it is mandatory to configure an active\ndirectory service, but the code was requiring to configure also\nthe `ldap` and `kerberos` services to be configured during the\ncreation of a CIFS share.\n\nThis patch fixes this, removing `ldap` and `kerberos` from the\nlist of mandatory security services.\n\nChange-Id: I34020da6a929f5c7ae1c63bc0f2ef338c49dbe10\nCloses-bug: #1942124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/531ad8c12ccd93eeda1edba6b4ecfc323452c177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/531ad8c12ccd93eeda1edba6b4ecfc323452c177"}]},"branch":"refs/heads/master"},"58031e369305337475aa20efb060b3061a5681cb":{"kind":"REWORK","_number":2,"created":"2021-08-31 15:30:44.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/64/806764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/806764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/806764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/806764/2"}}},"commit":{"parents":[{"commit":"b50a0bb9deb6842e45b40f30d56f18b31c9fdd03","subject":"Merge \"Add config option reserved_share_from_snapshot_percentage.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b50a0bb9deb6842e45b40f30d56f18b31c9fdd03"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-08-31 14:52:11.000000000","tz":0},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2021-08-31 15:29:39.000000000","tz":0},"subject":"[NetApp] Fix list of mandatory services for CIFS share creation","message":"[NetApp] Fix list of mandatory services for CIFS share creation\n\nTo create a CIFS share, it is mandatory to configure an active\ndirectory service, but the code was requiring to configure also\nthe `ldap` and `kerberos` services to be configured during the\ncreation of a CIFS share.\n\nThis patch fixes this, removing `ldap` and `kerberos` from the\nlist of mandatory security services.\n\nChange-Id: I34020da6a929f5c7ae1c63bc0f2ef338c49dbe10\nCloses-bug: #1942124\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58031e369305337475aa20efb060b3061a5681cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/58031e369305337475aa20efb060b3061a5681cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
