)]}'
{"id":"openstack%2Fmanila~812213","triplet_id":"openstack%2Fmanila~master~I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-16 16:02:36.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-09-16 16:02:36.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-12-14 18:04:58.000000000","reason":"Goutham Pacha Ravi replied on the change"}},"hashtags":[],"change_id":"I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a","subject":"[api-ref] Fix missing parameters in share network list","status":"MERGED","created":"2021-10-01 19:11:59.000000000","updated":"2022-09-17 00:41:32.000000000","submitted":"2022-09-16 16:02:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"812213","meta_rev_id":"414345546659c267bfae096a11625a7237b8cb58","_number":812213,"virtual_id_number":812213,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 16:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-03-23 19:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-09-16 15:52:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-01 20:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-01 22:13:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-01 22:31:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-02 01:22:18.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-08 22:57:13.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-12-16 11:44:33.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-02-01 14:05:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-03-03 22:00:10.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"63a4b51a486b948d0e43a1f9d571a2e765a921ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-01 19:11:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0ffcdc8b2fd80601cd62287cda3ceba3823ec87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/54c576b032f74344b271e36733f5f967 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1644bef04fb426dab0e2efe6e8368d4 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76eb6927778c4f1c9c1580ebd19eef5a : SUCCESS in 6m 34s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b213b294970f4fd5a47f896de366ae0b : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c0d5230830282ca0488f04dc725720ef772af092","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-01 20:14:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4512ed9c7eedd7edfcc48e03d9e4c916524d6bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 20:25:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bd397ba97d9544728bedb1402a346f92 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1beec353cf444eeb77a1388a93fd11f : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9676f91602cf49d3a6a634a9217a505a : SUCCESS in 5m 47s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d9adf76ea1d4e42a28fbcaf7e7f6e94 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1591b99e426eda7b14a4bcd29f178d9541fe0d9d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-01 22:13:47.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/13/812213/1/check/DellEMC_VNX_Manila/f43ee3e/DellEMC_VNX_Manila/219 : SUCCESS in 2h 11m 12s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"8e95da59ad68382ad8ef1969633add56a9241627","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-01 22:31:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/812213/1/check/DellEMC_Unity_Manila_dhss_true/834de6c/DellEMC_Unity_Manila_dhss_true/401 : FAILURE in 2h 11m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/812213/1/check/DellEMC_Unity_Manila_dhss_false/595c3b5/DellEMC_Unity_Manila_dhss_false/394 : SUCCESS in 1h 26m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e57af2980a56566e32afb6a6ba417ff4d740ceda","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-02 01:22:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-1/ns5_manila : SUCCESS in 2h 22m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"8de1a514c65af7537e01525b53a298fbc60efacb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-15 19:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-1/ns5_manila : SUCCESS in 1h 46m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"7777e5133e8bb6b8057d52d52096465c94429637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-08 22:30:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf9f36da17f38cbb470849651719e0cedadd1f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 22:40:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dd1a3754871c45efae66531265c2a6fa : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64e872ce92e4d6cb4ac82d97f1ad648 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232d32ea07bb476889e118cfef8714e0 : SUCCESS in 5m 11s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6401c9d9ba6c424ea84c6f516b6604c7 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"9f77eef177b7e999a06d8497d4d8887efeb04f8e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-08 22:57:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"be5036aaa8d52da4c4b8efae8ea56c8bd783e546","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-08 23:09:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b860d27de001420eba6d82d5c747d7a18d9a02c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-09 00:56:28.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-2/ns5_manila : SUCCESS in 1h 42m 9s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e344939d1c3b3a7f6b57dc1e3cada259836e7b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-13 19:59:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf3fad62d0e5c3d14b19b25c16391a6b53b22406","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-13 19:59:50.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"42d85bcd3513cd94b4ae6aa98c570e7bad540528","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-13 20:11:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"831b6d2754dfc0f9f3e9d927407f7dfbc048e1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 20:15:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/436885974fb84c608f7b00d7039b9987 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03a2c8f47f6457d9f2e47b7293f8fd9 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ed0b6b9f434e7c861e51c0ed3c5187 : SUCCESS in 6m 24s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f070169f88e4e70992d81e99bb935f2 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2db2b86f26bb4c7c774e42306684abbf971e1056","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-13 21:45:00.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-3/ns5_manila : SUCCESS in 1h 45m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"1d13ca9dd6e271d8736c38db485bf4454e0a3fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-14 17:57:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e3fb43b6a2a7a21b8612eab5aa38f035a7a003d4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-14 17:58:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e4873874fcc2cdc6c49b87dc434faba688ea7ab4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-14 18:04:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c0910df035a174fb119de267f0ebd6005f7ebfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 19:01:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/49c5d6bb19f84145938f2f525fbbd343 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e038e0151b4529978825c222f0cae6 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef7cd10325b4bd095621ad036ec6048 : SUCCESS in 6m 26s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/64b7495bedc94a6484081a4081b8f9c2 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":4},{"id":"bc602dc98cf1b0af99768ce0b5df3d4a82cdb35a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-14 23:43:37.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-4/ns5_manila : SUCCESS in 2h 8m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"0be19446e96d55c2999f60ce5986505f23d7d56f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 11:44:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"921e2fee0b319dd709c735ad2ea632f20c37d446","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-02 18:24:32.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0e4dd94fa8751026d58f995f0b79c3af4d864f1e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-02 18:39:08.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"79dc42052eada20d69f2595c9353dcaeff69165e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-02 18:50:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c5ee6d07d612924b5b4c3b323af68f73442639c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-02 19:24:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24187d2e9f55674750a1a1fe6d69b4d5b9be93bb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-02 19:24:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"96a60cdc243f7991bb95bfe7b6422cb33fa5fdd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 19:37:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f2ba6c1e84f64553b74eb055bfc50dce : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679f161f1e5747d8820aeab01372a754 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533122bb15874740a791013668a282ac : SUCCESS in 6m 55s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f259b6fb7924029a69e12f06ecb50c3 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":5},{"id":"d9c98b186687c1fd0c1ea78888f19032e08d4cf6","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-02 19:55:42.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"eee50422174da1ee0ba155dc155eaa2c6f5cc5c0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-02 23:29:13.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-5/ns5_manila : FAILURE in 1h 12m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"3acd93d50a48326517c64dab0e463e17ac505b8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-03 20:03:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9e16f35714bc14a0f46f9965b2bfb093d82074a8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-03 20:04:11.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cea983037f0ccea3bcabc69df5c15ffddf55c36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 20:21:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/96c0fb79bf704c4682196724fa6ba863 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f70c4ee345b489cb0dfa8771c932e98 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4ff5a3af7643bfaaf2167481055105 : SUCCESS in 6m 17s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/91a6e584a2d64e73825fdbb32cf1eb42 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":6},{"id":"7fce3aa285d0b641751c9fd2d410fcc319395cd8","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-03-03 22:00:10.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/13/812213/6/check/dsvm-tempest-manila-zadara-driver/355fcd8 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":6},{"id":"e717b635f543992521a6551ecdbc4139b4f16cbc","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-03 23:53:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f28dae5e710517aa60b8fed89eae5cf485bb38f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-04 00:08:48.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-812213-6/ns5_manila : FAILURE in 1h 9m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"b7cf5ff2a32e88789cca8f45adf64f7a0201f042","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-18 14:15:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"810a9bc6df324470e94cac9b5e1253ba7f24ec27","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-23 19:41:43.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ad7adebf197fa80e3047bc32c678975d93ea6cc1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-16 15:10:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2bf8805e4d0453aafca1bed79198100fc8b55816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:28:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ee3def6a8b4c34b2ef93e35db94423\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ed67d1f5f4414ac09021000d6e890eb6 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdde4825c3849f39e0c21b6310cf09a : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d1fa540a27149d6b9d310467ffda946 : SUCCESS in 6m 01s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/185cef67012f4b7fb0e70595adca17b3 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":6},{"id":"f52a77570d5a542d17fdfb502b8784470e6ddb64","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-16 15:52:15.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ca11e99ac7d91becb604cf7ef5c7aa615f469386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:52:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0775bb640bad0975f3f467bcd71b2573fd61ce2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:02:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c3d40819e1428593fb650154b671f5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2d0f18e8ab0f4a5fb18044f624db2034 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b640fe2feb410ebf4425c3793dc756 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63391609a3ae4bb5a79cdf9060093fd7 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":6},{"id":"a75bbe614d2ecd13828356817feb251600ec564a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:02:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6f42d138d714add3021bbcd562075dbc29424850","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:04:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9b6db8f26f447783700c3cd3f464d2\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7b65c129a6c84a018fc6320041735802 : SUCCESS in 57s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e902be9c2206445cb17c58b4b52c6ba1 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":6},{"id":"414345546659c267bfae096a11625a7237b8cb58","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-17 00:41:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/812213/6/check/DellEMC_Unity_Manila_dhss_true/280250a/DellEMC_Unity_Manila_dhss_true/877 : FAILURE in 2h 53m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/812213/6/check/DellEMC_Unity_Manila_dhss_false/bae28a5/DellEMC_Unity_Manila_dhss_false/933 : SUCCESS in 1h 01m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b82175f30d50a8a88ffaeafff6d47f322ee4f43c","revisions":{"63a3cbbf26a211fa1941456d5efd7c29cc6d2a27":{"kind":"REWORK","_number":1,"created":"2021-10-01 19:11:59.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/1"}}},"commit":{"parents":[{"commit":"33b6d8a7e49f3ea833694c1f44adcd1024903aec","subject":"Merge \"NetApp: properly use netapp_server_port config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/33b6d8a7e49f3ea833694c1f44adcd1024903aec"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/63a3cbbf26a211fa1941456d5efd7c29cc6d2a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/63a3cbbf26a211fa1941456d5efd7c29cc6d2a27"}]},"branch":"refs/heads/master"},"fe0db5925596d889ba1d3a4bd1b5024cc146d3fc":{"kind":"REWORK","_number":2,"created":"2021-12-08 22:30:03.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/2"}}},"commit":{"parents":[{"commit":"c09a5f08c45c9112b406dd171534cfb099e25c82","subject":"Merge \"Add api-ref for security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c09a5f08c45c9112b406dd171534cfb099e25c82"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-12-08 22:29:51.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fe0db5925596d889ba1d3a4bd1b5024cc146d3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fe0db5925596d889ba1d3a4bd1b5024cc146d3fc"}]},"branch":"refs/heads/master"},"57452a1b38d84fe8d465fdeda53e41a9983d23b3":{"kind":"REWORK","_number":3,"created":"2021-12-13 19:59:12.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/3"}}},"commit":{"parents":[{"commit":"c09a5f08c45c9112b406dd171534cfb099e25c82","subject":"Merge \"Add api-ref for security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c09a5f08c45c9112b406dd171534cfb099e25c82"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-12-13 19:58:55.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/57452a1b38d84fe8d465fdeda53e41a9983d23b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/57452a1b38d84fe8d465fdeda53e41a9983d23b3"}]},"branch":"refs/heads/master"},"6f2add1e234446fac2e299b8a77c74384f0e6028":{"kind":"REWORK","_number":4,"created":"2021-12-14 17:57:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/4"}}},"commit":{"parents":[{"commit":"c09a5f08c45c9112b406dd171534cfb099e25c82","subject":"Merge \"Add api-ref for security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c09a5f08c45c9112b406dd171534cfb099e25c82"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-12-14 17:57:38.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f2add1e234446fac2e299b8a77c74384f0e6028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f2add1e234446fac2e299b8a77c74384f0e6028"}]},"branch":"refs/heads/master"},"b8aab06b1adcdd57912004c11a2cb0878d41f24a":{"kind":"REWORK","_number":5,"created":"2022-03-02 19:24:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/5"}}},"commit":{"parents":[{"commit":"c09a5f08c45c9112b406dd171534cfb099e25c82","subject":"Merge \"Add api-ref for security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c09a5f08c45c9112b406dd171534cfb099e25c82"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-02 19:24:00.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b8aab06b1adcdd57912004c11a2cb0878d41f24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b8aab06b1adcdd57912004c11a2cb0878d41f24a"}]},"branch":"refs/heads/master"},"b82175f30d50a8a88ffaeafff6d47f322ee4f43c":{"kind":"REWORK","_number":6,"created":"2022-03-03 20:03:48.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/13/812213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/812213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/812213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/812213/6"}}},"commit":{"parents":[{"commit":"c09a5f08c45c9112b406dd171534cfb099e25c82","subject":"Merge \"Add api-ref for security service update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c09a5f08c45c9112b406dd171534cfb099e25c82"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-01 19:05:35.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-03 20:02:17.000000000","tz":-180},"subject":"[api-ref] Fix missing parameters in share network list","message":"[api-ref] Fix missing parameters in share network list\n\nShare network list API referece was missing some filter parametes.\nAdd the missing parameters so users can be aware of all the options\navailable to query share networks.\n\nCloses-Bug: #1945711\nChange-Id: I21b4f48d0c41f052aca1d7dcd0f1b640cfa3084a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b82175f30d50a8a88ffaeafff6d47f322ee4f43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b82175f30d50a8a88ffaeafff6d47f322ee4f43c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
