)]}'
{"id":"openstack%2Fmanila~815539","triplet_id":"openstack%2Fmanila~master~I9b23a47801abc8914cb406292780e423591e0573","project":"openstack/manila","branch":"master","topic":"metadata-apis","hashtags":[],"change_id":"I9b23a47801abc8914cb406292780e423591e0573","subject":"WIP: Metadata for User facing resources","status":"ABANDONED","created":"2021-10-26 18:33:34.000000000","updated":"2021-10-26 19:45:31.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23f12381417ebeb36ef425d7557db7b6d979325e","_number":815539,"virtual_id_number":815539,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 19:45:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"cbecd71fe181e1997f6e4e22fd8f21c1164b8fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-10-26 18:33:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"732dd6df18cb1fc7f30a5e99857d28a4601596c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-10-26 18:34:20.000000000","message":"Abandoned\n\nduplicate","accounts_in_message":[],"_revision_number":1},{"id":"23f12381417ebeb36ef425d7557db7b6d979325e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-10-26 19:45:31.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-39-815539-1/ns5_manila : FAILURE in 1h 3m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f248edbc3025c1aa4fd16be72119046a99a0f9a","revisions":{"4f248edbc3025c1aa4fd16be72119046a99a0f9a":{"kind":"REWORK","_number":1,"created":"2021-10-26 18:33:34.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/39/815539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/815539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/815539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/815539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/815539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/815539/1"}}},"commit":{"parents":[{"commit":"a659470852f920621e5473dc323751039a24da87","subject":"Merge \"Adapt CephFS driver to do not try to escape export ip\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a659470852f920621e5473dc323751039a24da87"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2021-10-26 18:10:48.000000000","tz":-240},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2021-10-26 18:25:17.000000000","tz":-240},"subject":"WIP: Metadata for User facing resources","message":"WIP: Metadata for User facing resources\n\nThis change adds resource metadata tables and\nAPIs for the following resources:\n\n- Shares\n- Share Export Locations\n- Share Instances\n- Share Instance Export Locations\n- Share Replicas\n- Share Replica Export Locations\n- Snapshots\n- Share Groups\n- Share Group Snapshots\n- Security Services\n- Share Networks\n- Share Network Subnets\n\nDocImpact\nAPIImpact\nPartially-implements: bp/metadata-for-share-resources\nChange-Id: I2e7d6edafc929e1c63425f6f4be02335fc263a34\n\nChange-Id: I9b23a47801abc8914cb406292780e423591e0573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f248edbc3025c1aa4fd16be72119046a99a0f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4f248edbc3025c1aa4fd16be72119046a99a0f9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
