)]}'
{"id":"openstack%2Fmanila~817549","triplet_id":"openstack%2Fmanila~stable%2Fxena~Ia99cc5313bed950e3363ae311b741395d60eb57b","project":"openstack/manila","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-11-12 17:34:35.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2021-11-12 05:18:15.000000000","reason":"yatin replied on the change"}},"hashtags":[],"change_id":"Ia99cc5313bed950e3363ae311b741395d60eb57b","subject":"Drop non-ASCII character from manila config","status":"MERGED","created":"2021-11-11 23:16:54.000000000","updated":"2021-11-12 20:01:23.000000000","submitted":"2021-11-12 17:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817549","cherry_pick_of_change":817565,"cherry_pick_of_patch_set":1,"meta_rev_id":"6a0f616bf0cc137d0fd66a05d88776253c4bc17d","_number":817549,"virtual_id_number":817549,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-12 17:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2021-11-12 12:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2021-11-12 05:18:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2021-11-12 12:13:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-11 23:16:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-11-12 00:09:15.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-11-12 01:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-12 03:55:53.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-12 07:48:25.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-12 12:13:58.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"e74cdfa2bff8bd9907ef24bacbfddb2b87632066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-11-11 23:16:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1d4594d9884e22868439bd045fbf0dba7985497","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-11-12 00:09:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/817549/1/thirdparty-check/pure-devstack-manila-tempest-aio/e520b1d/ : FAILURE in 51m 03s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/49/817549/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/0b04cb3/ : FAILURE in 39m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0bcb2949dddbdfad63afdc67e8fcd5088e99f91c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 01:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e811f6e0804d9b85ac6abb24cc34f5 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3040a018f9ca4f6b82cfdcea6298965d : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cee1d2fe7a547a4b2ce04535d395e07 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f6bf3226c14a4cb7a61afc25d8a1b9 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc4cb036d0d447688b9e1c5d1528990 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303abe8f93cc4e478cc30485e3fd3994 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7d0f8d2c808417abb18600dac3f82ad : SUCCESS in 8m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e06383520a14cfe9d8f3ab1be2ece35 : SUCCESS in 6m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6fbbb8de9fe84cd7b46875f94d9c3ba5 : SUCCESS in 6m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/67f48d57494e4a0b9b12cff8332a1953 : SUCCESS in 28m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/754db5f4359f4a4d9bb4c42d9996e0ae : SUCCESS in 34m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cdf4e8f701b744d486dbc864bbcbbe1a : SUCCESS in 1h 01m 29s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6db69306a44c46f7b1909c653c18a423 : FAILURE in 2h 00m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a5e89adf057546628481afc98efefcad : FAILURE in 1h 25m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d6d098577ce540d6bc8d0d91a8c3afcf : SUCCESS in 32m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d3ade38b2c6f4e3eb68d0d8564bd55f1 : FAILURE in 1h 39m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a21491ce0ae14c07b58f159a18cc650e : SUCCESS in 23m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/27082645e0564b1582a9fc8fef0c6770 : SUCCESS in 27m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6ee18426e7db433e99a98c10aff434e4 : SUCCESS in 58m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a373b043f00c4188a60404146c244104 : SUCCESS in 1h 03m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5afcb18200164ed6803130c3a9bc481d : SUCCESS in 25m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/42dd4435ee9342f68012a2f2f98d8152 : SUCCESS in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38d1034f4be62dc8cf40afc36aada835bc0d0cf7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-12 03:55:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/817549/1/upstream-check/manila-cDOT-ss/485ed35/ : SUCCESS in 1h 07m 54s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/817549/1/upstream-check/manila-cDOT-no-ss/ad1c2fe/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3190c05659d1695c00cb8cdd929e9b913f11ff1d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-11-12 05:18:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c91648c6096da5ef001fa6b4ff69de79009a765","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-11-12 07:48:25.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-49-817549-1/ns5_manila : SUCCESS in 1h 48m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"01ea077971d14df1fcf5983462c8650839b1908f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-11-12 12:13:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75248525532cc30c27a5432983c3e1b2bb834c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:14:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ddecff59e043ce71c73e8019294c0f13ae5cd24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 13:36:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40266e72c07042669c5ffc74cf15e084 : SUCCESS in 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2d146eeabe448f4b3fab0f08b6af324 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5764680991af4fb1b78517f045507bfd : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c2caa33ac34e109a229426f95fbcbd : SUCCESS in 9m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b66af409a4c4e1ab1635bf4b5fde9ae : FAILURE in 19m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3209b875e3754cb9a76969943d401dc4 : SUCCESS in 36m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/35ec84f8c75f49279daa0c4c28de910b : SUCCESS in 42m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dfa67f768b31440cb175af66ddb2171c : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7e3edcec1ab0376ce9944085724a2cc0d048c13f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-11-12 13:43:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbbdb1f566bcfa6b368a87949afeffc5b812ccec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 15:48:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97deedaff4348bebf7a965fad941a1b : SUCCESS in 13m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21ee5ebb08da4bcdb4c2b42c66915f00 : SUCCESS in 23m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da813396b3754c6ab1594a3b83b783ff : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acaf4e7bcb444a5aa0ecb3da21a9098f : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23095da8086540d6a90dd655d6a2d38f : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f3fad12b4364660b16558b3bbe87227 : SUCCESS in 12m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d4b496dc10f415a87771a2ea1ea823c : SUCCESS in 17m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1b1a5934e6c64422a7488f4ef7861b84 : SUCCESS in 8m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e817537cfe44aed8c46e132e60876f9 : SUCCESS in 5m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ca75fbd7ca164db3ae1f09d4f03dc366 : SUCCESS in 33m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/73ef601a98ec414992292f8d4f9af7d9 : SUCCESS in 35m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f4eaa3a6623e4db8bf3961537bdf1980 : SUCCESS in 1h 05m 10s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d058f3de6173479d8cab5aaadbf9d9ff : SUCCESS in 1h 14m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5d6a4c883b6b4d068081e4aee7b8429c : FAILURE in 1h 52m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/10033560d4fc47e3bd2b3f17d9852193 : SUCCESS in 49m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/01bf55e862b745a9a07d4037a8587b37 : FAILURE in 1h 47m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f5a3bc78a13a4fdca444b2c62637616f : FAILURE in 12m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/78bcdf6e84354702813d93a002786c2a : SUCCESS in 23m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f5ea945aad3f439f800ed507f0fb8a3d : SUCCESS in 1h 08m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/03bac1d6e1064a3fbc12c396cc005f1d : SUCCESS in 1h 09m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/335a7d62ccd244f4a65f958dc6f6a045 : SUCCESS in 38m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6a38f134fd014fa7aaed61af4dc4d356 : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7111f960c80a3a6e9bb5fd96dd33d8b5c852a738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 15:49:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d96511c998da6f739d3f249fd6befce346483d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:34:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7be2f5286547e9a51d568e1a35d5c9 : SUCCESS in 13m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/253956ff4e56483db4a73494480fc5a1 : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d17056f65e4da5b9bd6246b4465e79 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adb6aff491bd42e2a56b8cfb297dae9a : SUCCESS in 17m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3e3a4b573ae4260b0cc5e626ee92711 : SUCCESS in 12m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c3067d1a50004552a7781ac720fe0085 : SUCCESS in 33m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/150bb9af9c5f4e6cb5ef030fa40c16a2 : SUCCESS in 39m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/04a2dba2ad80427e9ddfeb84535040dc : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"4cd3d5dea03fdaf157f3a29ff75e34117962ea0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:34:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e2cc454c56d15f6845a754f1bec8fdcee40a9fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:38:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1026822b4b84a378ca0a98cd096cf77 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6a0f616bf0cc137d0fd66a05d88776253c4bc17d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-12 20:01:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/49/817549/1/upstream-check/manila-cDOT-ss/055f5fe/ : SUCCESS in 2h 53m 48s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/49/817549/1/upstream-check/manila-cDOT-no-ss/abbf8ee/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fb10e6c5db11f2a980819db476d3fda80f521a2","revisions":{"0fb10e6c5db11f2a980819db476d3fda80f521a2":{"kind":"REWORK","_number":1,"created":"2021-11-11 23:16:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/817549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/817549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/817549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/817549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/817549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/817549/1"}}},"commit":{"parents":[{"commit":"a1f8f4b293d6bb656109367f3531faea5f322402","subject":"Adapt CephFS driver to do not try to escape export ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1f8f4b293d6bb656109367f3531faea5f322402"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2021-11-11 14:06:25.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-11-11 23:16:54.000000000","tz":0},"subject":"Drop non-ASCII character from manila config","message":"Drop non-ASCII character from manila config\n\nNoticed the issue with packstack deployment\non CentOS 9-Stream[1]. The issue is raised\nby puppet when non-utf8 encoding is used like\nLANG\u003den_US or LANG\u003dC.\n\nOther option is to force LANG\u003den_US.UTF-8\nwith the deployment but just for this\ndescription help(all other services working fine),\ndoesn\u0027t look good so proposing to drop that non-ASCII\ncharacter.\n\nIssue got caught in CentOS 9-Stream only as it\ndisabled locale forwarding in OpenSSH by default[2].\n\n[1] https://review.opendev.org/c/x/packstack/+/816796\n[2] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2002734\n\nChange-Id: Ia99cc5313bed950e3363ae311b741395d60eb57b\n(cherry picked from commit 5bf206285de3aa61b56ee6c83ffd36e9c1617860)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0fb10e6c5db11f2a980819db476d3fda80f521a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0fb10e6c5db11f2a980819db476d3fda80f521a2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
