)]}'
{"id":"openstack%2Fmanila~820197","triplet_id":"openstack%2Fmanila~master~I0be32487c880ee811172f39106e4bcefdec2b4e4","project":"openstack/manila","branch":"master","topic":"scope-config-opts","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"9003":{"account":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2021-12-21 10:10:49.000000000","reason":"Change was abandoned"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2021-12-21 10:10:45.000000000","reason":"Bogdan Dobrelya replied on the change"}},"hashtags":[],"change_id":"I0be32487c880ee811172f39106e4bcefdec2b4e4","subject":"Set service specific rpc control_exchange defaults","status":"ABANDONED","created":"2021-12-02 16:44:24.000000000","updated":"2021-12-21 10:10:49.000000000","total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7030937548b77ffbcbdaf5085b03107f8ed38be4","_number":820197,"virtual_id_number":820197,"owner":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-08 11:23:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":-1,"date":"2021-12-12 00:27:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-02 16:44:56.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2021-12-02 16:44:56.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2021-12-02 16:45:24.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2021-12-02 16:46:26.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-12-02 17:31:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-12-02 17:50:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2021-12-02 18:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-02 19:30:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-03 07:03:08.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-12-08 12:42:06.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-08 14:15:42.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-08 14:40:20.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-09 19:53:40.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2021-12-09 19:53:40.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"CC"},{"updated":"2021-12-09 19:53:41.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"}],"messages":[{"id":"63150f94943075ba0c4315ccbbbd58f7cce54ec3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-02 16:44:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6e7c87b69a7a5c70a0321d0626772987372f1f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-02 17:31:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/97/820197/1/check/DellEMC_Unity_Manila_dhss_true/cd1991a/DellEMC_Unity_Manila_dhss_true/139 : FAILURE in 46m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/97/820197/1/check/DellEMC_Unity_Manila_dhss_false/dc17565/DellEMC_Unity_Manila_dhss_false/132 : FAILURE in 43m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"5b496ee10a1faa7cfd57111dc7db1c827d34fd7e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-02 17:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/97/820197/1/thirdparty-check/pure-devstack-manila-tempest-aio/ef10756/ : SUCCESS in 1h 04m 38s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/97/820197/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/55a0611/ : FAILURE in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"486bbbb61232476f3166b9eda5c5e58424463b1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 18:01:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/905c2ca0a6fe48f487c716a0ad8d48ac : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41968cc6ad442aba793b3110df317af : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0afd8b6e07614ba4bd480f06c0eb4cbf : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3964bfb516714c718d066183e8e3ddaa : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bdff21732324322894f7e75261fc863 : SUCCESS in 24m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec91056f18ba40b991fbd6edf4225e2f : SUCCESS in 25m 57s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a06bf69c560b4dc89443f5c2c78857ad : SUCCESS in 5m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d31e795b732842be9e5cc7dd3a78f742 : SUCCESS in 5m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2c54f6b0250145e39b5aa6822d0b00e1 : SUCCESS in 31m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e9d4049f7ecb4958b8d1c5cdd3a22d42 : SUCCESS in 38m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e8f5ebebc5ff474e946f98b9fbb4f6a7 : FAILURE in 31m 03s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dd12be6bc85e45f0b37a4baa9a52c2dd : FAILURE in 30m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5b7ec7c8bfb4403ba2851936d38b934e : FAILURE in 32m 32s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6af8382d23fb48ef805971a58829637e : FAILURE in 31m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2be1fc5562c64385ad7b495fd9c7aa6f : FAILURE in 31m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fe02f4addd104f15bb0d8af0dd65a4ee : SUCCESS in 32m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a00b98ecbce44b03851d3351f8869c92 : SUCCESS in 30m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0ab7c598c0e8483a835abb6b64259988 : FAILURE in 35m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bf0dd9da36d3471b9d2234d01fd321bf : SUCCESS in 1h 08m 29s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/46be7e77023146afac79b45038af509c : FAILURE in 32m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c8fff1bfd0a64d30993b8764988a9055 : FAILURE in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a452a723455b165371f724abdbd12933a7b331da","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-02 19:30:17.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-97-820197-1/ns5_manila : FAILURE in 1h 19m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d545ba5e8a57f4a9f801ae611579e46dd87f0b11","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-03 07:10:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d5487579758e5dbe1a7720ad8743338d8d0210e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-08 09:11:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42d6b5f0dc2f35153091e3d5cfe2c9acad99c446","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-08 09:12:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a0ced89c9f8b78ca0c5ea2c8ee123fb0c4183a9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2021-12-08 10:02:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/97/820197/2/thirdparty-check/pure-devstack-manila-tempest-aio/4b67bc9/ : FAILURE in 50m 30s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/97/820197/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/27badcb/ : FAILURE in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c3ada3690c8cdda0ca389a8cf3be4c7f3f8cfb8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2021-12-08 10:10:23.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-97-820197-2/ns5_manila : FAILURE in 58m 47s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"e1381e72602fed3e246fa45acc4b9b163285857b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-08 10:19:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6250a9fe6b754308557cd1553f934fe5717d40d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 11:23:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f834b781f384c82a84137533e95da7c : SUCCESS in 12m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53828bfba4c642978379c806e0723b9c : SUCCESS in 39m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b205d7f75aa14d1eb70acf89f74ac979 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca3d468d1db46e0a8bcbe83d2944780 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe9c91272d0d481a9ab55ac44eda3284 : SUCCESS in 17m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4accb7cd4da41aebf6b33c45a1ab202 : SUCCESS in 17m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58bbe6260bf64b8a8bca18ad2252d726 : SUCCESS in 6m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/953cc986aa1748d6a4a93147f5625c49 : SUCCESS in 5m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9e5d0827a0a94dc08fe41850d88849bb : SUCCESS in 5m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a14acc8f41b2489d84585b13fd9f17de : SUCCESS in 29m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b49208f3644140838cbe2410f01d962a : SUCCESS in 34m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4b298b9450b74422a00e1474558d03ab : SUCCESS in 1h 05m 17s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cbcd13c563cf41d5b08a90fa53b4102b : SUCCESS in 1h 33m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2ce371e696cf4a37906495a0f3c4b187 : FAILURE in 1h 49m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bf703a2317d24058b7bb803a86627def : SUCCESS in 45m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a3a91db278b8491a947f2ad8ead8d213 : FAILURE in 2h 10m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4cf8f49815cf4cf7ae35f34fb134eed4 : SUCCESS in 26m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bc4df33642f14ddd9b24efacd9005b6d : SUCCESS in 21m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/de346bc5ba0c423da59c9b347ea6f78f : SUCCESS in 1h 27m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3ddce8fe32f14cf7ba197bfdadc25fc9 : SUCCESS in 58m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3d1b9db9f66e4534a631ba0d2bf40fe1 : SUCCESS in 30m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f9c8a4df20a84c4ab6cc9b8f2b661a9d : SUCCESS in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7dc5009cf527ae1fb29c36a927fc97ed897fbf65","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-08 12:42:06.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/97/820197/2/check/DellEMC_VNX_Manila/ce6543c/DellEMC_VNX_Manila/69 : FAILURE in 1h 17m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"890d1d2577f21f599ae61eb8014b844dc1f43698","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2021-12-08 14:15:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/97/820197/2/check/DellEMC_PowerMAX_Manila/5a8025f/DellEMC_PowerMAX_Manila/47 : FAILURE in 2h 51m 31s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"04e96111af9c61c4e455ac4f3f0d18c14f4b981b","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-08 14:40:20.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-ss https://logs.openstack.netapp.com/logs/97/820197/2/upstream-check/manila-cDOT-ss/2e9a764/ : SUCCESS in 2h 00m 55s\n- manila-cDOT-no-ss https://logs.openstack.netapp.com/logs/97/820197/2/upstream-check/manila-cDOT-no-ss/9a9e8d7/ : SUCCESS in 3h 15m 52s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"7e5f45a389b342cb386b8fae1b3face90f207180","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-08 16:56:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/97/820197/2/check/DellEMC_Unity_Manila_dhss_true/7401142/DellEMC_Unity_Manila_dhss_true/154 : SUCCESS in 3h 22m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/97/820197/2/check/DellEMC_Unity_Manila_dhss_false/f8e8175/DellEMC_Unity_Manila_dhss_false/147 : FAILURE in 16m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"855fe4b4e5217ef2e3e93756cc1c016464e77cef","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-12-12 00:27:09.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7f9363da15ac461114fc892ae027913d7a904cdf","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-12-12 00:27:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"184d8b76f232214326a05ee7c195eb6a29bf486a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-15 20:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84064e60e94d7de8d3788c7a201771a0ab1d07b4","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-21 10:10:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7030937548b77ffbcbdaf5085b03107f8ed38be4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-12-21 10:10:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be61b315683cda3669d92f1a98c1c16a82cbed5d","revisions":{"4195f0b7f847244c677f4489b88eac30d1076725":{"kind":"REWORK","_number":1,"created":"2021-12-02 16:44:24.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/97/820197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/820197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/820197/1"}}},"commit":{"parents":[{"commit":"72ecaf0fab3b65789d7c7fae57510f11521f66a3","subject":"Merge \"[api-ref] Fix datatype of cast_rules_to_readonly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72ecaf0fab3b65789d7c7fae57510f11521f66a3"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2021-12-02 16:43:32.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2021-12-02 16:44:22.000000000","tz":60},"subject":"Set service specific rpc control_exchange defaults","message":"Set service specific rpc control_exchange defaults\n\nThe RPC code from Oslo registers the control_exchange option\nas a generic for openstack, but provides an API for changing it.\n\nThe generic \u0027openstack\u0027 value may conflict with other services\nthat configure the same default exchange with for its own use,\nlike enabling amqp_durable_queues.\n\nChange the default control exchange to \u0027manila\u0027, so that it could\nas well configure the control exchange for its specific use without\nconflicts.\n\nblueprint: scope-config-opts\nCloses-bug: #1953050\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I0be32487c880ee811172f39106e4bcefdec2b4e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4195f0b7f847244c677f4489b88eac30d1076725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4195f0b7f847244c677f4489b88eac30d1076725"}]},"branch":"refs/heads/master"},"be61b315683cda3669d92f1a98c1c16a82cbed5d":{"kind":"REWORK","_number":2,"created":"2021-12-08 09:11:03.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/97/820197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/97/820197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/97/820197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/97/820197/2"}}},"commit":{"parents":[{"commit":"72ecaf0fab3b65789d7c7fae57510f11521f66a3","subject":"Merge \"[api-ref] Fix datatype of cast_rules_to_readonly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72ecaf0fab3b65789d7c7fae57510f11521f66a3"}]}],"author":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2021-12-02 16:43:32.000000000","tz":60},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2021-12-08 09:10:57.000000000","tz":60},"subject":"Set service specific rpc control_exchange defaults","message":"Set service specific rpc control_exchange defaults\n\nThe RPC code from Oslo registers the control_exchange option\nas a generic for openstack, but provides an API for changing it.\n\nThe generic \u0027openstack\u0027 value may conflict with other services\nthat configure the same default exchange with for its own use,\nlike enabling amqp_durable_queues.\n\nChange the default control exchange to \u0027manila\u0027, so that it could\nas well configure the control exchange for its specific use without\nconflicts.\n\nblueprint: scope-config-opts\nCloses-bug: #1953050\n\nSigned-off-by: Bogdan Dobrelya \u003cbdobreli@redhat.com\u003e\nChange-Id: I0be32487c880ee811172f39106e4bcefdec2b4e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be61b315683cda3669d92f1a98c1c16a82cbed5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/be61b315683cda3669d92f1a98c1c16a82cbed5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
