)]}'
{"id":"openstack%2Fmanila~826513","triplet_id":"openstack%2Fmanila~master~Ifbe3c64e4268fd646a23dfa25d9fed9a374d4ac7","project":"openstack/manila","branch":"master","topic":"testing_container","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-01-26 19:31:02.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ifbe3c64e4268fd646a23dfa25d9fed9a374d4ac7","subject":"DNM - test parenting image-elements-job","status":"ABANDONED","created":"2022-01-26 19:17:37.000000000","updated":"2022-01-26 23:46:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"8c17e2917ad50af9c773c651a225806ea70c04f9","_number":826513,"virtual_id_number":826513,"owner":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-26 19:57:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-01-26 19:19:04.000000000","updated_by":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"reviewer":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"state":"CC"},{"updated":"2022-01-26 19:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-26 23:46:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3849b24b9cae7be68f80e05be0fd19f5a9aca8d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-26 19:17:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94e2bf4faea3cb9a430792e8e7189c9d115e41e5","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-01-26 19:19:04.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ee1feebe9fa70831bc41d6d1d11dfdef12361c4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-26 19:31:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c378d3b5ce5ed59e0f5e26adcb11262bdcc51f05","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-26 19:51:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b61031e41a40ab0548812e6b2a57344152d51f3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-26 19:53:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d3db89f8bb3e4f891221acc029071d33fd18557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 19:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e255663a326a4d00a44f1dbdbac2705c : SUCCESS in 12m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf1db69adb844fc8416dfb0ae106419 : SUCCESS in 36m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a22000734264168bd086f3fb712e488 : SUCCESS in 26m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/623ffe63018141d195f9418c4b6932c2 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6ad70af4e8748d4a5dd6ce860faed78 : SUCCESS in 21m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/228a4911809a4991b54295b5a269bbf5 : SUCCESS in 18m 10s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4d3584b3e2604156b50425a5d77c1604 : SUCCESS in 6m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/68dd126ab022489e9df72217e2905722 : SUCCESS in 5m 40s (non-voting)\n- manila-tempest-plugin-container-build-test https://zuul.opendev.org/t/openstack/build/816c6c2fed9d46e4abea7025bace32ba : RETRY_LIMIT in 15m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48530f1ed4664b7fb9998d76769f646389ed152f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-26 19:58:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8c17e2917ad50af9c773c651a225806ea70c04f9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-01-26 23:46:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-13-826513-1/ns5_manila : SUCCESS in 2h 10m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"435156b8735ac5cd39675c99a6f02b100432ba4b","revisions":{"435156b8735ac5cd39675c99a6f02b100432ba4b":{"kind":"REWORK","_number":1,"created":"2022-01-26 19:17:37.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/13/826513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/826513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/826513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/826513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/826513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/826513/1"}}},"commit":{"parents":[{"commit":"d1509256dac8d2e66d873e4b9d38e4e06f344b59","subject":"Merge \"OnlyHostFilter allows user to specify host during share create.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d1509256dac8d2e66d873e4b9d38e4e06f344b59"}]}],"author":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2022-01-26 18:59:37.000000000","tz":-180},"committer":{"name":"Douglas Viroel","email":"dviroel@redhat.com","date":"2022-01-26 19:00:24.000000000","tz":-180},"subject":"DNM - test parenting image-elements-job","message":"DNM - test parenting image-elements-job\n\nChange-Id: Ifbe3c64e4268fd646a23dfa25d9fed9a374d4ac7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/435156b8735ac5cd39675c99a6f02b100432ba4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/435156b8735ac5cd39675c99a6f02b100432ba4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
