)]}'
{"id":"openstack%2Fmanila~830835","triplet_id":"openstack%2Fmanila~master~I34b316f85af2cee3aa22d9111764e2c8af1bbde8","project":"openstack/manila","branch":"master","topic":"bp/manila-share-support-recycle-bin","attention_set":{},"removed_from_attention_set":{"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-07-14 21:53:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34b316f85af2cee3aa22d9111764e2c8af1bbde8","subject":"share recycle bin, Fix follow-up suggestions","status":"MERGED","created":"2022-02-24 15:02:27.000000000","updated":"2022-07-14 21:54:42.000000000","submitted":"2022-07-14 21:53:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830835-bp/manila-share-support-recycle-bin","meta_rev_id":"d061354352d996775b54745817580bae9dbdcbbe","_number":830835,"virtual_id_number":830835,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 21:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-07-14 20:50:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-02-25 19:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-07-14 20:50:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-24 16:01:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-02-24 16:25:26.000000000","updated_by":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"reviewer":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-24 17:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 17:49:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-02-24 17:56:24.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-25 03:36:49.000000000","updated_by":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"reviewer":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"state":"CC"},{"updated":"2022-02-25 19:01:19.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-25 20:03:06.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 20:50:13.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"363798364d5da15cfbbb4c1efa2ff8c4428346f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-24 15:02:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd5c3ef9e4ff0679a7d0ff3c851e39d1a76e5e2d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-24 16:01:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/35/830835/1/thirdparty-check/pure-devstack-manila-tempest-aio/4c2d7ce/ : FAILURE in 58m 10s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/35/830835/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/87389c0/ : FAILURE in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"884f54a280aba9f8c68d886788865d2782b8c11b","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-24 16:25:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/830835/1/check/dsvm-tempest-manila-zadara-driver/9c8580f : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0624a736ca655c47d09cbc6bb38012a463f51dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 17:08:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/002d1d47afcd4e72889104cbf8688fe3 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1e251cb399402386133b191c906b82 : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92dd5fdba13746f785373ebeb2f60aaf : FAILURE in 32m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0edd31ec0dd741338a4c4b981a1d7f09 : FAILURE in 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c962db87eef845f69a040ef428404734 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1265470dc374850a04ba97d6571ea38 : FAILURE in 25m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/567bbd8f5a86457aaa4b96955b20f47d : FAILURE in 16m 13s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3121b215c01141029f7081f218fbb002 : SUCCESS in 6m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/95ac58f97f0d431bbb33837b8eed4762 : SUCCESS in 6m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5c2f16ccb97444a3840791a19cbf0988 : SUCCESS in 29m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5b6bb98dae4641e4906099836d187a8d : SUCCESS in 45m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/76a12886c7624924bb2c5dade84da643 : SUCCESS in 1h 14m 25s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fa37f49e1e4849e281e75d9d9f378049 : SUCCESS in 1h 14m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0b5a77c6997644c2b13b5a1fed678aa5 : FAILURE in 1h 53m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/02d387907f7e4d13b6252a17bc23e4c9 : SUCCESS in 43m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/99023c1f4d1744dcaae529d6c0ac7036 : FAILURE in 2h 03m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/03087428d0e3432fa1ece34be0b6db3c : SUCCESS in 43m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/36cfe1fa86dd4050a8d05e6cc2e65644 : SUCCESS in 33m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8175856b19854bf4a901b5d60d5d434d : SUCCESS in 1h 06m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e154314863b84b0a8b8542b119d718e5 : SUCCESS in 57m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c9af9ac9c31344b1a8e0e62c6cb4dd96 : SUCCESS in 39m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fbc8a341bc5a4eeda7f569662e2afeb2 : SUCCESS in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a57e274612de025a97f4324acc826be8bde844fb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 17:49:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"190755306cea4d5d9f42fdd30cc010de90ba26ca","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-24 17:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-830835-1/ns5_manila : SUCCESS in 1h 51m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"38e1e936b144e608cf2f47db61b6bfc22aabf32f","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-02-25 03:36:49.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b7bbda428dbde3df74be1764c4c2d87ad15ddaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-25 08:17:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d55a12318c73bdb3b8e93e8ad2b7c7ead9ca4f33","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-02-25 08:33:13.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a999fd7fdcb00090c544c77c12e0501875a1c83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 10:25:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8abad9a85ad4454fb566d9d83f158d85 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/076498aa598c48088a229c58934c44e2 : SUCCESS in 10m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15e928c10e7c4784951ba8d763cc60a2 : FAILURE in 35m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67d99cb345cc4f0d8ace3f14d908c824 : FAILURE in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a39638776b477597373de2d621ea72 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75717a846a664b1c8b380cefb604c0e7 : FAILURE in 23m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f87370691d34669a90c58a9eeba776a : FAILURE in 25m 58s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2e5aa70733fe4932810117a078796447 : SUCCESS in 5m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ce00640a387d4dacb7567b901221f408 : FAILURE in 7m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec3d2805bcd74cc7b450500bd10122b9 : SUCCESS in 39m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d37fbe5c64ba4981bf9fb7880ec9774e : SUCCESS in 33m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b25bf5d140004b0aa4a67d52e983a4b5 : SUCCESS in 1h 09m 32s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dfebf03448634653962c6837f98e54f0 : SUCCESS in 1h 52m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c9323f682d64bc989579ead3ce06eab : FAILURE in 1h 43m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b4762bfa7e1c47ce8d4c89d55c1d0d61 : SUCCESS in 48m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2a337c8080fa4b14b88c7e68081f2879 : FAILURE in 2h 00m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b413c4a431984ae2aa865d946022fda6 : SUCCESS in 35m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/152b9855ba114905a5b9d2ccbeeeea47 : SUCCESS in 22m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6d24f2adbfe846edb3f74110746efa01 : SUCCESS in 1h 15m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8d447431c0cb43d684e3365cdf3dbcd1 : SUCCESS in 1h 00m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1af05866e7094cc19ffe25fa82aa1766 : SUCCESS in 39m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fda983d4af624f09ae74c7999b2e0520 : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"480b228e56feff667e59b28e183635fc612dcccc","author":{"_account_id":19933,"name":"Shlomi Avihou","display_name":"ZadaraStorage VPSA CI","email":"openstack-dev@zadarastorage.com","username":"zadara","tags":["SERVICE_USER"]},"date":"2022-02-25 11:00:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- dsvm-tempest-manila-zadara-driver http://openstack-ci-logs.zadarastorage.com/35/830835/2/check/dsvm-tempest-manila-zadara-driver/781fe52 : SUCCESS in 46m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b892c85a0c2d830b8638668210ee69131b6f6adc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-25 11:06:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0cc04c0a91434d6ea679606a6dc21bef7187a800","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-02-25 11:08:20.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e182a8db651086628932458cc18c240c101b839","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-25 11:49:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-830835-2/ns5_manila : SUCCESS in 1h 55m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ada169850bc419e81b2b5a8dae00c91ac1bd4c2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-25 12:01:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/35/830835/3/thirdparty-check/pure-devstack-manila-tempest-aio/1052251/ : FAILURE in 53m 53s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/35/830835/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/7a78e6d/ : FAILURE in 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c882c56a44ff1ea4502c41c7f049719d8a3186d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 13:13:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3d118b6d62104442a979908141ff3e48 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c054b86b5f451ca98e17d6dc47869b : SUCCESS in 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/986a6e58a65540008967c9fecacef37d : SUCCESS in 30m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68d348211e144f1a8ab5a00195ec726e : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe997b143dd4ee884c9a0df44f593f8 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c27012249f7f40a9ac7522e65b2a2941 : SUCCESS in 16m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/538b5148c70e4f5bace3b40a1e231a1e : SUCCESS in 16m 12s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e115e9a8aafd45498664283c2f147c48 : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e786fdbeaaaf4acb8255da2a9433fdeb : FAILURE in 6m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/20656f921fc14b78a25e318456f3f94c : SUCCESS in 34m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b484eb7528d74454a5b020f5cac7c777 : SUCCESS in 37m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/233f8daa4de44e06abd65a43dfb491c1 : SUCCESS in 1h 23m 08s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/66fd4c2bcf804532b821fd737c4c134e : SUCCESS in 1h 29m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4ff0e44cee0445fb910811c5ba28335c : FAILURE in 1h 27m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4aa681e7f0e34ec39389db0fc6b0235f : SUCCESS in 56m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b7ad660783834c30960abd4185994583 : FAILURE in 2h 03m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c4b209e206b6462cb2fbdbf23c648005 : SUCCESS in 31m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/abcc9de887e1484cbe2932e35c3acb8c : SUCCESS in 28m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0a86a213b7544bea0363c5ad5fee64e : SUCCESS in 1h 12m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c16a31ca72af484cbc7bebc373d916a6 : SUCCESS in 46m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fe5fdcad65094bc6af9d8f3161a9d746 : SUCCESS in 44m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5e9d3844d3784ef4b084cbdfa70cc0f7 : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7ee2cc0c373c636e80fab8d074fffa35c3f16c2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-25 13:59:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-830835-3/ns5_manila : SUCCESS in 1h 50m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"e9c0dd2c1ae91a0c29587d15e11b8b06772bae7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-25 14:57:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1a8f843d7c0796bc26bfa7ef25c9a47e05c96c2b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-02-25 15:51:11.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/35/830835/4/thirdparty-check/pure-devstack-manila-tempest-aio/8a3371d/ : FAILURE in 51m 53s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/35/830835/4/thirdparty-check/pure-devstack-manila-tempest-fa-aio/b39620b/ : FAILURE in 36m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db67a368f85ad4edc9358223fab87d65b80d93a6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-02-25 16:41:46.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://openstack.deiter.ru/aaf864307c6ff20c3b5cc1ef422b55c2/refs-changes-35-830835-4/ns5_manila : SUCCESS in 1h 43m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"3ab0ee7e009cfe3d50fdf1939af33cda079a3bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 17:50:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/036871e405ff40b48a17f8f37c6420f9 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7cd24edc9a94c1cb7ea5e06812fadf3 : SUCCESS in 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e11c2ace7b04fc68da7752d1f25091c : SUCCESS in 29m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff147adbf43847cb97dd6701f256bede : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b6008f3c8146c9858e293d54c99ba8 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/911fece3cc7a433ca446517025b67f24 : SUCCESS in 18m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a988998826e4836a987c4d9a5509c75 : SUCCESS in 22m 00s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ca4dd4a7ac2d48dc9d942e0d305528f0 : SUCCESS in 8m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/df0d0df837034756960f43b795978073 : FAILURE in 8m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0c9279f893cc4412829ad4c00bb2823d : SUCCESS in 25m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd157db43b4147b5af35c911b88024c2 : SUCCESS in 32m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/38fa83d87fcc4b2ba01919862aa2db93 : SUCCESS in 1h 03m 57s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e699fe78af0a463499de4b605307b208 : SUCCESS in 1h 16m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a1730aa7ca6644f397e60d9e22c7a334 : FAILURE in 1h 33m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/47334cad8c284f83b68cee815d48d0f8 : SUCCESS in 58m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dacd4a387aaf484bb09ebf7caabb49ab : FAILURE in 2h 37m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f3d89a7650fd462ea6fb6a73b80e4b92 : SUCCESS in 40m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/08f376a1e1714f47b2179ed9e4ea30cb : SUCCESS in 31m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/189d4dfab0bb466b8de9d14c9f79168f : SUCCESS in 1h 27m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8bd6cce231034d7b873ab02720309b35 : SUCCESS in 1h 11m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dc6271e806e14b51a3f4adb969f7bbb0 : SUCCESS in 37m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1691a076979b466f8f2a6952e768878e : SUCCESS in 43m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc3f847243f1613c27a2eb376c16674e856e08b8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-25 19:01:19.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72227058603f44ae685844946bf7e077b878b61e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-25 19:01:35.000000000","message":"Topic set to bp/manila-share-support-recycle-bin","accounts_in_message":[],"_revision_number":4},{"id":"fd72c08d46799b9d84fd57d807404faf2f3f9ac8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-25 20:03:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/830835/4/upstream-check/manila-cDOT-ss/f4e839e/ : SUCCESS in 2h 09m 59s\n- manila-cDOT-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/830835/4/upstream-check/manila-cDOT-no-ss/f67d249/ : POST_FAILURE in 2h 11m 46s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ecbf459d7405038e988219ba6000229b39fc626c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-14 20:50:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7a1b86b346c12d7e23d4f52a967044166e9242f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:50:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2ecc4e0c3b77acf4ab73b9c8cab0356dd52649c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:53:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d81666e230542b38cc8f957cf867068\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23ec0f0117cd4168ab0fa98964c4f078 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430c41fb02df4b37863cb41445f760ef : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9d8cb713ac40d58aa0e33188d8277f : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35e4982bffa640a9a975b19f59f207d9 : SUCCESS in 9m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f2319b59f5d43979932add72ff138eb : SUCCESS in 9m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e58ec8031aa344f9ba46f8b1f5aeb416 : SUCCESS in 59m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/19d30f508b2146ae8a9ae757efec68f7 : SUCCESS in 32m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3932ed28d5249ba82fd57f647c97900 : SUCCESS in 43m 47s","accounts_in_message":[],"_revision_number":4},{"id":"aaeefab4b01ec5e28acd80f7969cc73132f8c980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:53:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d061354352d996775b54745817580bae9dbdcbbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 21:54:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0add8b8eb4104d6385aaff566a2a0480\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/e943bb80d3164b21b90bc69bf95c08bd : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64221dd5bc1646a682036a4a551e6c73 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e3f53c590a6069247312e7bf17b40f5f0dd30d3","revisions":{"09c71722d091e23a27311708a815f82e2541551c":{"kind":"REWORK","_number":1,"created":"2022-02-24 15:02:27.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/35/830835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/35/830835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/35/830835/1"}}},"commit":{"parents":[{"commit":"1dc17d9e141da7c9ac6fea967af6602dca9cbe79","subject":"Merge \"change to valid error status in test_generic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dc17d9e141da7c9ac6fea967af6602dca9cbe79"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-24 15:01:58.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-24 15:01:58.000000000","tz":480},"subject":"share recycle bin, Fix follow-up suggestions","message":"share recycle bin, Fix follow-up suggestions\n\nChange-Id: I34b316f85af2cee3aa22d9111764e2c8af1bbde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09c71722d091e23a27311708a815f82e2541551c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09c71722d091e23a27311708a815f82e2541551c"}]},"branch":"refs/heads/master"},"5eb79d7102a73ec210abafd79818aeb8552bff15":{"kind":"REWORK","_number":2,"created":"2022-02-25 08:17:20.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/35/830835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/35/830835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/35/830835/2"}}},"commit":{"parents":[{"commit":"1dc17d9e141da7c9ac6fea967af6602dca9cbe79","subject":"Merge \"change to valid error status in test_generic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dc17d9e141da7c9ac6fea967af6602dca9cbe79"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-24 15:01:58.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-25 08:17:02.000000000","tz":480},"subject":"share recycle bin, Fix follow-up suggestions","message":"share recycle bin, Fix follow-up suggestions\n\nChange-Id: I34b316f85af2cee3aa22d9111764e2c8af1bbde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5eb79d7102a73ec210abafd79818aeb8552bff15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5eb79d7102a73ec210abafd79818aeb8552bff15"}]},"branch":"refs/heads/master"},"3bd3da29b32f0214eeb97be496199534df4cd7c3":{"kind":"REWORK","_number":3,"created":"2022-02-25 11:06:22.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/35/830835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/35/830835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/35/830835/3"}}},"commit":{"parents":[{"commit":"1dc17d9e141da7c9ac6fea967af6602dca9cbe79","subject":"Merge \"change to valid error status in test_generic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dc17d9e141da7c9ac6fea967af6602dca9cbe79"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-24 15:01:58.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-25 11:05:56.000000000","tz":480},"subject":"share recycle bin, Fix follow-up suggestions","message":"share recycle bin, Fix follow-up suggestions\n\nChange-Id: I34b316f85af2cee3aa22d9111764e2c8af1bbde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3bd3da29b32f0214eeb97be496199534df4cd7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3bd3da29b32f0214eeb97be496199534df4cd7c3"}]},"branch":"refs/heads/master"},"4e3f53c590a6069247312e7bf17b40f5f0dd30d3":{"kind":"REWORK","_number":4,"created":"2022-02-25 14:57:48.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/35/830835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/35/830835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/35/830835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/35/830835/4"}}},"commit":{"parents":[{"commit":"1dc17d9e141da7c9ac6fea967af6602dca9cbe79","subject":"Merge \"change to valid error status in test_generic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dc17d9e141da7c9ac6fea967af6602dca9cbe79"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-24 15:01:58.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-02-25 14:57:02.000000000","tz":480},"subject":"share recycle bin, Fix follow-up suggestions","message":"share recycle bin, Fix follow-up suggestions\n\nChange-Id: I34b316f85af2cee3aa22d9111764e2c8af1bbde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4e3f53c590a6069247312e7bf17b40f5f0dd30d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4e3f53c590a6069247312e7bf17b40f5f0dd30d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
