)]}'
{"id":"openstack%2Fmanila~832876","triplet_id":"openstack%2Fmanila~master~I163e2691105468a0e821c8a5991f8c20e85ff599","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-03-11 00:43:53.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-03-11 00:43:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I163e2691105468a0e821c8a5991f8c20e85ff599","subject":"Add missing deprecated messages to policy","status":"MERGED","created":"2022-03-10 00:08:56.000000000","updated":"2022-03-11 00:45:53.000000000","submitted":"2022-03-11 00:43:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"832876","meta_rev_id":"8cdb92790532dcf30606f7b2942b4e56cf128f30","_number":832876,"virtual_id_number":832876,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 00:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-03-11 00:45:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-10 21:19:14.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-10 21:19:14.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 00:10:40.000000000","updated_by":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"reviewer":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"state":"CC"},{"updated":"2022-03-10 00:31:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-03-10 00:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 05:27:04.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-10 14:21:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-03-10 19:35:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-10 21:19:14.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"9e457b00b5fefe776ae014d9b1f06e0e67d110a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-10 00:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7328cf1cc42877199d0d2f6cb16936b8866d781","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-03-10 00:10:40.000000000","message":"Patch Set 1:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ad057efd820ca7494e307a9d79784dcf365eb96","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-10 00:31:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/76/832876/1/thirdparty-check/pure-devstack-manila-tempest-aio/9729ddb/ : FAILURE in 20m 56s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/76/832876/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/5e3a848/ : FAILURE in 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36144d697cd260fccaa5fe4d4678e7d752387713","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 00:58:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aaf4f69f3204d599f5418654bd99793 : FAILURE in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42f9cd528cde45409fe115bb62c0eb14 : SUCCESS in 16m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d492f2be3fd1492689e43dc19368f875 : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479a6fed03f041eca4c6ddd7f4417d42 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf359b5c3de44fca938a991f662a6fad : FAILURE in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faa2712f04344a6bb0b9fd738cb872b3 : FAILURE in 6m 45s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/41c1eec7399a44829591fe1b0d2fcc25 : FAILURE in 7m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80acc604ed05480f9ab066644ac33c38 : SUCCESS in 6m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d57449acb9f74c8e82737062e6fcba71 : FAILURE in 29m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00868f11a72c4bdfb5a3385b09be693a : FAILURE in 19m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ab644d4c1650407eaa5e88bc86691a05 : FAILURE in 27m 34s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/37be8dc4fcef4a43a88200db18ee9ced : FAILURE in 29m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/226c329317e242159e25274ba9625c9b : FAILURE in 29m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4e52e7c44f1d470bb4d2caa2b42d5b88 : FAILURE in 27m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/93732aa232704d9582c18dd3420f875a : FAILURE in 28m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fa436cae71ea4d769e67392190b26fd9 : FAILURE in 20m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f4b6d4b8f06f47799fcce6e9c56970a4 : FAILURE in 20m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9823a1564b647b4a4d6aca5c622a820 : FAILURE in 27m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1d42ee5380b34810931ed7b9e2953d98 : FAILURE in 43m 56s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f6f31a82029a456dbdb394eb33b56020 : FAILURE in 30m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/256d2ded02904889af8f5327271ff5f1 : FAILURE in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66b9635a16a3ba0417acc42189e8e819e0191910","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-10 05:27:04.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://93.100.254.243/nexenta/refs-changes-76-832876-1/ns5_manila : FAILURE in 1h 13m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd492c09c12df9abc052f629f41d43d8b69bdcc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-10 12:21:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4687ea5bccd7083fd110b6ee4067c4086ee0e432","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-03-10 12:26:26.000000000","message":"Patch Set 2:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0477ade0c01f414701e441ff5a0b178b102c9189","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-10 13:27:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/76/832876/2/thirdparty-check/pure-devstack-manila-tempest-aio/a8d40f3/ : FAILURE in 1h 04m 32s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/76/832876/2/thirdparty-check/pure-devstack-manila-tempest-fa-aio/f3bfb7a/ : FAILURE in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a36f4b3ec0aefc13c7f10d2300a16ae70afc8aa7","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-10 14:13:27.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://93.100.254.243/nexenta/refs-changes-76-832876-2/ns5_manila : FAILURE in 1h 51m 34s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"ca9270b2ea7fa38d8b164fc2e9ee4a3b9789c19f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-10 14:21:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09e9c1b95b0f5c8961bf41a502a4d1cd4c99c5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 14:35:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23937dc1e59f4fa0b2d0bdcf10d26465 : SUCCESS in 12m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0931ddebb7284628a7fe5d97d14ad578 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/113f5b0eb2914c9e87b566d9990f1d82 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ea7442984a45a386ab2e10a4edf95c : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af59a3949618449bb584c4b2b84c1f4a : SUCCESS in 13m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93a7a5f9ac3b4332a3a80239c4f8de38 : SUCCESS in 16m 17s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec220ab18a354078b40c1b10115ee474 : SUCCESS in 8m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/379cc1442413474989f1d9fe461354fa : SUCCESS in 8m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0748b5f205a341ab8b5e153186082722 : SUCCESS in 31m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/947f6efbe51442ed8ca1cc13cf1108a5 : SUCCESS in 34m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9aed1b81dc24674828d6e77c671bfe6 : SUCCESS in 1h 20m 58s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/22c5e28a47c34f05b9d20c4c0d2594a0 : SUCCESS in 1h 35m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f55d93d1fcab47eb812a06d2471c11ea : FAILURE in 2h 03m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/08ddc9faec594422b08e30ad2e424daf : SUCCESS in 1h 02m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7be396a6c0ec4037b03b836693dbc519 : FAILURE in 1h 29m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c2b1ed1dca2c480bbb6388d4fa2d092a : SUCCESS in 35m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0a2db5f790f43dca0742e5f5f7778cb : SUCCESS in 29m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ca06915ddaa9467cb02537bba2f51795 : FAILURE in 1h 25m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4a43431841cc48c0bc5e7823f9af8857 : SUCCESS in 1h 16m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0ceb9ed5834048149df56f5f9214ae02 : SUCCESS in 38m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/443da1a09dec41ae90231d70b63619da : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a61b85e3e15409ff028dd1f5145e8299d59e3776","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-10 19:33:29.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b85efd885e4fd036ba3adb7836e859f5eba892e3","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-03-10 19:35:00.000000000","message":"Patch Set 3:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":3},{"id":"60345c375048f6c413843afba468b661014e9f85","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-10 19:35:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"05870a657f92d18bafb6f9d4ff4eb5defe9b16cf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-10 19:36:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"840d40cf1fb02320b27fba151bf6881cd83dd7af","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-10 20:30:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/76/832876/3/thirdparty-check/pure-devstack-manila-tempest-aio/2088957/ : FAILURE in 55m 31s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/76/832876/3/thirdparty-check/pure-devstack-manila-tempest-fa-aio/ef68bbd/ : FAILURE in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"11a6fdfe3716193f7b9f5f8fd81a2fd40a279603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-03-10 21:17:58.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"22bb1a5db1ed333b17ac9482257bc4e43749e1e1","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-03-10 21:19:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"aa95bbee7d7cad3e68cd316512eca40aaa8a3727","author":{"_account_id":33693,"name":"yuehui gan","display_name":"fzzf","email":"gan769410938@outlook.com","username":"gyh"},"date":"2022-03-10 21:20:28.000000000","message":"Patch Set 4:\n\nStarting check jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9fade7210e7f50d9c41d214cf0d00eece74f164e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-10 21:47:16.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://93.100.254.243/nexenta/refs-changes-76-832876-3/ns5_manila : FAILURE in 1h 55m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"085ff8d8ca8ca484e1c54982a07d15daf44cef82","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-10 22:32:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/76/832876/4/thirdparty-check/pure-devstack-manila-tempest-aio/ab356e5/ : FAILURE in 51m 17s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/76/832876/4/thirdparty-check/pure-devstack-manila-tempest-fa-aio/5b90c7e/ : FAILURE in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"705ddd6cafc0be7c31831c5ef4f46b4651f376e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:28:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d882e2f6984b0fb910022b82fc9eb0 : SUCCESS in 11m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3a9d2f231204f5b9336b2502156179a : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f48f6f50062a45b1bfaf4d5ca40812b7 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b473a2b7f647d881fe28e756626dfc : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/924d0043baa44a42ae464dd4bff8ff7f : SUCCESS in 9m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7362c34ba5424006857e979ad4906722 : SUCCESS in 12m 22s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dce38be505ab4bc68eae40f452d9d447 : SUCCESS in 6m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3878ff53c5ff444ab6738401df6f2179 : SUCCESS in 7m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/529c6d1258e74f6db6773b9295cfcc2a : SUCCESS in 29m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/880aac1e4d8f4106b662479f3e51e28a : SUCCESS in 34m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cb73ad73156b4965a64bcfc75a577fc9 : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d631e598e3514ad9b1f1d80523997f6f : SUCCESS in 1h 27m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8eb92205f6da49e8ac24df439dbd7295 : FAILURE in 1h 37m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/170480106e3e4bc5b34918c7ccbca29f : SUCCESS in 58m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ccd6979d85ca4e64bd54527681d755b2 : SUCCESS in 2h 04m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f22ffdac483142f8862c8e6d0ef5e2c7 : SUCCESS in 41m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fa44f1b91c8646dda49282c679d18dba : SUCCESS in 22m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4dc6948ccfb342c487e827534be6ef8c : SUCCESS in 1h 01m 38s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/dfd7b159ecd2490eb16b4ae2522645ba : SUCCESS in 1h 05m 38s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bcef12ab826d499fbf38a11d7d24aa75 : SUCCESS in 37m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/86120c029654418d9914119edf1139fa : SUCCESS in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf1270924fabd6b427ed2198db92fb34766c85f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 23:29:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"49a3edfb8d36663ed09d52138c40cb81dd3d967c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-11 00:37:46.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://93.100.254.243/nexenta/refs-changes-76-832876-4/ns5_manila : FAILURE in 1h 45m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"fdeda3010a9dab1c86cd9dac046378226bca88b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:43:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76affaac05e46e781be1c55f7745bac : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a9ac2ef03ad4fd89bb9a2dbbcb71be6 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce6423b3b05451ca5dc4ce0ba58df73 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6bf549da9f440ea3cc7031aea58bab : SUCCESS in 12m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fe36a1f88ea40aabec6c6aa9ab244f8 : SUCCESS in 13m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/20ed5590dad546f4b244a2d4ef5e90e3 : SUCCESS in 31m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4c0ce1dc9b7e45868fc794c8163280cd : SUCCESS in 36m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6879b8200b0e4e3191b564e6b3269d20 : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b3ff5426bcd3d9ab1205288fbf06ea6b6cc506a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:43:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"811432abf4a8994defa88bf4ab20dd0e575dc2dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 00:45:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844e88c22dac4ddbb0b1b37900af8e93 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"48afe551a0d162fc9f9cd18a715e71ac5ecdf8a5","revisions":{"25dd8aa45c76782814eb83cd8ced714ca5bd0405":{"kind":"REWORK","_number":1,"created":"2022-03-10 00:08:56.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/76/832876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/832876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/832876/1"}}},"commit":{"parents":[{"commit":"cf872f576f7e977386386dab9bbe79dd7b5d8a56","subject":"Merge \"Metadata for Share Resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf872f576f7e977386386dab9bbe79dd7b5d8a56"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-10 00:07:14.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-10 00:07:14.000000000","tz":-180},"subject":"Add missing deprecated messages to policy","message":"Add missing deprecated messages to policy\n\nDuring the Yoga cycle, we\u0027ve introduced a new API to check if it\nis possible to add share network subnets to share networks that\nare currently being used. We missed the addition of the deprecation\nmessages to that new policy though. This change serves as a follow\nup to add these missing items.\n\nChange-Id: I163e2691105468a0e821c8a5991f8c20e85ff599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/25dd8aa45c76782814eb83cd8ced714ca5bd0405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/25dd8aa45c76782814eb83cd8ced714ca5bd0405"}]},"branch":"refs/heads/master"},"458e17248676844e5520ea6fdabef9e731354753":{"kind":"REWORK","_number":2,"created":"2022-03-10 12:21:24.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/76/832876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/832876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/832876/2"}}},"commit":{"parents":[{"commit":"cf872f576f7e977386386dab9bbe79dd7b5d8a56","subject":"Merge \"Metadata for Share Resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf872f576f7e977386386dab9bbe79dd7b5d8a56"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-10 00:07:14.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-03-10 12:21:24.000000000","tz":0},"subject":"Add missing deprecated messages to policy","message":"Add missing deprecated messages to policy\n\nDuring the Yoga cycle, we\u0027ve introduced a new API to check if it\nis possible to add share network subnets to share networks that\nare currently being used. We missed the addition of the deprecation\nmessages to that new policy though. This change serves as a follow\nup to add these missing items.\n\nChange-Id: I163e2691105468a0e821c8a5991f8c20e85ff599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/458e17248676844e5520ea6fdabef9e731354753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/458e17248676844e5520ea6fdabef9e731354753"}]},"branch":"refs/heads/master"},"e64754feae87e72aa1d498db53094575b57968c1":{"kind":"REWORK","_number":3,"created":"2022-03-10 19:33:29.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/76/832876/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/832876/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/832876/3"}}},"commit":{"parents":[{"commit":"cf872f576f7e977386386dab9bbe79dd7b5d8a56","subject":"Merge \"Metadata for Share Resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf872f576f7e977386386dab9bbe79dd7b5d8a56"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-10 00:07:14.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-03-10 19:33:29.000000000","tz":0},"subject":"Add missing deprecated messages to policy","message":"Add missing deprecated messages to policy\n\nDuring the Yoga cycle, we\u0027ve introduced a new API to check if it\nis possible to add share network subnets to share networks that\nare currently being used. We missed the addition of the deprecation\nmessages to that new policy though. This change serves as a follow\nup to add these missing items.\n\nChange-Id: I163e2691105468a0e821c8a5991f8c20e85ff599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e64754feae87e72aa1d498db53094575b57968c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e64754feae87e72aa1d498db53094575b57968c1"}]},"branch":"refs/heads/master"},"48afe551a0d162fc9f9cd18a715e71ac5ecdf8a5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-10 21:17:58.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/76/832876/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/832876/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/832876/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/832876/4"}}},"commit":{"parents":[{"commit":"3d04487fa89ee01433c08b41ea367d0d5090cf8a","subject":"Fix sporadic test failure in manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d04487fa89ee01433c08b41ea367d0d5090cf8a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-03-10 00:07:14.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-03-10 21:17:58.000000000","tz":0},"subject":"Add missing deprecated messages to policy","message":"Add missing deprecated messages to policy\n\nDuring the Yoga cycle, we\u0027ve introduced a new API to check if it\nis possible to add share network subnets to share networks that\nare currently being used. We missed the addition of the deprecation\nmessages to that new policy though. This change serves as a follow\nup to add these missing items.\n\nChange-Id: I163e2691105468a0e821c8a5991f8c20e85ff599\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48afe551a0d162fc9f9cd18a715e71ac5ecdf8a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48afe551a0d162fc9f9cd18a715e71ac5ecdf8a5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
