)]}'
{"id":"openstack%2Fmanila~836039","triplet_id":"openstack%2Fmanila~master~I81569723adb5ff33076ec96b9ec3216b177379f3","project":"openstack/manila","branch":"master","topic":"bug/1967326","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2022-06-10 14:11:35.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-06-10 13:09:02.000000000","reason":"Carlos Eduardo replied on the change"}},"hashtags":[],"change_id":"I81569723adb5ff33076ec96b9ec3216b177379f3","subject":"Add Guru Meditation Report to wsgi","status":"MERGED","created":"2022-03-31 16:18:43.000000000","updated":"2022-06-10 14:12:55.000000000","submitted":"2022-06-10 14:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836039-bug/1967326","meta_rev_id":"e8366a55e2f49a57c9930269b20bc1185a1baa79","_number":836039,"virtual_id_number":836039,"owner":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 14:11:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-06-10 13:09:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-06-09 12:57:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-06-10 13:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 17:12:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-03-31 17:20:14.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-31 17:54:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 18:02:45.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-31 20:02:47.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-20 07:58:09.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-06-09 12:57:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"aeca47beee0fe0e0dc6d5323bb87d859963c78ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2022-03-31 16:18:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"326655230ae2e5b063ae054c664c7e65ad1c8c1b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-03-31 17:12:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/39/836039/1/thirdparty-check/pure-devstack-manila-tempest-aio/0cd8c26/ : FAILURE in 52m 42s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/39/836039/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6606dd7/ : FAILURE in 37m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59211a69b212d75d4cb18b4cc54defab2c727cef","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-03-31 17:20:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-39-836039-1/ns5_manila : SUCCESS in 1h 1m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a599945e496ea1a6a3338021fa52261089b2d586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:54:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bfd6a917aa441ab2764fd1d15497b4 : SUCCESS in 9m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/331a6bdd201e4f40a7da2a04c5a20076 : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a66601e671884b4ca1ed7d81e31fc1a4 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da029409ba6242cca737ad2bf6b4ef87 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12b87617b9204a0da6e8343afd3cf885 : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e1da315aa09433f8eba5b692aff6196 : SUCCESS in 18m 37s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/aa4b0a8947344556b95c584c80b31a3b : SUCCESS in 6m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5308630b49b44cf1bf67475b0211d396 : SUCCESS in 6m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f373da168c1242a4b9c8c0b5892a92c3 : SUCCESS in 35m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c2f664e3bb1349b985fba67a7fe929b8 : SUCCESS in 34m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e8dee1a273254dbd8c9c9c43039f8644 : SUCCESS in 1h 32m 11s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3a14d48abef64db3845b2ea264d981a0 : FAILURE in 1h 15m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c05ca58a728f4638874bf15760cb29e2 : FAILURE in 1h 30m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ea87f85b8eb443fa507832a3b211abd : SUCCESS in 57m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e3b678b40c184958b6dba2a89658a50c : FAILURE in 1h 12m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/aab9e4405d77429c9fa7c59bf940d263 : SUCCESS in 38m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/93f6ce568fb14469a43991823e2df718 : SUCCESS in 29m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a4ffc118545d4317bd9805a97098975e : FAILURE in 1h 25m 29s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/88671b4fcd9a469984a583300b674abb : SUCCESS in 1h 04m 06s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3c3b502f6e18435c9b3325700a723122 : SUCCESS in 1h 05m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7e4a1f67b77d44e79e67dc89bafb1928 : SUCCESS in 38m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a60b1b9da7784afcbc485baccb9f3bb3 : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a55bf284dc549c26290539955d76af3d18b98397","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-31 18:02:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/836039/1/upstream-check/manila-cDOT-ss/4b705c1/ : FAILURE in 7m 27s\n- manila-cDOT-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/39/836039/1/upstream-check/manila-cDOT-no-ss/d5616fb/ : FAILURE in 7m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f17758fa8385fd69f088fe4c7a591f2368da5567","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-03-31 20:02:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/39/836039/1/check/DellEMC_PowerMAX_Manila/10137da/DellEMC_PowerMAX_Manila/190 : FAILURE in 2h 07m 42s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"760e831156597ddb996945c1301378bb56875bc1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-09 12:57:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95a917f4a0381263ddf542e8a4b2f31108eceed4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-06-10 13:09:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"369b2c313e06f089c5bc46a54440033666cae38a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 13:09:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59f2f33ed65fc69c9e046a776e924022f3f4e98c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:11:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f606c85f7dea4c2b83f8b79882e006fe : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f23d562bc5465bb9f0ea1ac4ed58fa : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4411072105de4784a1ed0d976a509bf2 : SUCCESS in 14m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94ad06462b444aa78befd037e06fdc87 : SUCCESS in 14m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/65281cec8ae04df086a2619739d2a899 : SUCCESS in 27m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aa8738d9346648ae8d6179dfffc20ba4 : SUCCESS in 31m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb1569c590a749e09540dead1ee0d76c : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4c24c1f8c95506f89b91fb904980ce297ea51147","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8366a55e2f49a57c9930269b20bc1185a1baa79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:12:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835feefd74a847f48c26c6868a2cc2db : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8e8eaf586ef1f59dffbcbc590bf60e3fca21728","revisions":{"f8e8eaf586ef1f59dffbcbc590bf60e3fca21728":{"kind":"REWORK","_number":1,"created":"2022-03-31 16:18:43.000000000","uploader":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"ref":"refs/changes/39/836039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/39/836039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/836039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/836039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/39/836039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/39/836039/1"}}},"commit":{"parents":[{"commit":"ac672387ab87b216cf163b4473e26aafa523374a","subject":"Merge \"Pure Storage - bump driver version for Zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ac672387ab87b216cf163b4473e26aafa523374a"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-03-31 12:35:26.000000000","tz":120},"committer":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-03-31 16:17:31.000000000","tz":120},"subject":"Add Guru Meditation Report to wsgi","message":"Add Guru Meditation Report to wsgi\n\nbest use it with the file_event_handler config of oslo_reports\nsince using signal handlers is discouraged on wsgi\n\nCloses-Bug: #1967326\n\nChange-Id: I81569723adb5ff33076ec96b9ec3216b177379f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f8e8eaf586ef1f59dffbcbc590bf60e3fca21728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f8e8eaf586ef1f59dffbcbc590bf60e3fca21728"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
