)]}'
{"id":"openstack%2Fmanila~841149","triplet_id":"openstack%2Fmanila~master~I8e98541d8e457e9e4609410853b50d6156465f61","project":"openstack/manila","branch":"master","topic":"bug/1927823","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-09-28 14:26:25.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-09-28 14:26:25.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-09-28 14:26:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e98541d8e457e9e4609410853b50d6156465f61","subject":"[NetApp] Fix lack of aggregate pool home state","status":"MERGED","created":"2022-05-09 16:55:17.000000000","updated":"2023-09-28 18:33:06.000000000","submitted":"2023-09-28 14:26:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"841149-bug/1927823","meta_rev_id":"7ceeb2e1fe7436952388a6db80495d2c090e16ac","_number":841149,"virtual_id_number":841149,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 14:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-22 18:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-08-05 17:38:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2023-09-19 15:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2023-02-10 15:19:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":1,"date":"2023-09-25 06:47:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-09-28 12:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 17:53:14.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-09 17:58:26.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-05-09 19:13:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-09 19:14:46.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-09 19:48:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-11 17:40:11.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-05-13 08:53:19.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2022-07-07 17:03:04.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-05 17:38:26.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-09-12 22:39:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-01-25 13:37:47.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-01-30 16:03:24.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2023-02-10 15:19:55.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-09-19 15:05:32.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-09-19 17:19:14.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-19 17:25:50.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-21 12:29:00.000000000","updated_by":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"CC"},{"updated":"2023-09-22 18:35:42.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-09-25 06:47:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"7452a17f3f16a6ef363df9e430159b070a8ea2ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-05-09 16:55:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ce2414a437161b7f40462c48a47acf4426a278","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-05-09 17:53:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-841149-1/ns5_manila : SUCCESS in 57m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe462af9154c341ff50cf33912e74c2294b4559a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-05-09 17:58:26.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/841149/1/thirdparty-check/pure-devstack-manila-tempest-aio/f4784e6/ : SUCCESS in 1h 02m 06s\n- pure-devstack-manila-tempest-fa-aio http://openstack-logs.purestorage.com/49/841149/1/thirdparty-check/pure-devstack-manila-tempest-fa-aio/6558316/ : FAILURE in 39m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a920f288c0397aee0b691f5bd76785b275fa9362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:13:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f62e25972ee4862b70834a6cba27425 : SUCCESS in 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2790d65faa4e6da4665ef5f97b10ae : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72f28076b50848bc96e0671b31f1cd4f : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432908a9749149c1be25171461ae7230 : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56b0185de2f3465db8e026baf03e9b5e : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5941ff1fe7cb4e7dbccca19563dbaee7 : SUCCESS in 6m 01s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/59d357e7cd6e4d8889d893b0c0c29789 : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/76e25c56c2e04b7d9b3259d5d8753e3f : SUCCESS in 5m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/799c2646b1214779bcbb530c1cfcad38 : SUCCESS in 25m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab3afdaff776405aad46ec335e371154 : SUCCESS in 28m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7efadb63f3eb48b78248515b83addec1 : SUCCESS in 1h 37m 25s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0b068bd43e3644fcbde81ae2159d18c7 : SUCCESS in 1h 14m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2ab350298ad34fd6aec5dd7ada6f1c90 : FAILURE in 2h 11m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b99b7465936945bcab35b0b08bfa067b : SUCCESS in 1h 05m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c22d0a6aae4748a28040e8cfbc895615 : FAILURE in 1h 55m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8c4815b6da2c499fb9e3c506ac5e6740 : SUCCESS in 44m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/524d058e3bb8412a87aeb34100c857c0 : SUCCESS in 28m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/140feb933be743d4b064fe7862f79096 : SUCCESS in 1h 08m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5d11d32d43bf41909d1ade6a60911579 : SUCCESS in 1h 05m 08s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/68f70055a67e4cd09a631c168fcbfc34 : SUCCESS in 1h 16m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/650b3b47147941d6af1ac2f689462cfc : SUCCESS in 34m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e8627b3eb3da438fb691578b48502bc7 : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f3846cde79595142e1504aaffcdb677cf4ea459","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-09 19:14:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/49/841149/1/check/DellEMC_VNX_Manila/054c414/DellEMC_VNX_Manila/259 : FAILURE in 1m 10s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d38bea0238cd23d01824947de1be8c5fe1194dc1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-09 19:48:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/841149/1/check/DellEMC_Unity_Manila_dhss_true/70040ca/DellEMC_Unity_Manila_dhss_true/466 : FAILURE in 2h 52m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/841149/1/check/DellEMC_Unity_Manila_dhss_false/90a5622/DellEMC_Unity_Manila_dhss_false/487 : SUCCESS in 1h 12m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1023862ec8888fb84908be5ddce030e2f649767","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-05-11 16:07:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac937f9e7d69e1845abb6914fd5f72d84d57a215","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-05-11 16:17:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdd07822d4bc5721b5909cbc698c2dcb435af0c9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-05-11 17:40:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/1/upstream-check/manila-cDOT-dhss/d5ba0a1/ : SUCCESS in 1h 21m 11s\n- manila-cDOT-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/1/upstream-check/manila-cDOT-no-dhss/ec4605f/ : FAILURE in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"891e1b7b3c2e084ad9494ef50a345815bf816b45","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2022-05-13 08:53:19.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0ee01bbca4b5e96f5a0f2d7af251f1fdf7df92e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-07-07 14:59:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17cde65ef952a80685dd5bfbc22289ffeeb5c5ea","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-07-07 15:00:57.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"92f30750934c223d2cca114937e6c297dfdba869","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-07 16:05:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/841149/2/thirdparty-check/pure-devstack-manila-tempest-aio/3f5ec5b/ : FAILURE in 1h 04m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bc37762811fa0c288f276a263b742dd117116357","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-07 16:40:32.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-841149-2/ns5_manila : SUCCESS in 1h 1m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"631b5836b6a6d0739b311803ae370a6c24704729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:56:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b94345fab24c929758ca73032b8fab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea48560fb3c45919d26992e8ca5bc74 : SUCCESS in 8m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b00461b97c8435a819440cda5e89131 : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1285fb8b5be046c2a8458549dc868593 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd905b25abe4e72bf6bf6ca0feeb728 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b326cacab834478aed60f9fcc1fb6e6 : SUCCESS in 9m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd2468ea687c48c2a99740b800353d8c : SUCCESS in 8m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1464c212d2754ef6a725ca30300c6d35 : SUCCESS in 5m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0670f1162cbe4524b39f00739bbef9e8 : SUCCESS in 1h 01m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1e338aa1790a466c964c3faf85feaac7 : SUCCESS in 1h 50m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7461b68212a9414cb63a365c3f4a4c13 : SUCCESS in 1h 01m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/da8ab02120c94308a0e1e4101e39c866 : SUCCESS in 1h 15m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/80bc4277e96d4757af549daed24b6d10 : FAILURE in 1h 51m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c0151c6757d6422c9dedb9840b0de8c9 : FAILURE in 1h 02m 27s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e3c5e70ad2b340158c38c41055810967 : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50d87bf54e574c06835354d19eec234e : SUCCESS in 5m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1463d26ca5004642a4511a3d5c0ec479 : SUCCESS in 26m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2b40949837db4ab1b9d3ca3dbf5db516 : SUCCESS in 27m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8972d6cabe714d98a886cded7e6e8f86 : SUCCESS in 42m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d05bc0cac5d84e15bc46b9f43f73b21b : SUCCESS in 24m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c2e5205181fe44afa2e4f8afd3e1c299 : SUCCESS in 22m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e6d9bc4598a3412a8001a1a90b7d4daf : FAILURE in 1h 40m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9af4511aee174d63b5474337102d072f : SUCCESS in 1h 03m 29s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4ad40eda01d942ddaa3a96a1500b8f3c : SUCCESS in 51m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bbc01152218c42c3a8cf494bccd73387 : SUCCESS in 29m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f693866ce6f74d6fbd86a289c38ff5f8 : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6995f07a132d683ddae8f8f26b168e91b67b03a3","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-07-07 17:03:04.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/49/841149/2/check/DellEMC_PowerMAX_Manila/7247930/DellEMC_PowerMAX_Manila/260 : FAILURE in 6m 37s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"b48b847a29454c66f3ff7c0108a3f3ccf55d677f","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-08-05 17:38:26.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2a11c5bccf5eb125de15e853c5682562714bccf7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-12 22:39:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f149223b954fa2089af8ccf9235c9a42f67feeb","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-09-15 10:41:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0d3c4397f5a1235df62b212d9135ff37ef422e4","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-09-15 10:48:03.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4e06a497d165dceb00b78e6dd3fa63678a232413","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-15 12:35:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss/ce7512b/ : FAILURE in 1h 52m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/4d0cc14/ : SUCCESS in 1h 11m 48s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"58383fa52eaf4809f45c2bcad54570223ef69123","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2022-11-04 09:08:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e0ebb28ec9b66c8cecd48fb4664f2ee91e575a4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-25 13:37:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2265f0a70e24dff15787cf9d7d69df59720dc683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:46:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c45582fab554ff9b1cff4d87e85c7cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473fbdd8bd65433cb87e5c0efe9bc544 : SUCCESS in 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d19dce18816344968e1291c9a899d8bd : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a3d36c40ba643049b64f79eb58f14d3 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9247134ac1e4b5e987a359e1ce2474c : SUCCESS in 18m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ae6bc2984a64a9587b531947f2f89cb : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1584adf1468040c0b73dae90e7121acb : SUCCESS in 6m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c604f2d342384b00ab0f317007047ded : SUCCESS in 1h 15m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/da614247e2a54b0194a00dbf195e904f : SUCCESS in 1h 56m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b0258b7623ed417db523342c6a41edbe : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a5e09af1f82d4ef0a577ff9868f3f2c6 : SUCCESS in 1h 50m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a7da1f378a5e4c0687b80a78242161fb : SUCCESS in 1h 58m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f22302cbf9f8413ea6e45c85677a6c73 : FAILURE in 1h 02m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/57b91795d57942b99e9e00b92ec3e9e6 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86060542f5cf4d168773c92550769182 : SUCCESS in 7m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2fa4170b581e4658b58fcbb26ba0fab9 : SUCCESS in 28m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80fd3bf216d74df4a42ab24f86e8f06a : SUCCESS in 25m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7018bbeaf30b40d1b61470b27b47c3e7 : SUCCESS in 35m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0423014166f24e669a9c117bf66454a4 : SUCCESS in 42m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f674e74eddb941f5b4b6c72de98e2394 : SUCCESS in 31m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2d1144797087485ba43141d3cd08c077 : SUCCESS in 1h 46m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/28d1ebc954fe447089b7f0961a7f7afe : SUCCESS in 1h 00m 40s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/730bd646acc74629a09199d0818fc989 : SUCCESS in 1h 15m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ba495584cd5940bfb44d387f1215e63f : SUCCESS in 43m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/548e3577ae7f4963b56d7769f37475be : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee74799ff0dd07974bceffa2b8b29b42fc60e8f4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-25 21:00:18.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss/bb1f29c/ : FAILURE in 2h 12m 42s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/281965b/ : SUCCESS in 1h 31m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"95d6ee20efbbb7c389ebe25f9808ddf555e3bfa6","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-01-30 16:03:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bd10e7925f31e495651a9fd45c56d7915b3fb85","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-30 17:39:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss/d2b3059/ : FAILURE in 1h 35m 06s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/4ac11c9/ : SUCCESS in 1h 24m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0b8b74dbb257d25245070b7488cae0dea98721d2","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-02-10 13:00:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1a706a2f5efbdddc7607e108ac650c797185d41","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-10 14:31:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss/218fef9/ : SUCCESS in 1h 30m 04s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/f11b188/ : SUCCESS in 1h 14m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"14d70d13108cbe926e97a6060ba343b557f6d05c","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-02-10 15:19:55.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19b27e966428acfbc1f4f8eebe8d78cb660de035","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-01 17:59:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0842504bb1929b1dafe1a6f2595ecd4b90ee7d35","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-01 19:34:19.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check-experimental/manila-tempest-plugin-ontap-dhss/197e6c3/ : FAILURE in 1h 33m 32s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check-experimental/manila-tempest-plugin-ontap-no-dhss/38f6b5a/ : SUCCESS in 1h 19m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"efdc93bc855da4feb3ede9c115218bc9219e9168","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-19 15:05:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aeb48e838e2a5905b46af8a70762f77d6d34a5ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-19 15:06:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"828c303a6201c1a1d2548529b9b853c3071d4506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 16:49:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df71e709ad2e4cd297222f03ec9e6d1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720106a29a464bb4a50a6ab6baddaa6d : SUCCESS in 11m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/813b2efcd36840afa20d0e2318ebd8f5 : SUCCESS in 26m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1efeca3ad59428ea70177074ae8df91 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36d7deb08d1c4d33bba4696c7252c30c : SUCCESS in 10m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c42d990b0a14cb682c2c1eb81269c58 : SUCCESS in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/441916e9d6b44150879e1458cd4264e8 : SUCCESS in 8m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30c74fa0dbaf4a7b8b0eea8eb2377a30 : FAILURE in 8m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece3242c2ba148aa9c8f3dd2a25efef1 : SUCCESS in 7m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a08ee510e5f46ee84493afd4ea6f870 : SUCCESS in 47m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/82e63161486b4a54a7ee90a690b54b8d : SUCCESS in 1h 42m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ddae3126027f48daaa570c7271a43a18 : SUCCESS in 1h 14m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a8e55b30a1da447a80077bb840ce9fe3 : SUCCESS in 1h 14m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ddcc2048f91844e394155ec0fb164e21 : SUCCESS in 1h 17m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c5aca9d47ad4da2b78092e260816d7e : FAILURE in 55m 21s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3bf8ad8b9c2f4350811c3d5cbe6c1f11 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c36632b387b4947ab34ef985d7b4009 : SUCCESS in 5m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/01848e708769479db16b7fc00fccfdab : SUCCESS in 40m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d967a3f38224435bae8232f5e9bac47c : SUCCESS in 29m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d18a24bca8944b70a34d0db22f064eba : FAILURE in 41m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f2503a79637f481f9d800c2e0c9618d6 : FAILURE in 24m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dd1e4703814449849ea6a868bd5dc3ca : FAILURE in 24m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1d58632021cb4a6e9e762d4129f1f242 : SUCCESS in 1h 06m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7cd9332821b3414a91cb2307f6d032c3 : SUCCESS in 39m 49s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e333f06734d0497a8cff4e24adade26e : SUCCESS in 51m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ea0f62a8cc324689bef78815e6655c96 : SUCCESS in 26m 25s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/52b450dc0d5940018eb403d84fc0ad31 : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b26e602e2e0705b2cc2411fd6462f6ca5e37b757","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-19 17:19:14.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/841149/2/check/DellEMC_PowerFlex_Manila_dhss_false/7e9457a/DellEMC_PowerFlex_Manila_dhss_false/410 : SUCCESS in 29m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"312fa23b157394aa2120f1a3b97bfbb606699314","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-19 17:25:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/841149/2/check/DellEMC_PowerStore_Manila_dhss_false/c1f7406/DellEMC_PowerStore_Manila_dhss_false/176 : SUCCESS in 35m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"33cc38d2d2bbdbcfdbc8ad6841c51bc452b41ae1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-19 20:13:41.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss/af9a2fb/ : TIMED_OUT in 3h 22m 33s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-dhss-rest/3867dee/ : TIMED_OUT in 3h 19m 52s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/dd75195/ : TIMED_OUT in 3h 10m 15s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/841149/2/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/10984b3/ : TIMED_OUT in 3h 08m 46s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"4dd2e8e67a8318f890904327da53526bc5c7b74b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-19 20:39:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/841149/2/check/DellEMC_Unity_Manila_dhss_true/d408761/DellEMC_Unity_Manila_dhss_true/608 : FAILURE in 1h 29m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/841149/2/check/DellEMC_Unity_Manila_dhss_false/6ec549e/DellEMC_Unity_Manila_dhss_false/638 : SUCCESS in 1h 16m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b44a519e30eb899a5b0bfc189c8bdec0088f5f2a","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2023-09-21 12:29:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2e0b233a671c92c5de4185f4905a4860b6d27293","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-22 18:35:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9297bab1c70dbf95854458f98c3bfd43e2805f70","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:47:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4a5cbc73e11c95882b3db6743c9e2586001a6838","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-28 12:50:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c95ef7189a712f100a9a8026bb9b88f15c096b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:50:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e50ca704d24e1e02baca9dcf6d4575ae83cd984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:26:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140df44472574f989906c33a30c24f30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7602379b2f46029c302628345b63d5 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e79d916b6f449f826987e98ca0c99c : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c5bef37fdba4f7a8d6e55e24b853c4b : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/278e63289e8f4c4f883f304025f3b6b1 : SUCCESS in 13m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d6ead37f46b436f82bfa9a41ef3bb29 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cae0382bab642a69680774e179282c9 : SUCCESS in 6m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7266b6a9c54047c6aaf2e8b2bca3b3b2 : SUCCESS in 40m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b945c15c89f04d7f9c6eb858d748abee : SUCCESS in 35m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7aa6b705b4774562a5f96a95c86d5b29 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d6d1fa081f3bb39576e8332f3f0375fc1f890c44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:26:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"285de2b5a5c9a0c9ae706c6319764f70c2c93bca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:27:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c2ed8aa26a14161ace6b7cf8bedb419\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3956c1d8e11410dac3379e3e32681d5 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/401b5362322f439bb69b5106029bda68 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2},{"id":"7ceeb2e1fe7436952388a6db80495d2c090e16ac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-28 18:33:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d57c33d7a870924c3de6e4830d6facf71d47b098","revisions":{"4b1720b9998b58a8e3c828b0426063b2c34342aa":{"kind":"REWORK","_number":1,"created":"2022-05-09 16:55:17.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/841149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/841149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/841149/1"}}},"commit":{"parents":[{"commit":"9ba393af276dbae0444a7a98978d2426b7ebbe35","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ba393af276dbae0444a7a98978d2426b7ebbe35"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-05-09 16:50:40.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-05-09 16:50:40.000000000","tz":-180},"subject":"[NetApp] Fix lack of aggregate pool home state","message":"[NetApp] Fix lack of aggregate pool home state\n\nThe NetApp driver is not reporting the home state of the aggregate\npools. This information is useful during maintance tasks, since\nnot home aggregate cannot create shares.\n\nThis patch adds to the report netapp capabilties the boolean\n`netapp_is_home`.\n\nCloses-Bug: #1927823\nChange-Id: I8e98541d8e457e9e4609410853b50d6156465f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4b1720b9998b58a8e3c828b0426063b2c34342aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4b1720b9998b58a8e3c828b0426063b2c34342aa"}]},"branch":"refs/heads/master"},"d57c33d7a870924c3de6e4830d6facf71d47b098":{"kind":"REWORK","_number":2,"created":"2022-07-07 14:59:11.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/841149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/841149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/841149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/841149/2"}}},"commit":{"parents":[{"commit":"9ba393af276dbae0444a7a98978d2426b7ebbe35","subject":"Merge \"Drop lower-constraints.txt and its testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ba393af276dbae0444a7a98978d2426b7ebbe35"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-05-09 16:50:40.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-07-07 14:58:04.000000000","tz":-180},"subject":"[NetApp] Fix lack of aggregate pool home state","message":"[NetApp] Fix lack of aggregate pool home state\n\nThe NetApp driver is not reporting the home state of the aggregate\npools. This information is useful during maintenance tasks, since\nnot home aggregate cannot create shares.\n\nThis patch adds to the report netapp capabilities the boolean\n`netapp_is_home`.\n\nCloses-Bug: #1927823\nChange-Id: I8e98541d8e457e9e4609410853b50d6156465f61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d57c33d7a870924c3de6e4830d6facf71d47b098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d57c33d7a870924c3de6e4830d6facf71d47b098"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
