)]}'
{"id":"openstack%2Fmanila~848016","triplet_id":"openstack%2Fmanila~master~I8eda1c849d0d1916345959178f32756e8a1e9c0e","project":"openstack/manila","branch":"master","topic":"bug/1976370","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-05 16:41:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-05 17:48:46.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-07-05 17:48:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8eda1c849d0d1916345959178f32756e8a1e9c0e","subject":"Stop logging sensitive login information","status":"MERGED","created":"2022-06-28 17:45:12.000000000","updated":"2022-07-05 17:50:42.000000000","submitted":"2022-07-05 17:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848016-bug/1976370","meta_rev_id":"c26b9d57f3fe4344558bb0732d177e8bd731ac5f","_number":848016,"virtual_id_number":848016,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 17:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-07-05 16:47:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-07-05 16:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-28 18:10:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-07-05 16:47:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 18:10:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-06-28 18:46:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-28 18:48:31.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2022-06-28 18:52:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-06-28 19:30:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 19:37:59.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-28 22:11:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-05 15:11:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-07-05 15:11:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-05 16:47:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"5be46daba7a7f5013bebabec593ec3427a33c0b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-28 17:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2104e0d3c63d081340e453c5814f0667722ad5f5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-28 18:10:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf92215d6be8e2d5a3569b736fc4cf91084b7231","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-06-28 18:46:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-16-848016-1/ns5_manila : SUCCESS in 1h 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a9c00dd249721cf65a11e0b8e2c9de32ab5f07ba","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-06-28 18:48:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/16/848016/1/third-check/macrosan-manila-dhss-false/c86514e/ : FAILURE in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"af99e613c810aa008b8fc6de268bd42d5b3e0670","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-06-28 18:52:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/16/848016/1/thirdparty-check/pure-devstack-manila-tempest-aio/a1b5a8f/ : SUCCESS in 1h 06m 20s","accounts_in_message":[],"_revision_number":1},{"id":"cc9115bc070f8077651c92b9be704ea8b26bc9ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 19:30:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db9112a67a44e8eb7344c7bfa7bdf6c : SUCCESS in 10m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90ed46c6cea8403f8f6fcce7812a3f7d : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88402396ce02451bb656e35330786c6c : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ccd8a521c6d46459d45dfd6c72e33c9 : SUCCESS in 11m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2af4b0f7fb844c0db8fa0654f363272f : SUCCESS in 12m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b930fc63c5c4aa48014d7a40e85570e : SUCCESS in 12m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa54b14660ef4e5f8490da578ddfe22b : SUCCESS in 6m 04s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f5c29e72bdc4141975d2112e33f020d : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/35643518f5d5498484e86073f80b9552 : SUCCESS in 4m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/272c0f95874949ffaa478f321ae40bf3 : SUCCESS in 28m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea723720405f4378b8ae66c162cd8189 : SUCCESS in 30m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7dfe568a3ee44ca89945f6430ef69d74 : SUCCESS in 1h 15m 39s\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e70970f65c8046bba1a493e155840f77 : SUCCESS in 1h 44m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6b543cfd1179415dbcebf3e8c4e27d99 : FAILURE in 1h 15m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6fac2611703a4d17955dd4193108f188 : SUCCESS in 49m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6133f69af7c045ab9bae2c72b55aebd9 : SUCCESS in 1h 06m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/23d99efb0226481cad9a8a35ced92e35 : SUCCESS in 27m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c24db1769a54ece8b1895558865dc51 : SUCCESS in 19m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d31e51cb0eab41fab3873bc60244873c : SUCCESS in 1h 08m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b3b95aad8bb141a396fe3e231534365e : SUCCESS in 53m 22s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/09ccb26b2cc24c61b66e7fe31e8c2a4c : SUCCESS in 58m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/636fa7e0726b4861a115e3ca9da2f0fd : FAILURE in 31m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6f67d019c3e64a69815ec5ce472f0861 : FAILURE in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ddad21191aa9d8a78a85fcce41424b0ab61aa2d","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2022-06-28 19:37:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/16/848016/1/check/DellEMC_PowerMAX_Manila/ec18e1f/DellEMC_PowerMAX_Manila/252 : FAILURE in 6m 51s\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"3e974ca4a1e91f8f3e99e859c23b946136fa4077","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-28 22:11:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/848016/1/upstream-check/manila-tempest-plugin-ontap-dhss/42e958e/ : FAILURE in 1h 26m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/16/848016/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/4a3723c/ : SUCCESS in 1h 41m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"2e0f977b66d597dbb5cfff3e951623ff101a4555","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-05 16:41:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be3938da5370ab085676305abbe6263e7420ec57","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-05 16:47:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"369a7215508b8bcf4f11d3595fb54ca5e16c9d0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:48:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7c02107465ba11c224b281cd87d335203572d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4164831819be48c0912543827b4f3d9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbf58cf8e454e66968369d8c22931a2 : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56bee98482bc40a7a6e9fa6a3f31dea6 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd4a7b316b84e25909b7f11d88931e0 : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f158b1d759fc4b41ba2e70fad25f803f : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f19102860f06480fa681da5bed80731b : SUCCESS in 7m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8152ddb384da4e2584651a7018492715 : SUCCESS in 58m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f002520954884b3287607dc886ff721f : SUCCESS in 32m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ab2587fae602414599ad431f206078b1 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c95b237c7951877f063d0e37cb41d5f26b86a345","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c26b9d57f3fe4344558bb0732d177e8bd731ac5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:50:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1025405efb1a461a9ec24f56f1c8ef9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8458cf6dcced47a49fcaaf6c01e7e94a : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68ade7f6023f43fea575036e2372bc61 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"517966292af2eccb4f98520aba4374f463de21c9","revisions":{"517966292af2eccb4f98520aba4374f463de21c9":{"kind":"REWORK","_number":1,"created":"2022-06-28 17:45:12.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/16/848016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/16/848016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/848016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/848016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/16/848016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/16/848016/1"}}},"commit":{"parents":[{"commit":"8bc19b7171a0e283ba4eeee2f640a2e3f6b117ad","subject":"Merge \"Simplify composition of processutils.execute kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8bc19b7171a0e283ba4eeee2f640a2e3f6b117ad"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-28 17:42:48.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-28 17:44:46.000000000","tz":330},"subject":"Stop logging sensitive login information","message":"Stop logging sensitive login information\n\nusernames and passwords must never be logged\nby the service, even in debug mode.\n\nChange-Id: I8eda1c849d0d1916345959178f32756e8a1e9c0e\nCloses-Bug: #1976370\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/517966292af2eccb4f98520aba4374f463de21c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/517966292af2eccb4f98520aba4374f463de21c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
