)]}'
{"id":"openstack%2Fmanila~848366","triplet_id":"openstack%2Fmanila~stable%2Fxena~I4c10bcdfc499452d4b05dfbee8059be35dfe5586","project":"openstack/manila","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-07-05 18:50:42.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-04 13:20:52.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-05 17:50:43.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I4c10bcdfc499452d4b05dfbee8059be35dfe5586","subject":"Force disk wipe when running lvcreate","status":"MERGED","created":"2022-07-01 16:40:25.000000000","updated":"2022-07-05 18:51:53.000000000","submitted":"2022-07-05 18:50:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848366","cherry_pick_of_change":817206,"cherry_pick_of_patch_set":4,"meta_rev_id":"53a6247d63ea8520dd8556d1a77453b2795d3563","_number":848366,"virtual_id_number":848366,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 18:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2022-07-04 13:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-07-05 17:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-07-05 17:50:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 17:29:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-01 17:41:19.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-01 18:24:05.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2022-07-01 18:41:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 13:15:24.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-04 13:15:24.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"4f30b0dd8dfe7e03f1f0107888eab4892f9bcb79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-01 16:40:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5f6606bf823bdcb002f044a7563d9dffa1281271","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-01 17:29:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/66/848366/1/thirdparty-check/pure-devstack-manila-tempest-aio/d2f4903/ : FAILURE in 48m 14s","accounts_in_message":[],"_revision_number":1},{"id":"850f9aba8906ea997ad6545db0b7051fb5d53711","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-01 17:41:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-66-848366-1/ns5_manila : SUCCESS in 1h 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce82ab374b99704bf7604020e1b374322f4e0ed4","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-07-01 18:24:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/66/848366/1/third-check/macrosan-manila-dhss-false/5dfff08/ : FAILURE in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1ee5bdb61e11ff6145f0099f4f386752449cdc59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 18:41:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b72721521642689289f665169f6e96 : SUCCESS in 8m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3ed2fee4cbd49d899c23c4732c87fee : FAILURE in 30m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5aef870288422da232cc2091249866 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fae3b91fdc284941a968e570908d11b6 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa69c4c6ba7f422bb31874d8985f2bf5 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc1d382d8e6d41bbad5364634731c2d4 : SUCCESS in 8m 09s (non-voting)\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/83ff4631df124f87835eb6b848cab4aa : SUCCESS in 50m 42s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/8f4fe8a265324c278ca88d2e58a9ebfc : SUCCESS in 1h 54m 59s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/59f9546511d946d3bbf1e836e429448f : SUCCESS in 56m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/349bfde9ccff4cef88bff41c5152370d : SUCCESS in 1h 00m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/cd4bd07397fb43e394601a9ab81e7e33 : FAILURE in 1h 29m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/2e7a0b014fc748558605259c4eb1876c : FAILURE in 56m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4611ecefaf5643bbb4289d77d2c25501 : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb784583782c4cbd8f659645c633eba8 : SUCCESS in 4m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/00b45ce2d83249b4b7acf1209886ad12 : SUCCESS in 24m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0531bae326644e659c5f20344b3f6d4b : FAILURE in 26m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3d95aa9d58fa415ca3bb7f8253a01042 : SUCCESS in 40m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/61bd4d7222b2469b965768d8ea4baf58 : SUCCESS in 23m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d7b609ee9117401c98f9ddbdf7989156 : SUCCESS in 22m 18s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/91f34e612810446e8f38630055dcb864 : SUCCESS in 1h 07m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7bfe4df88d3d4a20947369f0d29948b3 : SUCCESS in 27m 19s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/11f7c88d3e344085bdc494560780336c : SUCCESS in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fbd905177346dfcda00ad00d447c9b89282df81a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-01 18:47:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78b64e57423a6b1a1327fc05aeac027339c0198f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 20:36:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ce3f91cbfa4a24a6e27f4ea33067aa : SUCCESS in 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f03a22d0094d0c83d494120e00d423 : SUCCESS in 27m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e6bf8e363344195aad5468cef207cc7 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/645744cf2f1e4e98b32557b2440fd42b : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34c54f3800bb4b59a27c8b468cf38471 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60c1ebdd0b484692b8177e19f78af5cd : SUCCESS in 15m 00s (non-voting)\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/83914198385144c7864ead280b8ac028 : SUCCESS in 56m 28s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/1737ddda378e4d8ba6d58bc3f8a7056c : SUCCESS in 1h 41m 58s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/704060e1344b4c45b61456a9aaf502ae : SUCCESS in 1h 02m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/f1980b720a4b43b09d8d4cc46d7a21f1 : SUCCESS in 1h 12m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/e691471683fd426bb8ee37a3574608bb : FAILURE in 1h 47m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/26600f37bb964fa695cd99502a6601c0 : FAILURE in 1h 00m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8ee84f622c804423a7cfd1569fc866e4 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8f2fac7c22448588bdab4f6af56ad81 : SUCCESS in 6m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/11ad70f0ae4842fb86bf7c3387d3e7f3 : SUCCESS in 28m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/043bed34bf324a919957c83d36f6d1b2 : SUCCESS in 32m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2d53dfda68384845bfde41c98335ddd4 : SUCCESS in 42m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d5377764d55440c7ac6a6cb594df5ac6 : SUCCESS in 27m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d9a8f0a284214637b6561cce02907dd9 : FAILURE in 28m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a71689d97055411dbd204415faf67f6c : SUCCESS in 1h 06m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/570911330bdf46a4a461dc651c6df0ee : SUCCESS in 30m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e23fd266194e485a92748332e9122e06 : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c21548ff7cac5a299171f25191109a11e9da503c","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-04 13:20:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73f9ad97d5b576b4368c6857731c40665b78c87a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-04 13:22:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bc91c388dde95d4a0d569504ddbef95dedf7a75","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-05 17:50:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0641b8bd7ded2fe36ff09fe022ba16bb0d43608f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 17:51:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"794ea4c7d3e842405cac2807ca681dc495b5a436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 18:50:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece0bc7e18f944f2b69b7b44661fd151\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3f2e2322ece49a3b73a1a526a4bd0d5 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4678dedb84984f6bbaf764467af04361 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d18025873e4b4fba5897fafe565eca : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d58cccd37034062bd922ead4b722a94 : SUCCESS in 14m 22s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/5d70b6f7471e42e0b1fb1dd585c1bf18 : SUCCESS in 58m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a6bce1f954ef4d839bb32cc5db7aae1c : SUCCESS in 27m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/54c9e8dda72145fe9dec402c67737a11 : SUCCESS in 33m 56s","accounts_in_message":[],"_revision_number":1},{"id":"564e823b8567b61f91f7334e267eeb88089b2821","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 18:50:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"53a6247d63ea8520dd8556d1a77453b2795d3563","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 18:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905642c941a3475087d13a2aaa35cd6c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0aa7a8f8a6346b5b90d266573155ba9 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63c3757951db1a259b4cd52ee09d985574117bcd","revisions":{"63c3757951db1a259b4cd52ee09d985574117bcd":{"kind":"REWORK","_number":1,"created":"2022-07-01 16:40:25.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/66/848366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/66/848366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/848366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/848366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/66/848366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/66/848366/1"}}},"commit":{"parents":[{"commit":"3e4d1292ff21b93d8fdbc0b88c41839d643970ca","subject":"Use defined project config of manila-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3e4d1292ff21b93d8fdbc0b88c41839d643970ca"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-09 13:42:47.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-07-01 16:40:25.000000000","tz":0},"subject":"Force disk wipe when running lvcreate","message":"Force disk wipe when running lvcreate\n\nThe LVM driver may fail while creating new volumes using the\nlvcreate command. Sometimes it can say an existing ext4 filesystem\nwas found in the destination, but we shouldn\u0027t be worried about it.\nThis change modifies the behavior and forces the lvcreate command\nto wipe the space if a signature was found.\n\nChange-Id: I4c10bcdfc499452d4b05dfbee8059be35dfe5586\n(cherry picked from commit 89028725d1ac6890332f4c06a30c2416ea951aff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/63c3757951db1a259b4cd52ee09d985574117bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/63c3757951db1a259b4cd52ee09d985574117bcd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
