)]}'
{"id":"openstack%2Fmanila~848864","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~I8eda1c849d0d1916345959178f32756e8a1e9c0e","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/1976370","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-07-12 19:17:41.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-13 12:39:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 13:51:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8eda1c849d0d1916345959178f32756e8a1e9c0e","subject":"Stop logging sensitive login information","status":"MERGED","created":"2022-07-07 14:48:14.000000000","updated":"2022-07-13 13:52:36.000000000","submitted":"2022-07-13 13:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848864-bug/1976370","cherry_pick_of_change":848856,"cherry_pick_of_patch_set":1,"meta_rev_id":"16f76aa996418d9f74d6d77d566716e64697c838","_number":848864,"virtual_id_number":848864,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 13:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-07-12 19:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-07-13 12:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-07-13 12:39:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-07 15:19:18.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 15:36:42.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-07 16:35:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 18:51:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-12 18:51:20.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"7ea745c72bf091109e16fa1f58f8a2d2f1b14c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-07 14:48:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"87de90b294c028cc1a397518b985cebd2fd305d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-07 14:48:19.000000000","message":"Topic bug/1976370-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"7b02c92847f24a4e317cae133f5c17bf1e37c339","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-07 14:48:21.000000000","message":"Topic set to bug/1976370","accounts_in_message":[],"_revision_number":1},{"id":"b48cbc587bc6bf850f4275239aa36d75f84d5145","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-07 15:19:18.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-64-848864-1/ns5_manila : FAILURE in 25m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"e4ccad9a659f4e444bda07c51f03c3cdcafcdf62","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-07 15:36:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/64/848864/1/thirdparty-check/pure-devstack-manila-tempest-aio/60fd2be/ : FAILURE in 47m 29s","accounts_in_message":[],"_revision_number":1},{"id":"148e5c0c0699345b9d6bc1bee8690474e3804fce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 16:35:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c33c45b6ab4a89af677bd00b6380ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f14db37fae4e75b9b59246c3fefc25 : SUCCESS in 8m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef31590973064facb40d2eb19390647d : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508d6b50eeec4c31843d4219029eadc6 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8304a80b574b4c4d8609487613b6ffcf : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1e02887ffb74c74b43f43cafe765436 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7feffe8d19e43da8c9fe88057086ea3 : SUCCESS in 9m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f9a9755aa6f44ba87361ed2e6ce01f3 : SUCCESS in 6m 18s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/2578c002929d4f518c2acba1ea8af4f8 : SUCCESS in 55m 14s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/30ae5afa63da4f0683fb256ca56ce219 : FAILURE in 1h 45m 11s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/7fa4ce2fd39f4c198cfea33067d4486a : SUCCESS in 1h 14m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/891cdd1188854b759d12a0d448b56092 : FAILURE in 35m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/93935a7cd80e4902a6f7e24d9ebc43ac : FAILURE in 40m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/dd31ae1258c4480e9790eaa36bc468d6 : SUCCESS in 1h 06m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0b8ec08b2954420ca9775168d52068e5 : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0c4133fffd841b693f5ab26ca0c17b2 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9bba68d647894bacb7ff731facd6d8f7 : SUCCESS in 24m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5dcbd3147ef844719102236a10c6a556 : SUCCESS in 46m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05f2539e1a3f4c66bd48da4e3cd9ef74 : SUCCESS in 35m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3cad94973c3b4a7cb252b25e709a5186 : SUCCESS in 34m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/68d41d7cf612401a94992930a62099ee : FAILURE in 27m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/abd994337c274e628fd6d445a131c4a2 : SUCCESS in 1h 07m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d6c55b34a39345cab681338db0c9dcf7 : SUCCESS in 43m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c39952371c704d9bb3212a5bbb4237a9 : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22d6ae2e42a683808ead438208fa6fac710e66e2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-12 19:17:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b41931d6f8962babe527378f3e74eb0a42a6bd7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-13 12:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60ad5bb3fb7ef9c0e5b6cd82c12b53ca641c620b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 12:40:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf146ed838dc3a9e5e97b93206629ce6ecd3ed7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:51:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ed9fffe02249c9adfc044e6cb70d51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4a9d5389df40ada7c6930fcfdfbd4c : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a919eb7966524d0998c86fcdcda607de : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5fa64c7ab844fbdb2db8423aa926fb0 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a31c12a778784a95b9e2f878f334d10d : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe5706b7b424487b223589d857c22ca : SUCCESS in 6m 34s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/5a84d94e243b4ce7bcb4cb8ba33ad3e0 : SUCCESS in 1h 05m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2f3ccf9191424b29a1ea87e352cfa4cf : SUCCESS in 34m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/20f5f256907a4ebd8ad338ca37d1c11e : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":1},{"id":"24e97a59ac451fb435aed84d4bc5070a5bde6968","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16f76aa996418d9f74d6d77d566716e64697c838","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:52:36.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bac06ae38a264f6197cfda726c8ae7d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fb029de09aa49b9a65fdd94502ef598 : FAILURE in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521d2a085303471c9852ea734327d8e2 : FAILURE in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"732bf38f5d7caf00489901b0a5f09187c72f4e6d","revisions":{"732bf38f5d7caf00489901b0a5f09187c72f4e6d":{"kind":"REWORK","_number":1,"created":"2022-07-07 14:48:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/848864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/848864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/848864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/848864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/848864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/848864/1"}}},"commit":{"parents":[{"commit":"71c906e6015bbf91ee9363b38030b1561398baef","subject":"Use defined project config of manila-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71c906e6015bbf91ee9363b38030b1561398baef"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-28 17:42:48.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-07 14:48:14.000000000","tz":0},"subject":"Stop logging sensitive login information","message":"Stop logging sensitive login information\n\nusernames and passwords must never be logged\nby the service, even in debug mode.\n\nChange-Id: I8eda1c849d0d1916345959178f32756e8a1e9c0e\nCloses-Bug: #1976370\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 517966292af2eccb4f98520aba4374f463de21c9)\n(cherry picked from commit 48f4bf6a168a8ec59ea26e44e1bdcf262ae96e90)\n(cherry picked from commit b7190fe3e72a9f9fa4ec5adfb1842d0fd85529d0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/732bf38f5d7caf00489901b0a5f09187c72f4e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/732bf38f5d7caf00489901b0a5f09187c72f4e6d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[]}
