)]}'
{"id":"openstack%2Fmanila~849849","triplet_id":"openstack%2Fmanila~master~I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","project":"openstack/manila","branch":"master","topic":"bug/1982808","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-08-17 14:55:58.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-03-18 02:32:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-08-17 14:55:58.000000000","reason":"Change was submitted"},"30555":{"account":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"last_update":"2023-08-17 14:55:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","status":"MERGED","created":"2022-07-19 09:49:36.000000000","updated":"2023-08-17 14:57:38.000000000","submitted":"2023-08-17 14:55:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849849-bug/1982808","meta_rev_id":"1c3ff5d5d6261dc2fb4e24c6e8952460a60c1c25","_number":849849,"virtual_id_number":849849,"owner":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-17 14:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2023-03-02 16:54:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-08-15 22:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2023-08-16 02:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-17 13:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2023-08-17 13:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 10:50:37.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 10:58:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-07-19 11:50:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 15:09:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-28 18:15:08.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2022-09-13 17:12:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2022-11-04 16:46:33.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2022-11-04 16:46:33.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2023-01-26 07:43:33.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2023-03-02 16:54:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-03-02 16:54:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-03-17 13:40:30.000000000","updated_by":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"reviewer":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"state":"REVIEWER"},{"updated":"2023-03-18 04:04:52.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-04 13:54:21.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2023-08-16 02:16:39.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"ff5efe5b0bbc5f5f2e9326ebcd1d15a385f04a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-19 09:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a5136410ddea27ed14762957ffb3689a2959bd0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-19 10:50:37.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-1/ns5_manila : SUCCESS in 1h 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"201b11cc992dedffcf01db82f275390518325f59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-19 10:58:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/849849/1/thirdparty-check/pure-devstack-manila-tempest-aio/484f061/ : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"308757834785c8345b9ab7df1d8aa5cf1f757248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 11:50:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9772005a3d484afcbd55acb233de0ad6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae879a9b6c224ea496a00900bc368e4d : SUCCESS in 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42445a5ff47d41c88ee07becc195be56 : FAILURE in 21m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b6a71d419843acab846e04c0d7ea5f : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39153b4219224f26a841f3db737456d7 : SUCCESS in 8m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dba9afffe6749df8f5dc022906a8a78 : SUCCESS in 10m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/651d06b172ed478e9ee01dd55679583a : SUCCESS in 11m 20s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b1e3b618ac94fe0aa810e25a6589ee0 : SUCCESS in 1h 12m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c186cdecb9ce49579e54b04b0041fe74 : SUCCESS in 1h 58m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/009ede2fe3cf4762a225e3c883dcc550 : SUCCESS in 1h 10m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/453d5291cbae4a64bcd38d08990a78a3 : SUCCESS in 1h 23m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fbe683a6b40a4deab6397dc6b2a57c30 : FAILURE in 1h 43m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/07a2ed38bf514e8c941d39ab96d25167 : SUCCESS in 1h 16m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/93dee4db3b1a4e0f98de86119bff3cad : SUCCESS in 5m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bf76ce6784b543b2a3d3bf4f0f217a27 : SUCCESS in 6m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ee4dc92277d4e2f95dd4c88df7b1d7f : SUCCESS in 34m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a0f949f34ae84cb9b8677bd94b019f7c : SUCCESS in 42m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5ffacbc27754ae9b82fad0ce046c263 : SUCCESS in 44m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5863fdabf9c7436e81ff61ad5d59c603 : SUCCESS in 41m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/07b5a8bdd8f04e5982a0c8bc9dfe57ad : SUCCESS in 25m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/42e29aa33c794905b2bd40d63f265875 : SUCCESS in 1h 37m 59s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/57f80319372c4b6a88a1af7c64af2963 : SUCCESS in 52m 45s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f6b6bc6e07824fee893a75150832b572 : FAILURE in 33m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/450ba0e2cf98425aada751669755727d : SUCCESS in 47m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e989f038e5aa41b6b5855ca9d6b65902 : SUCCESS in 49m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afe0b5b6b2273cca5d3506eef8f28d78cba4de97","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-19 15:09:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/1/check/DellEMC_Unity_Manila_dhss_true/4f65d6b/DellEMC_Unity_Manila_dhss_true/647 : FAILURE in 1h 22m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/1/check/DellEMC_Unity_Manila_dhss_false/0a0cf50/DellEMC_Unity_Manila_dhss_false/680 : SUCCESS in 1h 00m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"66d83caeb540ecce378ef3e7e535a8cad88741bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-26 06:53:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a90f5eef5bccac9c1cf7a616d71c2d9c7dbb764","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-26 08:04:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/849849/2/thirdparty-check/pure-devstack-manila-tempest-aio/15a8f1c/ : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b7482d98a170db1e858a8279643c5601e19f37d5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-26 08:07:27.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-2/ns5_manila : SUCCESS in 59m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"f64bed3cc59f4ea60a5d0027b327ee0cfe229e8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:57:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87e156c2a81743ba92d8f1ad73fb1e98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6de707a59a457db460e5f0bc6daae2 : SUCCESS in 9m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5880d6553d24fdf867e1ae559d36219 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3534bef62142fb914a7a02d9873d19 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa83957ba4e1482cbf0fa5ab36733729 : SUCCESS in 13m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a87682a228004186b06db6fd4f743701 : SUCCESS in 12m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c1fce8a6f4c4b1f807526e1fdd251d1 : SUCCESS in 13m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c8a64befeee4e3db3516ab3cac74ff4 : SUCCESS in 5m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/224308650e68411f82c243d386f05a2d : FAILURE in 1h 17m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/43d5c4acd7014d68bb58dbb6171f0988 : FAILURE in 1h 55m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e0939336b1784d1288c0c064035129d3 : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7b61e3a0b4c54b55ab85d3c8fdb4a7d2 : SUCCESS in 1h 44m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/169cf5f166e2402e9c9160fdee023208 : FAILURE in 2h 01m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/abc7c487704541fe81c600e72d4cdf2c : SUCCESS in 1h 14m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5d7520dcf1a74d16a43f6d6f8fbbb049 : SUCCESS in 4m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7b1f4a3c95c4c82a13b26000d68fa88 : SUCCESS in 7m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c7ce66f2497b40d8b4714103010bb3df : SUCCESS in 40m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/24241926e87044caa996e599ced3db4e : SUCCESS in 40m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fc603e14533d449eae647e2b3a9102b1 : SUCCESS in 57m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d66e6ab40ee1452eaae07f9042f7e903 : SUCCESS in 40m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ce8436743b2144a5a5e54200a09a6c13 : SUCCESS in 33m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0c1b4ec3297c42a19bc5ae84fd60b6d5 : SUCCESS in 1h 29m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/11b8964072954a088bdf839932685cc6 : SUCCESS in 1h 07m 06s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ab85cbd91552436ea317b15c31b26d61 : FAILURE in 42m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/756d8aad92e24f259755e0dc3882a14d : SUCCESS in 46m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0417359397ae45ca915606c7fadb74f5 : SUCCESS in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d54f6ec89a750348ba7d45a9b4a6e919b38ef72c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-26 10:34:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/2/check/DellEMC_Unity_Manila_dhss_true/7b53917/DellEMC_Unity_Manila_dhss_true/657 : FAILURE in 3h 34m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/2/check/DellEMC_Unity_Manila_dhss_false/c77b39b/DellEMC_Unity_Manila_dhss_false/690 : FAILURE in 15m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7b7d6713d7c6f60e925876f3cdd7f7a11316f5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"date":"2022-07-28 16:20:15.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c25cc2998eeea117fcdd9b6553401a8ca9acde1d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-07-28 17:31:49.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/849849/3/thirdparty-check/pure-devstack-manila-tempest-aio/889f4a6/ : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":3},{"id":"6bb8dc27cc829e5b379298f4165f821224f3c826","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-07-28 17:41:07.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-3/ns5_manila : SUCCESS in 59m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8cd2f36d0472f7599725df9a15bc16bb0b3227fd","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-07-28 18:15:08.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- macrosan-manila-dhss-false macrosan-manila-dhss-false : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"d5962720a2940bab6cbbe6b572dbf451f4135dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 18:35:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3249e0e6d5c44da294d60ee8ddf6cde9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74915011c72542fc9e38e26ef06037a8 : SUCCESS in 12m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43965cc9ee44453b9d325d172d1a1887 : FAILURE in 31m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02be773917cf47598aa2dd2f1a8a8c0e : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d468069a2bb433f843a5fa14b52f9ff : FAILURE in 18m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0674fca63c22445494596e466a2ff8e9 : FAILURE in 18m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b5c912ee25c44dbba0208e56d41bbc9 : FAILURE in 19m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3810c5fb092d4ac5812a98f103ab33e6 : SUCCESS in 7m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1dff1f21dca4438585e7341020715f3a : SUCCESS in 1h 26m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a0382612e570480ea7dcb9a2dda97528 : FAILURE in 2h 13m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf45245a44dd4bd7ae4c6775b1462354 : SUCCESS in 1h 20m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c2cc43ce93504627a737ced392ceee16 : SUCCESS in 1h 33m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b809163d421e4fb3bd0430174f21f9e9 : FAILURE in 2h 02m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/853d1e564dae448cb3534cfa82f213b4 : FAILURE in 1h 38m 25s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/690b20f7736741649949f351eac1e012 : SUCCESS in 6m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/163f1fe6f0db4d589fe42c4f4132e575 : SUCCESS in 5m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/50ee6718a96e4f7b802e7eea5a722684 : SUCCESS in 42m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e66816a67ff84be6b45c207b99ee16ea : SUCCESS in 44m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29b5efcec57149048803d2580fdd3e6c : SUCCESS in 52m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a63805378f9b43b8bb8dce95aa6aa405 : SUCCESS in 39m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/195b8b79a76a41839ea63f13d212311c : SUCCESS in 34m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8edee4d60959462c8bca3ca4cc21ed93 : SUCCESS in 1h 37m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5f22f03ed45b403dbc6e4d7e6f44573b : SUCCESS in 1h 08m 57s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/adde0dba92cd4ffebb39b95e2ca8f982 : FAILURE in 42m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b809937628784cbb82082bc4df720f3b : SUCCESS in 44m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d5f64b854ac5421596e2a7cc4f375f82 : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a59f50b3a96779707391abddd045bf6c4a3f902","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-28 20:18:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/3/check/DellEMC_Unity_Manila_dhss_true/bfdbebd/DellEMC_Unity_Manila_dhss_true/664 : FAILURE in 1h 18m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/3/check/DellEMC_Unity_Manila_dhss_false/bccc21c/DellEMC_Unity_Manila_dhss_false/697 : FAILURE in 17m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"2d8194a05a3979394e67b22650c94339bc0aaf8b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-13 17:12:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a7e7477fa3eed783dcc1ef43e31dd3f3944ac56","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-25 14:08:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"64438e9fec970b9da2d3127f335a69c538b95fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-25 18:55:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e924f0ac795422484da7b1792fda0cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431b59c5119e4697b1dadb481caf00dd : SUCCESS in 13m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccd90ea865da4d89baf8c4dbf69ca8cd : FAILURE in 37m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73edaf8d9bd14874b13a436272c44f75 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bacb1efd86445189646c6a3027b3433 : FAILURE in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdf20d76eea84815a63622ab3f415067 : FAILURE in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/426dc0082c6f483bbe3f3acc4ffc672f : SUCCESS in 6m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/acb117feb8614bdb9869eccff116954d : SUCCESS in 1h 08m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2d31bc1a2dd945b38d7f19e37bf8f67f : SUCCESS in 2h 08m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fb3ba7b8825d40e283562554a576bc2f : SUCCESS in 1h 08m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db5175ecfb78459b8093b6430d07165f : SUCCESS in 1h 15m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b73ed2c304e45c99285a95706436b67 : SUCCESS in 1h 37m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/142e361389a349489fff275b3fb931d6 : FAILURE in 17m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9032ef6c7d340379be08eb191d7d4ce : SUCCESS in 5m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecad3bf809ef4eafa331da906b85d147 : SUCCESS in 5m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c25e67660d1b4937a223361d0f7d96fa : SUCCESS in 25m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c12abcdaae2a4053b680d4653e51dcbb : SUCCESS in 39m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8d5f510c2c764603abb1f6fddf553445 : SUCCESS in 48m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/42d59326da0440a2ba98bae4f3d80671 : SUCCESS in 28m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d03fab20c8974ca29e36b94eb39e3d32 : SUCCESS in 22m 01s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e839fe695d8d41d3aef637bc9241185e : SUCCESS in 1h 17m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5ff673694e7e4085862d2597c994d607 : SUCCESS in 1h 05m 51s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/eb687f5289294a7385605540a7169a82 : FAILURE in 1h 18m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c032d51d2e0e49349a5294c2712eb002 : SUCCESS in 37m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/08142b951ac14d048d7a943412d82925 : SUCCESS in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41f43ecbe01b212acfc36db4c0e0226ced6b511a","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-26 07:43:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8ed637def4f9d16b25a875883f3c72aaff70026c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-01 18:15:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d75d18df42ff7a6b01e95de18bc000b33d7d07b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-01 18:36:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"153af0c1b8d101f2fcef68de1aca5b9596cf3b2d","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-01 18:37:22.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"70b407d00be283858b039e311791a57082ba07b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-01 20:01:07.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/849849/5/thirdparty-check/pure-devstack-manila-tempest-aio/85b231d/ : FAILURE in 1h 22m 28s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/49/849849/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/32b9a1e/ : FAILURE in 1h 02m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"31982e29294ac248e774dd1ea665e286640592ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 20:53:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f257aa215a0f43e78e8196f9e73c0ab8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a345da6f0eb47dab507a639c98c9955 : SUCCESS in 9m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c37654b2cbd347d7bbc7f7114c46a8de : FAILURE in 18m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628b18c8187e4a82ba8357224d7a7a0f : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d406ad81364ca68d4ce59c7bfc6f9f : FAILURE in 9m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07b9a0828b134860b8da41779688a0f9 : FAILURE in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/324f601355bd43c691cf1f7e6c23b0a0 : SUCCESS in 5m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1ba0c425177e4132bc2dfb78774d1eae : SUCCESS in 1h 05m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e5604ea788aa4f6a9cb46e6b3044fb89 : SUCCESS in 2h 10m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/36eed07624e34a0184586d35a5053980 : SUCCESS in 1h 08m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1c16fc3bd5f148c292fe7cad247a2ba3 : SUCCESS in 1h 37m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/feb755cb916246c39e11216acdab2c1a : FAILURE in 2h 06m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e97e7c08be28486da16584704696dd62 : FAILURE in 1h 21m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/81b3751483444bb3aaf238c3f9ae7d4a : SUCCESS in 4m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acae86e831274f81a7b57de7bc7e0e93 : SUCCESS in 5m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cf09d68cf8a24cb1ae579057d3fc0b4a : SUCCESS in 29m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c78d2a64091d426f95bdb6832f63cf94 : SUCCESS in 34m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d27142f9d3bf4c3fb3cb906cce31f686 : SUCCESS in 52m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ae3293371484420db62617ad9b8be418 : SUCCESS in 35m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4ad59051a7e8483dacee63542fd059f8 : SUCCESS in 31m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0f18ca2871d14c30bedd363ac83377a9 : SUCCESS in 1h 33m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/63723ca36a734297a86c57667faf239c : SUCCESS in 1h 02m 48s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c1edd2a0f574d3d882494797079b31e : FAILURE in 1h 10m 40s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2e7b30c0121440e8a8fd798a35731cd7 : SUCCESS in 28m 19s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/190693bd75834353bce44783be92d2bc : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e6692f85362bcad7aac184b3d32ecfbb548f181c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-01 21:03:04.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/5/check/DellEMC_Unity_Manila_dhss_true/51ad8d0/DellEMC_Unity_Manila_dhss_true/1329 : FAILURE in 51m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/5/check/DellEMC_Unity_Manila_dhss_false/d346d17/DellEMC_Unity_Manila_dhss_false/1429 : SUCCESS in 52m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"290547a5bf98f718070d246c9ac5c1c83db9ef76","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 21:28:31.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-4/ns5_manila : FAILURE in 1h 2m 22s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"25b5800420aaab65d2584aa705c717f063844139","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-01 21:42:59.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-5/ns5_manila : FAILURE in 31m 46s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"f03617198fb30ccbd6199ba718d4a99777f9256b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-02 11:33:46.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1f96bd8092ec9313f0f0e5cc4f1126291c60d928","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-02 12:13:37.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-6/ns5_manila : FAILURE in 39m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"0039ac59e2bcec4a9fb0e98226a14cb9a5e31188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-03-02 12:24:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3104f541552b0686f3feb169096ca2617857ace8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-02 13:23:07.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-849849-7/ns5_manila : FAILURE in 57m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"5926e78224ee43b01d842bfbb5d6c000569370f9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-02 13:39:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/49/849849/7/thirdparty-check/pure-devstack-manila-tempest-aio/5a8918a/ : SUCCESS in 1h 12m 20s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/49/849849/7/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/ff0b5a5/ : SUCCESS in 1h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d38605af2ef9807c5addbfae7cdfc8809aea76b7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 14:19:49.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/7/check/DellEMC_Unity_Manila_dhss_true/6c0f469/DellEMC_Unity_Manila_dhss_true/1339 : FAILURE in 51m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/7/check/DellEMC_Unity_Manila_dhss_false/db24a3d/DellEMC_Unity_Manila_dhss_false/1439 : SUCCESS in 1h 07m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"d67ec61a71aea3da659e325acc9b2f34bbb7a66c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:36:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ffdba17cfd14a4e902271bf9cdf847e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa644653cc54fd786475204fecdeeeb : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c5b3e0c07194ced85315e707a19fbe4 : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2bb340e80242e0b18f16245940c9e6 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4633c10687e46fa84d02b38c88dab9a : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89284876679b4de299732248fea97ec7 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6bd016415c45ccaddf52b914870adf : SUCCESS in 5m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c53d1455ad304a39b8c3762a1e1d3a1f : SUCCESS in 1h 27m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/91980ea242e144e8888dc080aaad0bed : FAILURE in 2h 00m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/24a4fb41d15446f0b627d73eb6980f26 : SUCCESS in 52m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/35db0bf8c68943c6b51141e07da40927 : SUCCESS in 1h 23m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3890b9efd99148da8da1fbe9b8a3f310 : FAILURE in 2h 09m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec2ea64797364639a35bfe0893a6f631 : SUCCESS in 1h 10m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/747f1d1336bf4d76bb26d95b31fe4819 : SUCCESS in 4m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/07da3a63e87d485da5d4b5e72c20aaf8 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b4e2a5defa5f49cd81714969fb195e6d : SUCCESS in 36m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ee654de8d004d3481971059d6ededd3 : SUCCESS in 42m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/74ab51121b8b401e84e75a706a5f7241 : SUCCESS in 52m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/48a619d219434eae8be30a268ffa2158 : SUCCESS in 37m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/006ff9cecc074fb29e3ebe208cca271a : SUCCESS in 31m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1d73221c5c824b25897664a75286d617 : SUCCESS in 1h 21m 29s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4b85fff6a34645599b567bfa7bdadf5b : SUCCESS in 1h 02m 15s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d61e7c5d4d764403aa14c175274279da : SUCCESS in 57m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e279adb39bf14c5f8f7e9457b688a24e : SUCCESS in 30m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/056133131c5b4a088d2d59d8ac74e31d : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cb753bb9bc31bd9f0c696235905f0945a7d07c6e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-03-02 16:54:17.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b2e100d274190cd41d960cc151c34c00f3cbcb6c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-02 17:00:19.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/849849/7/check/DellEMC_Unity_Manila_dhss_true/7fe7974/DellEMC_Unity_Manila_dhss_true/1342 : FAILURE in 54m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/849849/7/check/DellEMC_Unity_Manila_dhss_false/f3e7e5d/DellEMC_Unity_Manila_dhss_false/1442 : SUCCESS in 49m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"524af42117c3dbcd87bdda42409709a0844fa0a0","author":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"date":"2023-03-17 13:40:30.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"31c262f70c2d889afd54a375357c864c3f272994","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-18 02:32:09.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0aac6a8ea351a0e70c2df48acdeb289ea32073be","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-03-18 04:04:52.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check-experimental/manila-tempest-plugin-ontap-dhss/2fe4855/ : SUCCESS in 1h 31m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check-experimental/manila-tempest-plugin-ontap-no-dhss/7212812/ : FAILURE in 1h 25m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6f78925a0265e83aec22137117c1f0775950815d","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-08-04 13:54:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9e54375e68ea14c7c0a66fbd5d8e20bfcbe65df7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-04 17:15:56.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-dhss/fa4b49c/ : SUCCESS in 2h 18m 53s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-dhss-rest/28b3541/ : TIMED_OUT in 3h 19m 15s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/7cf9c59/ : TIMED_OUT in 3h 10m 51s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/de40903/ : TIMED_OUT in 3h 08m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"f9208b5c304f7a6e28025fb2e6e73fd343adecc4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 22:58:24.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"82d31c9dcefce8be686687f68e6467d8f21b39ff","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-16 02:14:27.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-dhss/17ccbb2/ : FAILURE in 2h 11m 41s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-dhss-rest/86a90f2/ : TIMED_OUT in 3h 15m 20s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/66d4bc3/ : FAILURE in 1h 34m 45s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/849849/7/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/29ab996/ : TIMED_OUT in 3h 08m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"891bf3db4aab20c7cda08674e5323d22393b0f28","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-16 02:16:39.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"7becabbf5cbf0e50fa964213cd3b3e4d0f00cb7e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-17 13:36:02.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"9a72e57344fabe27592ccc993aab6ec4cdd19f3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 13:36:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1c5848de3bd0d82a17ea235db4488b5794b65e75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:55:57.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e2019911d54bbabf59016dbea406d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b2d190eff44430a783135dd45f3a1c : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04a47fc52f6f4bdd8786db7dd53a956f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a51a06b003844a02aaa83d4c9aa228b0 : SUCCESS in 11m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/724e33968eff4a23a2394da74a2c1846 : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe211aea14374793bf4c3a9b2264551c : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dada10738e5c4429891b2edf6824db44 : SUCCESS in 7m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/19f47f0c7d8d4a1c983eb08bc3074faa : SUCCESS in 1h 13m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aaec574070644bb7bfbfaa1b60f46ae8 : SUCCESS in 30m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aa1b5762d98b4ffd8e8afde0bbcb15d6 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":7},{"id":"a9571ac8c8a56a32229e6e04f59222bcf4f00c5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:55:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1c3ff5d5d6261dc2fb4e24c6e8952460a60c1c25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 14:57:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e690624d5b9043c48b2e2ecdee780e62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d867291feda04713905bd0d31d1fc5cc : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ced0c886c14408397d24f142ebbe402 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"38b13bb40b879adcc06cef0b0fa629e4b00f7122","revisions":{"f71c5369293407a0cc15fd727d16404a8de98944":{"kind":"REWORK","_number":1,"created":"2022-07-19 09:49:36.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/49/849849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/1"}}},"commit":{"parents":[{"commit":"205d716bca8a60e9ca24989f52aadf7cffdd7042","subject":"Merge \"Remove admin context check, update unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/205d716bca8a60e9ca24989f52aadf7cffdd7042"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 09:37:44.000000000","tz":-180},"subject":"[WIP] NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"[WIP] NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f71c5369293407a0cc15fd727d16404a8de98944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f71c5369293407a0cc15fd727d16404a8de98944"}]},"branch":"refs/heads/master"},"262097df51e6d381ff3dd1a8d99b189e959b8568":{"kind":"REWORK","_number":2,"created":"2022-07-26 06:53:55.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/49/849849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/2"}}},"commit":{"parents":[{"commit":"205d716bca8a60e9ca24989f52aadf7cffdd7042","subject":"Merge \"Remove admin context check, update unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/205d716bca8a60e9ca24989f52aadf7cffdd7042"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-26 06:53:00.000000000","tz":-180},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: 1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/262097df51e6d381ff3dd1a8d99b189e959b8568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/262097df51e6d381ff3dd1a8d99b189e959b8568"}]},"branch":"refs/heads/master"},"dd2b43ecf4b3681b049730dbf5e0dee8dd72a47d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-07-28 16:20:15.000000000","uploader":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"ref":"refs/changes/49/849849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/3"}}},"commit":{"parents":[{"commit":"205d716bca8a60e9ca24989f52aadf7cffdd7042","subject":"Merge \"Remove admin context check, update unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/205d716bca8a60e9ca24989f52aadf7cffdd7042"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-28 16:20:15.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd2b43ecf4b3681b049730dbf5e0dee8dd72a47d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd2b43ecf4b3681b049730dbf5e0dee8dd72a47d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7acbb07e4fc0b4dc5e93448d60c220e7ab874e3c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-03-01 18:15:26.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/849849/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/4"}}},"commit":{"parents":[{"commit":"f075b1bd7d01516818693c930f9974227de50363","subject":"Merge \"[NetApp] catch errors on getting vserver when reusing share server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f075b1bd7d01516818693c930f9974227de50363"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-03-01 18:15:26.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7acbb07e4fc0b4dc5e93448d60c220e7ab874e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7acbb07e4fc0b4dc5e93448d60c220e7ab874e3c"}]},"branch":"refs/heads/master","description":"Rebase"},"1751ad9bd8fc7ce964b4f3d92c38311961e171ca":{"kind":"REWORK","_number":5,"created":"2023-03-01 18:36:43.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/849849/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/5"}}},"commit":{"parents":[{"commit":"205d716bca8a60e9ca24989f52aadf7cffdd7042","subject":"Merge \"Remove admin context check, update unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/205d716bca8a60e9ca24989f52aadf7cffdd7042"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-03-01 18:36:28.000000000","tz":-180},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1751ad9bd8fc7ce964b4f3d92c38311961e171ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1751ad9bd8fc7ce964b4f3d92c38311961e171ca"}]},"branch":"refs/heads/master"},"a4e00d4585578548f5f39062af903fc069b2f22a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-03-02 11:33:46.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/849849/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/6"}}},"commit":{"parents":[{"commit":"51f75da64eb4878de9c656032790e4ed3579a3b8","subject":"Merge \"update capacity filter during extend share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51f75da64eb4878de9c656032790e4ed3579a3b8"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-03-02 11:33:46.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4e00d4585578548f5f39062af903fc069b2f22a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4e00d4585578548f5f39062af903fc069b2f22a"}]},"branch":"refs/heads/master","description":"Rebase"},"38b13bb40b879adcc06cef0b0fa629e4b00f7122":{"kind":"REWORK","_number":7,"created":"2023-03-02 12:24:57.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/49/849849/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/849849/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/849849/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/849849/7"}}},"commit":{"parents":[{"commit":"51f75da64eb4878de9c656032790e4ed3579a3b8","subject":"Merge \"update capacity filter during extend share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51f75da64eb4878de9c656032790e4ed3579a3b8"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-03-02 12:24:30.000000000","tz":-180},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/38b13bb40b879adcc06cef0b0fa629e4b00f7122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/38b13bb40b879adcc06cef0b0fa629e4b00f7122"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
