)]}'
{"id":"openstack%2Fmanila~852004","triplet_id":"openstack%2Fmanila~master~I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63","project":"openstack/manila","branch":"master","topic":"bp/netapp-ontap-rest-api-client","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-02-16 06:58:05.000000000","reason":"Change was submitted"},"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-02-16 06:58:05.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-02-15 22:17:52.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-02-16 06:58:05.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-15 22:53:32.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-02-16 06:58:05.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-02-16 06:58:05.000000000","reason":"Change was submitted"},"34489":{"account":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"last_update":"2023-01-19 16:16:47.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63","subject":"NetApp ONTAP: Implemented REST transition Client","status":"MERGED","created":"2022-08-04 13:01:57.000000000","updated":"2023-02-16 11:50:45.000000000","submitted":"2023-02-16 06:58:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852004-bp/netapp-ontap-rest-api-client","meta_rev_id":"52876391cd5b5af5e00d42b33a1bfd6a3c40bf89","_number":852004,"virtual_id_number":852004,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 06:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-02-15 22:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2023-02-15 22:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2023-02-16 06:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}],"reviewers":{"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 14:02:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 14:21:32.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-04 15:25:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 18:48:29.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-05 00:48:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-08-30 21:46:35.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-01 00:04:23.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-09-15 19:30:30.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-09-29 13:10:47.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2023-01-05 12:02:35.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2023-01-30 16:09:22.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-02-09 13:47:00.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"CC"},{"updated":"2023-02-15 14:13:41.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"CC"}],"messages":[{"id":"8d5c928d646f7d91231eafe71de1101c060a6521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-08-04 13:01:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"963358acf159c255f48e6341f6b6f45960c4aa17","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-04 14:02:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-1/ns5_manila : SUCCESS in 59m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"0be3830bd3685bf148aaebcf7607ec43b52a7640","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-04 14:21:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/1/thirdparty-check/pure-devstack-manila-tempest-aio/42e59bd/ : FAILURE in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9d398124682318dbfe702dd974f0715e03c6fca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:25:10.000000000","message":"Patch Set 1: Verified-1\n\n(16 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61c9d59e350046e58d87efe311bd35da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec241a5a4ab647cd8c4d89d18c3fb310 : SUCCESS in 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f79358306c824af889a3b93e2c6c2d0d : FAILURE in 21m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f84558cf15a437db9c0411f70675e6a : FAILURE in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b2e21e6f61a4fdd8b90210546fff01a : SUCCESS in 10m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f690098165084c98856d3476c116768d : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19a80d8cb9d44af4ada3ef895dda21b1 : SUCCESS in 10m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cf5f34581174a6595847ceb011b6035 : SUCCESS in 5m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5858ebca60fb472aad185e6094defa40 : SUCCESS in 1h 08m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6ffe72c948ba4c719a0ff546d0bcfcf0 : SUCCESS in 2h 13m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/187fb137655c4650b32526016b54244a : SUCCESS in 1h 10m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d8cd1d4b6e4d4889b8359d51f57b2393 : SUCCESS in 1h 47m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9e82efb67104433ca0074c1577dae2ba : FAILURE in 1h 52m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/25f93fd7bd334892bee73249b60dcd6d : FAILURE in 1h 30m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed758fa27a814e7b910d57ca156db8f5 : SUCCESS in 5m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dac9711b2cb74c0fa37a0f1fd5f2c0fc : FAILURE in 6m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0c6ea68de2004c679e2479caf7b77af4 : SUCCESS in 33m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a1d2dd14e41430daeb14e8cbf18f34e : SUCCESS in 35m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1a8a2dc1aba41a6afa0c21f49715da9 : SUCCESS in 54m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d4e2a69dea874d05b386bd0b07e86fb4 : SUCCESS in 36m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a99c5d9a39474064bb3f44063c3af770 : SUCCESS in 25m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/34ec8701372c482ea79641b1151db098 : SUCCESS in 1h 30m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/874769e6171744aaa35e4997f1d7d13c : SUCCESS in 1h 03m 04s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f47601709f084b63b62a9f508e20bcdd : FAILURE in 34m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c0bf228b379d45f1acc0491bc121ecc4 : SUCCESS in 35m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/89ad93bd242d4f3db01ba699b1f6dca7 : SUCCESS in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a5ecf9ab4f9427428120d71971cfd4b0d32e726","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-04 18:48:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/1/check/DellEMC_Unity_Manila_dhss_true/1c764d4/DellEMC_Unity_Manila_dhss_true/684 : FAILURE in 3h 19m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/1/check/DellEMC_Unity_Manila_dhss_false/b7b13da/DellEMC_Unity_Manila_dhss_false/721 : FAILURE in 13m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"4cbf26e1cf209a746b153d802a4e30a2f9a59fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-08-04 20:09:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4c8c8b39660fdcf0a0d248f70e5098f100e908f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-04 21:27:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/2/check/DellEMC_Unity_Manila_dhss_true/2e06ca3/DellEMC_Unity_Manila_dhss_true/685 : FAILURE in 1h 17m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/2/check/DellEMC_Unity_Manila_dhss_false/45275e7/DellEMC_Unity_Manila_dhss_false/722 : FAILURE in 17m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"83369269a96e9e40112a777296dfe44934600124","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-04 21:39:39.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/2/thirdparty-check/pure-devstack-manila-tempest-aio/7465776/ : FAILURE in 1h 29m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8336c638da5e0e9f9e91331f6ace12eb973657ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 22:19:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa0516f0eb34188bda4c6ce75e247c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a9a1d63ef104aa397f1907fe3907e37 : SUCCESS in 11m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46f4bd5e2dd141bfbf9c607bfcd9723a : FAILURE in 19m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258099ffa3854393b8aea2da921d2d83 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e9edc121f8249e8a57e333901458210 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17ec93ee832d40438702afdc33e37c5d : SUCCESS in 13m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69c9ecf8817c481d8d1e070d08bcaed8 : SUCCESS in 13m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f52ee309ed80465ea26233db85b46c9e : SUCCESS in 6m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f58ebb21ecd748869b6edeefb6d1bb16 : SUCCESS in 1h 07m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9f3faeff93fe45729ffa0142cddc1360 : FAILURE in 2h 08m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/066bd8fc3ee142888e4b5c39204ec5e9 : SUCCESS in 1h 17m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/727b137f87c94018a945a417889345c4 : SUCCESS in 1h 12m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/21ba73dffc664e58b014f24b142275fe : FAILURE in 1h 22m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd6795e0506b44eababecc28a2aeb8b7 : SUCCESS in 1h 11m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f5047c189374efcab91d5c139417c32 : SUCCESS in 7m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8cabca1d74b147d4a1a420b063a8d728 : FAILURE in 6m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ff4152147c6945e9ae4e24b14d202b09 : SUCCESS in 31m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b07a42f98c514677ab3b868c1590defe : SUCCESS in 42m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e026e960cbb74d0cb2823d30b1d741cc : SUCCESS in 47m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/169173b2dd874006a1403261137b84ef : SUCCESS in 33m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/47a34c60cda94734b71faa3d5f49919f : SUCCESS in 34m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/52da69a93fad4d24a2f84b5332d4d3a3 : SUCCESS in 1h 11m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e0e3fc1e797d40c2bafcb59fb956b85b : SUCCESS in 1h 06m 09s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1ca431f58c9941e3818a22fe7001250b : FAILURE in 27m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3ea2e30095a84dc992a21f9f372930b7 : SUCCESS in 42m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e19c709bb0014f9f9092910c9b57f3b0 : SUCCESS in 53m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4e4bbd7fd34fbe2f546685541e2a63d76741d284","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-04 22:53:48.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-2/ns5_manila : SUCCESS in 1h 33s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"5a28f12d1b50ac40715c94edbf8a0eab1e6f6a37","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-05 00:48:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/2/upstream-check/manila-tempest-plugin-ontap-dhss/59cbc95/ : FAILURE in 2h 27m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/6092538/ : FAILURE in 2h 14m 07s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"448b1e6588aa7756270284eb696853aa95b5d978","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-05 04:24:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/2/check/DellEMC_Unity_Manila_dhss_true/0672195/DellEMC_Unity_Manila_dhss_true/687 : FAILURE in 3h 21m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/2/check/DellEMC_Unity_Manila_dhss_false/17e90bc/DellEMC_Unity_Manila_dhss_false/724 : FAILURE in 17m 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"fa79f5e06dbe149c7434156deceac788e0622af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-08-10 15:49:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50a494f371d92f16336c94202b665a36fd609c72","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-10 16:49:00.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-3/ns5_manila : SUCCESS in 58m 43s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"ed6333cb1c14f8cd3aab6a266105aae0e916c78c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-10 16:56:06.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/3/thirdparty-check/pure-devstack-manila-tempest-aio/a4ee395/ : FAILURE in 1h 05m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e01be3f710423a7f65ca796370d5a3cecc3206e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:46:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80dc10c7b0a46b9b8b3b06fee5eee55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d2617d20d94dd7947d18e9d60ad580 : SUCCESS in 11m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aa4e1f141a342458cc5df584cb8ccdc : FAILURE in 17m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308f543490594b46b0b2d7e25d61596f : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eb65b42a9814b768c87c26a71dbc141 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0cbc14d27974144b3e3865b516abbbe : SUCCESS in 10m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3100888240854b25a0c9d7669b15b4c8 : SUCCESS in 10m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f74507016ca4160aa00186d0741bd38 : SUCCESS in 7m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3f23873a1dd94761adbff96dc4fbb487 : SUCCESS in 1h 07m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/34ced3eb74d94f0abc5410680d415c84 : SUCCESS in 1h 54m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b3316d7de12c4b5f899b667a08c2e268 : SUCCESS in 1h 17m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db01344a1a874ab787aed409b7eb187e : SUCCESS in 1h 33m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b1a72155248d434c9fc5ac9ec0416b38 : FAILURE in 1h 54m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d9c10a61ea5249f09fc1e46d8413ccb1 : SUCCESS in 1h 06m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/72b50fb0bdf64a5aa705e635527c42cf : SUCCESS in 6m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/36b147a7c32c4b1e88133ac5063815dd : FAILURE in 6m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6c5c514832974e89be8a44af950a479b : SUCCESS in 31m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7a4d5d9f710a4da2abf0d4655a9b16ed : SUCCESS in 33m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/75b7724231fc4178a77c209dffa717a5 : SUCCESS in 48m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/bd245531023d432d84a0b5418a1a1cff : SUCCESS in 36m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/737e4cac32ba4bdaa71d53921643409a : SUCCESS in 26m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8c5fa54b8427416aa6369e5d39ea0ec7 : FAILURE in 17m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/64399dba887a481893daf9749d939ca6 : SUCCESS in 1h 05m 33s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c2bab3fc3014df999007e11a968b6f8 : FAILURE in 34m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/02c01853a08c4fac9c11faeec2924687 : SUCCESS in 37m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7e7773e1f37143a79952f5ca5e9e0355 : SUCCESS in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff49c361d1be760df6a40f0d647d40d6213e477a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-10 20:56:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/3/upstream-check/manila-tempest-plugin-ontap-dhss/63f298a/ : TIMED_OUT in 3h 08m 57s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/6ee9d48/ : RETRY_LIMIT in 8m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"dc31dd7af9841f6f10116c0b3c847643c27150ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-10 22:31:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/3/check/DellEMC_Unity_Manila_dhss_true/581dfb0/DellEMC_Unity_Manila_dhss_true/694 : FAILURE in 4h 20m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/3/check/DellEMC_Unity_Manila_dhss_false/85a95aa/DellEMC_Unity_Manila_dhss_false/731 : FAILURE in 13m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e4c75d8c57044a55cbe4445df1da93bc57496a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-08-16 19:24:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5abcc34de4ef5c074acf6cd5a1a6e17508df60b5","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-16 20:26:52.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-4/ns5_manila : SUCCESS in 1h 1m 59s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"660572a02b82ea0ac30abbab7f20e6f43f25ceb3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-16 20:43:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/4/thirdparty-check/pure-devstack-manila-tempest-aio/2dd1adc/ : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":4},{"id":"67a00ee6f4f5785995a58676fe26217237a19121","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-16 20:50:03.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/4/check/DellEMC_Unity_Manila_dhss_true/47de4ed/DellEMC_Unity_Manila_dhss_true/709 : FAILURE in 1h 25m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/4/check/DellEMC_Unity_Manila_dhss_false/bca3186/DellEMC_Unity_Manila_dhss_false/746 : FAILURE in 17m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6eac104a1ad519a447128271a004c9683b0d5638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 22:33:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df20a8f495564cf39bf0c9f1ddf692ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35063211342f4d08b8371095f15014d7 : SUCCESS in 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6df311720c154ae0b2303550b3e2b898 : FAILURE in 18m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2861eaa93f8f4581b51bac36a4a15dbb : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a81f120567464263b02932a1d51c8b43 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/137d430480a24b2d9dba395a3597e9b6 : SUCCESS in 8m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc9c436f90144d2bb6ca75344ee739f4 : SUCCESS in 10m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1fca2d69cca4c44a70b421445eecdb2 : SUCCESS in 5m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1d1494dd63aa4b3f8c555e982c34d0e7 : SUCCESS in 1h 10m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2e5c4e87332c4f5eb15a3de81733df35 : SUCCESS in 1h 44m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da80dc8afef843b9b59b85a057039637 : SUCCESS in 1h 09m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/147dff32c3d74787b6409082496f0156 : SUCCESS in 1h 12m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/575f2e33fee84df588bc1c6b9d434d9f : TIMED_OUT in 3h 07m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/582435454b34411caf70b137ac9e9ac5 : FAILURE in 1h 04m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2e12d97b6068402087c74e3a623ee1e4 : SUCCESS in 6m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/daf4bf65dbcf4dc5a4ec3d09e6ad46c5 : FAILURE in 5m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e258030b81f44ca2a64c9338006a3530 : SUCCESS in 37m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/65f0bc4a1e2f47bb992a0a65606e1faf : SUCCESS in 35m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/98a04ae047c4490fadd7628be9e5e705 : SUCCESS in 48m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1ef1f805b5804e84ad7ba4597e110b8d : SUCCESS in 44m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/73a10b2082d543b79a68f9b6f2a87847 : SUCCESS in 25m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2cf670785d8a4338b0029668b7dc96f6 : SUCCESS in 1h 28m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/681d2ee1d76240068b2321fa775c2f39 : SUCCESS in 1h 14m 43s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/eb03bb28fa4b476cb23f66c9d26aa743 : FAILURE in 33m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5d65c838b5a14865bac2f21ed0bb5280 : SUCCESS in 43m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3911ae1a18934854b778387aa0e6ffd6 : SUCCESS in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72e165b4de6cf420db82550905cc5df9372de52b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-16 23:48:54.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/4/check/DellEMC_Unity_Manila_dhss_true/4112af0/DellEMC_Unity_Manila_dhss_true/711 : FAILURE in 1h 15m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/4/check/DellEMC_Unity_Manila_dhss_false/04991ed/DellEMC_Unity_Manila_dhss_false/748 : FAILURE in 13m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"50da682f11be10e821dee04c99ebf9b2d49817c0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-17 01:06:55.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/4/upstream-check/manila-tempest-plugin-ontap-dhss/788d75d/ : FAILURE in 48m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/f6ca9c3/ : FAILURE in 1h 30m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1846f255b3a27e82a518bbabcd62ad240c6fa7da","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-08-30 21:46:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e480182ffedeb2532515eecbbbfb8731137331e6","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-08-31 17:10:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c7270613e9a1cc7fac49ac5ed3cb33d3f59b520a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-31 19:17:18.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/4/upstream-check/manila-tempest-plugin-ontap-dhss/7eaa39c/ : SUCCESS in 1h 43m 50s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/638edd4/ : SUCCESS in 2h 00m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1354445bdfa8afc88fbaa95a797647eb4c97fbf7","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-09-01 00:04:23.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a2a39bbe806d4e81229b1117c4be72e39fbce1a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-15 19:30:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"89296d17bc6c6897d4c13f83bcfc597a49459b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"date":"2022-09-29 11:58:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d911da37f09aba1bcb0c39223208a86e1837d2fc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-29 12:26:37.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-5/ns5_manila : FAILURE in 28m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"856cd45393858ccea241143afcc7c0d89f287703","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-29 12:58:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/5/thirdparty-check/pure-devstack-manila-tempest-aio/7e72f33/ : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":5},{"id":"ace67282343522ccd432a14fb156e6b7a83a406e","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-09-29 13:10:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/04/852004/5/third-check/macrosan-manila-dhss-false/4edd090/ : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":5},{"id":"d25cccf1dfc447119d92f3cbdaf68b9f40b89e36","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-29 14:48:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/5/check/DellEMC_Unity_Manila_dhss_true/e73e242/DellEMC_Unity_Manila_dhss_true/941 : FAILURE in 2h 49m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/5/check/DellEMC_Unity_Manila_dhss_false/e181623/DellEMC_Unity_Manila_dhss_false/1001 : SUCCESS in 1h 14m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"d8fdc2bc32b4cc58260c86794c64d95ee1ef68ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 14:57:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442fd9f2d9ff44d182525f24843748a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef22ca2874464313bfbebf7bf579a480 : SUCCESS in 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73961ed450b640339326b58a67c4209f : FAILURE in 20m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1499974b2c524f06b11d1232898c5138 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1bff6aff459463eaca7e3668bb5ce8d : SUCCESS in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/116fa83df744436289430b165f18189b : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caa7b8c8ba1c4f859f3aaa68014c37be : SUCCESS in 6m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/352942f1ef8442b6bd17983bf8dfcde1 : SUCCESS in 1h 27m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8062f1cf2a994f60bb579e53e6752192 : FAILURE in 2h 40m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c8cefd84691240cf85cf774b37015791 : SUCCESS in 1h 25m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d862bee97f864ebc9675ecbbb54acfb5 : SUCCESS in 1h 34m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/98272de13d024a67b59551f736cc5ece : FAILURE in 2h 54m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/988fa12ca6a145dd9a245c417d52e7fa : FAILURE in 1h 27m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bae0adea318845fea4b4a098ed26fbe7 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3cc8efeed9f4f35ad1d2e4001a933b9 : FAILURE in 5m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a468fd201a024a899befe2ed216260e2 : SUCCESS in 37m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7248ed4fac0d478da3f225780d09b3ec : SUCCESS in 39m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7f9fc530dd3744bdb80ab18d943191ff : SUCCESS in 58m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7f43e5934557496c8d06973972c26c5a : SUCCESS in 38m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/22ed9a0ff40a43f2b9ce8f1b31403e4e : FAILURE in 28m 05s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f0d9d74f100547c3b84905329885b6a0 : SUCCESS in 1h 32m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b5be7f6802f84ba1b8052e05365509b8 : FAILURE in 1h 05m 14s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e4bc110b5b5f49599ce0a7e7dcbe4484 : FAILURE in 41m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/83775435c3bd432f877b97fe8cbf6ec1 : SUCCESS in 42m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c36c94ef60bc4c848257ec0b4df4c359 : SUCCESS in 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a97b92d330bae6aefc69f2e9795e51155da496d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-29 17:06:25.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/5/upstream-check/manila-tempest-plugin-ontap-dhss/25133bf/ : FAILURE in 2h 08m 02s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/abdd260/ : SUCCESS in 1h 57m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"dda41dbd0972d25f8cba0720bc7b3023f785504c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-29 23:11:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/5/check/DellEMC_Unity_Manila_dhss_true/1a92400/DellEMC_Unity_Manila_dhss_true/944 : FAILURE in 3h 02m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/5/check/DellEMC_Unity_Manila_dhss_false/7c021d9/DellEMC_Unity_Manila_dhss_false/1004 : SUCCESS in 58m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"cadc5db5b9a2ebe5857c3c045832da2ede67a8db","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-01-05 12:02:35.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ea482010ff990add577288c37fe099c25daf7d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-01-19 15:02:22.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f4b582c07f6c949ef285ae847e02e6e1a42230cb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-19 16:16:47.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-6/ns5_manila : FAILURE in 1h 13m 53s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"1d1261b742616f9997da94e1b827e5470e89aef1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-19 16:27:33.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/6/thirdparty-check/pure-devstack-manila-tempest-aio/5998a41/ : SUCCESS in 1h 23m 30s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/04/852004/6/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/a158e9a/ : SUCCESS in 1h 00m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"960191bbe94f27e306b8810ff891e2a69068b1ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 18:22:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c9cbc321da4f06a0910ad4cf750170\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0173d69ead05470d80017520b2d0e887 : SUCCESS in 9m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb5b2277db040819a3cdf73f2e3f51a : FAILURE in 32m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb8dd5aafa04b8ebd93af506c442c62 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1bfe9a72a4343338071cbeb48064baa : SUCCESS in 9m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e629271d3d34ce1807cbe03f78da212 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6e5de773a5742538272d88135202d5f : SUCCESS in 9m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6b437687de7d49c59301b164fba6cd00 : SUCCESS in 1h 02m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6eb53b90f72a403b9f3ec0928c5c8109 : TIMED_OUT in 3h 03m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8c943322f8ee4135967b6419b4bf28bd : SUCCESS in 1h 09m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6d2482dd46184124b36161aa902fc808 : SUCCESS in 1h 37m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a55ad7399c7d4bf0bf506c24e3a532ef : FAILURE in 1h 57m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/93dafc90602d4e35b46579fe39612e5e : FAILURE in 1h 19m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5acca7b7fba346e6865903946faec827 : SUCCESS in 4m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9bd37d413a7341ed932ce1a1cdb90acf : FAILURE in 5m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c57c173f0b874e58addd4e6705bfb9ec : SUCCESS in 36m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/badb58acd423410f9d24852bcd37e8cc : SUCCESS in 39m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b3857980a644aae9fe58fb8e6200c6c : SUCCESS in 59m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9ea1c5becace4f9286fdbaf102f53fe4 : SUCCESS in 29m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8f7b17aaa6054fcaaaf0d6468241490d : SUCCESS in 22m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b92c5d2bce974b8abe42c60a5123b4d8 : SUCCESS in 1h 13m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4e60914e14974c49b779fa7d562f7d21 : SUCCESS in 1h 01m 46s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dddce837b5e84ec5a664cca8df179af2 : SUCCESS in 1h 04m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fc1ee05e9e77420d83124657a9875369 : SUCCESS in 39m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3b5f2c841e5b41809324c7206cee6a50 : SUCCESS in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5633fc36ac288fd96f715178c49395fbdcbab32f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-19 18:32:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/6/upstream-check/manila-tempest-plugin-ontap-dhss/fd9d44b/ : RETRY_LIMIT in 7m 35s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/3040d7f/ : RETRY_LIMIT in 8m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"45135ea7af188df432da88d9c5adbbaf82d1297e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-06 12:10:23.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"bfd6895d17af2f3b193f1d7870254f170a4914d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 13:18:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/7/check/DellEMC_Unity_Manila_dhss_true/6be50c5/DellEMC_Unity_Manila_dhss_true/1185 : FAILURE in 1h 06m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/7/check/DellEMC_Unity_Manila_dhss_false/2e97320/DellEMC_Unity_Manila_dhss_false/1284 : SUCCESS in 1h 08m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"c21ffc35668e3be5a929acf231e5fc9ee86c0974","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-06 13:39:05.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-7/ns5_manila : FAILURE in 1h 28m 11s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"c3d55a70184d0fdef1e755c96ec58277954cc844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 14:02:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7637f681c614059a3116994340efab3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698499d0157c49adba84a59387603e37 : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acb4c354e95f4047a18adc6749c447d7 : FAILURE in 28m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7aec291d3a4b2c918a25f8c4b3094b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2467ab08ca9c4559ae9a99db9303841f : SUCCESS in 16m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b0e69dc0ff3467eb5fd76827b7f6100 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a54f43a65cc45da997ddcff5a1f0a53 : SUCCESS in 5m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dff1a0a30e904030a431bbf5ba2cabca : SUCCESS in 1h 09m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bbf758ed1d23491dae88222f9d406af9 : SUCCESS in 1h 48m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33040025cf484de384dba6b356632052 : SUCCESS in 1h 10m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/189b0df717ef4242a8ed6e8b8b9c3e75 : SUCCESS in 1h 07m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e220cd6de1f542788b1bb74832d04d13 : SUCCESS in 1h 38m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/81699521ede44ee3b23a17c701b480e5 : FAILURE in 1h 06m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f6916be73a74aeaaa314fcb9d387cc3 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/21354c7de25041978e6f3114ef7c7704 : FAILURE in 6m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f96d7afdb9b44a93a086b80a70c476b1 : SUCCESS in 34m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21c2ebbd0e0c40a39006516c4a15388b : SUCCESS in 34m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8c56ee0770c14eda8d548fd6f0b8ef3f : SUCCESS in 53m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a499165b87f44e2391ba8affa81bafcf : SUCCESS in 33m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e1a97602b37f47ccb6bf0c3a936d6ebe : SUCCESS in 26m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d8e3c19386d14e27be34b43bddd6ba70 : FAILURE in 1h 06m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/402ebb6010a14522bb0540a005ef0176 : SUCCESS in 1h 03m 46s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/47cfa296386944bfa27218c5f12aa80e : SUCCESS in 1h 00m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/73547563d00646e283d10b45074d2916 : SUCCESS in 38m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c5b516503fda45ec84cce3f10c793ab9 : SUCCESS in 41m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a2ce8fdd081d97958cedf1ac474bb4aff8efef07","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-06 14:12:26.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/7/upstream-check/manila-tempest-plugin-ontap-dhss/62a801a/ : RETRY_LIMIT in 7m 53s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/db4646a/ : RETRY_LIMIT in 8m 19s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"d5149d29005938840e54437873294c0eccf89169","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-06 14:23:08.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/7/thirdparty-check/pure-devstack-manila-tempest-aio/d0403c1/ : FAILURE in 1h 40m 44s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/04/852004/7/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/f7881a1/ : SUCCESS in 2h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cce86405003af869def902c8fb1c404b7c872778","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 19:16:08.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/7/check/DellEMC_Unity_Manila_dhss_true/06ae80c/DellEMC_Unity_Manila_dhss_true/1190 : FAILURE in 1h 08m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/7/check/DellEMC_Unity_Manila_dhss_false/4b18bb4/DellEMC_Unity_Manila_dhss_false/1289 : SUCCESS in 1h 12m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"e6442b6fe1012d2a24e87dce0ce64f68614d218a","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-02-09 13:47:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"78a969bfd732401f37f68065bebf6c223f55c964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-13 02:02:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b606eac0d87451d9be344942e6b0415199992571","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-13 03:14:08.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/8/thirdparty-check/pure-devstack-manila-tempest-aio/88b311b/ : SUCCESS in 1h 10m 05s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/04/852004/8/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/37028b1/ : FAILURE in 1h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"784913d4ecb6863743bf836c9ad2e2cb04b32b56","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-13 03:14:32.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/8/check/DellEMC_Unity_Manila_dhss_true/9a03ea2/DellEMC_Unity_Manila_dhss_true/1216 : FAILURE in 1h 11m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/8/check/DellEMC_Unity_Manila_dhss_false/2e6183d/DellEMC_Unity_Manila_dhss_false/1315 : SUCCESS in 1h 01m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"c67f3c1629bcfdb52fbcea25e61ed6a5b364792f","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-13 03:16:41.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-8/ns5_manila : FAILURE in 1h 13m 35s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"deb5fcf47bea99996075d66d29ec195d0cfdefeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 07:23:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fe40be9d8544257b37160826ae2f21b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5c49f1627d47f5804769315a75f767 : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf28b66e31c94169a0297959f5c98565 : FAILURE in 19m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b2584de1ab40f8847beb11579d01df : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5327e088a4354d26a457130a2d3d2dac : SUCCESS in 8m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64557388d7544728b2378240c6fb8965 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28e519a2b8e040328d952eaa4a67e659 : SUCCESS in 5m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c7e123b65f654edeb871cd833c59e1e6 : FAILURE in 2h 10m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/84b104b16ad047868b5f315678c524d4 : FAILURE in 2h 11m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a15d53f62b749d895ce0b09cebc7b23 : SUCCESS in 1h 22m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e7073f44eb7744eca282e290a6f4d177 : SUCCESS in 1h 13m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a6c10084d98446e99cf1478a5bde0174 : FAILURE in 2h 06m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e8e156c65a564e81bee3932be934534f : FAILURE in 1h 15m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/384c9f5127b44602b2548a13685eca1d : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1d8b673927034a84b5407f028accca63 : FAILURE in 4m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/97ddff9fbe4c4db492091d04ed6c80be : SUCCESS in 31m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d8364b4cfb56450ca635452332178225 : SUCCESS in 35m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a4a72064fad40fd8cafd15759859133 : SUCCESS in 45m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5ced4d9976b7444284352d5fba393476 : SUCCESS in 36m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e3ab0af9458f4e0fab71ce0b1727db39 : SUCCESS in 23m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/61cf3b929d2549dfb00cde05b9b22da5 : SUCCESS in 1h 28m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2391986b97044870955a8dec641f5da6 : SUCCESS in 55m 40s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6d6b09948af34385980ba987ebe9cece : SUCCESS in 45m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ea3d38682ac4498293cbf1d3a1db28b1 : SUCCESS in 36m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0983fb3c72cb4ec6b47aba85dffa9c7b : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"89b39882024863436d01a6e7c17470514a66846f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-13 23:59:22.000000000","message":"Patch Set 8: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"eb7c6bed42230f3390064f2d310dfa468bd17220","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-15 14:13:41.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f83f1abc62a8929791221bbe87d1d91927721280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-15 14:13:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"256dc549f469cd252411ab7609457c1d87a08a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-15 14:44:02.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"06469430de5617dfec43a94c22dfde62b594112a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-15 15:39:59.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/10/check/DellEMC_Unity_Manila_dhss_true/a6801cb/DellEMC_Unity_Manila_dhss_true/1226 : FAILURE in 47m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/10/check/DellEMC_Unity_Manila_dhss_false/eeab2a4/DellEMC_Unity_Manila_dhss_false/1325 : SUCCESS in 55m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"c7ec38cbca6035a4b706e35e9a0d8e6417e2b32b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-15 15:58:44.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/04/852004/10/thirdparty-check/pure-devstack-manila-tempest-aio/a16c2b9/ : FAILURE in 1h 05m 22s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/04/852004/10/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/6cbe651/ : SUCCESS in 1h 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"493afb48128e6ec7e190c969a174ff84c11dddc1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-15 16:30:22.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-10/ns5_manila : FAILURE in 1h 45m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"ad0e9c87eda166349e37b54f5943b2add95498de","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-15 22:17:52.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"59ef8b829845e72f4009ce23cfb43bf8f499ca9c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-15 22:32:14.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-04-852004-9/ns5_manila : SUCCESS in 8h 17m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"c56c8848719fcd208f16f268c309083a2b34b419","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-15 22:53:32.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"970ec6a2509c345864f848d3567a8fc3c6660125","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-15 23:08:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e3492cd082ce6871bf620c46dcff09cc1e7b5e2c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-15 23:26:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d84386e7f3d425759d2683f954a3d80b667a73df","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-15 23:46:14.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"50faba5af952f246ecf7329fd1085a2828155769","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-15 23:46:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"10f6d18532a08dc3edf0fcb14fc592a6abd1ae9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 00:10:02.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb641bd6dc4b47de80ba3384831edea9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa4c6be239146cfa5cbc77ac92cc9aa : SUCCESS in 11m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf1618ce2a9a43638936fad7a2965988 : FAILURE in 31m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1331626b449458a8ea45263b7d0492f : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/243cca47718f4b4880b4735f42dd976c : FAILURE in 18m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50cff91c06cb4d139f29888b789738dc : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f4cb65cb31748799bdaae682bd3fc53 : SUCCESS in 5m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/befd2c4215494069a9065214ec45f661 : SUCCESS in 1h 20m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c3e10a227c2f4a58b58de78ed846e270 : SUCCESS in 2h 02m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2c5af91293b5495fb6505db3a81980be : SUCCESS in 1h 05m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/afa52f1180fc4d5b879ac90f9559ddb6 : SUCCESS in 1h 18m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1a1466cd99f040c8a319b591957103c0 : RETRY_LIMIT in 3h 01m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fbb26747663e4956af0bcb56ae0d47a4 : SUCCESS in 1h 27m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/52b5fc59819d4e5bbe4616755abe3e59 : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a6550373acea4abf8d53cdd45b48cbd2 : FAILURE in 8m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04c7be636c674f9d81f7f665f0dee34d : SUCCESS in 30m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cfea86c9400748cf85fbd349c3738082 : SUCCESS in 29m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bf8993828486445791c00df4042f85ab : SUCCESS in 44m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/24690af71afb486ba959b34fe5254623 : SUCCESS in 35m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8cac869c172b4674872d55e4f49fa40f : SUCCESS in 19m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/36c867f09868433897f75b4828afe402 : FAILURE in 59m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3956857e6aa648bdbae2f21aa6ad3d61 : SUCCESS in 59m 51s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/62b282daaf7b46d98fc8e8af97be0be9 : SUCCESS in 1h 19m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b9768eef5329419b98132eedf435585c : SUCCESS in 38m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d64f3472f64f4cd992a08fce58793166 : SUCCESS in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8057631420582453ed61a9866d1a78183f199c55","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-16 01:04:16.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"386a1f50a7e9b5bb7a7aff08bd294421b595a5f5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-16 01:30:54.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/10/upstream-check-experimental/manila-tempest-plugin-ontap-dhss/76eafe9/ : SUCCESS in 1h 43m 46s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/04/852004/10/upstream-check-experimental/manila-tempest-plugin-ontap-no-dhss/d0b68b3/ : FAILURE in 1h 25m 18s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"5a916ca3143c6962dcd82c122a58e65784315513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:21:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5535133e095d4328abd1a083f4a4e4db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681c9bd410aa42c78bac6e4490624cd3 : SUCCESS in 9m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c102517d3e04ba988278bb3903dfd19 : FAILURE in 20m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffc299df37f4016a3342386f11c3b44 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8dc11d98737437da3e2e4af3f152053 : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56e2273d7f714402bc9fd58528464b2b : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c9c7aa6762740ae83724506789ec071 : SUCCESS in 5m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/89bddaeb82b74a35bc5d48495b0b367f : SUCCESS in 1h 12m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b04e652301f64a70a2b78053ebd31c00 : FAILURE in 2h 11m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e0a62fc0d014df09c3a23ef66eacf35 : SUCCESS in 1h 17m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7a6fb7d85fde4a6188d3ebad2c877aca : SUCCESS in 1h 08m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bef4c41d37c84b1d8ecf234485c2ab02 : SUCCESS in 2h 03m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ed4c7f8d8039441591517ef193e4b201 : FAILURE in 1h 02m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3fd16eb34fcf4301868761f9155e0862 : SUCCESS in 5m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e8dbec020874b5192be4f87c771b42c : FAILURE in 5m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4c4b6524107949799f8987584eadbdc9 : SUCCESS in 25m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be2a83f51a5e484089c9e50ab7a30c73 : SUCCESS in 30m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/afb02ca9e3c74f5aae57ea252df290d8 : SUCCESS in 48m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4347720580ac471a834f54217ec1e9a4 : SUCCESS in 31m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3e18316c846f48cd86287e65574671bf : SUCCESS in 22m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/05e572634a6b47e28a5b47568d1148ab : FAILURE in 1h 04m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7a279545ccb1466684faa5807d94f4bb : SUCCESS in 57m 22s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8124d342d3bb490fa6b19a716e65c074 : FAILURE in 30m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/eafe2d3ed9d742daa120dd5c95de3717 : SUCCESS in 37m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2981768ea7604d4996923db55994c650 : SUCCESS in 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8d3bc41d953b12000f54614298342cf07b83f614","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-16 06:02:04.000000000","message":"Patch Set 10: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7ce96bd713884b94c6ca9480e7d1e624300624ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:02:41.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"df6c88c39eb38250b61f96e48adcf86e9c279746","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-16 06:35:57.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/04/852004/10/check/DellEMC_Unity_Manila_dhss_true/f7dac90/DellEMC_Unity_Manila_dhss_true/1230 : FAILURE in 1h 14m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/04/852004/10/check/DellEMC_Unity_Manila_dhss_false/41b67eb/DellEMC_Unity_Manila_dhss_false/1329 : SUCCESS in 1h 02m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"4bf4c0545c089af872f02d80642f0459edac757c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:58:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc846727d4e7465eab235d1b9277e322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe4f580f4f34b03b7cd007545e6df8b : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3704d36275041a2b794d5c1b3f70022 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f4c509b9a4b4f3b95d01c463eb3ce52 : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f982251044749e1adf3b73d172e8684 : SUCCESS in 15m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a11aa9718c5c4ad18e4773b09c44219a : SUCCESS in 7m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/36f2d2d690834c32b80f8b537f112819 : SUCCESS in 53m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a0e2ca25618425fa39ba4a212206481 : SUCCESS in 26m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8098fb28ed44418d8f98c3b33d96dc0f : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":10},{"id":"c8e46c4635454c6f2ab586db420ed77d29f2c516","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:58:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"9333c33c6ea7e4fe6efaa738bf3857c6eeba5181","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 06:59:18.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c062e24ed83144f7a0711febd05a9a7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0aee1b412c14f2ca5ea1af75845aa22 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d4bf03748b14f5696879502f31dfbcd : SUCCESS in 49s","accounts_in_message":[],"_revision_number":10},{"id":"52876391cd5b5af5e00d42b33a1bfd6a3c40bf89","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-16 11:50:45.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e3ef58f86bdf31ab0cecaefe0a6e7c656f7765a6","revisions":{"883787d0cc7980a86ef041f9f03ad6f60c95be2f":{"kind":"REWORK","_number":1,"created":"2022-08-04 13:01:57.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/04/852004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/1"}}},"commit":{"parents":[{"commit":"5866dadd4b1ddcd935c854b3190e6df5e12ac4e4","subject":"Merge \"Use defined project config of manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5866dadd4b1ddcd935c854b3190e6df5e12ac4e4"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-08-04 12:41:41.000000000","tz":-180},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/883787d0cc7980a86ef041f9f03ad6f60c95be2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/883787d0cc7980a86ef041f9f03ad6f60c95be2f"}]},"branch":"refs/heads/master"},"aefb6c677689331e8a6fb3ea9a5df2ab13f8cdb7":{"kind":"REWORK","_number":2,"created":"2022-08-04 20:09:37.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/04/852004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/2"}}},"commit":{"parents":[{"commit":"5866dadd4b1ddcd935c854b3190e6df5e12ac4e4","subject":"Merge \"Use defined project config of manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5866dadd4b1ddcd935c854b3190e6df5e12ac4e4"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-08-04 20:09:10.000000000","tz":-180},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aefb6c677689331e8a6fb3ea9a5df2ab13f8cdb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aefb6c677689331e8a6fb3ea9a5df2ab13f8cdb7"}]},"branch":"refs/heads/master"},"8506a991ef696f6ecdb383e9b30e24d1d0fcc231":{"kind":"REWORK","_number":3,"created":"2022-08-10 15:49:44.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/3"}}},"commit":{"parents":[{"commit":"5866dadd4b1ddcd935c854b3190e6df5e12ac4e4","subject":"Merge \"Use defined project config of manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5866dadd4b1ddcd935c854b3190e6df5e12ac4e4"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-08-10 15:49:43.000000000","tz":0},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8506a991ef696f6ecdb383e9b30e24d1d0fcc231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8506a991ef696f6ecdb383e9b30e24d1d0fcc231"}]},"branch":"refs/heads/master"},"647310021f3c7d46bd9801826958a54365308b9d":{"kind":"REWORK","_number":4,"created":"2022-08-16 19:24:24.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/4"}}},"commit":{"parents":[{"commit":"5866dadd4b1ddcd935c854b3190e6df5e12ac4e4","subject":"Merge \"Use defined project config of manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5866dadd4b1ddcd935c854b3190e6df5e12ac4e4"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-08-16 19:23:46.000000000","tz":0},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647310021f3c7d46bd9801826958a54365308b9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647310021f3c7d46bd9801826958a54365308b9d"}]},"branch":"refs/heads/master"},"97dfa20a995869becb89d7ec35e68d2a4240cd8f":{"kind":"REWORK","_number":5,"created":"2022-09-29 11:58:07.000000000","uploader":{"_account_id":34489,"name":"Matheus Andrade","email":"matheus.handrade15@gmail.com","username":"matheusandrade"},"ref":"refs/changes/04/852004/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/5"}}},"commit":{"parents":[{"commit":"5866dadd4b1ddcd935c854b3190e6df5e12ac4e4","subject":"Merge \"Use defined project config of manila-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5866dadd4b1ddcd935c854b3190e6df5e12ac4e4"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"MatheusAndrade777","email":"matheus.handrade15@gmail.com","date":"2022-09-29 11:54:42.000000000","tz":0},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/97dfa20a995869becb89d7ec35e68d2a4240cd8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/97dfa20a995869becb89d7ec35e68d2a4240cd8f"}]},"branch":"refs/heads/master"},"f426109dbc55a12ead017c7c1c78c9afe2b32b77":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-19 15:02:22.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/6"}}},"commit":{"parents":[{"commit":"c5cf2b98201386b5a67a45246067af50da72da85","subject":"Merge \"db: Migrate \"service\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c5cf2b98201386b5a67a45246067af50da72da85"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-01-19 14:58:20.000000000","tz":-120},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f426109dbc55a12ead017c7c1c78c9afe2b32b77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f426109dbc55a12ead017c7c1c78c9afe2b32b77"}]},"branch":"refs/heads/master"},"a0b7946f8c17f0cd8c2ae80c04af18162a989cd6":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-02-06 12:10:23.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/7"}}},"commit":{"parents":[{"commit":"f6f93ff9516f0ae24801d7b422c0e504bfebd0f5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6f93ff9516f0ae24801d7b422c0e504bfebd0f5"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-02-06 12:04:31.000000000","tz":-120},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0b7946f8c17f0cd8c2ae80c04af18162a989cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a0b7946f8c17f0cd8c2ae80c04af18162a989cd6"}]},"branch":"refs/heads/master"},"d490f70df5c0fba4056c2ee303c00e47de4f46f7":{"kind":"REWORK","_number":8,"created":"2023-02-13 02:02:35.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/04/852004/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/8"}}},"commit":{"parents":[{"commit":"f6f93ff9516f0ae24801d7b422c0e504bfebd0f5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6f93ff9516f0ae24801d7b422c0e504bfebd0f5"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-02-11 00:49:01.000000000","tz":-180},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d490f70df5c0fba4056c2ee303c00e47de4f46f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d490f70df5c0fba4056c2ee303c00e47de4f46f7"}]},"branch":"refs/heads/master"},"81985fe764dfa9249456874408c2f8888949c0d3":{"kind":"REWORK","_number":9,"created":"2023-02-15 14:13:59.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/9"}}},"commit":{"parents":[{"commit":"6d4fcb7609381296c1046a8f93ea998f6214454b","subject":"Merge \"NetApp: Disable tunneling in vserver_exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6d4fcb7609381296c1046a8f93ea998f6214454b"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-02-15 13:45:37.000000000","tz":0},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/81985fe764dfa9249456874408c2f8888949c0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/81985fe764dfa9249456874408c2f8888949c0d3"}]},"branch":"refs/heads/master"},"e3ef58f86bdf31ab0cecaefe0a6e7c656f7765a6":{"kind":"REWORK","_number":10,"created":"2023-02-15 14:44:02.000000000","uploader":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"ref":"refs/changes/04/852004/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/04/852004/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/04/852004/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/04/852004/10"}}},"commit":{"parents":[{"commit":"6d4fcb7609381296c1046a8f93ea998f6214454b","subject":"Merge \"NetApp: Disable tunneling in vserver_exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6d4fcb7609381296c1046a8f93ea998f6214454b"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-06-03 18:39:46.000000000","tz":-180},"committer":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2023-02-15 14:43:02.000000000","tz":0},"subject":"NetApp ONTAP: Implemented REST transition Client","message":"NetApp ONTAP: Implemented REST transition Client\n\nAs the result of adoption of a REST API by NetApp ONTAP, this patch\nis adding the basic structure to migrate from ZAPI Client to REST\nincrementally.\n\nTo avoid adding bugs to the pre-existent ZAPI client and easily backport\nthe new client to older releases, the new REST implementation is a\ncompletely new code on the client layer: a new client REST and API REST\nclasses. The driver layer should be agnostic from this transition, the\ninterfaces that the driver calls should keep the same as before.\n\nThe REST client contains a fallback mechanism, which will call the\nZAPI client when the operation is not implemented yet. This first\npatch is only implementing the get ONTAP version operation using\nREST, all other operations keep working with ZAPI.\n\nThe REST client can only work with ONTAP 9.11.1 and upper.\n\nA new configuration `netapp_use_legacy_client` is added to enable\nthe operator to switch between new REST client implementation and the\nold ZAPI implementation. The default is `True`, given that the REST\nclient is still experimental.\n\npartially-implements: bp netapp-ontap-rest-api-client\nChange-Id: I0bbc7609df66b72e9f8e26f4abb8092ae68cbe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e3ef58f86bdf31ab0cecaefe0a6e7c656f7765a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e3ef58f86bdf31ab0cecaefe0a6e7c656f7765a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
