)]}'
{"id":"openstack%2Fmanila~852005","triplet_id":"openstack%2Fmanila~master~Ie78a67a699005398c9834aa8942224d58deef3f2","project":"openstack/manila","branch":"master","attention_set":{"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2022-08-05 02:31:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie78a67a699005398c9834aa8942224d58deef3f2","subject":"DNM: speed up the test_walk_versions","status":"NEW","created":"2022-08-05 01:34:20.000000000","updated":"2023-04-05 17:52:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9c4cb3745dc4040399a5aec98a9a1c14cf82361","_number":852005,"virtual_id_number":852005,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-05 02:31:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 02:31:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-05 02:34:48.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-05 02:36:56.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-05 08:02:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-16 19:31:03.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2023-04-05 17:52:47.000000000","updated_by":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"reviewer":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"state":"CC"}],"messages":[{"id":"10b65a0a53dae3eb8a6d0dfb5fb815f6ed082b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-08-05 01:34:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fa83151a438dbe19c80fcc77b60b423b5cd4522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 02:31:23.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb65d6917e02471e9e7cc2cbc58ac9d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35474672222e4852b32f07a3b1050365 : SUCCESS in 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66885601bb0f49e7b18a2ed5e735c237 : TIMED_OUT in 52m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f696db3c7a4135a7b7a7b6302282ed : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc14b00787b046909e455578af0aca88 : TIMED_OUT in 40m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b98772d27f74b4ca3b3b78337a2acfd : TIMED_OUT in 41m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e63cbfec23040fe94a3059b19c17604 : TIMED_OUT in 40m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/480570278fa949f3808265a40472d07f : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/218a41cd16464a1d88eda27f7847a753 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e33726e61d874f8bddb12ab4fdf93679e582b4fa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-05 02:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/05/852005/1/thirdparty-check/pure-devstack-manila-tempest-aio/15b58e0/ : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":1},{"id":"08532803644074f8a4d9475c56d77028a7da4ca0","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-05 02:36:56.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-05-852005-1/ns5_manila : SUCCESS in 59m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"5bf85c65c58e7b05fdd6b6e2a352fb2c8364c4ce","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-05 08:02:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/05/852005/1/check/DellEMC_Unity_Manila_dhss_true/506ddeb/DellEMC_Unity_Manila_dhss_true/688 : FAILURE in 3h 37m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/05/852005/1/check/DellEMC_Unity_Manila_dhss_false/977e0f5/DellEMC_Unity_Manila_dhss_false/725 : FAILURE in 18m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a750572db4c7ce775a7cb82347e489063431a6a5","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-09-16 19:31:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75fc237fb1a2180e6538e70c894638758bc62465","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-09-16 20:17:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"381227568032afa3a1285d584e2f72803ec432b7","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2022-09-16 20:19:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9c4cb3745dc4040399a5aec98a9a1c14cf82361","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2023-04-05 17:52:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9547086137475bf336f9a4c390e56577849e6d6","revisions":{"b9547086137475bf336f9a4c390e56577849e6d6":{"kind":"REWORK","_number":1,"created":"2022-08-05 01:34:20.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/05/852005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/05/852005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/05/852005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/05/852005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/05/852005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/05/852005/1"}}},"commit":{"parents":[{"commit":"a3b48cc5c19dc89d99963f5a0d364e71978cdfe6","subject":"Merge \"Replace abc.abstractproperty with property and abc.abstractmethod\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a3b48cc5c19dc89d99963f5a0d364e71978cdfe6"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-08-05 01:13:13.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2022-08-05 01:13:13.000000000","tz":-180},"subject":"DNM: speed up the test_walk_versions","message":"DNM: speed up the test_walk_versions\n\nTry to use the neutron approach [1] for improving the execution of\nmigration test_walk_versions.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/770935\n\nChange-Id: Ie78a67a699005398c9834aa8942224d58deef3f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b9547086137475bf336f9a4c390e56577849e6d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b9547086137475bf336f9a4c390e56577849e6d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
