)]}'
{"id":"openstack%2Fmanila~855220","triplet_id":"openstack%2Fmanila~master~I8e21e9170eace134a51efed84de1ccc58eb7eaaa","project":"openstack/manila","branch":"master","topic":"bug/1988146","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-02-16 01:11:46.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-16 15:32:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-02-17 03:56:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e21e9170eace134a51efed84de1ccc58eb7eaaa","subject":"Add \u0027default_ad_site\u0027 field to security service object","status":"MERGED","created":"2022-08-30 15:37:34.000000000","updated":"2023-02-17 04:46:55.000000000","submitted":"2023-02-17 03:56:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":69,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"855220-bug/1988146","meta_rev_id":"4adfb57ab12806f77a9633b9fa6cf3e3c10b0920","_number":855220,"virtual_id_number":855220,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-17 03:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-02-16 11:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-16 15:32:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-02-16 15:32:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-08-30 15:39:06.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-08-30 16:53:53.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-30 16:58:39.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 18:25:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 23:07:46.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-08-31 05:46:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"8e27c82f74ae7ae4626a5732d10a01af3cecbf29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-08-30 15:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ab5f3b2971c3a8b8dc089174873d10bf1f5aacd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-30 16:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/1/thirdparty-check/pure-devstack-manila-tempest-aio/bd84764/ : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c9fd09e6f925611557a8eb9ecec62e3fa4e3793b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 16:58:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-1/ns5_manila : SUCCESS in 59m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c20da51daaef123e2a5fe177c2647aa63e069354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:25:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d4e58adf8d54743b3ca453f0f1f2e9e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/498be38259a44fecb810b249f93cb21d : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9935703405454ba292fa2cb2746377d7 : FAILURE in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f171da9b2c5410ebb94bf63516bf861 : FAILURE in 20m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537e1af31938485393a5becb442cdfaf : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0271353fe97340b491edd84fcc2ec472 : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6e799e7aa1b468f882fd82f03bc9694 : SUCCESS in 9m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40ca0aebb2744c0fb1d690526b65f017 : SUCCESS in 10m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a7765061b5464d9f31e9bcb9867105 : SUCCESS in 7m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/01f6eee5363143458ebbb1f24deecb9d : SUCCESS in 1h 26m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dccf8359b213446c9ac23497a9abd20a : FAILURE in 2h 40m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e1e17a4766f0476ba9cd9f356b8bfb2a : SUCCESS in 1h 12m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54a2ba598bf6485abf14dce722065b9d : SUCCESS in 1h 33m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/78b060c6fc5042a7a399a4d3b67fe940 : FAILURE in 1h 59m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f28bb4205414429db232602ee17c9567 : SUCCESS in 1h 14m 35s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1e90e160ba07423c9aa9bbc84fa93d0d : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ce51bbb5826499896a30f4f8501ce8a : FAILURE in 6m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2c7ff486c6434c4489d28adf58e66d9a : SUCCESS in 36m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed1ef67730c742cc9d4ec8106ad22b47 : SUCCESS in 37m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8626f50c33794e56b674007c3041b630 : SUCCESS in 1h 04m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a01c6e6a45d044438d926a49b73e242d : SUCCESS in 36m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2ed377dc4b084b0b9b0dcd586d8c260f : SUCCESS in 28m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/c1821bc807614246bb97d368a169c399 : SUCCESS in 1h 49m 53s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/daaa818793c44ef999d3314cea8e0ae4 : SUCCESS in 1h 02m 29s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f9a8e85726764f6faa03c4323ef33d41 : FAILURE in 38m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b354d3aee8f74e36baf0802467f0bb1a : SUCCESS in 40m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b4a99d37185e4a639e4f459236e3fc87 : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32b34ff2ffe1675c3826f74d42f59178f3d092d9","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-08-30 18:45:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eab3f227352b763f95b6f5cabf0a4a878419d490","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-30 19:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/1/thirdparty-check/pure-devstack-manila-tempest-aio/47785f7/ : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":1},{"id":"20677c588662f171ec83ec905eb77a3d14a86682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 21:21:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3b7d54ff2de4a648479d88fa4be6a23\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/279cf78bf6334d0cb3c0c155f9259e20 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ed9da786864471aa6fefb6bb33f628 : SUCCESS in 12m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a3d6ea8a924d47b6f9a5f474be385b : SUCCESS in 27m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1393c3465a7b4b8591ce125290e32b48 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d741564e00a49e38dbd505b957a62b2 : SUCCESS in 17m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d741f28d83bf4468a6bc2acaff14c192 : SUCCESS in 16m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4045f455c78a44e7bb9c4990fa26c933 : SUCCESS in 15m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4709d3a25030438ca8c3d4ac8ffef886 : SUCCESS in 6m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1dbcac2cba3448168174bca91b749766 : SUCCESS in 1h 09m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/feb49d28c8dd485ebb80cb59a5eae20e : FAILURE in 2h 30m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f82d9d62c1f54098baa9bb1f4e4d8baa : SUCCESS in 1h 14m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2cd1135edad946299b535e8bda870acd : SUCCESS in 1h 34m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca28965df52f40329bd98b86c05edb10 : FAILURE in 2h 11m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44af4f3ea6f34abd84bcc5f7c29d190a : SUCCESS in 1h 20m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/164528e3a486414ba035f3bad5ec5adc : SUCCESS in 5m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4dd1617ce3d34215aa31ecb120fb5afc : FAILURE in 6m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0376d9021d364de6b0b74c4342ad6c44 : SUCCESS in 30m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/420fc23f076345d0ab9f61d44facabff : SUCCESS in 34m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3eeb30ab01874180a72787791c66e9a0 : SUCCESS in 1h 01m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4bb947940b0f482293160301c6be02e2 : SUCCESS in 39m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7af50c77f8d64cc99b99667fb89102bd : SUCCESS in 30m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/357268eb3fce49f3a30431a645a28bd8 : SUCCESS in 1h 23m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3274cdf6668d45699cb6ca1c66821aa6 : SUCCESS in 1h 08m 59s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9f325f55367b47a9af0958f7bf7f1723 : FAILURE in 33m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8ab5962dfcab4f24b98d7012d772b72a : SUCCESS in 44m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5ea18215c5c4452fbe2609690d31cbd5 : SUCCESS in 41m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00108a6ca379fde677c1b0054d776ffb59dfee4a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-30 23:07:46.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/1/upstream-check/manila-tempest-plugin-ontap-dhss/d29c5da/ : SUCCESS in 1h 37m 26s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/2f2515d/ : FAILURE in 1h 45m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5f14839ca9cc4a5b86a27fabf5546215cd5e3950","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-31 05:46:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/1/check/DellEMC_Unity_Manila_dhss_true/a874a7b/DellEMC_Unity_Manila_dhss_true/764 : FAILURE in 2h 10m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/1/check/DellEMC_Unity_Manila_dhss_false/01a0d3d/DellEMC_Unity_Manila_dhss_false/804 : SUCCESS in 1h 08m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"e5f9159df1772c1c6efb0b39888921530ada88a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-14 10:34:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e32c18bbb143e8d2956c8d812ca4a307990b7047","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-14 11:35:54.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-2/ns5_manila : FAILURE in 1h 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"a235ce2659f67b494abbd925c0991bb8da727086","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-14 12:05:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/2/thirdparty-check/pure-devstack-manila-tempest-aio/c8074df/ : FAILURE in 1h 30m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8f73d70c2263cca596da01dae180c31db3807022","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 12:22:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/2/check/DellEMC_Unity_Manila_dhss_true/0f2cc93/DellEMC_Unity_Manila_dhss_true/851 : FAILURE in 1h 18m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/2/check/DellEMC_Unity_Manila_dhss_false/880c383/DellEMC_Unity_Manila_dhss_false/900 : SUCCESS in 1h 02m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"530d45b9ae8d1ad87179b95b41b695e901d00680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 12:59:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7798e2216a614288aec67ec854aa2d4b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0c99c4608baa460bba019d5a36c6abc1 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5633dfda790f4676987d88fd784f69a2 : SUCCESS in 8m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/380deaabcec4443e8df48b136c83d3aa : FAILURE in 16m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8f7d9fa9484b9ebccd07778bb73d4e : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab9b6dec9d4f4154b4e50099d5a1a4ee : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/528697f3518942459e10049720346195 : SUCCESS in 9m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d6a423af5314acb906864e2815ea55d : SUCCESS in 10m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3ec5ccf78fa4ff88bd8e11255a9435e : SUCCESS in 7m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/515090dc9f1e42b7ad54723677cb2b0d : FAILURE in 1h 20m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6fae5cd4bf004bb2ad3223bfaab1d77c : FAILURE in 2h 19m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/be9debd4b4e84d5c9eb16aec0072f890 : SUCCESS in 2h 03m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cd3b09fc43fa451b898a6ab183f10004 : SUCCESS in 1h 15m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/41d00044cf09457686b9849e34556f0b : FAILURE in 2h 12m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/af1e469443bb49d4a4bf3b213bf982db : SUCCESS in 1h 30m 35s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/90a5b6178abc497191f99874dc3ea73a : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/44015ce52b2346418314fd78829481b0 : FAILURE in 5m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c6c9ecfc59ec46eba75eb6724f7237d4 : SUCCESS in 36m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0a73776b09ed4754b139b195b654f8ec : SUCCESS in 38m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1534849d54494a66a4104a964fa73d5f : SUCCESS in 1h 01m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0db6b176f3394ca992b34d2a805d488b : SUCCESS in 35m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c6f2c151ea594c8b870954cb150da9c4 : SUCCESS in 38m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/18c0b8e0fd384d2a8637d2d82d382a00 : SUCCESS in 1h 28m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2fc6fa05e0314a7c9e6c3513e2c46ad0 : SUCCESS in 1h 09m 39s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2d42ec83d28247f486ae6b5f66650edc : FAILURE in 33m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/40f2408bbc3f49eda3b2972b9c28076c : SUCCESS in 46m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/beceb5879ca840edbce1bce18f56cce3 : FAILURE in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cee80e99ef13252329e07dd373c505341e1db33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-15 06:54:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"89d47b9b397a38628775a8d05caa81b2183ef06c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-15 07:58:29.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/3/thirdparty-check/pure-devstack-manila-tempest-aio/15952d5/ : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":3},{"id":"1d4901187261af4b4759641aef2fdcf856cd0ec9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-15 08:07:47.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-3/ns5_manila : FAILURE in 58m 24s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"59f1371b4b18984d0bfe05a33c144788db952ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 08:46:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1108c8dbf9f42929b4e42678b6ccd24\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8cca2454a4a9468bbf186c7d3c7aef41 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/225ba3a4062b4c45874b208d0ec9fd3d : SUCCESS in 8m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e333e47bfceb4d65950426a7b198bf08 : FAILURE in 19m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428320bb391448e8b78adcf4fa03a7d9 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d755cc3adfb34fd6a00f35eb2664c36a : SUCCESS in 11m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/630a69edf3f7491187666fc1a227a869 : SUCCESS in 9m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c3200bb1ee645f89e16bb7ca01fb7f8 : SUCCESS in 12m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e85057c8a57243a09ebf7b35cfda5af0 : SUCCESS in 6m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/01f946cb3ba5408b8d80913d17b7a8f4 : SUCCESS in 57m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/faea9ca72785488caff53bda5d0ce6d4 : SUCCESS in 1h 49m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cfe401ff9bfd4621ab92925aa2bd3f44 : SUCCESS in 1h 14m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9d809811507643c4b62368c2a3f5ac34 : SUCCESS in 1h 11m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/443a02795d564a568a333e626e195fba : FAILURE in 1h 16m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/10afc19b6ea14659b2f618c0aa023138 : SUCCESS in 1h 25m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/42d35c7db0da493a8ddb9ef188d44f49 : SUCCESS in 5m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d53b875dce8c4e3ba3a6c4f2bf35aca5 : FAILURE in 6m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/376f18c26cdc49beb24fdf8d836a4523 : SUCCESS in 32m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/19929e50b7444ccd8fcbb7c076e0bf6a : SUCCESS in 38m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/58988bfdca4642159848b2022056b361 : SUCCESS in 53m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1b929a0a46294ff59cc7736d55423799 : SUCCESS in 37m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1e21661b2d824253aa2ecb11342dddc5 : SUCCESS in 24m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1dc554af85b34440a4e7d8f327df6ddf : SUCCESS in 1h 26m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07ca9d1ceae54bb18114d9e29526691c : SUCCESS in 1h 01m 51s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/32c6688db94d4328b591a3ff211dfd18 : FAILURE in 39m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/82ee3d72fff64763880f265dddbe630a : SUCCESS in 35m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/472098440d2642b8ac4b1d4d33bc59d0 : FAILURE in 51m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e46664e00fbde204de743b5edaaf164063a1cf6e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-15 10:38:01.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/3/upstream-check/manila-tempest-plugin-ontap-dhss/c72480d/ : FAILURE in 1h 50m 04s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/ac612f7/ : SUCCESS in 1h 13m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"685148a46ec7c198ec35c61ee6ca828cae9443e9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-15 11:15:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/3/check/DellEMC_Unity_Manila_dhss_true/7b1e7ad/DellEMC_Unity_Manila_dhss_true/863 : FAILURE in 1h 21m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/3/check/DellEMC_Unity_Manila_dhss_false/1b4d11e/DellEMC_Unity_Manila_dhss_false/914 : FAILURE in 41m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e36200cdaa65ec4421925beccf0ae14a6e5f4dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-21 09:42:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"54729d413be558272b3649b5947ef2c234aeeeae","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-21 10:38:56.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-4/ns5_manila : FAILURE in 55m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6603167b59c730083bd64ce1b40095777623d0e2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-21 10:45:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/4/thirdparty-check/pure-devstack-manila-tempest-aio/806d2a3/ : FAILURE in 1h 01m 22s","accounts_in_message":[],"_revision_number":4},{"id":"caa15b7e57e30d54be1b0574e88bb10f257de022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 12:05:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76076a35277c4a28830ca9b37d891b0a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6173e157f7f4790976f11596dacd3bf : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87e541ca72cf4c188080eec847b81800 : SUCCESS in 11m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08b6d7511a6c458c8f6379e735500c55 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/652f94ec1bea41c4b81d5858ab321273 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a830e72f152f4becb5c0ae6f29795769 : SUCCESS in 12m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c07cc08ce2934d799952ffec878f67fb : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfdc8290136146688df1fbde37484c19 : SUCCESS in 6m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32bfcda7a53d482da89aae4c231c8b67 : SUCCESS in 1h 15m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9a691bdff8a94ea7a88c90e3b9748a77 : SUCCESS in 1h 50m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c815316393024543a21fc5544ac4c67d : SUCCESS in 1h 22m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/15eb50d0358b4407b9c2a810551ddbe6 : SUCCESS in 1h 26m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/38ca40c918e549fca3e767ee2dba5f1d : FAILURE in 2h 19m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8d3c435249e445e783e211b57cc59061 : SUCCESS in 1h 20m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f79e80ad263c4c548fd1d45657df82f6 : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8ec92d921a34438a5008a2f065e34cc : FAILURE in 6m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d2f32a7b298a4e3fad404bec7e6379b3 : SUCCESS in 40m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/875ef962bd314b87a3961ebf6cc75bb9 : SUCCESS in 41m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f1326eb5037e4b7cae7eab0366f85bd2 : SUCCESS in 50m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/08e5b771c06944bfac437a89bcef4844 : SUCCESS in 41m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/af2d0251a0254a8bb96119621e6d0d5a : SUCCESS in 24m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/87c9351dbc5b4f33922dd1a71a4c1ca8 : SUCCESS in 1h 48m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/00754bfe8d45461ea942785d27463c3f : SUCCESS in 1h 10m 30s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/50ea20bd36d0465db617182a6f3e89f7 : FAILURE in 35m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/deb46c149fcc42a58de60443b8e02666 : SUCCESS in 35m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/39c201f394b54e38943e810eed37bc5b : FAILURE in 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9bbb52546904a9946d8ceafcccdeabd3f44d6616","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-21 14:03:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/4/upstream-check/manila-tempest-plugin-ontap-dhss/8a6e518/ : SUCCESS in 1h 34m 21s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/19f8987/ : FAILURE in 1h 52m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ae62fdfb46b88996533e11fa4772e2294697e9b7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-21 14:39:21.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/4/check/DellEMC_Unity_Manila_dhss_true/eb2179c/DellEMC_Unity_Manila_dhss_true/907 : FAILURE in 2h 25m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/4/check/DellEMC_Unity_Manila_dhss_false/f88d13c/DellEMC_Unity_Manila_dhss_false/966 : SUCCESS in 1h 06m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"204950a4721691a2e9f2139bae7919e9d51e34b0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-21 19:24:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"67ad84da053bbde50325a32a21da1b2347dac10b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-22 13:29:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"54cd8f3cf277e1f39a2acc765c92e5119c386768","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-22 16:35:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"335d3b87656c5b7abe1a63850e2a88dcd9fc11ab","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-22 16:44:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5216598cd2b45ee7d3c604db17a2848ce31dc4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-03 08:45:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9368cb33abf8a4bf58ccf3997c344497e72e3986","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-03 09:54:52.000000000","message":"Patch Set 5:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-5/ns5_manila : FAILURE in 58m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"337ba20517872ada4c96dc9ed729f2d67b08bb13","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-03 10:26:26.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio pure-devstack-manila-tempest-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"2b7062238ba94583496b27b1c0b95b90d321a95f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 11:15:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b62bd1181c4af5a3a55d4f2d9fc50c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e4e467d44da47279053a411b403108c : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4204ac77ccbe46afbb32a45022884a1a : SUCCESS in 12m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9783035af5b4e4689d0a40d5d102471 : FAILURE in 21m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3579668e264d1e85cb7f9c7c38a7e9 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/687d4cad3a1a45b8ae768dd6f353d385 : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84d0fe2e27b54345833239f00f2e71d4 : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51fd8b1e8344cfdbf4ad7298363c14c : SUCCESS in 6m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f4dd0951739245718fa6c4b448bead27 : SUCCESS in 1h 15m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cda6dc4db73d472a97a819f04bf74515 : FAILURE in 2h 25m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3e286cd99dce4c3d897758eaf06a1568 : SUCCESS in 1h 19m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dd40f66d5e874fea80f99d4099c33581 : SUCCESS in 1h 32m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d2e783b2110443a78536b755f08bce0e : FAILURE in 1h 56m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5e859569aae14e368ee9152d3d5ec296 : SUCCESS in 1h 22m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a38bc1a349a46f4b95f3e8dea848d01 : SUCCESS in 5m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a20637431348448dbbacf357be406c58 : FAILURE in 5m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/52d7a413ce5649ecb33e13aea3ffc001 : SUCCESS in 40m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7615625c829e46d894867acdd3ef78cc : SUCCESS in 49m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bc4134d7b4f24200b3caacbd8dcd25eb : SUCCESS in 51m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0d5904a8f7ff418494e23268e3d62470 : SUCCESS in 37m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1ae59e804a904e9aa8450c401db4b662 : SUCCESS in 35m 01s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5e00a351140c48f98d947f4566d7e680 : SUCCESS in 1h 39m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cf91d2b75b9b4829abfb8f41a61aa084 : FAILURE in 58m 30s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5855ae3943824d5bbbf6498837eaa0dc : SUCCESS in 1h 10m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/21c92ba57536484fbeae7b976f572fb3 : SUCCESS in 51m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/61101672d3f942f1a6294a38ca5bac58 : FAILURE in 41m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3c26acb6e70e938a7906cd885be99f8429fa300d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-03 13:27:16.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/5/upstream-check/manila-tempest-plugin-ontap-dhss/a8104a8/ : FAILURE in 2h 06m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/375b9b1/ : SUCCESS in 1h 32m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"36470242e88ae5654ac2cc18aea62f484a040eb5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-03 18:50:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/5/check/DellEMC_Unity_Manila_dhss_true/387f0a4/DellEMC_Unity_Manila_dhss_true/958 : FAILURE in 2h 34m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/5/check/DellEMC_Unity_Manila_dhss_false/9c3cdaa/DellEMC_Unity_Manila_dhss_false/1019 : SUCCESS in 57m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"eaa79473ace67fe5da55bfd929d3b7e5a5cd9784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-10-15 14:56:22.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"272cc5e0619cc1dc2ee42c1a27e7e4eb302c1fe9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-10-15 16:09:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/6/thirdparty-check/pure-devstack-manila-tempest-aio/991f55f/ : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":6},{"id":"597f15400b15339cb1f0529b80fc2a83dd48e0c3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-10-15 16:30:54.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-6/ns5_manila : FAILURE in 58m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"972133f1f9ac0b5ffc5b80eaaedec456946c9697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-15 16:41:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4996f07993e49a8a35b0c49574af316\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/918b4f284b894977988490782fbf1f14 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27d5afbb36294a318872defc49ac19f6 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ffea54ce646467a91beccd26fbcb053 : FAILURE in 18m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44538d31e68d45cb963caf7008e61765 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ccc96e433c47c68cf57b077f175c57 : SUCCESS in 11m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e540d74ebd97490d80f6cdc21bc7ffa5 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4056d5158a45669b06caa26da27a38 : SUCCESS in 6m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5f80b3e9078642fa9cfc563e2aeba4d7 : SUCCESS in 1h 00m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/21d5e10912154e53b936747e913347d2 : SUCCESS in 1h 37m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5877f9af1c1c474eb62734db31d44266 : SUCCESS in 50m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b9a4f47b8daa4528aab66c24b6a78bb3 : SUCCESS in 1h 22m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d733dee9559543f48adb5a85bb8850e8 : FAILURE in 1h 37m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9140722118134659b0e9122fc4d5685f : FAILURE in 1h 05m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/30ad3291b1a0493184fed34db5ee24a7 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2cdccf8c27734f3bb736783b6dbfd416 : FAILURE in 5m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aa7b7b3294974f03b90dc451e8f703e2 : SUCCESS in 26m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/88efe964870841258ec6cb2d04522ac0 : SUCCESS in 28m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dfe5be93081c4edc94a3e24e97a60998 : SUCCESS in 38m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/30f1492f69464f3f8657ff922f3bfd37 : SUCCESS in 29m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/830d5c4bfdec41a8b6313cdf8be7b2c7 : SUCCESS in 17m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/005ecdb1e2c94d0b97d8d553aca29d2a : SUCCESS in 1h 15m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d21618e0bba241a8b1eb001d7234e5ed : FAILURE in 42m 12s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ada2216330544d94a23216ed8238272c : SUCCESS in 56m 23s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0c6525c3545b4d67a5892ec26f454d9e : SUCCESS in 32m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4ba5bf48b1044b679fe939648d4b192a : FAILURE in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a224e6bc11e24b80e543baa26482ecb92dbcb972","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-15 19:13:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/6/upstream-check/manila-tempest-plugin-ontap-dhss/016bae8/ : FAILURE in 2h 26m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/482584b/ : SUCCESS in 1h 19m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f4c416f0b2426f92d7d422fc7ec4bb4992ac24c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-16 01:33:30.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/6/check/DellEMC_Unity_Manila_dhss_true/37174ba/DellEMC_Unity_Manila_dhss_true/975 : FAILURE in 3h 43m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/6/check/DellEMC_Unity_Manila_dhss_false/25da2f1/DellEMC_Unity_Manila_dhss_false/1037 : SUCCESS in 1h 20m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"861197b4d8c95ee906a54a68483ca40edb1b29a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-16 14:41:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2ba0ff6a5ccfd12fc5abfb8f2db38492031c2fcb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-16 15:44:15.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-7/ns5_manila : SUCCESS in 1h 2m 10s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"867e93bf89d22c9eb63f7902d2fe1c4a1a880967","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-16 15:47:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/7/thirdparty-check/pure-devstack-manila-tempest-aio/25f0882/ : FAILURE in 1h 04m 55s","accounts_in_message":[],"_revision_number":7},{"id":"57d8d311ce6881c12f7ff959dc82427e6ed4b350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 16:53:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33217e7a5884360b7d21810f6774816\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/652e8c77089c4f999107895ee0c123a9 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfabe0d9cfd4adfbba8c24fca9aec1f : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a83049c6626747ae992dba44c5717e81 : FAILURE in 17m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a7b2fd549eb407ab64fcd21687aaad3 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfea7016971b46c1bdf9a07abe86e0cd : SUCCESS in 11m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/808e5282cdee4bd6a62446e662777d90 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c78642ed009048f6bb9ea7b4a718fa48 : SUCCESS in 16m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e91ed92ccd78409e8f9b303655b358dc : SUCCESS in 1h 03m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/96cfef846bad4df683702d8fd97b54fa : FAILURE in 2h 06m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c114df3ea2d045388cead4dab89d9887 : SUCCESS in 1h 20m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/256795e927ff4d59af44586e6559c4db : FAILURE in 1h 09m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ce0f94da0fa044198e43bf848a58daf9 : FAILURE in 1h 34m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/71e26637806a4e0182fcfb85d233b98a : SUCCESS in 57m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/92bb86107a784e86a7081c93e1c97998 : SUCCESS in 6m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1fc9edd513045818d515019d4490893 : FAILURE in 6m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/967c78f41e7943588972ea54365adc8d : SUCCESS in 28m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c5a8646d5a1949e699218493e5d7bc53 : SUCCESS in 27m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f0e335f20174493f8751b0a2b3614cce : SUCCESS in 52m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/88c472df0d424e42a8223f0b9fc05889 : SUCCESS in 33m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/278b24c4de7442d48334c73c8cbc560b : SUCCESS in 21m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9f9735e11cf14b08ab926131634bc538 : FAILURE in 1h 25m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6114cc6ed893461ca02ebb0c99eae53f : SUCCESS in 52m 16s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/35a2f95348804561976f578d12b96eb2 : SUCCESS in 1h 07m 45s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cff24d156c5a4f4cada0b083cda5a236 : SUCCESS in 33m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ff7607811463445483a8d001744f22bb : FAILURE in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"95265cd6a04a74d081a04130f9429ba15a73e830","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-16 18:55:52.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/7/upstream-check/manila-tempest-plugin-ontap-dhss/a4ae235/ : SUCCESS in 1h 59m 26s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/9fe8b0b/ : SUCCESS in 1h 18m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"6a52b1b45e69a2301bad5c5f8db40612d171c04e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-16 21:59:24.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/7/check/DellEMC_Unity_Manila_dhss_true/b49f914/DellEMC_Unity_Manila_dhss_true/1024 : FAILURE in 3h 32m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/7/check/DellEMC_Unity_Manila_dhss_false/793804d/DellEMC_Unity_Manila_dhss_false/1091 : SUCCESS in 49m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"187a504c2f2e10372e60eb931f6f510b993e8a17","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-11-23 00:59:28.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1710bd8df47a4328568e65b35b53281f62f7ff64","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-23 08:04:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54bbbdf8c3ca11a59695074d7c4345b5be5e2f8c","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2022-11-24 02:15:52.000000000","message":"Patch Set 7:\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"636be9b905a5864e5ba30787e8da62b6b8faf0b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-28 15:40:09.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4a4531c2df2c80dafd19f56d043db8b9b342278a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-28 15:40:27.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1001d0f7148ab8c721dc36af67def22a1382d559","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-28 16:41:11.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-8/ns5_manila : SUCCESS in 1h 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"68e8ebc0e619693dfe9f231af4c461989d5f4be1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-28 16:41:59.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/8/thirdparty-check/pure-devstack-manila-tempest-aio/935dc7f/ : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":8},{"id":"b263e3502af3f058b2931e296162b543581aa6ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 17:44:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1729aa170d424a3384412edbb30da77e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4c89da5ddf6142e7b50f4cbd42e32466 : FAILURE in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a821612fb87447feb5cdf4bf5fe1bd99 : SUCCESS in 19m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63f7b041b661420baf66b769e7aaf2ba : FAILURE in 20m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51685d3c22484973b4a281ab96e57769 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f245babc20b24ff2bd2e54d2e06a0109 : SUCCESS in 9m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34bcac81597a42998c7182f581365260 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc64fa5fa3bb4c12a066d7d287dff608 : SUCCESS in 6m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ea82fbaf96a14ae4ae8edadd13d479ef : SUCCESS in 1h 11m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8ff9ef10911f4188a82eb8fac1a63071 : SUCCESS in 2h 01m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4389d86c69fb4af99114dbe81daf0508 : SUCCESS in 1h 03m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ad2524d82bc8475981652254a3c8e1da : SUCCESS in 1h 09m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f32765237777450f8682908b4f39d94a : FAILURE in 1h 37m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/98b5154b71c7479cbac494ed40fe985c : FAILURE in 1h 00m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/febe213db56a4a6887bb2f1d0c36b14f : SUCCESS in 4m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5d17e78fef114bb89376d4c0ef9485ff : FAILURE in 5m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2e24b8886a24cd3acf5042be60be0e4 : SUCCESS in 32m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f3005fbb736c47fab1de8cad15fb68d1 : SUCCESS in 37m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0ee40767c42b4aa7b4fa385fb4c8a490 : SUCCESS in 42m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f80719134cbf43d1bee666c6a58e8471 : SUCCESS in 35m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ebd341a20f2a4d9287ad26602f42507b : FAILURE in 32m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/41c658462cd8408db1df5780c9db04c2 : FAILURE in 1h 35m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8e41addcad404addba7c2d453fb0b633 : FAILURE in 1h 00m 01s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ef35bd446c424e678eaad169c83d2193 : SUCCESS in 59m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/78e29a0a66eb4bbd9bec91d276390ad5 : SUCCESS in 30m 54s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/567bb6932c12418dba250f21cfb67ddf : SUCCESS in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"affa054bee7acdf6021eeb96fb7fd9f6cc4c1197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-11-28 20:17:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3555583fb5c87114aff44e56af4d37565b1c13f9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-11-28 21:18:11.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/9/thirdparty-check/pure-devstack-manila-tempest-aio/8b4c128/ : FAILURE in 59m 33s","accounts_in_message":[],"_revision_number":9},{"id":"12569def3e3122c813e39b3b8c6869792286c25e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-11-28 21:18:59.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-9/ns5_manila : SUCCESS in 1h 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"bc98b0b14800ff21c61c605049e28f61e75dc434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 22:21:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c16a72e397f424693f7968d290b79c8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5f59b542135f47628f57a6828c2aaac5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ebbad1bac648dc9f06dd5c37273b11 : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d984c9a6c8ff42dcb0f890e47a3b32f8 : FAILURE in 21m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a911e537c2a1483092a397d98c8581ce : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c30e5dda99dd47998a59c30beec5e946 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/896c03679a19463791f9020dcfc3edd1 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c18c775cafb4c3d8c998da80e2ab12f : SUCCESS in 6m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49201f1158884064857a38af06887db1 : SUCCESS in 1h 20m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8117827ee2a443b0858c9d1c82e676e1 : SUCCESS in 1h 59m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/18de230b09604f528851b890d2f239d3 : SUCCESS in 1h 37m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6775b95402094eac9fe81e1e232971dc : SUCCESS in 1h 20m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1bfba36c98e4458d94586b04014c5c6d : FAILURE in 1h 50m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/129eb07c727b4aa3830ecd625e9fca46 : SUCCESS in 1h 23m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d82027709ba8420cbc06de6160c715e4 : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ed47119d06ed4618b8d9fc1b97b5938c : FAILURE in 5m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cee231c468bc4ed287571549cf292a13 : SUCCESS in 31m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/99870d21bd4d409f8196ba8c9426882d : SUCCESS in 40m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5b50b50c1fe84600b3a98eed5bb6a02b : SUCCESS in 54m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3a38e10648414c70bcc3cee4466e5f9a : SUCCESS in 38m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/94c2b956c31d40ca9801ee3b184c661f : SUCCESS in 25m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f2fa83dffee24dbd801f57ac8c52358d : FAILURE in 1h 32m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3e7934748a79429aa04e70f3d7c57675 : SUCCESS in 57m 44s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8dd3aa93d5984afbb223d1a99f47483a : SUCCESS in 1h 09m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d0df3f348dc8441b929fa588c26e2b80 : SUCCESS in 36m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2efe0cecb13b4dfe9b4493f15e43c409 : SUCCESS in 42m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"79cfad08ae5bb60ca51f822d6a673f3659038f22","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-30 00:55:22.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/9/check/DellEMC_Unity_Manila_dhss_true/7ffeefd/DellEMC_Unity_Manila_dhss_true/1055 : FAILURE in 2h 44m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/9/check/DellEMC_Unity_Manila_dhss_false/0b3d7b0/DellEMC_Unity_Manila_dhss_false/1127 : SUCCESS in 51m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"92eef919ca54659d21153b851f65e0d36aa7be0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-12-01 15:13:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"118b4551b66eabb2656ac0d33cdf740cc54da678","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-01 16:32:10.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/10/thirdparty-check/pure-devstack-manila-tempest-aio/ebfe8f5/ : FAILURE in 1h 17m 57s","accounts_in_message":[],"_revision_number":10},{"id":"ca99d1561367430ee782fea744aa64fbb8701177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 17:36:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/053ba5f554584fa787c13da1848c4e56\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5de3c4d1a7e940b6b435109c754381f7 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6dd15efa4f84805be1ab7982277d88b : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45d741e909da4d1dafc7ae32274b9d04 : FAILURE in 18m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a06019847c743b59e00473076dd9c83 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48a4a529dd9a4beda8dfcef2cade85ec : SUCCESS in 9m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca282235c570407eb404875688027f75 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6782e7785f74996a3188ca22e30d38a : SUCCESS in 5m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9961ac67d9114062aa15336a7d1bfa0d : SUCCESS in 1h 18m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/076eaa2c00f34583be4a6c3a7bba601e : SUCCESS in 2h 16m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3caf4d2845904053b6e65c9e000ce660 : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6bb3b8b7dfc64515b6876410f96ba261 : SUCCESS in 1h 27m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ed9bc6a4e9fc4e76a8464341bd4abda2 : FAILURE in 1h 25m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/669ba500725b4c3e856625e059cceec1 : SUCCESS in 1h 21m 01s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d2e9c617f3314d4ea4e5ed6aa682c006 : SUCCESS in 4m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba23c744e3bf40f28e07f126c02ad528 : FAILURE in 5m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/921f1152f2cb40eeb25ea1f94ddf90a2 : SUCCESS in 24m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/383ad69e51204026bde90e4bbb36a055 : SUCCESS in 40m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/255a31153fac4f0db1e06f7c700bdfc3 : SUCCESS in 51m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d9519579da4b4c6cb47fc235c3188d54 : SUCCESS in 29m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2939124612464d0da7b68c1726d73447 : SUCCESS in 17m 26s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/da52f72dfe9c450ea76051b59cfa14bb : FAILURE in 1h 34m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/de7fa60ad7fc492baaf07d421efc7163 : SUCCESS in 57m 57s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3eb9ef4289664cf6a938d6fd79381516 : SUCCESS in 1h 05m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9fa16b256caf48c989d4f54f8c34b8c0 : SUCCESS in 39m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9af3781ad91a4f45b610e23d5d708341 : SUCCESS in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"54cb020a04c04b5997d30caff8bd5941ac0f9216","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-01 18:15:24.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-10/ns5_manila : SUCCESS in 1h 40m 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"d3f707ef4a8134028078753db7ca1b02b11a2111","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-01 18:20:18.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/10/check/DellEMC_Unity_Manila_dhss_true/91150af/DellEMC_Unity_Manila_dhss_true/1062 : FAILURE in 3h 06m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/10/check/DellEMC_Unity_Manila_dhss_false/fecec71/DellEMC_Unity_Manila_dhss_false/1134 : SUCCESS in 54m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"65a62bf404a4926c50fabbb0dee68e00d60eeec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-12-19 19:00:23.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"1208b62f4163a54a96e884561a39d84ebc91f416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 20:56:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad3934197ed4857a9c947d7d96edf34\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0c40a79409024b0dafae47df7d64539a : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d6bf1d17184454be7d23650fb795ba : SUCCESS in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b80ae02d3f4b20a874a406d77cdd36 : FAILURE in 16m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5b5b0f2c1e4bd986a566ec28d6e5d2 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/741c787b8cfb41fbad2fd368377786b9 : SUCCESS in 9m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77d08185351641b7be9c92782abee420 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/920bcc922fd6488aa6cc97854150f1f9 : SUCCESS in 5m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cec9b0cd9da9444d9e85c4d08ca93017 : SUCCESS in 1h 26m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/90cd220d74264a7ba01e0c388b4171d1 : SUCCESS in 1h 40m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/36f2cc12928a4f2682d216fd7082df16 : SUCCESS in 1h 20m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f1573f0727084932b54838c6679b44fb : SUCCESS in 1h 23m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2a7a815318442a08b8866d95c54000c : FAILURE in 1h 50m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b15ffe7e851f4d75933bbe5f8fb29148 : SUCCESS in 55m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/598c3b9fac1d4d31864f30e6a3772014 : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/acec9d07513743f897c30f6a5e187f25 : FAILURE in 5m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ecc362d0e68b43b698df1bf3daa4c55f : SUCCESS in 39m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/30c001ed286641888d0f565ed18472e2 : SUCCESS in 42m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e341ae7b23e64844a9c5c76ba4ed0b3f : SUCCESS in 42m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7da8f2a6313d4c04b51b1be3681617d9 : SUCCESS in 24m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b7120a8f4a8b4a1aa5d07851abf3e106 : SUCCESS in 27m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/41296e45d3e04c6da1dacb39ad7550f1 : SUCCESS in 1h 27m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8808d21f91ab4482af524196b66277bb : SUCCESS in 51m 59s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/118e15aced984a33acba3758c5f604f3 : SUCCESS in 46m 01s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/af592920117245a38c76eda2c94453cb : SUCCESS in 36m 19s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dee26b907d5f41878bbfc725a879e5e0 : SUCCESS in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b993a4978093ec619f6d91a458fad1fdcbb80a59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-19 21:01:39.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/11/thirdparty-check/pure-devstack-manila-tempest-aio/824a454/ : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":11},{"id":"7b49a244d25f9b8ba3e9bc03f8cbfe59bf51817e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-19 22:03:53.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-11/ns5_manila : SUCCESS in 57m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"4b1f33e9c9568fe0d10fb1dbc55976ea2d6bbbfa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-19 23:16:30.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/11/upstream-check/manila-tempest-plugin-ontap-dhss/c394387/ : SUCCESS in 1h 37m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/8dc8a6d/ : RETRY_LIMIT in 9m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"84dd8373bee98b759db3c8d5ca4df8a8ecf48f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-23 11:05:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"279341314a897448c5ea5c20b9e8d5808134148c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-23 11:33:58.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-12/ns5_manila : FAILURE in 28m 20s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"357e77f758849a78b3056f3b114a23cec89ceca7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-23 12:17:33.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/12/thirdparty-check/pure-devstack-manila-tempest-aio/8280f96/ : SUCCESS in 1h 11m 31s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/855220/12/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/e5f572e/ : SUCCESS in 1h 09m 12s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1d0427fa0b6c96c2ea39867521f8ad6acbfc3361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 14:13:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c6079a116245048344b6ec17ac7dba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/654d866b15544ac29873d0a3608125a7 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284699ee3cd54b549dadf04be22bf76e : SUCCESS in 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f0f798f114d4b1888fb507332ae84e9 : FAILURE in 28m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0106335dacb4473b3a118315ebc421b : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7721d25be8b4643ba4a64af55f322ef : SUCCESS in 10m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32cc2846814341728a8d561f649ccbf3 : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5674214c478143a09381b9fccc969cdb : SUCCESS in 6m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4fa2cfb41fbc4029baa938a1acd13662 : SUCCESS in 1h 18m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb62cc331f7f41cabe9005c4ced58a92 : TIMED_OUT in 3h 02m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4fd7f4b1049b4070a2963365b1d87eaa : SUCCESS in 1h 14m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7cd80477414643738e22e0f4eb422f9f : SUCCESS in 1h 36m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2ce4ff47100348e78386c238946e02f7 : FAILURE in 1h 46m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/040b2d4475ba46c9b6a39433f31552d5 : FAILURE in 1h 09m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a09d2b20ee6b40f1a675ebdf8ec031e3 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b4f76353fce4d1783798db2739ac74b : FAILURE in 5m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b305e3a49804b58a4935730c6f64e0a : SUCCESS in 33m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e04022c5ca5a4e019007d39e146dfcb1 : SUCCESS in 49m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bb6f25ad5c0c40b3ac35b7e1eb58054b : SUCCESS in 42m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/90afcb57a6f34a92b90881d2162be85c : SUCCESS in 25m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f7b0cfb548f343028c44886e0055f13e : SUCCESS in 27m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/06cc8a301f9f40deadf42ddc92a3f43b : FAILURE in 1h 06m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ef64bd1c864e4c6099d9dc48bd4b448f : SUCCESS in 46m 22s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e7516b8ea8ac42beb969c9658fe5716c : SUCCESS in 1h 15m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c62b5667c0ec4222b1481f71b82616fa : SUCCESS in 26m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c115daa1981e40019b5dcb7de33d16c9 : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a1b764b2e1d7d96760b09ec47dcb913eb3fb4415","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-23 15:46:18.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/12/upstream-check/manila-tempest-plugin-ontap-dhss/8538146/ : SUCCESS in 1h 31m 29s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/12/upstream-check/manila-tempest-plugin-ontap-no-dhss/88810b8/ : SUCCESS in 1h 26m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"cc7e457f996b32fd8a08fa25d60d833adda6fb52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-09 09:38:01.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"d6f0125399f66120ca44cab66bcb9862729256b5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-09 11:02:42.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/13/thirdparty-check/pure-devstack-manila-tempest-aio/f4e258b/ : SUCCESS in 1h 05m 02s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/855220/13/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/ff44af0/ : FAILURE in 1h 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a63b0cc5d9dd3ad64124aca4fb90789c29077cbb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-09 11:12:04.000000000","message":"Patch Set 13:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-13/ns5_manila : FAILURE in 55m 44s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"e93c88ef5bfa14c2f7757d8079ff07cab0406cb1","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-09 13:00:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e9b37ad08b30d236807fe571cfc5db126043dad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 14:59:06.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92d33c17b7974da58004a7625442c0e4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e1c7ae7a767d428392e5c485dc4b233b : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91fc825057e6437c87999a0fcb6db6d1 : SUCCESS in 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3f8264f9cf74f7ba2f600f61a44085a : FAILURE in 17m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a2ce6bf14c426abbe568f536245095 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b097e461958d4e528b803f91ab3d7396 : SUCCESS in 10m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41793b2122b844f58666346e7dcd00b2 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d81955eb767451a93494f484c4b3e3a : SUCCESS in 6m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/433ad56934e2478d8deb6101f77b6129 : SUCCESS in 1h 12m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/181d63f581844f6d9f305050f5b8881c : FAILURE in 2h 06m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f8d279666e248bfbb99ddab1876b8a8 : SUCCESS in 1h 18m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bd5008c56dd24cce8795fa02db329b79 : SUCCESS in 1h 25m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d9ea1e4bfe9b45f795a6021378630173 : SUCCESS in 2h 08m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3f07946e752946dab4c44ca848a4bad2 : FAILURE in 1h 10m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/944f4a1533ee4360bf0500f06b3b7781 : SUCCESS in 4m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9032c0a33baa40abac37967a338d8460 : FAILURE in 22m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fdb99966efbd49b98f24f572b4a4dfe2 : SUCCESS in 30m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5bc05aa38d6f40a298e91feb9ec57920 : SUCCESS in 30m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/854b26d095e14f8aa13846b249f5f638 : SUCCESS in 45m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/956b5976e59b4a2498ec1d82b3dacc34 : SUCCESS in 28m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8042d23f77d54f4faf492f484959f61d : SUCCESS in 29m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/472fbb1eba12481e97183d55081b241b : SUCCESS in 1h 29m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/623bbb6471f949a98cfb21039477dc0b : SUCCESS in 1h 02m 07s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fcf6cc9d8d2743de9bc6b0892dea99c8 : SUCCESS in 49m 43s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1cc349aea2d84b49b7667880f15f8934 : SUCCESS in 31m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1014afd4514d4c7ab5a430e05ad4aefc : SUCCESS in 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c272f6c524ceab321a0e17d01902b8c6a10f198e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-09 16:42:40.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/13/upstream-check/manila-tempest-plugin-ontap-dhss/c0bb5b9/ : SUCCESS in 1h 42m 50s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/855220/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/44a7365/ : SUCCESS in 1h 17m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"d5f967d55dee9df94f677b9563a0947c6df15c95","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-14 22:59:20.000000000","message":"Patch Set 13: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":13},{"id":"c619282c09638723e8f0525f986ea86f23cffbbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-15 16:10:26.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b86b90dbdb028776c7a1a488d88f0524fbe1b5ad","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-15 16:10:27.000000000","message":"Patch Set 13:\n\n(8 comments)","accounts_in_message":[],"_revision_number":13},{"id":"62583a0d7fbfb1afde3305f9d4424f82b44955cd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-15 17:25:41.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/14/thirdparty-check/pure-devstack-manila-tempest-aio/be74f80/ : FAILURE in 1h 14m 14s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/855220/14/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/50f8bcb/ : FAILURE in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f8fb4438f3e789c33e3d6e5f555bfb724b7979f6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-15 19:38:40.000000000","message":"Patch Set 14:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-14/ns5_manila : FAILURE in 1h 53m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":14},{"id":"96fbe136878e405a1e4f14486acee65a151e8315","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-16 01:11:46.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"3a544d2e30874f4e829cb139529fe8661d416aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 01:33:14.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fb1f1bad3d546d2839911d05a835b2d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c8aca334d3384f279746b5a102dbd9b7 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082777116b8a4d0db91ac8c986e59594 : SUCCESS in 14m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c8544241db243d39a80ca7cba86f4b4 : FAILURE in 18m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e957e56f2dab41649817a653982af94b : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffe4b55162e3464ab62d12750c072b2f : SUCCESS in 8m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c20bcfa8e8fa4fb5a2df1853c9ba5e14 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4878fa24d4bf4bdba553c74e20b1af92 : SUCCESS in 6m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/00e391b6cf73484288b303f684ea5423 : SUCCESS in 1h 20m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1c789b92aac14a60b714799d77c97d5c : FAILURE in 2h 10m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/203e90cd3c314b6ba0c9b58146d5c14a : SUCCESS in 1h 03m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c221e106a5c6403eb4e44ed8ea54612f : SUCCESS in 1h 18m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bbfc7ee353824e72a30605cc42ac9efb : RETRY_LIMIT in 3h 01m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/453b603acf66477f92b5d63aa90c6ecc : SUCCESS in 1h 16m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fcf90a5dc39744ac918dc93846cd65c1 : SUCCESS in 6m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/03fc21c2c5b740ce9ce310cd80be39d7 : FAILURE in 20m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e07e749dc9445078b302df4e038a855 : SUCCESS in 26m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/169501ab6fc349cc9676ac45bad1ef03 : SUCCESS in 37m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56c3bb26540a481dbea946b37945e492 : SUCCESS in 45m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/773f9f7bb3994bffa487fe972f357361 : SUCCESS in 35m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6daae7ecf3ba45e4b0acc17b8f89e8e0 : SUCCESS in 21m 34s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/afebfed09b344c2aa3acee0a1c49de1f : SUCCESS in 1h 17m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8c07ff1278344d37b4cd03874e8bf1d7 : SUCCESS in 49m 26s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f86da8fb16274dccaedf3f536b8b7304 : SUCCESS in 59m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e60659f7649240109ee71cdf29d22b85 : SUCCESS in 45m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3f3dd7e113f5476e8df117897f4cc6a7 : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a55ce01085c7d775003d17668a58ba2d6b025d90","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-16 07:15:42.000000000","message":"Patch Set 14: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":14},{"id":"62586079705d55c9cd766b574dae17eda8f0a357","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-16 09:31:36.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"156acaba580068d5d8577e27114758787dd8af23","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-16 09:31:57.000000000","message":"Patch Set 15:\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"df3034206f723fcdabb9810287669118b996956a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-16 10:52:45.000000000","message":"Patch Set 15:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/20/855220/15/thirdparty-check/pure-devstack-manila-tempest-aio/1e9f0d7/ : FAILURE in 1h 19m 49s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/855220/15/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/0f1c2a9/ : FAILURE in 1h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"acf92fdc360f8243711cc0d77bb64f8929ee7689","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-16 11:06:02.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_true/61c14bc/DellEMC_Unity_Manila_dhss_true/1232 : FAILURE in 1h 10m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_false/2c25e20/DellEMC_Unity_Manila_dhss_false/1331 : SUCCESS in 59m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"3f0b8d485676ce31ed2710add727b40d64479880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 11:45:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/878644ef88474fe68061dd9b58350a73\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a8857790c3aa4e6e97c2a7cccfc95ec9 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2c711e32764714948394259931889b : SUCCESS in 10m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd651779c2354e08a9b4fbd8ed708076 : FAILURE in 19m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06072ae75a946a7a4f017481826f8a3 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/769068de93db459aae0d23632caedb21 : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d76cc2297c01485a9f1ac865836371dc : SUCCESS in 17m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f9b7961e1884a46979dc0ae71eda3de : SUCCESS in 6m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fd216cf765dd4d039a3033d368b81c92 : SUCCESS in 1h 14m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ddad3107ee54485aa8349dcf7c48481e : SUCCESS in 2h 11m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/40a8152ebe784d7795c3ef0dd2834acc : SUCCESS in 1h 26m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/598df7f598b64ab98e66e154910b46c0 : SUCCESS in 1h 25m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/17d1c7f719ba4c009247b26de15d7f76 : SUCCESS in 1h 31m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4644b1868a1e46a6a85e6327f8db0587 : FAILURE in 1h 07m 36s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc7613c4c5fa498889b120ceff7d8e25 : SUCCESS in 4m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b9c077bf4a1f44f7bfa8fc5462da2375 : FAILURE in 23m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/81058ba3d7384d91aa3869af7b5b17fc : SUCCESS in 32m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed07bffee1d34db8953a972c371f65b1 : SUCCESS in 47m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f1b1521e1c2d41bb83eb4cf948e481b4 : SUCCESS in 1h 04m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/feaf72fcedeb433ba7ac50ff75fb7336 : SUCCESS in 36m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f8fefffbe4df4c8b96736795dac6c31a : SUCCESS in 20m 11s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/673a1e4b7e3e487fb6f388d6bd2d24f4 : SUCCESS in 1h 25m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7a1d8587731545a592a9d6fba147cd39 : SUCCESS in 59m 50s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b7371e8b09f2474181553e61c6e1d7bd : SUCCESS in 1h 01m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ba469ce9634c4d579659767a02994ea2 : SUCCESS in 32m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ad23479344454c25913079e5c8c5cd4d : SUCCESS in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1f2c673ab1b95f0a1332bc7ff41cc54b829f0f3c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-16 11:48:42.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3c7c392c4e164b379247bf93ee5ff0e32efa98aa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-16 13:59:38.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_true/21606e7/DellEMC_Unity_Manila_dhss_true/1235 : FAILURE in 1h 01m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_false/97c6b93/DellEMC_Unity_Manila_dhss_false/1334 : SUCCESS in 53m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"6dd12db44c43ab84cadbe0a835af4f7c0b2ce4c6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-16 15:32:37.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"30947a333e408e276e9a65c364c216fd4ae2da37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 15:33:20.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"511d527f6fe7026d14b6c8ea56bfb15777623308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:45:14.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d17316328ef04afa8eb9d40e13d96b55\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/051ac432b4d142f186a3d02cd37fe485 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09bc3c1da894ec3a568d7747129bde1 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9981c1f3d1471f87cd723ff2d92151 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/142af5c9da614f17bb46ed915e84354b : SUCCESS in 8m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3bfd49e077d40b3954a6b4ff34dc85d : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea43d3a7eb444296a65f9b2c5da71e9c : SUCCESS in 5m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3af1cc9afa0d4b36a9aca31679e0940f : SUCCESS in 1h 07m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f63514c3e1514ad480bfaa926043f3c8 : SUCCESS in 34m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e5dc76f6f5f475eac31efba38587ca0 : FAILURE in 13m 31s","accounts_in_message":[],"_revision_number":15},{"id":"a15dc48133385c1d031353cfc1b546f41ae300db","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-16 16:51:30.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a00c68b862499e074cc582d155d3e376dc242924","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-16 17:16:34.000000000","message":"Patch Set 15:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-20-855220-15/ns5_manila : FAILURE in 7h 5m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":15},{"id":"2a74ab3091f3a68eb0dc05b94ea753e5640fbd19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:22:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19856ac0d1634800b848e417f280e2c0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91556400da264599bcd579e9e1306c2c : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc29a8fcf064b46b8c7a974bddf2848 : SUCCESS in 10m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5947b2923390487ba00f6cc17b103e8a : FAILURE in 19m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6012b5f5134fd0b29a277c2d7ea1a4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74450397d15541c38e943a314f29036a : SUCCESS in 9m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c6934c92d8749b3aa2079bf4f7b8bee : SUCCESS in 13m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a916112ea944241b2fcf870c9b04dda : SUCCESS in 7m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/771c26e04f8947618260d502c65157cd : SUCCESS in 1h 41m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/79f1e423ef464350a6631fb63e5efc4e : FAILURE in 2h 01m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/23313818dfa84948a516ecb2630bda7e : SUCCESS in 1h 11m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4ab50428b9f54630baa80b94430aae75 : SUCCESS in 1h 13m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/71db6ae820f1441d8e6ca5547dd15f54 : FAILURE in 1h 53m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/150a327d7331423aac116e4fcec7aa9f : SUCCESS in 1h 08m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e0c20fb6b25401ebc43242635b3d5ed : SUCCESS in 5m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ce4e7ef9586455eaaad4346e5d158e6 : FAILURE in 24m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9c9de31660d641dfb898e03e847eae2e : SUCCESS in 34m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9cc36d0ed0b44a3e962ab107a9b4ef45 : SUCCESS in 44m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2a7fab0da8e744cfb745c009aba4140f : SUCCESS in 48m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/616bb3c4db4b45c184886d0ef362f74e : SUCCESS in 45m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8e76989939d748b7957f6ea5d069fcd2 : SUCCESS in 29m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dcc5c84adb204acda4dd4a4949b7b3ee : FAILURE in 1h 17m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/24c174350cc8451c823db7c6063b94ba : SUCCESS in 1h 04m 38s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6a3c9a54b8d04dea81e835e002a44903 : SUCCESS in 1h 07m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ecc0cf88b6aa4236b7de261607574f7b : SUCCESS in 31m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/64f404fa9c6a468ba6c75fb5bd49e6fa : SUCCESS in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a3c55eb598a4307c46d054d977fdf10ea6f4e036","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 22:23:03.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"f3e9daf89db8a4b2754f77e34bebded724f3f767","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 23:36:59.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2f79cf4e589448da67ea0c035e160eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8e71da69b93d468e9f128b0034d86081 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5ecfb6384e4e489c64bcdfbc6dfec8 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7037d57ced21447c84a49936f1bad13f : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49772dbaca4c46739b6ea0adea97a13d : FAILURE in 19m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4f713f006144b4d982f5cf5792ee815 : FAILURE in 20m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08bbb828769542a88249d92cc7b72900 : SUCCESS in 5m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ff003b78fc554f0c84d638183af31a04 : SUCCESS in 1h 12m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a8323c65e5484a6d9b5c1b90bb409f2d : SUCCESS in 21m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c15715744c20443c874f68dbfb6d9ec0 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":15},{"id":"0c3810623adf56af40849536b3478a81d48b7d50","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-17 00:08:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"8bcb4aacd9f76a6f62a7e931af2b5a253d969b88","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 01:22:03.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_true/2438a5d/DellEMC_Unity_Manila_dhss_true/1244 : FAILURE in 1h 07m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_false/10ffa0d/DellEMC_Unity_Manila_dhss_false/1343 : SUCCESS in 1h 12m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"88b93e37ed447b81130f8db8e2f6ada818993ab7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:14:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9335ccd1a1c9492499a9a8cabfa4bb11\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e3758962d9314615b49e43a05e5a5847 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6ea99b26d64a39b2790760790e479c : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c2983bcd5f74ae7bc77eaaf132b562b : FAILURE in 23m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd974311e8484c59a22a15a65e8628de : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c97f1f6841334b5196f98194cc053d14 : SUCCESS in 10m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e4fa16b1f974a8f8ac8758a93921d8c : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adcd0660b77b4afd9dbbd915b925c796 : SUCCESS in 5m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f86052a5b9214363a91fc77a88098fa8 : SUCCESS in 1h 18m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e80e9b94db5045c380dcb34a12538c72 : SUCCESS in 2h 00m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/df6c60da44704e46b08821c3812949c6 : SUCCESS in 1h 09m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e1088a6681d74c94b2432342154d3adf : SUCCESS in 1h 15m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6d2969525cba4d8190e3f3c755f28839 : SUCCESS in 1h 36m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b8fc3ccb51b4cea817858f7c4f14bd1 : FAILURE in 1h 03m 35s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5c71a30b153a4d47b8a4aa8a1b184316 : SUCCESS in 5m 01s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d19d9cf300f046a497dfd2b09390a1f7 : FAILURE in 21m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d2537cc8ce9947b09558609fe552530f : SUCCESS in 28m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/96d6e34d81314e9688dd34e7d01147b4 : SUCCESS in 35m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/50cbf16f4a4f4defa517baeebf35a928 : SUCCESS in 41m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ab933f37f1af4f65913fd545be296216 : SUCCESS in 33m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/49e4019c209240e798aeca1a891827cf : SUCCESS in 20m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/fee338347ce343a0bdf2313cb696a29c : FAILURE in 1h 07m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8e7b3780f3ac4e70b18cd2ae04dc7b8e : SUCCESS in 1h 05m 24s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8a047e8ad7214ab6afb86d91d6b90124 : SUCCESS in 54m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9c29c1cce3e24279b3c27975f9051cf5 : SUCCESS in 36m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7a2f24d618794a6e9e7724d8847f766e : SUCCESS in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"78c4bb6b4e87cf43edbf21390d790a516d31c00c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 02:15:43.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"167c36d84475bb38d8a0fc84863357fd7b9bd750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 03:56:53.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ed6ff317f048b5bebb76e5223d177a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/07f2bc62172844bdafce21dafae5f588 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45085a6f40b48bb8f5f51845637c3bc : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30839c152d394bfeaf103b7b3b175cc4 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21a539cdbe974560995f88d4c32b3aa3 : SUCCESS in 10m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f046d07cc19546da951b799d2465185a : SUCCESS in 14m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f2aa0a5b1934ce7967d2e6ebed9d087 : SUCCESS in 7m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/45f8eaa5543c44808206cf213c765845 : SUCCESS in 1h 19m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/639c12738120409f8b9a436373ad3f53 : SUCCESS in 32m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff524fd2ea734b0f803398e003cccc12 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":15},{"id":"0e0633cdfa8f4010537ecc80d96036e4c9f9fd01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 03:56:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"8845142c8e971b3fea3b6f7cc71a4a3fd4737980","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 03:58:10.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2ec648c521f4667a89eccded7e78586\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ddbcbc690ccc44b290fba046b08dd9b8 : SUCCESS in 50s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/defe83be328041a3acf188e5685f8cd3 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2d96a0c27c440799512c7631fdd2d6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":15},{"id":"4adfb57ab12806f77a9633b9fa6cf3e3c10b0920","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 04:46:55.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_true/a89c4ae/DellEMC_Unity_Manila_dhss_true/1247 : FAILURE in 1h 04m 01s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/855220/15/check/DellEMC_Unity_Manila_dhss_false/daa6fb0/DellEMC_Unity_Manila_dhss_false/1346 : SUCCESS in 56m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"858939c19042bca563f9241325b92074ddf61e08","revisions":{"722f390c1f582668906f891adfb61ef2add2979f":{"kind":"REWORK","_number":1,"created":"2022-08-30 15:37:34.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/1"}}},"commit":{"parents":[{"commit":"281788715b618384168e69b1c1e5b17b10e37bb3","subject":"Add scheduler_default_extend_filters option.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/281788715b618384168e69b1c1e5b17b10e37bb3"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/722f390c1f582668906f891adfb61ef2add2979f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/722f390c1f582668906f891adfb61ef2add2979f"}]},"branch":"refs/heads/master"},"6437b18c6bf88c661b8ce036e395f228ba970836":{"kind":"REWORK","_number":2,"created":"2022-09-14 10:34:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/2"}}},"commit":{"parents":[{"commit":"3f431ef9eb54c8ba2c2b546f5bcbc6258e935915","subject":"Merge \"Fix DriverFilter/GoodnessWeigher string evaluations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3f431ef9eb54c8ba2c2b546f5bcbc6258e935915"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-14 10:34:20.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6437b18c6bf88c661b8ce036e395f228ba970836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6437b18c6bf88c661b8ce036e395f228ba970836"}]},"branch":"refs/heads/master"},"c6f63c5d08f82a89f058ed55c787e2f832c4bd21":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-15 06:54:28.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/3"}}},"commit":{"parents":[{"commit":"6e8ed59a91934096a4ddc9b184ef50a633cbd1ec","subject":"[NetApp] Consider last transfer size and error for replica state","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e8ed59a91934096a4ddc9b184ef50a633cbd1ec"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-15 06:37:17.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c6f63c5d08f82a89f058ed55c787e2f832c4bd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c6f63c5d08f82a89f058ed55c787e2f832c4bd21"}]},"branch":"refs/heads/master"},"c567778fd23416cba175ed63485363590a54a4e8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-09-21 09:42:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/4"}}},"commit":{"parents":[{"commit":"e597c4b5e611382126add9586cbba4db9f291169","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e597c4b5e611382126add9586cbba4db9f291169"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-21 09:40:15.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c567778fd23416cba175ed63485363590a54a4e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c567778fd23416cba175ed63485363590a54a4e8"}]},"branch":"refs/heads/master"},"71805c1ee5d2756c2b5c2c8ef96e1d6e3af661e7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-10-03 08:45:07.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/5"}}},"commit":{"parents":[{"commit":"4aa84c722f83d842b6d0bb1a60a4ce3c636af591","subject":"Merge \"Raise an exc if no VIP/backend is available\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4aa84c722f83d842b6d0bb1a60a4ce3c636af591"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-03 08:44:45.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71805c1ee5d2756c2b5c2c8ef96e1d6e3af661e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71805c1ee5d2756c2b5c2c8ef96e1d6e3af661e7"}]},"branch":"refs/heads/master"},"5a7358c7daf80b8cfa419d8d3cc7affd19b57d30":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-10-15 14:56:22.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/6"}}},"commit":{"parents":[{"commit":"031fab2c8fab31c56462390706557d2c37ba0b89","subject":"Merge \"Fix replica quotas allocation during share migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/031fab2c8fab31c56462390706557d2c37ba0b89"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-10-15 14:52:24.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a7358c7daf80b8cfa419d8d3cc7affd19b57d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5a7358c7daf80b8cfa419d8d3cc7affd19b57d30"}]},"branch":"refs/heads/master"},"a65b330f8706e5a700e80a501bce08de2f9df0fd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-11-16 14:41:33.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/7"}}},"commit":{"parents":[{"commit":"605925e8f4a70bb70a9921d6879101d30b7ad8b9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/605925e8f4a70bb70a9921d6879101d30b7ad8b9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-11-16 14:28:54.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a65b330f8706e5a700e80a501bce08de2f9df0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a65b330f8706e5a700e80a501bce08de2f9df0fd"}]},"branch":"refs/heads/master"},"343b019327bfd7fa14c73ac54e4c26031060454e":{"kind":"REWORK","_number":8,"created":"2022-11-28 15:40:09.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/8"}}},"commit":{"parents":[{"commit":"605925e8f4a70bb70a9921d6879101d30b7ad8b9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/605925e8f4a70bb70a9921d6879101d30b7ad8b9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-11-28 15:23:41.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/343b019327bfd7fa14c73ac54e4c26031060454e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/343b019327bfd7fa14c73ac54e4c26031060454e"}]},"branch":"refs/heads/master"},"074d271b34b20e3625de49f7498b48cddfdec815":{"kind":"REWORK","_number":9,"created":"2022-11-28 20:17:46.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/9"}}},"commit":{"parents":[{"commit":"605925e8f4a70bb70a9921d6879101d30b7ad8b9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/605925e8f4a70bb70a9921d6879101d30b7ad8b9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-11-28 20:17:36.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/074d271b34b20e3625de49f7498b48cddfdec815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/074d271b34b20e3625de49f7498b48cddfdec815"}]},"branch":"refs/heads/master"},"2e841781e1e86aec7d8bf824d524a4acc90058f8":{"kind":"REWORK","_number":10,"created":"2022-12-01 15:13:07.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/10"}}},"commit":{"parents":[{"commit":"605925e8f4a70bb70a9921d6879101d30b7ad8b9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/605925e8f4a70bb70a9921d6879101d30b7ad8b9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-01 15:12:58.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2e841781e1e86aec7d8bf824d524a4acc90058f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2e841781e1e86aec7d8bf824d524a4acc90058f8"}]},"branch":"refs/heads/master"},"e977e77ce367d287cce1e52bb057fc7158f6fd76":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-12-19 19:00:23.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/11"}}},"commit":{"parents":[{"commit":"4408ec4f7cfac834c454537388fa23a13915e3f9","subject":"Merge \"[Infinidat] fixed host assisted migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4408ec4f7cfac834c454537388fa23a13915e3f9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-19 18:41:19.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e977e77ce367d287cce1e52bb057fc7158f6fd76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e977e77ce367d287cce1e52bb057fc7158f6fd76"}]},"branch":"refs/heads/master"},"c9517d8a4f2e64ae22d43364c8c5f198aed3e43f":{"kind":"REWORK","_number":12,"created":"2023-01-23 11:05:08.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/12"}}},"commit":{"parents":[{"commit":"c5cf2b98201386b5a67a45246067af50da72da85","subject":"Merge \"db: Migrate \"service\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c5cf2b98201386b5a67a45246067af50da72da85"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-01-23 11:05:00.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c9517d8a4f2e64ae22d43364c8c5f198aed3e43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c9517d8a4f2e64ae22d43364c8c5f198aed3e43f"}]},"branch":"refs/heads/master"},"b24b96c608f5fcb6608efa9a2da9a3f3536e584f":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-02-09 09:38:01.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/13"}}},"commit":{"parents":[{"commit":"f6f93ff9516f0ae24801d7b422c0e504bfebd0f5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6f93ff9516f0ae24801d7b422c0e504bfebd0f5"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-09 09:37:29.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24b96c608f5fcb6608efa9a2da9a3f3536e584f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24b96c608f5fcb6608efa9a2da9a3f3536e584f"}]},"branch":"refs/heads/master"},"26227bd88beff3e8f94d58992169e13aec59f4ef":{"kind":"REWORK","_number":14,"created":"2023-02-15 16:10:26.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/14"}}},"commit":{"parents":[{"commit":"6d4fcb7609381296c1046a8f93ea998f6214454b","subject":"Merge \"NetApp: Disable tunneling in vserver_exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6d4fcb7609381296c1046a8f93ea998f6214454b"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-08-30 13:33:40.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-15 16:02:44.000000000","tz":0},"subject":"Add defaultadsite to security service","message":"Add defaultadsite to security service\n\nAllows to configure optional field \u0027defaultadsite\u0027 which if provided\nset domain controller discovery mode to \u0027site\u0027\n\nAPIImpact\nCloses-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/26227bd88beff3e8f94d58992169e13aec59f4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/26227bd88beff3e8f94d58992169e13aec59f4ef"}]},"branch":"refs/heads/master"},"858939c19042bca563f9241325b92074ddf61e08":{"kind":"REWORK","_number":15,"created":"2023-02-16 09:31:36.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/20/855220/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/855220/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/855220/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/855220/15"}}},"commit":{"parents":[{"commit":"d94d1ae7bd3513d3d45c66bd48495c5d06e0ff90","subject":"Merge \"NetApp ONTAP: Implemented REST transition Client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d94d1ae7bd3513d3d45c66bd48495c5d06e0ff90"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-16 09:23:07.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-16 09:28:28.000000000","tz":0},"subject":"Add \u0027default_ad_site\u0027 field to security service object","message":"Add \u0027default_ad_site\u0027 field to security service object\n\nAllows to configure optional field \u0027default_ad_site\u0027 from version 2.76.\nRestrict to make sure either server or \u0027default_at_site\u0027 provided, but\nnot both.\n\nAPIImpact\nRelates-bug: #1988146\n\nChange-Id: I8e21e9170eace134a51efed84de1ccc58eb7eaaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/858939c19042bca563f9241325b92074ddf61e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/858939c19042bca563f9241325b92074ddf61e08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
