)]}'
{"id":"openstack%2Fmanila~855234","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~Id9faa4f73996cb7467b7361d0406c980f06905c2","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/1915237","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2022-09-16 17:23:05.000000000","reason":"\u003cGERRIT_ACCOUNT_33648\u003e replied on the change","reason_account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-12 20:05:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-10-25 16:07:58.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-10-25 16:07:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9faa4f73996cb7467b7361d0406c980f06905c2","subject":"NetApp OnTap: Fix compatibility check for share migrate","status":"MERGED","created":"2022-08-30 19:32:57.000000000","updated":"2022-10-25 16:09:13.000000000","submitted":"2022-10-25 16:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855234-bug/1915237","cherry_pick_of_change":847161,"cherry_pick_of_patch_set":1,"meta_rev_id":"821dcb366041491acb0e53b41c70dbeb3fe6778a","_number":855234,"virtual_id_number":855234,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-25 16:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-09-12 20:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-09-16 17:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-25 14:26:19.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-25 14:26:19.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-30 19:32:57.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33431,"name":"Fábio Oliveira","email":"fabioaurelio1269@gmail.com","username":"fabiooliveira1"},"state":"CC"},{"updated":"2022-08-30 19:53:08.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-30 19:58:44.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-08-30 21:50:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-12 19:05:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-09-12 19:05:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-12 19:05:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2022-10-25 14:26:19.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"77542de15b110db24a49f48ecbeefe595131a875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-30 19:32:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"9ed568ca1d885366f5cabe65c2815d83cea4a43f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-30 19:33:03.000000000","message":"Topic bug/1915237-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"1087605da52398f31849e9b6d6e5b9555e1b5ee6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-30 19:33:05.000000000","message":"Topic set to bug/1915237","accounts_in_message":[],"_revision_number":1},{"id":"1288553d6e8e79b846bd182e5f0a18fbf4bc2cc6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-08-30 19:53:08.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-34-855234-1/ns5_manila : FAILURE in 19m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"3d645c0bb54fdf80ad7550c03fef7db723942c6e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-08-30 19:58:44.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/34/855234/1/thirdparty-check/pure-devstack-manila-tempest-aio/89e11e0/ : FAILURE in 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c90f391a2f82cadd088a1fb9d95490d04c0c82c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 21:50:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98224b3d7fc5418da35aaea9bf400dff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea633ca2faf4baba97cd55a23ea23e6 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299874eddfc94616b5a833623db6c33e : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdfddcd8866f4755a8f457a8eeb96bcf : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b7624bbea684e8a8a867680c97a5a6c : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da6517b8dca44e9cb9971ea795ed9722 : SUCCESS in 11m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102fd1b478f24eb6b42cb80d8c56cad5 : SUCCESS in 14m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/532ef5d9126b4c51a5aca62c36b5e2fd : SUCCESS in 6m 28s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/819dc296f9a84b08983e1a12992ecf50 : SUCCESS in 1h 00m 38s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/090fdb0667f445cf96e81a80a619b484 : SUCCESS in 1h 44m 33s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/3eea78ed735e45589acd5f387e68ead9 : SUCCESS in 1h 00m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/0f06d63b102a465bb2025b01ac315860 : FAILURE in 52m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/f9cd93e0a9ff4506a76d02d2824ba63f : FAILURE in 1h 03m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/cb4f78638b794d9da90a4e06d80fc080 : FAILURE in 54m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ade2d66c2fa04b89bd409c3676039a9c : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2614a6e659594283b22afaf81ab1704c : SUCCESS in 5m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/703a7feb625544aab86c370a8a11f148 : SUCCESS in 39m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/17c09140436b4e74b6f3ab8a7d7cbbe7 : SUCCESS in 33m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6300fac336cb4c7c9dca33e886f18448 : SUCCESS in 46m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5e19947c80fd445192b95823a1a1a772 : SUCCESS in 27m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a54c5ba669a14c828b4c04804e50caf3 : FAILURE in 31m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/2dfd98a96f1746f9bb254e184fceb573 : SUCCESS in 1h 32m 56s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/74709723a6fe47078088deef640b25f6 : SUCCESS in 1h 11m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/86d2700ddd1b461386ba64411d5c21f8 : SUCCESS in 35m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c5a795642f2945d79701617c51767fa4 : SUCCESS in 37m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c399082af2ac8456b368dac6560dcc254b3409f8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-12 20:05:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"baba5eb5d587e925430610c01de76d15994d1265","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-09-16 17:23:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d9e5f4e690a14e33f287dae0967a99d8d547523","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-10-25 14:26:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0121498c81a9514de2db8d8214a35ca9495c6f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 14:26:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71977a283ced4e205d0c4ecba0b3057c6db67ebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 16:07:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b04ce287a24044840205f251ead5cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fda50c6835b845bab8648ba23ec1b3f4 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3d2f849c014bbba8fc057d051e7040 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/899cb5739b1a478daef26be1d06edc32 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bc81849da664ab8aaea719d83bfcb64 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc462b6d0ab4b3f8434142d529a8647 : SUCCESS in 5m 53s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/f970963e3ad645dbb5d3439df05fe186 : SUCCESS in 1h 17m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec7fda33b2e44646b6e4efd829bfa20b : SUCCESS in 29m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00a9ea087d494ffdb9f23c48a7f1af33 : SUCCESS in 31m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8d63beb88fe119a83dc9ec9d2b78191f335f4378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 16:07:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"821dcb366041491acb0e53b41c70dbeb3fe6778a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 16:09:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5804053d2558440f9cee36d7966aef81\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6141121a20c2442a8ef022b13216d9e4 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01d2ea94da8442dfab76b10e9e5dd936 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee7ede8a363044c20c0145caef438aa9ca8743f9","revisions":{"ee7ede8a363044c20c0145caef438aa9ca8743f9":{"kind":"REWORK","_number":1,"created":"2022-08-30 19:32:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/34/855234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/34/855234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/34/855234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/34/855234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/34/855234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/34/855234/1"}}},"commit":{"parents":[{"commit":"bcf07d40f66ce8ea41170c843addf37e94397835","subject":"Merge \"Add FIPS testing job\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bcf07d40f66ce8ea41170c843addf37e94397835"}]}],"author":{"name":"Fabio Oliveira","email":"fabioaurelio1269@gmail.com","date":"2021-12-07 14:13:29.000000000","tz":-180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-30 19:32:57.000000000","tz":0},"subject":"NetApp OnTap: Fix compatibility check for share migrate","message":"NetApp OnTap: Fix compatibility check for share migrate\n\nThe issue happens when the parameter \u0027encrypt-destination\u0027 is sent to\nOnTap devices with version lower than 9.0 or OnTapi lower then 1.110.\n\nThis patch adds a condition to check these versions and ignore the flag\naddition if needed on \u0027_send_volume_move_request\u0027.\n\nCloses-Bug: 1915237\nChange-Id: Id9faa4f73996cb7467b7361d0406c980f06905c2\n(cherry picked from commit 9417ece45629d4103d2b37a593dfbcc7e774362c)\n(cherry picked from commit 27d79965f79fea4b34809fcd449f007ed6cc5ec5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ee7ede8a363044c20c0145caef438aa9ca8743f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ee7ede8a363044c20c0145caef438aa9ca8743f9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[]}
