)]}'
{"id":"openstack%2Fmanila~856144","triplet_id":"openstack%2Fmanila~master~Icfe4b225e216dfda46366af1eb3820b193befa0f","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-19 01:40:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfe4b225e216dfda46366af1eb3820b193befa0f","subject":"db: Migrate \"service\" APIs to enginefacade","status":"MERGED","created":"2022-09-06 14:59:10.000000000","updated":"2023-01-19 02:07:08.000000000","submitted":"2023-01-19 01:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"856144-sqlalchemy-20","meta_rev_id":"84bd21c5204f158775b27fefb56a4360bd3c6706","_number":856144,"virtual_id_number":856144,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-19 01:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-01-13 20:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-01-06 00:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-01-13 20:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 16:08:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 17:02:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-06 17:04:14.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-07 03:12:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-07 13:23:02.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2022-12-21 16:06:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-06 00:22:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-01-13 20:30:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"1c5a925baed7dd20d9c2113d2acca40d8e9eef1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 14:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ad1a573e5a300e02272ffcd71aa47beaf55c7a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:08:17.000000000","message":"Patch Set 1: Verified-1\n\n(237 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41839906a186475aa06f8d3462e9051f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfeed9f5a744631be5cf0d47d9cb6a6 : SUCCESS in 13m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86cf0de1afe643a6ba3ef58dc6cb3359 : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194a2634588243fba15e0f0e321f5e87 : FAILURE in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b56164cda1e49a587edd6b10c35d138 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b8428eb6e73496e83b5435247c6d62b : SUCCESS in 11m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f960af63389848dca5e312b9baf33458 : SUCCESS in 10m 11s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/88c1653c072641fcacd884a132b975ff : FAILURE in 32m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b9fc115f7e38428aa42d5ef9f704978b : FAILURE in 39m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bd78a2c73b2647caa1577b7d5aa8b801 : FAILURE in 45m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54e0fc770eb0486d8039a3ef0ffc282d : FAILURE in 29m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0f8e3e4ae45c410a90e3b6afccb0b08c : FAILURE in 34m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a27348483b0441faaa791cafa7e6891 : FAILURE in 26m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/925538079cab42feafce28f2db45e621 : SUCCESS in 6m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ead4a138eb42441c9381a8c25a82dc55 : SUCCESS in 6m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6eb6a4a046d24a8ba1f14d953ff83eff : FAILURE in 19m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6d4f02b2fa65440ea8f15635c55dcd97 : FAILURE in 18m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d49a38b28cf40da9d326b56a0b7d418 : FAILURE in 29m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/adcb69e71793448aa416546bd4cd0b83 : TIMED_OUT in 1h 06m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c95b66c97024cc08d3a493f7170ced8 : TIMED_OUT in 1h 05m 10s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/c0a6ca5c316a489b9bd32fd868837b2e : FAILURE in 44m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6fe8ea95251742da81366b0b71638d39 : SUCCESS in 51m 44s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5abfe6e7ceb34082a2346e00e006efd3 : FAILURE in 35m 54s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0556dc0085b74c6a85efac997f4be4d1 : FAILURE in 40m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e91c870a2a724cfa9bdcca9a71139fa3 : FAILURE in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e11dbd9a0f0ba470f85a813bee036a095b4f31dc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-06 17:02:31.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-44-856144-1/ns5_manila : FAILURE in 32m 32s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"d6d9be59bd62b130c901e5dce9bc0ae22594b643","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-06 17:04:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/44/856144/1/thirdparty-check/pure-devstack-manila-tempest-aio/213ddeb/ : FAILURE in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a054ca4ad2d2308ae642c8c8d45c61176cd8fcae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-07 03:12:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/856144/1/check/DellEMC_Unity_Manila_dhss_true/ea5e05f/DellEMC_Unity_Manila_dhss_true/796 : FAILURE in 2h 13m 17s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/856144/1/check/DellEMC_Unity_Manila_dhss_false/e792586/DellEMC_Unity_Manila_dhss_false/837 : SUCCESS in 56m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0401517c20f3b60860d9d8d22a1885d6b0c5f184","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-09-07 13:23:02.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/44/856144/1/third-check/macrosan-manila-dhss-false/b69495e/ : FAILURE in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"17cd3f6028177a4bc88970d5789eb4c2e977e07d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 23:08:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"aa8143acb3063033e32c9256f260cc9fb64cb024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 00:19:22.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38b9f191cb2841d38004c6cda8b6544a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8652f1160aef4a2da32dbed038d801cf : SUCCESS in 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d983c1cd492d4b49b0fda4a7663b4ac1 : SUCCESS in 20m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34d5cb13f504841a77db4daa0f69700 : FAILURE in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce539f46ca4c428a8e08280e84f5a3c9 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3995a3929d614658b406bdb1d215fded : SUCCESS in 10m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e42fe7bc64443f5a2470a4b7a2fe96b : SUCCESS in 10m 31s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/15faf23c12e34bbb8a99c4c3b4fd51b9 : FAILURE in 40m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f228687d8fb9426986dc272f2554cc1c : FAILURE in 47m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d4f6604664aa4b969a2c65c7f65d2da9 : FAILURE in 42m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b1b1b731d0f143479153c37563638139 : FAILURE in 38m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9af6e4bb92324940b22fd521f939b56a : FAILURE in 37m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ac278aa6599a4fb6ad01ada57f028bfc : FAILURE in 31m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8b772c043a644b44ab5e498f4beaa151 : SUCCESS in 4m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ea3ebc85f2594a6aaf54dc37d8b892c5 : SUCCESS in 8m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cee0c03c8ba343bf91e6d20dea2763fa : FAILURE in 17m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f917c79c10764e3b86ddc03a2d2edd6a : FAILURE in 23m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa31b889d2dd462c9ab0ced95a29c362 : FAILURE in 32m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e366ec367482490482506b212cb0ad24 : TIMED_OUT in 1h 03m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/686f97a0f0524699a3782778e4991449 : FAILURE in 22m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3968197963a44f3896ce2aca7cf0a2d2 : FAILURE in 41m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/089a312e378a44a28d94082df2ac1f4e : FAILURE in 52m 38s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1e3ebcb9a4ad451e8f08dc0b950a1809 : FAILURE in 28m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cca2a1230c254bcfb3bf1f2bc5bff6d0 : FAILURE in 29m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c442459322274d0f83daf6ea72f3e005 : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"943e2f9824136c0b0cda1e6209cd4ad2d213429c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-08 00:21:53.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-44-856144-2/ns5_manila : FAILURE in 27m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"200c0cce6b92ef754f5b3445c46f7e092e8b8a10","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-09-08 01:55:44.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/44/856144/2/third-check/macrosan-manila-dhss-false/0cb416f/ : FAILURE in 52m 52s","accounts_in_message":[],"_revision_number":2},{"id":"c25fe6ae6cea9d2e58cc0b2220360903b3ecf9c0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-08 02:12:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/44/856144/2/thirdparty-check/pure-devstack-manila-tempest-aio/5d6a587/ : FAILURE in 2h 58m 17s","accounts_in_message":[],"_revision_number":2},{"id":"4221c44e34dcfd75b19e145902359bf340b63d3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-08 04:58:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/856144/2/check/DellEMC_Unity_Manila_dhss_true/9994a5c/DellEMC_Unity_Manila_dhss_true/807 : FAILURE in 59m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/856144/2/check/DellEMC_Unity_Manila_dhss_false/85ee75f/DellEMC_Unity_Manila_dhss_false/848 : FAILURE in 40m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9380c21a4295c1aa1a96a7b1bd19bcb3e615fec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 10:50:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b9f67f7b4ea5312f3955aabbbc0b3af79870d98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:58:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fef31e47333447a6a223ad6e56f3eb08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc40777320f34f208e31da033da92f14 : SUCCESS in 10m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e48d4f7bd494effb3428db72a613ec3 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5e973fb9c2742659340d50774373fdd : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fcd413ae2fb4b6f96237932b01f7143 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d64bcfd7245423396135ae06ef32833 : SUCCESS in 14m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0868bcb8ad6f4da5a91d9758be626b6c : SUCCESS in 14m 49s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dcaa00b0d5a546ffa02126088a5d6866 : FAILURE in 31m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d3119723fe564f3ebc2ae6e78d5a18fb : FAILURE in 34m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8f5a8316ee1944a1a86379ea12284072 : FAILURE in 43m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b214c00363aa42c4b74f6cd017067a64 : FAILURE in 38m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ce9f6b56b3fb427a874a44f6033d8234 : FAILURE in 30m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d52da656e74e4f98adce3f2e2faac87a : FAILURE in 27m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/65f5cf3556184066853286de658c656d : SUCCESS in 7m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/11837b483d8a49e4abd6794bc619705e : SUCCESS in 5m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4608bcd777934a46b25c3f3d3af51172 : FAILURE in 15m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fc1dba1b26d4ff9a5554382631e436f : FAILURE in 25m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e1a27af9e7594dfa934272c13170603a : FAILURE in 34m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a8c46e23ef8b4fac88584c442da9c80c : TIMED_OUT in 1h 04m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aca271a6bac54135832fb370217a44d0 : FAILURE in 25m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/13356d45cb9c408795e292e421317c38 : FAILURE in 37m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5aaa320f00014581baf48fa3e71e1120 : SUCCESS in 55m 37s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2c80e20c0eb4d038c7986781c55fd3a : FAILURE in 35m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ca00b4690f8c41ea950aeb6adcf5399c : FAILURE in 29m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e6917850c6b04279a0a28e552e72d9e8 : FAILURE in 48m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7acb8799c4ff445e2ba30e8211eccca5c6197aa6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-08 12:12:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/44/856144/3/thirdparty-check/pure-devstack-manila-tempest-aio/0d09d63/ : FAILURE in 1h 15m 45s","accounts_in_message":[],"_revision_number":3},{"id":"78da750ca3946f1da4e705c4f154d3fa53981536","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-08 12:27:25.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-44-856144-3/ns5_manila : FAILURE in 30m 55s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"a63d0b479a2601df65c799878bba82fb68c849a0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 14:11:26.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a26518ac6eb8e092c50ec78ad2139a078f419aa0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-09 00:04:47.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/856144/3/check/DellEMC_Unity_Manila_dhss_true/3f49275/DellEMC_Unity_Manila_dhss_true/817 : FAILURE in 2h 18m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/856144/3/check/DellEMC_Unity_Manila_dhss_false/9e6697f/DellEMC_Unity_Manila_dhss_false/858 : SUCCESS in 58m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ba226ef8b500ca5baf8ff5a219be3d9e9374b3fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 17:47:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9da225bd17bd597b57e66f28357943c55a043cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:10:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89fbbe3118bf4ff1a1bf072facc7e175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97086ed4736c49c087353575f1116f30 : SUCCESS in 11m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06dfcc635174db8921d5ebaba104954 : SUCCESS in 20m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b442f30cf943d593eda94eb85e26ee : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e6a255138a847cbbc8ba72d62a76798 : SUCCESS in 10m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc6596508ce34dad8bdff5747d9af80f : SUCCESS in 10m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d16cbefe030440a8fd32a41b7e38432 : FAILURE in 36m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9b949603efa41fc8f51341e0b1265bc : FAILURE in 36m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4998f2f5cbd14bcbb1e3c366b4a7693b : FAILURE in 39m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a801d3652ac4475985be9fb9838995fc : FAILURE in 37m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e3212635f95c4d80bd331101fe87788c : FAILURE in 35m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d6d6001666e24655832c6146d5a069a0 : FAILURE in 32m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/72e6eece79d5425d9f4b251dd9eff33c : SUCCESS in 5m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4eba383c7ff3446d96dd4291f181de75 : SUCCESS in 5m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c6ca70a5833f492bb61ac9d92c257b62 : FAILURE in 20m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4403c920dc9f4b7cb74df9e1bb52381d : FAILURE in 22m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d42e70253f2245409a46b31956c138b0 : FAILURE in 33m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5872fe573e80460b904519c6f8524082 : TIMED_OUT in 1h 02m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c315e4547fbb4b35a2561d3305c87f5f : FAILURE in 28m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7ccec2d4d49b4b998a25f45c17606a09 : FAILURE in 42m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/db7e5aa75d274e8a87a8f696a71334f8 : SUCCESS in 1h 04m 34s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8ed72efa4af54e789e649878ae9b9cae : SUCCESS in 1h 04m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0cccc2b1ab4a4f529934d9b3019d45b4 : FAILURE in 38m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/23cd6f473bf54edf9d2bf4076242915a : FAILURE in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09869ec1b24dd1cdd8d9448a62630c8e8c7a8670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-21 12:23:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a89c26f24d3ff2729c27a9b8abd972514d6a5cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 14:33:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd9eca951b440df9015df6aed1e1f4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6293ec5539d41998222c579b0169e63 : SUCCESS in 11m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb47a0fc5a1c45e2bca20c65c31c0596 : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d3ef4feec4e426384eefd9fb7f5db30 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39b064eb74c54ede9b963dd7329b44ea : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ec9339f477944e2811f5b6221801c9e : SUCCESS in 9m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1bfac3de83114689bde3b1a9599f5aaf : SUCCESS in 1h 53m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e68c4e81b3314c9aa696f2791c20cc6b : SUCCESS in 2h 04m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/315cf45c47e84da98951416a45e2427c : SUCCESS in 1h 03m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f9cae5bcf4cb4e86bdd10b5eb89b45fc : SUCCESS in 1h 18m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ecb0ed88a66644659bd465140db40de9 : FAILURE in 1h 58m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9e5c9996b2854fcbad7049399159f043 : FAILURE in 1h 12m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6afead296e2f45c2861d42c77d5ed62b : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c2b51f67255c48cc8d87b5c7f282561d : SUCCESS in 6m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/53d780f663814f7089c6b5aa60f12261 : SUCCESS in 34m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b1a5496e5e134f889c4d727c98c67552 : SUCCESS in 33m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1373189e6f5c45b1ba880a3ee00bde77 : SUCCESS in 47m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e12a548cbc924b7e963068f01d23b8ab : SUCCESS in 34m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5b8e1a77aa5f4d73a98e9331824fdf95 : SUCCESS in 29m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/42bd596496394828912bfef52699fc2f : FAILURE in 1h 48m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ea09ade9b51f412c9c8b175dd804a463 : SUCCESS in 58m 43s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c677d11c6f24774b1229856675b1d7a : SUCCESS in 1h 01m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7e508acd3c794dfda1ccb96f5a1942b8 : SUCCESS in 37m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/56bb6ef8536340e68fe6292c2919ea94 : SUCCESS in 41m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"245f8fa3586640b1eb872bf9d5a1ce9d5f0a3ad5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-21 14:48:58.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/44/856144/4/thirdparty-check/pure-devstack-manila-tempest-aio/53e38e1/ : SUCCESS in 1h 06m 50s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/44/856144/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5c5cdf5/ : SUCCESS in 1h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ecc343393737a1a89dec9a136942d1f38f6d44d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-21 14:55:19.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-44-856144-4/ns5_manila : SUCCESS in 1h 54s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"a9035c90568a8a3c3f732cc74acea8da88a4a16e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-21 15:08:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3dddd3326bbfc42471c07485e725e7851ec88c82","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-21 15:11:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"17a34a31c97fdcb902c7707b32d2a755615f1da9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-21 16:06:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/856144/4/upstream-check/manila-tempest-plugin-ontap-dhss/7faa799/ : FAILURE in 1h 31m 45s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/856144/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/4effcc6/ : RETRY_LIMIT in 10m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"9cedd81c00375a27cae8095e3ba8f2698bad5816","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-06 00:22:14.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"604742b6ba51f990499b060b4e414ba9a224a1bc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-13 20:30:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a22b59ce6eb1940bdd67d145da043da19208d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 20:31:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e3df185decdb00e900d67124c87182b1dd0eed99","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-13 21:56:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a9ce00e4cce637bab348cdbf56b15aad4d2f6ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 22:19:01.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e1632cf097d4fb58901b7dc71e2a45a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db005dd520f413e82c26551098ee613 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02ba57519e84e64be9031b45c649d93 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64dfafb6ad6d4f4abc6ecac58cbd74ae : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49c7ff4a1e77434094b9759c82a70bef : SUCCESS in 10m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/88c73aa4841d48a79517a15a73eafd99 : SUCCESS in 1h 45m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0bbd27e9d9a0482ca8e1b90a4b13cb72 : SUCCESS in 27m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c098937f193842ff9873ca7ece1b4882 : FAILURE in 38m 10s","accounts_in_message":[],"_revision_number":4},{"id":"17f263f4d2b434ce403ca609dfc0f390e6159e6d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-17 01:20:12.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9cfda48b137e23b8a64004c9c29e5e02cd448252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 03:37:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/417741bd8aaf43e99f7cacb0042eb4d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8993526e99b4746ab9cf572f2ff3b1b : SUCCESS in 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b74e9d1fe72e40999d79cb018a594f1e : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb02519d73e84f719030b7f417a2ee74 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45a337d6d4c49d1a4c618b67f62aac7 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2560ed05fd842b59b696c2e4279c166 : FAILURE in 20m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c4db0f6f838246768ff94baa4b0e04a4 : FAILURE in 56m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e287d73929fa407fa64898a5f8685aab : SUCCESS in 2h 12m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/112d93b9d150476a857d0d4187e724f7 : SUCCESS in 56m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/526bf038b3ae42cc95f712c26de78c6f : SUCCESS in 1h 24m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/805641da925c443f97eb39cdf27393f5 : FAILURE in 21m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7449af7201e54af0b8fd089f772d9772 : FAILURE in 1h 19m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/78d0c77732394d7d8c16fca3b35da973 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e454426cfbfc456f89762cd1445d4b5f : SUCCESS in 6m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43fc4299709446dbbe248a7615a71f57 : SUCCESS in 26m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a406add3b30c401f9557a4a9de975e31 : SUCCESS in 36m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9d75f7def2b046368711e35331563f98 : SUCCESS in 44m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/bd60f6a8b3354c57a49e5bd6c5cdccdd : TIMED_OUT in 1h 02m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2fc0519924e84ac787dd0479ec71fd10 : SUCCESS in 18m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9e6e6db13e174f1991a1569ba25bb553 : SUCCESS in 1h 51m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d253a178811b46fa8b9b866239ae29b0 : SUCCESS in 1h 05m 36s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9eff563e55cc4999bcc5728ddd231c99 : SUCCESS in 57m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/887ef5aa078f47efa9a26deb28fe6956 : SUCCESS in 27m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/61c0fc7b03ca4c0182b760dd540b5606 : SUCCESS in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"612753cd27f1d836141f8a7ff7129bbf166207cd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-18 22:38:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9580cace6345a168f3e211fdf4e7a2832db60fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:39:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2795a460e46c42b59cbdcff4cdf19cfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27849cd4d2cc4423bbb2dc397fd5e483 : SUCCESS in 10m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecac73cae28c443aa6c4457a791861d8 : SUCCESS in 25m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf15b87808040e88c6b9c34d44cce25 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27bda0e569904c35894daaea49f73214 : SUCCESS in 14m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2824ea090f2a421d846dda1fe1d9bb0b : SUCCESS in 14m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2bf63ceb6f6840bd9b71b84102fccbda : SUCCESS in 1h 02m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b3ebfff3bb2b48358a6105b831e9bb0f : SUCCESS in 1h 58m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fb8a27d3b2994c728c1ef0ad802279ba : SUCCESS in 59m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a23b225ba48a41c295b792c999f36b0a : SUCCESS in 1h 12m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/de5406de881249529456c6ba28ddc2a8 : FAILURE in 1h 34m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6fcc56fe655a4b8a8f17ddd5d95ffbb0 : FAILURE in 55m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0bee4c0505f642f198bc4038c73f2fef : SUCCESS in 5m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/19168ff94fd24c2dbea954f4d4a23d12 : SUCCESS in 5m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e8b4ebeb12a04d4ca164f716ebf6c4d6 : SUCCESS in 29m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f79ba6e2a54947f989df58a72f7b23bb : SUCCESS in 41m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29132d8c482b485c85b90a9b23585530 : SUCCESS in 47m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a82d47cf44f24669b2aaa2b8a433b58c : SUCCESS in 31m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/85d9b4a382294a979093e588445b6cfc : SUCCESS in 21m 59s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3423174ef3194812b4a2892372cb1dd1 : SUCCESS in 1h 16m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b7346307b01041b1bb07f0ce9e822d18 : SUCCESS in 55m 08s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cd15b66bbcb94f4b8b3bd575a2edd1a3 : SUCCESS in 1h 01m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/10d893e408c04a40aa187de956ef5be3 : SUCCESS in 32m 44s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2d105730858d4d50aa1cd49996fb5c7c : SUCCESS in 48m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9de98bbb21aa3a229e88ad0c2f3b1c9c5c3ee0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 00:39:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9c7a985923568d906cf42d62977ded68caa6c570","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 01:40:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a17f8b11dd42f1aa32f2d9a496dd8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8bad5e6b6749b69da9f09e6c451937 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5df3878e634ce79432c7f00e295bf2 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b62a19e9f2cf48f5a73412b6fbb51875 : SUCCESS in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fd3e9f48fcc47cd97be7a934edf506e : SUCCESS in 10m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82b8589471a1401fb7e1b3a6d0f14518 : SUCCESS in 58m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1da8ae1d34874e7f86d44cb89f6f13a4 : SUCCESS in 28m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5d22eff4943f46b3b9b58bae6c9d8550 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":4},{"id":"369b6a85e3fed45f5c822eaaca454e08e4b5aee0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 01:40:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"367480374dfe824747736df464ce086637fdebf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 01:41:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/751d255c21c7401d8532021644bf476c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9187d5aa4bb64e94b8a453b1a99635c5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4},{"id":"84bd21c5204f158775b27fefb56a4360bd3c6706","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-19 02:07:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/856144/4/upstream-check/manila-tempest-plugin-ontap-dhss/5eba43a/ : RETRY_LIMIT in 7m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/44/856144/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/baa67eb/ : FAILURE in 1h 26m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"83a7fdf010ff11de3791bfaefe581e81bc2cce03","revisions":{"6b4917017469a9f09b1747a0589eaf190d58ef3d":{"kind":"REWORK","_number":1,"created":"2022-09-06 14:59:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/856144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/856144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/856144/1"}}},"commit":{"parents":[{"commit":"927f677821f7c556e666a4724dc53a8a896735dd","subject":"db: Prepare \u0027model_query\u0027 for migration to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/927f677821f7c556e666a4724dc53a8a896735dd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 10:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:32:41.000000000","tz":60},"subject":"db: Migrate \"service\" APIs to enginefacade","message":"db: Migrate \"service\" APIs to enginefacade\n\nMigrate service-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icfe4b225e216dfda46366af1eb3820b193befa0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b4917017469a9f09b1747a0589eaf190d58ef3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6b4917017469a9f09b1747a0589eaf190d58ef3d"}]},"branch":"refs/heads/master"},"224ad09c90c8523c96d7b5bef268df2f5da36a55":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-07 23:08:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/856144/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/856144/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/856144/2"}}},"commit":{"parents":[{"commit":"dc15ef83ddcce21d04530c1505d846cf0127ced0","subject":"db: Prepare \u0027model_query\u0027 for migration to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dc15ef83ddcce21d04530c1505d846cf0127ced0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 10:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-07 22:01:50.000000000","tz":60},"subject":"db: Migrate \"service\" APIs to enginefacade","message":"db: Migrate \"service\" APIs to enginefacade\n\nMigrate service-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icfe4b225e216dfda46366af1eb3820b193befa0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/224ad09c90c8523c96d7b5bef268df2f5da36a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/224ad09c90c8523c96d7b5bef268df2f5da36a55"}]},"branch":"refs/heads/master"},"727b84839617013589661d53243e4ab4eb003a72":{"kind":"REWORK","_number":3,"created":"2022-09-08 10:50:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/856144/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/856144/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/856144/3"}}},"commit":{"parents":[{"commit":"77941d17ec7874f2b1e14d7a97c0e2044df3ec1f","subject":"db: Prepare \u0027model_query\u0027 for migration to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/77941d17ec7874f2b1e14d7a97c0e2044df3ec1f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 10:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-08 10:20:14.000000000","tz":60},"subject":"db: Migrate \"service\" APIs to enginefacade","message":"db: Migrate \"service\" APIs to enginefacade\n\nMigrate service-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icfe4b225e216dfda46366af1eb3820b193befa0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/727b84839617013589661d53243e4ab4eb003a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/727b84839617013589661d53243e4ab4eb003a72"}]},"branch":"refs/heads/master"},"83a7fdf010ff11de3791bfaefe581e81bc2cce03":{"kind":"REWORK","_number":4,"created":"2022-12-21 12:23:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/856144/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/856144/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/856144/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/856144/4"}}},"commit":{"parents":[{"commit":"e1057ae66fb6abc3a3d411e1fd5028aa94bff046","subject":"db: Prepare \u0027model_query\u0027 for migration to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e1057ae66fb6abc3a3d411e1fd5028aa94bff046"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 10:56:28.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-12-21 12:10:32.000000000","tz":0},"subject":"db: Migrate \"service\" APIs to enginefacade","message":"db: Migrate \"service\" APIs to enginefacade\n\nMigrate service-related APIs from the legacy enginefacade to the modern\ncontext-based enginefacade. This one is a little tricky as it highlights\ntwo issues.\n\n* The models in \u0027manila.db.sqlalchemy.models\u0027 do not match those created\n  by the migrations. This causes us issues since we apply the former\n  rather than running migrations when creating our database via the\n  DatabaseFixture, which means we\u0027re not actually testing against what\n  our customers see. One gap is closed here in order to get newly added\n  test working as expected but there\u0027s a larger issue that needs to be\n  addressed at a later date.\n\n* We are using a pattern of \"try to create and retrieve on failure\"\n  rather than \"try to get and create on failure\" when determining\n  whether to create availability zone records. This works where each\n  operation results in a new transaction but does not when the\n  transaction is shared. We switch this.\n\nWith these two issues addressed, the rest of the migration is relatively\nstraightforward.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Icfe4b225e216dfda46366af1eb3820b193befa0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83a7fdf010ff11de3791bfaefe581e81bc2cce03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/83a7fdf010ff11de3791bfaefe581e81bc2cce03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
