)]}'
{"id":"openstack%2Fmanila~856147","triplet_id":"openstack%2Fmanila~master~Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-30 21:31:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d","subject":"db: Migrate \"quota usage\" APIs to enginefacade","status":"MERGED","created":"2022-09-06 14:59:10.000000000","updated":"2023-01-30 21:33:52.000000000","submitted":"2023-01-30 21:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856147-sqlalchemy-20","meta_rev_id":"3a927fc50bd8aa09b5a5f8ac9ede29f17b64582d","_number":856147,"virtual_id_number":856147,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-30 21:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-01-30 11:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-01-30 20:05:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-01-30 20:05:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 16:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 18:07:40.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-06 18:27:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-07 10:59:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-07 15:51:01.000000000","updated_by":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"reviewer":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"state":"CC"},{"updated":"2022-12-21 19:14:26.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-09 18:06:41.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-01-23 19:06:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"94cf7014b51d52b621cebe3c3f0a84a26f0d8a52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 14:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa9060c6ba4dfcdb5ca7a02e392d42aaecd14496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:23:05.000000000","message":"Patch Set 1: Verified-1\n\n(232 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67b9cdcfcf1d4ea7b1f10c5be1faead1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c9444504814ed3ad31181fed78f085 : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/297d22b29309491b9c380cd0d0c083c2 : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba22b33fffe46a4bfff56eac923f02b : FAILURE in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36fba8105214464eb2c8d5a6d9458e3c : SUCCESS in 9m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b26fde3b297c4564995c9d5ed24c60d0 : SUCCESS in 10m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/897c47d06a2c4cb499690dc3271c2ecd : SUCCESS in 9m 22s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a1d69af2c4ff45ccb9000b18c09c3a20 : FAILURE in 35m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3b1a801fcd1b4e8c851bbf7e9fc26dae : FAILURE in 49m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b39d1dd0b52f4321a9c1956f813b4a31 : FAILURE in 59m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/12b2e34872be4c75a9cc98f1ef0822c2 : FAILURE in 38m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7c05407e1d3a4a49b8e535e5523eec25 : FAILURE in 36m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/911eb495caa14a4a958ec7e347bc1772 : FAILURE in 37m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/791fb3f587dd4a0eaba213bc5dcdde60 : SUCCESS in 6m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2251a3f3bac141719f5644d632ff9670 : SUCCESS in 5m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fb84da389eb945fe9c27c58651bc0e5e : FAILURE in 23m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/767c5b4bebd7444c90ca8d00236ed633 : FAILURE in 25m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cd1a947ce36d454c979432709e4b8c91 : FAILURE in 36m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/41bd1e466c9945d38c847d6e65824796 : TIMED_OUT in 1h 04m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3fdbc40f46e648ffa57c963ccb61d6f8 : FAILURE in 30m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b69c977a4c1b48afa46310beaedb67a2 : FAILURE in 50m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/30bf5ad55fea4de786e495ea33c3907f : SUCCESS in 1h 14m 25s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c89f0f74bdd141fda81d991de1f36d6f : FAILURE in 43m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b3a74a9afc994632b9507f0d21fb0bd9 : FAILURE in 38m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b6706b25de8f4c288d9d51329271dd8a : FAILURE in 38m 42s (non-voting)\n\nWarning:\n  Comments left for invalid file manila/tests/db/sqlalchemy/test_api.py","accounts_in_message":[],"_revision_number":1},{"id":"b32e4dffc11a8877656b159b8c98f1766a040906","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-06 18:07:40.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-47-856147-1/ns5_manila : FAILURE in 30m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"b88d202245816c9986d2649b9c1e342203f31735","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-06 18:27:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/47/856147/1/thirdparty-check/pure-devstack-manila-tempest-aio/6818d36/ : FAILURE in 1h 10m 27s","accounts_in_message":[],"_revision_number":1},{"id":"7b02aa0397579ddd1c5535de1e1e6978b8a8b7d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-07 10:59:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/856147/1/check/DellEMC_Unity_Manila_dhss_true/829502c/DellEMC_Unity_Manila_dhss_true/799 : FAILURE in 2h 32m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/856147/1/check/DellEMC_Unity_Manila_dhss_false/bd566ea/DellEMC_Unity_Manila_dhss_false/840 : FAILURE in 14m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"95d564e5b13e91d690cd26244a24abe1049d7ad1","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-09-07 15:51:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- macrosan-manila-dhss-false http://47.111.113.211:8081/logs/47/856147/1/third-check/macrosan-manila-dhss-false/4f0d723/ : FAILURE in 48m 28s","accounts_in_message":[],"_revision_number":1},{"id":"963be3b9b3193a01b679c082a39ccd81fefc905b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 23:08:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"25e3ada8808887227faf05db27daaadf8f2b57d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 00:17:45.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58bc703447ba42aeb93c70c958b79eec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63f013c07f84b04b025f65d4535d9c3 : SUCCESS in 11m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db540b67c51b4fea8ac3c67adb57f416 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90913c543f394fd6bbc151229185de4e : FAILURE in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cf280828eba4eedae3039905118bd70 : SUCCESS in 11m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89691643ccc84b83bf84a0a4ee6085cf : SUCCESS in 11m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9473a5fbf26547a9a3c51147954e20f5 : SUCCESS in 9m 27s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b0fc584bd214a9a8be93a1d479a4ee8 : FAILURE in 37m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1137b49691c940b2a79fe397980fadc9 : FAILURE in 43m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/73fb88fc62334400a30f24648cd3b722 : FAILURE in 43m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3741ceaa6f774238890ac7037b14c4b6 : FAILURE in 33m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/93a4becdf66c4eb1953a98d960a68793 : FAILURE in 36m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/185c31da8b894660b223c58bc3e87287 : FAILURE in 30m 45s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/909ab5c6904f4673b7db3f87f3b089ff : SUCCESS in 6m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71e2cf7e39e04868815d3830fb4d50eb : SUCCESS in 4m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3a4a32ae714c43b0a1922f6d482569ea : FAILURE in 25m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/74aa8647373241e2bb0f2cb61ec76f9a : FAILURE in 22m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/68e640e95aef4e7b8da5b06cf58709e8 : FAILURE in 36m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6e0478a81bce4b9fb7b97020efeac1ba : TIMED_OUT in 1h 02m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fe59ca841ceb4b14b0b5dd3a1a2dfe76 : FAILURE in 39m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/eb40cb9ec5b04bbdb030ed2ca6e8180c : FAILURE in 49m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d11604be3e1a40108beae685f0d4ef85 : FAILURE in 54m 26s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/524eb126d24f4aec918de69f010311dc : FAILURE in 34m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/383091f25bc4442b9fcb4cf44d238f26 : FAILURE in 44m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b5fca55bc75e4024b3eb3a49e6237e92 : FAILURE in 34m 15s (non-voting)\n\nWarning:\n  Comments left for invalid file manila/tests/db/sqlalchemy/test_api.py","accounts_in_message":[],"_revision_number":2},{"id":"fad88f79a9e88f229d89ccf982f7efa2596e7f62","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-08 01:32:53.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-47-856147-2/ns5_manila : FAILURE in 31m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"75595e076592d7b856ce9132799fef689f13c1fc","author":{"_account_id":34894,"name":"Macrosan Manila CI","display_name":"Macrosan Manila CI","email":"ganyuehui@macrosan.com","username":"macronas"},"date":"2022-09-08 03:48:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- macrosan-manila-dhss-false macrosan-manila-dhss-false : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a71485f1bfb2c84db38be2ce995ca385b2bacfe0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-08 04:54:20.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/47/856147/2/thirdparty-check/pure-devstack-manila-tempest-aio/e5bc182/ : FAILURE in 2h 41m 13s","accounts_in_message":[],"_revision_number":2},{"id":"626c9b3a42c1433719e9c0d33aef29ef4b417c1d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-08 08:01:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/856147/2/check/DellEMC_Unity_Manila_dhss_true/650d67a/DellEMC_Unity_Manila_dhss_true/810 : FAILURE in 1h 01m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/856147/2/check/DellEMC_Unity_Manila_dhss_false/e9ec8b5/DellEMC_Unity_Manila_dhss_false/851 : FAILURE in 41m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"46275eaac8b1c8b7b0c018b891ccf3fb034f4328","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-08 10:50:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"26b660f9607fc597d2ea79d7462f2bf92fc3115d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 11:58:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24692929eaf1460a8edcb6803647f75a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ba00adb65c47e8b414c59095698ca5 : SUCCESS in 12m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8f2d2d0f0e043a9a52a3247f2419450 : SUCCESS in 19m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f1c355a6b144a687af38c5dbaedbf0 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58d5d40a78c849cbba337b92361a4416 : SUCCESS in 11m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/508d07eea9d54df5aefd875a7da715e6 : SUCCESS in 10m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0e6eff0c08f4f24b6b4e69232039ae4 : SUCCESS in 11m 33s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/95294e50fe3f4f80ad60ea0ba5ac48fe : FAILURE in 36m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4cf73b87b14445b786ecb21147987843 : FAILURE in 53m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2a23eabfa37246d2bd23d1c5a5078601 : FAILURE in 45m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ec8c84321f96436a9f5232b25c9c97a3 : FAILURE in 37m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a01fcce85841408c9d0ed58cfa42c576 : FAILURE in 41m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/993d2407576b4f9eb808704e654a652e : FAILURE in 43m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/962ea22886c74cc6bda48ab1095e740e : SUCCESS in 5m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fe17e29d11b14829aa3b7d343557b053 : SUCCESS in 5m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/429900318d6c415d90bdd3909a49353c : FAILURE in 19m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9a9f1352339b447781d1e42ac9f4272c : FAILURE in 18m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dcbca428871444faa5f4ff7e3197e79d : FAILURE in 28m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/836042f1335441e08851e364352809d6 : TIMED_OUT in 1h 04m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3dbcffc30bfa4182a86a0d1f78f8af46 : FAILURE in 21m 42s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3e3c266354eb4a20936cc210c6386f1b : FAILURE in 44m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/77674093f2704d20a6b4a836f9497fd4 : SUCCESS in 55m 03s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2fe0a8f8f8c478ca5b5facae6a584ed : FAILURE in 36m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1cc9f25827c249d783fb1f9500f92a0b : FAILURE in 30m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/61c83fbfdbf246bbbe13e22f042aa374 : FAILURE in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7cdb28c6e356920735e9e73f21b6dc5c0269ed7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-08 13:16:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/47/856147/3/thirdparty-check/pure-devstack-manila-tempest-aio/a48e3b9/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a0c2103682fb53452e35b5678b95ec7c3e4051fb","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-08 13:34:33.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-47-856147-3/ns5_manila : FAILURE in 30m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"8bcfc0331e116fdf559204432dd4be04a93f3379","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-09 08:38:41.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/856147/3/check/DellEMC_Unity_Manila_dhss_true/e712d07/DellEMC_Unity_Manila_dhss_true/820 : FAILURE in 3h 12m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/856147/3/check/DellEMC_Unity_Manila_dhss_false/9a4642b/DellEMC_Unity_Manila_dhss_false/861 : SUCCESS in 1h 02m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"33ccda50b702b6229aee933f81bc262108281efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-21 12:23:35.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a7979b0d4735a0e9e966781eb1e3562729ddcb83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 15:35:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f115b95a7f94453ab215cdd174debe1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5f1db83d074a8d90cb1e81c568684c : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5abad48397c541a38e9b8497c04378a9 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176d864ba9824ea799eaf2a1612547c1 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bca3afcd7cf4cdf8d57e21b1ef6580a : SUCCESS in 10m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0421f6e7d9dd47f383e7765b3254477c : SUCCESS in 8m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6005a22cbd34946920d8a0a31aeb772 : SUCCESS in 1h 16m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/035cd6cb60284638b9ef724b407cf76f : TIMED_OUT in 3h 02m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/885a764927b64c17aa395ea09ce40574 : SUCCESS in 1h 30m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c03e7bbe5d0e4e2791bfc33e32172943 : SUCCESS in 1h 17m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5e714d4c71ce4867a30e35d68d0a2a0c : FAILURE in 1h 47m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b4e225315cb4f409ce9d77ffd30e524 : SUCCESS in 1h 06m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a0db5304a1de4d46928358f69730bd2d : SUCCESS in 5m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cafde8a5359d4343904dbc15e15d40fd : SUCCESS in 5m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7e06f355affd44d583134241a5ebd100 : SUCCESS in 32m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21d2a9bc249c425bb31f927dc0c37699 : SUCCESS in 38m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6b2bf962e104be7a3e7e913a15ee2df : SUCCESS in 59m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b457990aac014c7ea0384c2f4367407a : SUCCESS in 35m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a023d55c48404c88b7cfec6f27136beb : SUCCESS in 26m 59s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4c20c9f6711245b4a39d41fc2be106da : SUCCESS in 1h 42m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0a205cde35dc404ba9dac8dc76c651ed : SUCCESS in 1h 05m 45s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/61e11ebe099a48bb8887685695444222 : SUCCESS in 1h 06m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/811dcbffef24405aaf1d5d711f3128fc : SUCCESS in 45m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a45e3365c31546dabed139969177c17f : SUCCESS in 48m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cbb9775d692c7772b30f300f26c33a6cb97a7f48","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-21 16:00:07.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-47-856147-4/ns5_manila : SUCCESS in 57m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"6693234dbec911e7239612271bcd5f686602432b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-21 16:24:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/47/856147/4/thirdparty-check/pure-devstack-manila-tempest-aio/a91e6f7/ : FAILURE in 1h 12m 45s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/47/856147/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/3f5180f/ : FAILURE in 1h 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c1b44188185195e25a7884c79abcb9fdca1b12b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-21 19:14:26.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/4/upstream-check/manila-tempest-plugin-ontap-dhss/d3381e7/ : SUCCESS in 1h 34m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/4b89df2/ : RETRY_LIMIT in 10m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"3a34813393f8f2cb925e1c0047c708c407badad5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-09 18:06:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d054ed4319d1f60c7a2e99ca7a074af669408533","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-23 19:06:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"665d1718e1cdc56509bf8b275fa9bda26a294cf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 19:07:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c5be844d6a97e87d6395ca9221ad43f8f195811b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 20:03:23.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/856146, https://review.opendev.org/856145 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"6ecbf5c4d27aef12539d033ed00337faf8fffff5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-24 09:42:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"58f5752b8cf7d47448756b7f66f27165b31dd663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 12:52:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbaf154363b54f8d9969eace4d455190\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27980df0cceb4e10985f4bd651544fab : SUCCESS in 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83594397faa5401d9909dd8f6b0969e7 : SUCCESS in 25m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/717be14874ce4744bbc39a5a9b8ebe1d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3db8b37f43c04572bf8e375d74c96a19 : SUCCESS in 10m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2507d50d14745349d3ce0c926d63234 : SUCCESS in 7m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d543412cc5714ccca3bb58a72a263472 : SUCCESS in 57m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1a8f33546eab4d368f38e4be1c39eb4c : TIMED_OUT in 3h 02m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/167b5a9102cc46da84dc3e827b97458a : SUCCESS in 1h 21m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6c394e4c10fe494aa035e47996a255f1 : SUCCESS in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8c562cdd2a974c529a9cb8b07d61168e : FAILURE in 1h 22m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/366cc188c2544ba4b82ad479a4dacd5f : FAILURE in 1h 05m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/87aaff8beaec4c699176796e2ce23b2c : SUCCESS in 5m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5785d716840e46288da0f8f6aa6c4fcd : SUCCESS in 4m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e572c242d3b142cd93391b31fd9f2349 : SUCCESS in 36m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12af1fd7eb3447dab3a687dc3ceb9763 : SUCCESS in 30m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/47fd1625cc0941abab8aec36a0d13588 : SUCCESS in 53m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/56ab6e1507b5436f94073c01caba6849 : SUCCESS in 27m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b7e7eee96e0d4838b30bb10c277d5a6f : SUCCESS in 32m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/00cb38a3302c405e83c2050e7c3938ac : SUCCESS in 58m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ae6a2591764447268057b4d8a182f1dd : SUCCESS in 59m 38s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8bcad4aea9564a569a063680043d0052 : SUCCESS in 54m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6681603fa45346d49c4e268f5b580663 : SUCCESS in 39m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7da46131ef1d44edb5a4028680efb5e0 : SUCCESS in 46m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"61878638981ab51fb5db0977be5c6e5c079b1878","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-24 15:31:52.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/4/upstream-check/manila-tempest-plugin-ontap-dhss/429eef0/ : SUCCESS in 1h 48m 03s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/d280a48/ : RETRY_LIMIT in 7m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"0f0dc43bdd2931092b285363cf48c115fa6b2f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-30 11:59:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"77c94acc011f4172a026f430c1ffe584cdd82ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 13:52:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0376fd45202f4fe58475ebc4e26fd4cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45deea1d72b7426d994c992d6411efac : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/035932d3bf3f44d9a07b3c6e020b007a : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f070f6c9513b40408062359980c28c57 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94f06e5c55b84c55a73060575478c42e : SUCCESS in 9m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23d93e10a66547dd896ea997bd354b56 : SUCCESS in 8m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d802fb698c984787ac734fd77cbea128 : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/34c329c03b794740aeea2e60822e3293 : SUCCESS in 1h 41m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5ae877b9d18e4f8783d031365f097f49 : SUCCESS in 1h 33m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/faa36088c1384e3ea343e18834072619 : SUCCESS in 1h 14m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8801a59ebc184ae2947ffcf8818f693e : SUCCESS in 1h 20m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd37ee093e9144688faad770507b8486 : SUCCESS in 1h 01m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ac80ceb540e94440924bc8c1fbff02a4 : SUCCESS in 5m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b31fa55f15ba4bf1a795d5c9b3eac2e4 : SUCCESS in 4m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/beeb5c31c54e414dacae3dea73ca2f49 : SUCCESS in 26m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/db409ea2428e4a569f574e870b2b6105 : SUCCESS in 41m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4ac5fcae585246849f6fff24f016c75c : SUCCESS in 39m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d0e91fe66ab14307b705afbccd30116f : SUCCESS in 45m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1182cb570eb945a8863a2c8b7913361a : SUCCESS in 21m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/767ff284516d42b59f08bc882dfef777 : FAILURE in 1h 08m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1fa7e07d36934bf0a7182020c71c0c49 : SUCCESS in 54m 40s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/21e79c62e11c40e585977a807dfff556 : SUCCESS in 1h 05m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a283a687c56c48f5b6031774dd315be7 : SUCCESS in 47m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6536a5fc5c5e4ae8a9e66fcf10a5b258 : SUCCESS in 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"77ba3411196d974bc58ae1872205956467d3b817","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-30 14:29:35.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/856147/5/check/DellEMC_Unity_Manila_dhss_true/370fa84/DellEMC_Unity_Manila_dhss_true/None : NOT_REGISTERED\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/856147/5/check/DellEMC_Unity_Manila_dhss_false/33d5f0b/DellEMC_Unity_Manila_dhss_false/1230 : SUCCESS in 1h 19m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"96d4f51a34d213d2d0ee1344e7341db82509720c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-30 15:36:18.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/5/upstream-check/manila-tempest-plugin-ontap-dhss/6c72b60/ : SUCCESS in 1h 42m 05s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/856147/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/9f14ee9/ : FAILURE in 1h 18m 02s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"ff6fdf5f8a0173896334b33d1bc5d2b602a27d76","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-30 15:54:55.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-47-856147-5/ns5_manila : SUCCESS in 3h 23m 37s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"29c5332d575660dede246cb366385f6b01ffb527","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-30 20:05:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"57359f848c01b4b41539cdf4728e793cf485a5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 20:05:53.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b1e1138633ee2b8bcd33b776aefc435107cd278b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:31:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73df405982f3411893cba1f04895a7d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74113ed84bc040e68d9a540c85f7ad33 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4294237ba05c449d80876294c9e1f9ac : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4bc2b3e75dc4c12a7eb45a04433ed0d : SUCCESS in 9m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0fdd6f1c2de442883c676cd3e64f654 : SUCCESS in 8m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49e36607f35b4bb9b5b4ca41cd62d6e2 : SUCCESS in 54m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0ad4581bc195479081bae8c6e2f491c7 : SUCCESS in 25m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6bd02400f0f540659943644054ebfed9 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":5},{"id":"d1eafdd9c5c51e9de4245070d8ebe46bfb538cf0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:31:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3a927fc50bd8aa09b5a5f8ac9ede29f17b64582d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 21:33:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97539493fc0490089e15a958f0b42ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01b583adbee4dbe9f138d89e07745a1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4042d7027385210a7c734d9f9eca519e7ccb2bfa","revisions":{"9e4c9b9e877cfdee052a4a2837102280785654e6":{"kind":"REWORK","_number":1,"created":"2022-09-06 14:59:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/856147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/856147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/856147/1"}}},"commit":{"parents":[{"commit":"375a83762f0995fa220930443ca6530fc5d6c6a4","subject":"db: Migrate \"quota class\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/375a83762f0995fa220930443ca6530fc5d6c6a4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"subject":"db: Migrate \"quota usage\" APIs to enginefacade","message":"db: Migrate \"quota usage\" APIs to enginefacade\n\nMigrate quota usage-related APIs from the legacy enginefacade to the\nmodern context-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e4c9b9e877cfdee052a4a2837102280785654e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9e4c9b9e877cfdee052a4a2837102280785654e6"}]},"branch":"refs/heads/master"},"4cb47d7ce88219df622a0059e2a71d480e09796f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-07 23:08:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/856147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/856147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/856147/2"}}},"commit":{"parents":[{"commit":"23417717ceb8f6db15f4e078aec42ecef3d4ecf0","subject":"db: Migrate \"quota class\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/23417717ceb8f6db15f4e078aec42ecef3d4ecf0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-07 22:01:51.000000000","tz":60},"subject":"db: Migrate \"quota usage\" APIs to enginefacade","message":"db: Migrate \"quota usage\" APIs to enginefacade\n\nMigrate quota usage-related APIs from the legacy enginefacade to the\nmodern context-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4cb47d7ce88219df622a0059e2a71d480e09796f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4cb47d7ce88219df622a0059e2a71d480e09796f"}]},"branch":"refs/heads/master"},"20ae0039b7693cc1750bbbd6b76563bc79e99161":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-08 10:50:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/856147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/856147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/856147/3"}}},"commit":{"parents":[{"commit":"ace3857395c23f0af31f0736209e8943fd84d7ca","subject":"db: Migrate \"quota class\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ace3857395c23f0af31f0736209e8943fd84d7ca"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-08 10:20:20.000000000","tz":60},"subject":"db: Migrate \"quota usage\" APIs to enginefacade","message":"db: Migrate \"quota usage\" APIs to enginefacade\n\nMigrate quota usage-related APIs from the legacy enginefacade to the\nmodern context-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/20ae0039b7693cc1750bbbd6b76563bc79e99161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/20ae0039b7693cc1750bbbd6b76563bc79e99161"}]},"branch":"refs/heads/master"},"cc280fef0d52f9981261df0e24cdc183895492a3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-12-21 12:23:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/856147/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/856147/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/856147/4"}}},"commit":{"parents":[{"commit":"e7a74a096b769e0d9621979aea1ed6026ff8790e","subject":"db: Migrate \"quota class\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e7a74a096b769e0d9621979aea1ed6026ff8790e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-12-21 12:10:58.000000000","tz":0},"subject":"db: Migrate \"quota usage\" APIs to enginefacade","message":"db: Migrate \"quota usage\" APIs to enginefacade\n\nMigrate quota usage-related APIs from the legacy enginefacade to the\nmodern context-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc280fef0d52f9981261df0e24cdc183895492a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc280fef0d52f9981261df0e24cdc183895492a3"}]},"branch":"refs/heads/master"},"4042d7027385210a7c734d9f9eca519e7ccb2bfa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-30 11:59:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/856147/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/47/856147/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/47/856147/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/47/856147/5"}}},"commit":{"parents":[{"commit":"ff2ffab964f11ea83d997bfdb969e1abd53f3b93","subject":"db: Migrate \"quota class\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff2ffab964f11ea83d997bfdb969e1abd53f3b93"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 14:46:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-27 12:39:28.000000000","tz":0},"subject":"db: Migrate \"quota usage\" APIs to enginefacade","message":"db: Migrate \"quota usage\" APIs to enginefacade\n\nMigrate quota usage-related APIs from the legacy enginefacade to the\nmodern context-based enginefacade.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ib7f5402b4a9c94290e14ebc9b76700cbf517de4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4042d7027385210a7c734d9f9eca519e7ccb2bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4042d7027385210a7c734d9f9eca519e7ccb2bfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
