)]}'
{"id":"openstack%2Fmanila~856756","triplet_id":"openstack%2Fmanila~master~I6d038244493583cc943063b50d731b8c1ef5ed28","project":"openstack/manila","branch":"master","topic":"bug/1989175","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2022-09-14 20:22:16.000000000","reason":"Change was submitted"},"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2022-09-14 19:14:23.000000000","reason":"\u003cGERRIT_ACCOUNT_33648\u003e replied on the change","reason_account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-13 22:01:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-09-13 18:51:48.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2022-09-14 20:22:16.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-09-14 20:22:16.000000000","reason":"Change was submitted"},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2022-09-14 20:22:16.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2022-09-14 20:22:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d038244493583cc943063b50d731b8c1ef5ed28","subject":"[NetApp] Consider last transfer size and error for replica state","status":"MERGED","created":"2022-09-09 11:53:52.000000000","updated":"2022-09-14 20:23:39.000000000","submitted":"2022-09-14 20:22:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856756-bug/1989175","meta_rev_id":"7c01ce8c12cc617d767cd403dbad94ba344eb15a","_number":856756,"virtual_id_number":856756,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 20:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-09-14 13:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-09-14 19:14:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":2,"date":"2022-09-14 13:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":1,"date":"2022-09-14 19:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2022-09-09 11:55:22.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-09-09 12:59:07.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-09-09 13:31:23.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-09 14:31:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-10 02:00:53.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-13 18:52:00.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2022-09-13 20:57:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"7a156b9279bd7c8e08a763dc36a095e048bed11b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-09 11:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"543580c580c4a46c8b73fbdf5b5f59d9abdb0bd9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-09 12:59:07.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/56/856756/1/thirdparty-check/pure-devstack-manila-tempest-aio/72ee3fd/ : FAILURE in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"11c79af9fa9a72af7d49b958d0edfe3c08ec54a9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-09 13:31:23.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-56-856756-1/ns5_manila : SUCCESS in 1h 2m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"063f69722ee0e3a3cdd3c563abe8537392de4bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 14:31:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/307bc0b9b6dc47b789bbd5333c06f2ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb177fc051e54c5e9c58f7d29d9bef1f : SUCCESS in 11m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cbe019e31c24736a6d2dac9250834dc : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be20fb598bd448686f76b2f4d438f30 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2702714df43e4438aa3d57de697d27bb : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c28530adc587409cbedc444dfd555e1c : SUCCESS in 8m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90ed1abad3ef468294d76ad4ed1349b2 : SUCCESS in 11m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62bf81a3f2ca4470b558f40304da3d24 : SUCCESS in 4m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6463eaaf43a94606bfc685085d290e44 : SUCCESS in 1h 17m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f25153c65f054dd6874cb9e8f9a5929f : FAILURE in 2h 07m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1326fde7a936431ca1b8f0773d146b65 : FAILURE in 1h 57m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/91979da239af469fbca2cebb511a8234 : SUCCESS in 1h 14m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ab6a4c2c99ac4f489ecbf8886b24bd4f : FAILURE in 2h 32m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9ac299f9b2354c80b01659b6ceef1aa2 : FAILURE in 1h 28m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ca6195f83f64f65a5bb6c02cac16666 : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd6625a5814746ffb5ac3a514f0e054b : SUCCESS in 6m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7307b1024633412e9cc8b1da6841c547 : SUCCESS in 41m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/55e1f3a6bb794ba6a190e7fd9bbaa0d1 : SUCCESS in 39m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dfb0c5449cf544339dd84dfac2bf43fb : SUCCESS in 1h 02m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/35c39e179ee244dc934f0428d6cb1bfd : SUCCESS in 43m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9babd7f75d2b49ef8415c80a79af99ca : SUCCESS in 25m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/beabd323ff3244368ee285439c5e00eb : SUCCESS in 1h 51m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9a7e7a2493e0445cb7e3c1a621f96af9 : SUCCESS in 1h 12m 46s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e53e9b15d8764b5cb0026fd3ed3d0a8d : FAILURE in 31m 43s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4c4614a331bf413b90b490ecafe967a2 : SUCCESS in 48m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b30508a5203a4234a98acc0c90205ccb : SUCCESS in 37m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcbcdca8403d4bbbe1f9849b1c884234d01ad60a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-10 02:00:53.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/856756/1/check/DellEMC_Unity_Manila_dhss_true/0b32372/DellEMC_Unity_Manila_dhss_true/827 : FAILURE in 2h 31m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/856756/1/check/DellEMC_Unity_Manila_dhss_false/5366b4e/DellEMC_Unity_Manila_dhss_false/873 : SUCCESS in 55m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7ddf1d94e383fb6c8aedf3ca06b43f9c95395120","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-12 18:08:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"38c5e8643797ec055f8c448b53afcb21b86ba01f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-12 18:10:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65c509ed33451b6098216d9dd62cc2cdc051934f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-13 16:19:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c823a9c2219b8dee5ccb9a9a533224110ecc2372","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-13 16:19:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf42331e4c3aaa37c9899c9328c718289888087e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-13 17:21:22.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-56-856756-2/ns5_manila : SUCCESS in 1h 1m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ca95d5597190f893ea3416163b9abb524b4a76f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-13 17:26:30.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/56/856756/2/thirdparty-check/pure-devstack-manila-tempest-aio/7c2e72c/ : FAILURE in 1h 05m 27s","accounts_in_message":[],"_revision_number":2},{"id":"de33f73dcf05213b5a866f88db5cfabd1f21a0e1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-13 18:51:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2ec89c04046cf333b072dfa8cc8caaed93c5749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 19:10:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a87256943434174a1abe9bb772a307e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a0b189f3af45e0b1a8dc9851fe42d3 : SUCCESS in 10m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5fe8f258de7422f98b42ccf531b473d : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc5ddecc268485985564a2f188efbac : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec283eb4a304eab98d499c289703fee : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6fb0610e1044b21ab72bb6f3939a35f : SUCCESS in 8m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa4cce7bb62347398825b40858bdf230 : SUCCESS in 10m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fedf3b9a50c4e3889c87d85103b5fab : SUCCESS in 5m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bfe54ce25e1f4914b3c2bf5e4b294c24 : SUCCESS in 1h 45m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/358f744d48be454ba5ed3f0114817319 : FAILURE in 2h 45m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5e50ac853e2f4c8c8f2a56bc475ef370 : SUCCESS in 1h 03m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3700421c7a8947718a2a58713ff3b835 : FAILURE in 1h 04m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4d236e49106d4c6897939a57c1e08316 : FAILURE in 2h 18m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bff1c7e9ec4e44938192bcd9db3435f3 : FAILURE in 1h 09m 30s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8eb9f7c9ffac44ed81f2e52d193040d9 : SUCCESS in 6m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/96ba312d722841f9b56fb990d8bfa147 : SUCCESS in 6m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dcf74029b0d742b48bd612b12578809e : SUCCESS in 33m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c311301fe634f2d9035b3375223abe5 : SUCCESS in 34m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/86bc7d43f49e405ebdb08d33013e0986 : SUCCESS in 55m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f8af081175f041ccb9cc4a54ff558f14 : SUCCESS in 33m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/72e53e88e03c4ccfa5b9fb2bb8277696 : SUCCESS in 24m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3c73039b46be46e7a092e1548c07a3f8 : SUCCESS in 1h 20m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/46dde51692ab4c089f24745428536bed : FAILURE in 1h 04m 34s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0703ae6a452e473d9168c2c3916428fa : FAILURE in 42m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5e8f36a0535546cdaa7070b9fb16e9fc : SUCCESS in 41m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/527183d1ea59441e870a817474eab075 : SUCCESS in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5cec2dabee6c16c635704aaf87d42386b647c83d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-13 20:57:32.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/856756/2/upstream-check/manila-tempest-plugin-ontap-dhss/22c6d6e/ : FAILURE in 1h 45m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/856756/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/d533ca8/ : FAILURE in 1h 16m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"615bb45f31226260bcd6da29bc73bf253e3e0d95","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-13 21:53:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/856756/2/check/DellEMC_Unity_Manila_dhss_true/62b895b/DellEMC_Unity_Manila_dhss_true/845 : FAILURE in 2h 56m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/856756/2/check/DellEMC_Unity_Manila_dhss_false/1f926ca/DellEMC_Unity_Manila_dhss_false/893 : FAILURE in 1h 10m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc46dcc51c40b71a96f23c0672755a462ab36c56","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-13 22:01:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30c7ffc33e36b55bbba27501c6a92524e4657f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-14 09:51:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f7bf6dee2a37f001e1d911b06ca9edf07c66b478","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-09-14 09:52:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16488cd97b47ccbdeb57776f34b13f2d5d6cb61b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-09-14 10:46:48.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-56-856756-3/ns5_manila : FAILURE in 55m 7s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"546cbcd26d7a99a597c89d354e88fbda644c4e0e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 11:04:34.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/856756/3/check/DellEMC_Unity_Manila_dhss_true/e4638e5/DellEMC_Unity_Manila_dhss_true/850 : FAILURE in 1h 13m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/856756/3/check/DellEMC_Unity_Manila_dhss_false/58b4f21/DellEMC_Unity_Manila_dhss_false/899 : SUCCESS in 56m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"acf7008dafd2b3a7dfd43e56b25a3341bd225ae7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-09-14 11:04:48.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/56/856756/3/thirdparty-check/pure-devstack-manila-tempest-aio/6933473/ : FAILURE in 1h 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"8bb135c285f4ac6fd8378c7aa3df3cde0e0d8f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 12:04:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2dffe2477b147b2ba5d70b864975a2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d66319fe054e02bc517343573543e6 : SUCCESS in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/119064ff67d54a0f886498b5632567b0 : FAILURE in 18m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8570730145e48f696352247455d26c9 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9064b52929f43f1b2f84bcd3bcd57c3 : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c468fc93df5e411ab7028aa92e55e24e : SUCCESS in 12m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deddf65c79ab4a0f9006c344aa28b61a : SUCCESS in 10m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5ab9897f3745308d09f05279e7b53d : SUCCESS in 7m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8097cafe00a04f0799d4937f38c60ead : SUCCESS in 1h 26m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5baf671f969a41b59766ce50acc16762 : SUCCESS in 1h 54m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/22376bf8aeff43a8947c9cfa50c6cf4e : SUCCESS in 1h 37m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2fae5d8b2a1b419d9d243908f59da01c : SUCCESS in 1h 21m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2880318636934127a87d771b260d4b28 : FAILURE in 1h 27m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9d482a07297c4624b5d102a76abc9ffb : SUCCESS in 1h 17m 06s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1a4cb21606104570b62c7a91b12d0705 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/99f8d0f97e0a42f8994cc1b1a9346e1f : SUCCESS in 6m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2956bc86773476fb07aec289995e352 : SUCCESS in 36m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad8fb3d39d36417590a8ad6bf221fb16 : SUCCESS in 32m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bc53aa67fbd4cc29cd464bcea7a37c1 : SUCCESS in 56m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ca8bcfacf3274fccb6b3a561007036f4 : SUCCESS in 34m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dbbf3154bf724573ab7a758332e5aae5 : SUCCESS in 29m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/03d22f84e3ec498aa73697eb7b014b0b : FAILURE in 2h 07m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/706f4e3547d745139fb219ae9e662879 : SUCCESS in 1h 02m 11s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/03a62d28fbad428db545e34ed2f0a7e8 : FAILURE in 38m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8b4b2ad14b7843559dc0cd8b7b9deed4 : SUCCESS in 39m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a38ec84cfbc64d54ba33c742269cd78b : SUCCESS in 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a917d463e8c9dc315783fbceafe8146a0567a844","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 13:35:24.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/56/856756/3/check/DellEMC_Unity_Manila_dhss_true/1d8a355/DellEMC_Unity_Manila_dhss_true/852 : FAILURE in 1h 12m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/56/856756/3/check/DellEMC_Unity_Manila_dhss_false/739994e/DellEMC_Unity_Manila_dhss_false/901 : SUCCESS in 1h 03m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bae7932b8121576e1a4c09426ed4cb7ff7dd9b3b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-14 13:40:05.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/856756/3/upstream-check/manila-tempest-plugin-ontap-dhss/e5e7872/ : SUCCESS in 1h 33m 41s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/856756/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/329191e/ : SUCCESS in 1h 19m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"62d26c1990ec2ad0adf77d8dbb6fb4cac4f89f26","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-14 13:46:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c29f63f5267db731ef7fb3c2cabefa8b5cbb9359","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-14 13:52:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"48ee5bb751a5c29905e5bf73190bef028bac99c3","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2022-09-14 19:14:23.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"76a82a478b6f140549a73c25e76cde01d40f7a2c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-14 19:22:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"097986d1ae592b492a25357972d81545000bdb29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 19:23:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9a6bc3856ea0b95160b4085346bb5dda8bd2ddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:22:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c29b39515b5c4a71a6a039514d19c3a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365c358459f44d52be669a585299575a : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f73d303c0b4472bd579d7590644ef9 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac4cc9131e564b4593f4fb0673bb7ce0 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61c94d29fdac4b54934006c14d62bffc : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d0f306677b541248d581dfc761dce74 : SUCCESS in 5m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a64f7e74da774fa29157f0e385feb658 : SUCCESS in 57m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cffe4ffcc226423f97ed4d781cb2f5c4 : SUCCESS in 28m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/447a5e8c29134631b409dde3c2d63179 : SUCCESS in 27m 19s","accounts_in_message":[],"_revision_number":3},{"id":"b62dc66d22b5e7b359fb591604d1cc8291cc54e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:22:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7c01ce8c12cc617d767cd403dbad94ba344eb15a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:23:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e345444166484772a2397d25076177fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e957ed5160949a8a263bbf3fbbaa2ed : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/490ab8c611234538852f58e54fb4adee : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e8ed59a91934096a4ddc9b184ef50a633cbd1ec","revisions":{"007c9a7f9311a1bf64fa473c4400245fc8e4f969":{"kind":"REWORK","_number":1,"created":"2022-09-09 11:53:52.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/56/856756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/856756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/856756/1"}}},"commit":{"parents":[{"commit":"584781338c6386a6042571c1564fa871aac4fe75","subject":"Merge \"Refactor the Ceph NFS driver to use Cephadm NFS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/584781338c6386a6042571c1564fa871aac4fe75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-09 11:31:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-09 11:39:27.000000000","tz":0},"subject":"[NetApp] Consider last transfer size and error for replica state","message":"[NetApp] Consider last transfer size and error for replica state\n\nIn order to determine replica state from snapmirror, in addition to\nexisting check of last-transfer-end-timestamp\u0027, also add new checks\nof `last-transfer-size` and `last-transfer-error`. New config option\n`netapp_snapmirror_last_transfer_size_limit` added with default value\nof 1MB. The last-transfer-size above this value or presence of any\nlast-transfer-error is considered as replica is out_of_sync.\n\nCloses-bug: #1989175\nChange-Id: I6d038244493583cc943063b50d731b8c1ef5ed28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/007c9a7f9311a1bf64fa473c4400245fc8e4f969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/007c9a7f9311a1bf64fa473c4400245fc8e4f969"}]},"branch":"refs/heads/master"},"cf3a9ae11085cca98e221e4896f015987c7644dc":{"kind":"REWORK","_number":2,"created":"2022-09-13 16:19:47.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/56/856756/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/856756/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/856756/2"}}},"commit":{"parents":[{"commit":"584781338c6386a6042571c1564fa871aac4fe75","subject":"Merge \"Refactor the Ceph NFS driver to use Cephadm NFS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/584781338c6386a6042571c1564fa871aac4fe75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-09 11:31:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-13 16:19:27.000000000","tz":0},"subject":"[NetApp] Consider last transfer size and error for replica state","message":"[NetApp] Consider last transfer size and error for replica state\n\nIn order to determine replica state from snapmirror, in addition to\nexisting check of last-transfer-end-timestamp\u0027, also add new checks\nof `last-transfer-size` and `last-transfer-error`. New config option\n`netapp_snapmirror_last_transfer_size_limit` added with default value\nof 1MB. The last-transfer-size above this value or presence of any\nlast-transfer-error is considered as replica is out_of_sync.\n\nCloses-bug: #1989175\nChange-Id: I6d038244493583cc943063b50d731b8c1ef5ed28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf3a9ae11085cca98e221e4896f015987c7644dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf3a9ae11085cca98e221e4896f015987c7644dc"}]},"branch":"refs/heads/master"},"6e8ed59a91934096a4ddc9b184ef50a633cbd1ec":{"kind":"REWORK","_number":3,"created":"2022-09-14 09:51:12.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/56/856756/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/856756/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/856756/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/856756/3"}}},"commit":{"parents":[{"commit":"3f431ef9eb54c8ba2c2b546f5bcbc6258e935915","subject":"Merge \"Fix DriverFilter/GoodnessWeigher string evaluations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3f431ef9eb54c8ba2c2b546f5bcbc6258e935915"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-09 11:31:34.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-09-14 09:30:27.000000000","tz":0},"subject":"[NetApp] Consider last transfer size and error for replica state","message":"[NetApp] Consider last transfer size and error for replica state\n\nIn order to determine replica state from snapmirror, in addition to\nexisting check of last-transfer-end-timestamp\u0027, also add new checks\nof `last-transfer-size` and `last-transfer-error`. New config option\n`netapp_snapmirror_last_transfer_size_limit` added with default value\nof 1MB. The last-transfer-size above this value or presence of any\nlast-transfer-error is considered as replica is out_of_sync.\n\nCloses-bug: #1989175\nChange-Id: I6d038244493583cc943063b50d731b8c1ef5ed28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e8ed59a91934096a4ddc9b184ef50a633cbd1ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6e8ed59a91934096a4ddc9b184ef50a633cbd1ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
