)]}'
{"id":"openstack%2Fmanila~868486","triplet_id":"openstack%2Fmanila~master~Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7","project":"openstack/manila","branch":"master","topic":"bug/2000171","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-02-15 01:45:47.000000000","reason":"Change was submitted"},"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-02-13 12:31:49.000000000","reason":"\u003cGERRIT_ACCOUNT_33648\u003e replied on the change","reason_account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-01-02 17:19:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-02-15 00:13:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2023-02-15 01:45:47.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-02-14 02:14:56.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2023-02-15 01:45:47.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-02-15 01:45:47.000000000","reason":"Change was submitted"},"33038":{"account":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"last_update":"2023-02-15 01:45:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7","subject":"Add quiesce_wait_time option to share replica promote API.","status":"MERGED","created":"2022-12-23 13:20:45.000000000","updated":"2023-02-15 01:46:57.000000000","submitted":"2023-02-15 01:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":58,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"868486-bug/2000171","meta_rev_id":"8d0bab3c948a0646475a9d8b198f5eb1ed7b654c","_number":868486,"virtual_id_number":868486,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-15 01:45:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"recommended":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"all":[{"value":1,"date":"2023-02-13 12:31:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":2,"date":"2023-02-14 01:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-15 00:13:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-02-14 02:14:56.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-02-15 00:13:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2022-12-23 13:21:33.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"state":"REVIEWER"},{"updated":"2022-12-23 14:19:48.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-12-23 14:34:54.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2022-12-23 16:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 12:45:22.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-01-12 17:31:00.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-01-31 14:59:35.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"}],"messages":[{"id":"af38775bdaba3d41d1af0e47683857a35bbf6f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-12-23 13:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2193cd0d3fd0269486dd24d1f25961921906ae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-12-23 13:30:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a54789978d0d102ed616f069404ed4105ad2ce5e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-23 14:19:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-1/ns5_manila : SUCCESS in 58m 31s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"c4d679492e44e0b1bb1b4c920e22fb4d9e520a93","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2022-12-23 14:34:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/2/thirdparty-check/pure-devstack-manila-tempest-aio/692e640/ : FAILURE in 1h 00m 16s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/8c5185d/ : FAILURE in 1h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6dc1dbf1a689a1c669a2413583a0b4cdf4709542","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2022-12-23 14:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-2/ns5_manila : SUCCESS in 57m 42s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"67319e9135e279a38347138d39ab21c0488a7018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-23 16:40:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0a5e6fe47db427ab72829fa9694ad90\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6d68ab52320a42ffbdc6ec38c03142dc : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489413a69e3c40ca8ec85ad1aa4e8147 : SUCCESS in 9m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b27702d31a044a52ab054aab1dd5761e : FAILURE in 4m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea3d02d30e7f4ea6820d20b6c86f9852 : FAILURE in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fefe4e2a939a431ab2b62d700339cc43 : FAILURE in 9m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49a8b621f33f4a4988c6ed993372a695 : FAILURE in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e47688d33874f40a82de11cad2d4f75 : SUCCESS in 6m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f78d67973794de68d495349bfcdda1b : SUCCESS in 53m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/73b101a358c849949f2d0d754cb1e9dd : TIMED_OUT in 3h 02m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d43e2be63783424ea50e56a18d01ed40 : SUCCESS in 1h 13m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/09c3a3b6d443410795f17044d9499917 : SUCCESS in 1h 18m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f1a84fab4c7544eb870e6b1b4e8bd062 : FAILURE in 2h 08m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c7a524f39b6d46858e2ba62c53ab9368 : FAILURE in 18m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1191f188052b4f8aa305bc692c2db14c : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ed2d70d72d846f4bc75e63119a7a495 : FAILURE in 4m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/036a8c4e7e954493870f876e68103a54 : SUCCESS in 22m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1cd26e0610c54179a55c9e57b8b89f3d : SUCCESS in 29m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4704ce4229e94903bb7bb08badfe8047 : FAILURE in 20m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ec007ee840da46209b8ce9f1f60746e7 : SUCCESS in 24m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9c2e771148904a1396361d8ce8157fbb : SUCCESS in 16m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d2d0b2cd34b1451990edb6a2aea69062 : SUCCESS in 1h 11m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/058847d6b58c45c1946227701adaff5d : SUCCESS in 44m 20s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b5e36ce30c8b43259eb75b86f0c5117e : SUCCESS in 1h 11m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2e4cbf321050442cb0f402fff571ac0d : SUCCESS in 27m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8f94ae3a277e43c6850715edd6a02f8d : SUCCESS in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6965b83637385f933e3595c042acd14d6f7de6cf","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-12-23 18:32:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45fcfcb28deec260021b190042589f18d228b02d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-01-02 17:19:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da81f8b39db2f4fe5acac057b1a2e5b11070b8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 19:05:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc271aaf9c2849c3abf2db10a5ff3bf5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd461d1f2a004eb49f9fd2a70e2bfa58 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c025759071104fccae06dd581fb7e22c : SUCCESS in 9m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ad8654f56a6408ca6d2b028fd7fd4f3 : FAILURE in 4m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255ff62a704a497db60b89a9e7818f19 : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/898eb08f08fb462f9b3498f52d4083e2 : FAILURE in 10m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c14f3fe0e8843a585e692c5a2d8eb5c : FAILURE in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7c322c6b4b443b296108c2af8579bba : SUCCESS in 5m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7a837ed16fa24b23b7daf7549b4e03d8 : SUCCESS in 55m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3aa2a151c52641dc8c13d7c07d4f4bef : FAILURE in 1h 43m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0873548a42104b5eb4d8e90de00ecddc : SUCCESS in 49m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b2edcf7422314aedac117e59075e72d6 : SUCCESS in 1h 13m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb80b9e9c62e49bcb495911b5fdaeb6b : FAILURE in 1h 34m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7aaa8a4ed3de4d2f897e1e1958c0378a : SUCCESS in 1h 20m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c213b8f4b36048c599b9a388016495f6 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ed49abf69814da3a5a6d91fc71705c1 : FAILURE in 4m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/07b73991f2ad44ccbef9dd8ed44d17f3 : SUCCESS in 30m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80fd220751764582a7c3db627843cf73 : SUCCESS in 41m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/487806cfc89248039c8682f421c34833 : SUCCESS in 38m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b97ff5a82734460ea60c74cd144058be : SUCCESS in 40m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/48b89a4ff05e43d6a79872d9297876a4 : SUCCESS in 18m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d4da225298354b7b95beca7e6514ee51 : SUCCESS in 1h 07m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d0dfe57948294debbd97deded04215f6 : SUCCESS in 43m 20s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2f3e6feae80d42d1b08bc38c4171a2db : SUCCESS in 55m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6857132082744fd193f91037bcdf4b75 : SUCCESS in 26m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bb112ae1e6f644be91b09ffb0156e3fb : SUCCESS in 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e8c5d3659fa20acba017fde3f6f7da60b976ab99","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-01-03 09:06:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1283cba1c9c6e658bcd7ff99fbebbce1a9dfa9c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-03 09:39:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"204fd9e19c7ea447c3807da6af2f13dd380891fd","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-03 09:40:00.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2d9da763265d0da6d48807e4a90defc4c0564135","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-03 10:52:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/3/thirdparty-check/pure-devstack-manila-tempest-aio/7328319/ : SUCCESS in 1h 09m 23s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/0c422e5/ : SUCCESS in 1h 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f5dfabac2243e1346954f04440ae22fe01faf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 12:09:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9b6efa5d43c4345a513e82e23133908\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7e03c69d52bd46c390843b3ac15ff31e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abebd9cf258c4b979cce93002a639200 : SUCCESS in 9m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e54e724bd15c4ef389e9e861b0bac3dc : FAILURE in 3m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36f9cbc5b3f2409bbb1fc598f4057527 : FAILURE in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25fae3f6c5ec44e99de3b5a5908b6068 : FAILURE in 9m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/934d63302adc4954bcadff93ec416308 : FAILURE in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc4752ff3cc4bdd8f9861f486c494bb : SUCCESS in 5m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e84464432649425388a53fdd4f12f5ed : SUCCESS in 57m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/12a420263b7a477f9e989080300157da : FAILURE in 2h 24m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/14701d82981a47ec8e54aef3bba78c31 : SUCCESS in 58m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ac1ebc7471fc49859eba4dc006a5fba2 : SUCCESS in 1h 39m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/351d03c345274841841e0c09edcea9d5 : FAILURE in 1h 21m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90b7f62a35fa4cf2ac7608b2d8e969ff : SUCCESS in 51m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cf9fd6722085475da5b0f10d5a020081 : SUCCESS in 3m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/360b7f6299d343f8a9f77ad1c405b3e6 : FAILURE in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e1428d7b56864ea59c46ed365524b0f7 : SUCCESS in 25m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d1fe9746b3e446ef9bd8967ab42a270d : SUCCESS in 28m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e2890c4f31874ece96054e226e14b3e4 : SUCCESS in 42m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0518b5ea7ca444f8918d8cfe175c8cf9 : SUCCESS in 24m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9ba9768922e248cbba135cc2a934d7e7 : SUCCESS in 26m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/aadd83f9692a402d9a230679531b0b18 : SUCCESS in 1h 30m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/219e95afc7064f2a834137bc8fa0f1fb : SUCCESS in 43m 22s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fbdd906182ae4becb4009bfe0e8a5371 : SUCCESS in 1h 17m 01s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/bb3d9d5b140f41588d7bfdd61091af22 : SUCCESS in 32m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2011e3d5aebe463ba5ee554636019e41 : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0e95bad980a2da22a8e0d6e51e53f8d536330b0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-03 12:45:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/868486/3/check/DellEMC_Unity_Manila_dhss_true/eaa8381/DellEMC_Unity_Manila_dhss_true/1087 : FAILURE in 3h 05m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/868486/3/check/DellEMC_Unity_Manila_dhss_false/f1d2ba0/DellEMC_Unity_Manila_dhss_false/1163 : SUCCESS in 54m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"40dd8a1519b649de393ea11edd387936dc8457c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-12 15:17:25.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8a7a80b7ea1b8fa55dbcc2cca983b96aa4d1428d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-12 16:22:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/4/thirdparty-check/pure-devstack-manila-tempest-aio/d7dc3fd/ : SUCCESS in 1h 00m 27s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/a5d34d5/ : SUCCESS in 1h 04m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"014fc64bb6db4012fbe1b2e9b0cb0ff4f9fc6ae6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-12 16:50:04.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-4/ns5_manila : FAILURE in 1m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"1c78a16f10708080d85e2da2093e353948326c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 17:28:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad1ab1bef8b4f09af6a3b40b9a8d380\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bbfc49155c4a43b880db05bad3ea33ab : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593d29b201c34316aadda7ed39751781 : SUCCESS in 9m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c61ecd6e811a410d808f150aa3a992c0 : FAILURE in 17m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/150841f24c9a43ebb3093721e4e153f4 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c687192583f14f49a99ebf65e7d238e7 : SUCCESS in 8m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/488d84ef1b69452791102075e31014a7 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce206dc860a04a1f9788767dc7206f26 : SUCCESS in 5m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d63f210cff79451492fb981ba00f69b8 : SUCCESS in 1h 13m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5adf1e1dc2364bfd8c6c080b086ca16e : SUCCESS in 2h 09m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/33f4584d6066497fa6c84e0cb18a1d1a : SUCCESS in 1h 16m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5c5e038450344241ac8c840917f5e257 : SUCCESS in 1h 14m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c1c1d85f86ff44358bc652fa7b5c988c : FAILURE in 1h 57m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3e24513d1c8a4f50b4ac64108ba470f1 : FAILURE in 1h 09m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f0e4bc46b1ab4dfba60965cada4b32a0 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a27a76c651984c0880b7b6ec39d27acf : FAILURE in 5m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c7fee741211b425dbb045dc32a1354cb : SUCCESS in 37m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a81c8d41826f4ee1ab4412daffcec915 : SUCCESS in 25m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cbae6a03a3a5498c8b43693aae55b31e : SUCCESS in 52m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9bb3283165d241c28b4cb9fcb3fe1529 : FAILURE in 30m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/29ea6c636cec48c69f30975fcb2676a9 : SUCCESS in 27m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/85a1497c421a4089bfaf47a6e6bd1b55 : SUCCESS in 1h 02m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f329d657ed4b4617b847ffdf7436e111 : SUCCESS in 57m 52s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ed11e4a59a8f4f45875ea6fd6fb7aa06 : FAILURE in 1h 04m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d260c93e73f14cee948cd7b72e7e5e75 : SUCCESS in 36m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cfb5327114ea455fa3d88bc570bdb802 : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"57d1397f601eb5e48206ba60d6413fcda9118dc0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-12 17:31:00.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss finger://managesf.services.openstack.netapp.com:7979/6cca6e3b67d1455c8c69d138aee02560 : RETRY_LIMIT in 1m 40s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/489b7e6c1b84453ba0f84df93e1f4fd7 : RETRY_LIMIT in 1m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"785e19515e2112e89d514ad0c97019fe51cf2862","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-01-12 17:48:49.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/868486/4/check/DellEMC_Unity_Manila_dhss_true/d8b6791/DellEMC_Unity_Manila_dhss_true/1140 : FAILURE in 57m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/868486/4/check/DellEMC_Unity_Manila_dhss_false/444f4a1/DellEMC_Unity_Manila_dhss_false/1210 : FAILURE in 1h 13m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a4d539ce8e60493f57a8d041ed94d4c1b11076c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-18 10:11:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca2154dd913fd6c70a1e1213cff64a4067a10de1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-01-18 11:22:55.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/5/thirdparty-check/pure-devstack-manila-tempest-aio/a73f902/ : FAILURE in 1h 00m 46s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/215d5f2/ : FAILURE in 1h 09m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"88452ca03fd78e07758884773d620f98249e3155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:25:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9dc098950f4e54b69fa8d734ea2fae\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8009ffca73924bf0b29637ef5b476aa5 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39bf8909d847497eb89a1774f8e00714 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6052ee76000f4a21a7160c3f05e0d481 : FAILURE in 26m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f504cfbf40c4b90abb92c3a92ae0013 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f8afaa321942a4ad378cabe7fc905c : SUCCESS in 13m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acf6399484c64da59222abfe4881a564 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f20cf015145c4fbab44cd86587dc8965 : SUCCESS in 5m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b543d44d4ea049d1b2af0bac3110d665 : SUCCESS in 1h 27m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2d8bb6461fb448f386391cb18be7f2dd : SUCCESS in 2h 09m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/be8a2127b1b24c75a70b075accd4f5a0 : SUCCESS in 59m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/149ef30046634225add1c12a86db96cf : SUCCESS in 1h 32m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3806ddf29a184735951de18938ce4816 : FAILURE in 1h 24m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a88e37078a444ca88a3cc83ca86ad1e6 : SUCCESS in 1h 13m 45s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1ca400a15cf24dce942e34e9d354bbda : SUCCESS in 4m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3e7257d7a3ad47ec8f1f977b5990b626 : FAILURE in 5m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/54e2868963794384b2d66375de02a6b4 : SUCCESS in 40m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b34f5ed20f7f48e18bd6226f6c13d55a : SUCCESS in 46m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/39f044872d5c4d6786b908641c35a5d1 : SUCCESS in 54m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7604f1a6e79e472b8d24494a527e5fdf : SUCCESS in 36m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cfd81c3fb2374db0b6c1ef2c42b0a15f : SUCCESS in 32m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ccfc5106001545e5b6ec6f38d50af3b8 : SUCCESS in 1h 17m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1b327c32589c4d57ba31ea0f8f4469fb : SUCCESS in 1h 04m 13s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/acaa0700095b426ca22847c44729b4be : SUCCESS in 58m 27s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/67263b5696a54db2a044d50f00a9faf1 : SUCCESS in 31m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3f862919f9a24140813ebba9db535853 : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"17b1d6c5cf7374b34d9ee9be0187c1b4e8a7c58b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-01-18 12:29:55.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-5/ns5_manila : SUCCESS in 2h 17m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"9a7ae4af79b4c6df98e99bb549af47074107c0e7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-18 14:06:15.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/86/868486/5/upstream-check/manila-tempest-plugin-ontap-dhss/d2ab113/ : SUCCESS in 1h 39m 42s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/86/868486/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/134ac55/ : SUCCESS in 1h 23m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"4abe323bcca8eb063b6308335b4275f4370465df","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-01-31 14:59:35.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"10d493a30dcc24d48a9c97696d7baccc69221b50","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-01-31 15:03:16.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"91e34f1292b91436d10f60b508b4d92c7cf6dd98","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-01-31 15:21:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eeba144c56de1e5fc371669e355b8164d81f2c00","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-31 22:42:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a960975cb3a848faa0041124015d547361466483","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-01-31 23:43:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"03a71a7306e5f68d0846865b766fc41619467a08","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-01 01:29:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8c8ad0f5478c003c82011c09e2c28e1e005d1430","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-01 01:30:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5c09ae3c55943e9edf487a6f68130c123976d3ce","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-01 01:33:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5def3dfff415d46c7b1f082a81e81cfb07a726a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-01 09:56:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0ecb9f9db13c135716d1711361934e0e9ebb50a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-01 09:58:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ebc5db8665b3069b1f337157a66bbf9950472e6f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-01 09:58:59.000000000","message":"Patch Set 5:\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9036019d532aacbd0a1ddace1505bc4a6035bac8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-01 10:22:36.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-6/ns5_manila : FAILURE in 25m 50s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"da592b144ea6f09f509f404306c2bf3d61dcbefa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-01 11:11:14.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/868486/7/check/DellEMC_Unity_Manila_dhss_true/11eb128/DellEMC_Unity_Manila_dhss_true/1162 : FAILURE in 46m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/868486/7/check/DellEMC_Unity_Manila_dhss_false/ac31bea/DellEMC_Unity_Manila_dhss_false/1261 : SUCCESS in 1h 12m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"762cc4d7533dfb27d91c5f8284ffb50e8c793b82","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-01 12:40:49.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-7/ns5_manila : SUCCESS in 2h 21m 28s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"b5c617913e26a15f2b17338b9dcf69d3798b562f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 14:40:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f84d201502b44c19689e85068ff78d2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c7d5f975b2f3423a9aa8771d1ac7f877 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69a3047e26d481191050ae575f75dc8 : SUCCESS in 10m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1d6eadc4c34d2cbdfc7c26afbd8dd7 : FAILURE in 30m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe427dcbf484a7c8a35ad50ef2a2c4f : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1d70f00dc747b28ea73fda6edd6796 : SUCCESS in 16m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bf8e219b1d54c1da2be4aebe66612bf : SUCCESS in 17m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7336af6d3b224885942f6cc37942cd7d : FAILURE in 3m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0832cd91af4a45b7acabe46773160e81 : SUCCESS in 1h 13m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a2e114a89bd940648f263e922cec66e2 : FAILURE in 2h 55m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a2f736410b8d4f859feb8895d9f1702f : SUCCESS in 1h 16m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0aeed4c5be8c4ecea5a5b0c41d8cb2bb : SUCCESS in 1h 42m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6f4eebb612464149a1865beb4ec28075 : SUCCESS in 1h 32m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3a69e7c45c11477e9c99f0a473af5e4c : FAILURE in 1h 34m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e1f2f4dfd3843b4bc353532c272891e : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a1fc2c5fee0c401a8e03b7cf59096356 : FAILURE in 5m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/57af1acc01de4a0abe516dad6c2347c1 : FAILURE in 35m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/13674b91c3884067a4cf5d3c57ef06f7 : FAILURE in 41m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/982db705e23448479d0be88124d38ffd : SUCCESS in 48m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e34b64883bad46cd848b261b3da479c7 : SUCCESS in 40m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a6aa010b0f6142c3a254b2b6fcf535f6 : SUCCESS in 28m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d3d8cc4fd1dd4fd183943fba73f857fa : FAILURE in 1h 05m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e278dc54991c43e1846db3b4eaa0884b : SUCCESS in 1h 07m 11s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8c27f0f2848940488608f3cc887886f4 : FAILURE in 1h 06m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/41a4b604c0db4f7f879d012c6e3138d5 : SUCCESS in 38m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/465a9f5d5e80404bad64c5da15793e74 : SUCCESS in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"12a0aef663532e391bf6b8067efb3feb1a891b41","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-01 14:55:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3ee15698a8f7493fe1106b1f09c758fb5b92420b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-01 17:14:29.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9596c77dc12944ec856c2f242d08d268\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb1fca55932e4f8f80b22c3461091142 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f6aacaefd7442796da09e52176cd98 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a611c294a4ca4b119e7c0eb37799b28a : FAILURE in 16m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72422d12dd414838915bbccc5988144d : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84b5f79546fd4f0c98cc209c8f6c5db7 : SUCCESS in 8m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21838a6935b94e08b69350f90194bf82 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5a619b5bceb4929a673f16d17f25a43 : FAILURE in 3m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/963f605d70324eeb81027214de85fe29 : SUCCESS in 1h 24m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5d4c735a95e342009d88eed3c9e05b35 : SUCCESS in 2h 13m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da20517fc6a540389a7215bcf34fe7f4 : SUCCESS in 57m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/de7dc22a33b540f78da0a075d7ec4ee2 : SUCCESS in 1h 13m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/618785b06a624e59b0b7f4c5e9a22845 : SUCCESS in 1h 50m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/91adbc88fc384a7eb665c7c8ae65818a : FAILURE in 1h 14m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bdf3c411be7b4922b0a9d86a711419fc : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0bbe65a679ef4d7385dd58f2f7c328ae : FAILURE in 5m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4dd02211d7d44fc8999ee4e96b871bf4 : FAILURE in 28m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2335b6b47c0e4ccb92042c5340e8bda4 : FAILURE in 28m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ea0fb726d37c4b9c80d06952ae35336f : SUCCESS in 56m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/527f82878576498b9154ab991372583c : SUCCESS in 29m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8430b2a0396a4a17bbf83009dae49765 : SUCCESS in 25m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ca8557c91258497db264c54d0a19d657 : FAILURE in 1h 08m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fe89acefeac1456daf1318eabecb90f1 : SUCCESS in 51m 00s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f5288326c04642018e747cf461525be8 : SUCCESS in 1h 19m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d02acd97a97e433b85291cb78c7ea790 : SUCCESS in 43m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/96269fa8f6c647d689a56a8fb1d26aef : SUCCESS in 44m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6bb7bbe0dd659b80bf6b71253e58415ae78eec02","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-02 01:13:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"58eb65c8db1e69d24f77a4f18ab6aa3489c959f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-08 12:32:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f2bd5cbfc628716c994cd1dfcef20f9267ff9976","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-08 13:41:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/8/thirdparty-check/pure-devstack-manila-tempest-aio/d3767f9/ : SUCCESS in 1h 07m 38s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/8/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/83e3213/ : SUCCESS in 1h 03m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a837e06333d96a3dab221cff3ccad1d59b3c807a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 14:28:58.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/868486/8/check/DellEMC_Unity_Manila_dhss_true/5a02584/DellEMC_Unity_Manila_dhss_true/1203 : FAILURE in 1h 03m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/868486/8/check/DellEMC_Unity_Manila_dhss_false/583a785/DellEMC_Unity_Manila_dhss_false/1302 : SUCCESS in 56m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d6cbcd20c5185ce562401906c9089fe8f499e299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 15:44:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e10f95f41ba8415d88a1210c218d2f12\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c6cd1513e05e450bb0756f9b70635cb2 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f2ad3b05534b99bd4dbe0737327bf3 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24d5b47cff1541f4b539e4687f53df91 : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a7f76998fb425580138fe16094a47b : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e06508cfe8cc4a63b808962c10ba3433 : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b77506c84b0c4cdab8bddf5f67aeed41 : SUCCESS in 15m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aff7c019903d480588c400ae75944b10 : SUCCESS in 5m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b727f09f7b824797bf3e10f306f8e94d : SUCCESS in 1h 24m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aa983a5e2ec044cf8ef9f1e532977141 : SUCCESS in 2h 43m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb4a0f93cd1e4855ade1a3a9cc9745c8 : TIMED_OUT in 3h 04m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f68cdddad0a144e190b70dd358dd18df : SUCCESS in 1h 08m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ea04854206a948c2913f68d9cfebc8fc : SUCCESS in 1h 28m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0127a54fcdbe4c26bfbb1afefe248cda : FAILURE in 56m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/93d1900eca284e7bbe639c23e25e8216 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/169ac76bba914a7ca30de46f76dfb4af : FAILURE in 7m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c98fc7ae582e40b1b89169628e6d7d68 : FAILURE in 26m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/85018ae19a554a24b11604f2146d2628 : FAILURE in 35m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c007b1d248744b4aabdd6a21ce7d1696 : SUCCESS in 49m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fa536ab7941a4eea87d70a3a9332fd2d : SUCCESS in 33m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0affbb2523f047c689e41f941c34e1e3 : SUCCESS in 19m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f5bbae185af741d09e9a38d12fe2fc54 : FAILURE in 1h 10m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/036de56a525c4e71ad8204fdf5ffe721 : SUCCESS in 1h 04m 52s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/413113d723324b26be2217f60d915d8b : SUCCESS in 54m 58s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e81270edb7e14c858c7825988dcef028 : SUCCESS in 34m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/44ebb7e259104ac4814a18339cb6a6c3 : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eb7e14b89bd977e1f960e0a8a925a64fe048606b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-08 16:07:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7e204171da0df014d03221521194d3fb0e5f685d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-08 17:15:44.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/9/thirdparty-check/pure-devstack-manila-tempest-aio/c936595/ : FAILURE in 1h 05m 38s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/9/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/576ed89/ : FAILURE in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ea8d510b0f0057f4180e4d767f1191f8a30a55af","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 17:18:17.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/868486/9/check/DellEMC_Unity_Manila_dhss_true/260e635/DellEMC_Unity_Manila_dhss_true/1205 : FAILURE in 1h 10m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/868486/9/check/DellEMC_Unity_Manila_dhss_false/af78421/DellEMC_Unity_Manila_dhss_false/1304 : SUCCESS in 57m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"c482c8f25bb2c30e17567387d3a9a792d94cb1a6","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-08 18:53:13.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-8/ns5_manila : FAILURE in 6h 20m 29s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"6c55cd1a680c37f9d25829a28799f77570f7ca83","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-08 21:06:44.000000000","message":"Patch Set 9:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-86-868486-9/ns5_manila : FAILURE in 3h 24m 16s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"89762110ad1cde1841c0250b61f6612d80e5d975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 00:06:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e108a110023432e96ec36e31559965d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11a4fce5d3604347a253db910019a34e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7007922093b4f0e89f4ca2f15f9bf0d : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd85947e7c8454c929d57f3565aef4d : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc57bb2f5de404290670ef89b0a9938 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/149c3ada6c144b1f9b9f02f5e40c7f22 : FAILURE in 18m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bb209e7ddfb4ec1a3427519134ad7e6 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a26f80cebce47deab06c7c618ecb5a4 : SUCCESS in 5m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7923914d95e34a5ba561545c037403cb : FAILURE in 1h 07m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/749f02698fb64e27ab5f0051e7554e26 : FAILURE in 2h 37m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8756d724bd2842a7af92450a57c00ce7 : SUCCESS in 1h 34m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c71b2ddc4cd34165af7eed1b359b94e2 : SUCCESS in 1h 45m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9445bdb7d65144cd8a40144383c919ec : SUCCESS in 1h 33m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b390aab812d84910a90125ec38a73da3 : FAILURE in 1h 11m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2e5969ff4ce64534909831df1e42fda3 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d31ad939d01146119369ff8e801584bc : FAILURE in 5m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/467b49da339048aa93cb11d5c2c88e95 : SUCCESS in 26m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6c041dc1af5a420db585dfb6647692cd : SUCCESS in 1h 00m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3d250bedcdd406085c1eb093ef65c0a : SUCCESS in 48m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6819e66009f44e7c986006503c14b9f0 : SUCCESS in 30m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/83fe0bca89ba43f696833320aed6c92e : SUCCESS in 20m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4888138d3c3249a997388784486bfa2d : FAILURE in 1h 18m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/39c9f5c513414229af2fa7165f5d146a : FAILURE in 14m 44s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/98ade7d63a6a4f688cd7c3545b6f7525 : SUCCESS in 1h 04m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b23a9f38d3e14c59a145a0d146e7cb64 : SUCCESS in 35m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7b5cd4669410491db8068b32f5752e96 : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4d89add7809f7a176da45ec9bbbd442b21d33520","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-02-09 08:35:53.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d2a4e887e8174a323a62b1104d557ac782489e09","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-09 09:50:06.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/86/868486/9/thirdparty-check/pure-devstack-manila-tempest-aio/6527104/ : SUCCESS in 1h 01m 42s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/86/868486/9/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/b7740e7/ : SUCCESS in 1h 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ee54828372dbaeaf3d824ecc66f81bd92aa7e2bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-09 10:57:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9d9de2ddd148b594358f843d911a49\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b6908bc67ce24837a8d6b46191a8c21e : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300ab63307eb4297bfcdbb33bc8a6902 : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe42a80a5a34ed4b6c407cb5c1bb564 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b62b056004145e5bb25468e13d532c4 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77985e576639427fbb71d4a01de5c4f1 : SUCCESS in 10m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7de713d43e6a4d619f0ecb418bb4da43 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a262bbfa8ec42b29d281b4232580d19 : SUCCESS in 6m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9ab47195ea2f47e49d2f82e8c42abd4a : SUCCESS in 1h 11m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7096cac7e9ca41699302a76730e9ed4d : FAILURE in 2h 03m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7e577c913a50405ba27afaf4f33fb12a : SUCCESS in 1h 38m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2d2d9de5eccc490e80063a72c36cefea : SUCCESS in 1h 15m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/54af0040bf37403eac4f71780fb0973e : SUCCESS in 1h 32m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8996110d77064f0cb40541bf72d04411 : SUCCESS in 1h 15m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/51f2fc34f3e543179fe14ec5ae442c6d : SUCCESS in 4m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fbafc0f1a9ea45fba9ed40e884ba17b4 : FAILURE in 4m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eabb8153c80048438f6886463d8dbc6d : SUCCESS in 30m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d1ede79afd7427db55fca014e03734d : SUCCESS in 41m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/270539f145134a0e896b7432155a86ef : SUCCESS in 48m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8759e8527e8942c59ed2d822a7552843 : SUCCESS in 33m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/166660708d99421aa5837cfcf344b032 : SUCCESS in 19m 07s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/86120437231248dcac3e6511a8f54530 : FAILURE in 59m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f085d69648b9498d9f3d244a2e4ae306 : SUCCESS in 50m 07s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f5b56f3f72cc4189a99b7c69cf336846 : SUCCESS in 59m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/de6eca09688543a2924f5482a1763fa7 : SUCCESS in 38m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3f855165cbad496a85fb24a35f57902f : SUCCESS in 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7947d910f5f5f3938a9163374b78956b3814f248","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-02-13 12:31:49.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7c45982c6179d967c4b43d353b7adca97ea96102","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-14 01:51:25.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"feac0c18020dc893cd7d95272553c54f0f40428c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-14 02:14:56.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0911c9bb09f2659d3c6c8fa7405d1c0d0e2c1f05","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-15 00:13:37.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2e3c88d87aa645767982dac8b565d6e61fee14ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 00:14:11.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ef80c03a1c2321e8a26a696b30a5fef7e711052c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 01:45:46.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c928c1556a8d4eb8bfdb2eea24795b51\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/de94fa4bcbec47c79bc05e4e8dce7eba : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a6a69001aa4eeeaaefd5e54f85d93f : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d695e1f6bbdc46ea87f1ee8d4cac0477 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f572bc1738147febc79846dde500329 : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a27ebfcc5844467dafff9749804a5918 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cd89926aa2940bfb4a186677132f55f : SUCCESS in 5m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8fac4cc77624f379670cf8d880aeccc : SUCCESS in 1h 26m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cbd7ba8670f24e7883599474e5edc552 : SUCCESS in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c32179833d5648ad96c5aad9bd27553e : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":9},{"id":"33face4d009c5baf2baff81de719ffe2d7c490a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 01:45:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"8d0bab3c948a0646475a9d8b198f5eb1ed7b654c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 01:46:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4797c6fa32734e47886c09dc5ec8b3b7\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/d76e8bf00b434b66ba4ddf547bbed04d : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e57e2ddd6e4b18b73bae67cb3949a0 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5290999eb55423898f9fc74b5f73f22 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5f0f14a7f8b99ac008be4604bcaeec2dbc149d06","revisions":{"4a91fdc8b1b903ae9eeb648f3737b8f888270724":{"kind":"REWORK","_number":1,"created":"2022-12-23 13:20:45.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/1"}}},"commit":{"parents":[{"commit":"4408ec4f7cfac834c454537388fa23a13915e3f9","subject":"Merge \"[Infinidat] fixed host assisted migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4408ec4f7cfac834c454537388fa23a13915e3f9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 13:20:38.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a91fdc8b1b903ae9eeb648f3737b8f888270724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a91fdc8b1b903ae9eeb648f3737b8f888270724"}]},"branch":"refs/heads/master"},"7b72342bb316198f848c7bfe3601de7946e5176a":{"kind":"REWORK","_number":2,"created":"2022-12-23 13:30:42.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/2"}}},"commit":{"parents":[{"commit":"4408ec4f7cfac834c454537388fa23a13915e3f9","subject":"Merge \"[Infinidat] fixed host assisted migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4408ec4f7cfac834c454537388fa23a13915e3f9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 13:27:52.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b72342bb316198f848c7bfe3601de7946e5176a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7b72342bb316198f848c7bfe3601de7946e5176a"}]},"branch":"refs/heads/master"},"8a90a37946e3019039c1c6136c0c211d66408529":{"kind":"REWORK","_number":3,"created":"2023-01-03 09:39:54.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/3"}}},"commit":{"parents":[{"commit":"4408ec4f7cfac834c454537388fa23a13915e3f9","subject":"Merge \"[Infinidat] fixed host assisted migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4408ec4f7cfac834c454537388fa23a13915e3f9"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-01-03 09:39:23.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8a90a37946e3019039c1c6136c0c211d66408529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8a90a37946e3019039c1c6136c0c211d66408529"}]},"branch":"refs/heads/master"},"9d55858aa675896c68a985d2b1e34a9951ec2286":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-01-12 15:17:25.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/4"}}},"commit":{"parents":[{"commit":"73848870a5811d42fbd04f6e43efa3c78e037b82","subject":"Merge \"db: Prepare \u0027model_query\u0027 for migration to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/73848870a5811d42fbd04f6e43efa3c78e037b82"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-01-11 08:05:08.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9d55858aa675896c68a985d2b1e34a9951ec2286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9d55858aa675896c68a985d2b1e34a9951ec2286"}]},"branch":"refs/heads/master"},"dcaaf1bb60d093113bfce2d852510965a276355d":{"kind":"REWORK","_number":5,"created":"2023-01-18 10:11:56.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/5"}}},"commit":{"parents":[{"commit":"a76bdac62d263e644e2cd1f13095071e2aa75eae","subject":"Merge \"Add config for netapp asynchronous SnapMirror schedule\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a76bdac62d263e644e2cd1f13095071e2aa75eae"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-01-18 10:03:40.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dcaaf1bb60d093113bfce2d852510965a276355d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dcaaf1bb60d093113bfce2d852510965a276355d"}]},"branch":"refs/heads/master"},"c34bd816e17cda74ea9cfc4651ba860ec2f5adea":{"kind":"REWORK","_number":6,"created":"2023-02-01 09:56:18.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/6"}}},"commit":{"parents":[{"commit":"1d4beed95b1384ae0615ab735706e989c46ebddd","subject":"db: Migrate \"backend info\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d4beed95b1384ae0615ab735706e989c46ebddd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-01 09:53:13.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c34bd816e17cda74ea9cfc4651ba860ec2f5adea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c34bd816e17cda74ea9cfc4651ba860ec2f5adea"}]},"branch":"refs/heads/master"},"e5ee2f7fabeead6e2c401b97fce1cbf28310e7ea":{"kind":"REWORK","_number":7,"created":"2023-02-01 09:58:57.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/7"}}},"commit":{"parents":[{"commit":"1d4beed95b1384ae0615ab735706e989c46ebddd","subject":"db: Migrate \"backend info\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d4beed95b1384ae0615ab735706e989c46ebddd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-01 09:58:13.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e5ee2f7fabeead6e2c401b97fce1cbf28310e7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e5ee2f7fabeead6e2c401b97fce1cbf28310e7ea"}]},"branch":"refs/heads/master"},"b64d7e29fd65b15f14d1f5a29897940a6f7d306f":{"kind":"REWORK","_number":8,"created":"2023-02-08 12:32:14.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/8"}}},"commit":{"parents":[{"commit":"1d4beed95b1384ae0615ab735706e989c46ebddd","subject":"db: Migrate \"backend info\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d4beed95b1384ae0615ab735706e989c46ebddd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-08 12:32:02.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b64d7e29fd65b15f14d1f5a29897940a6f7d306f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b64d7e29fd65b15f14d1f5a29897940a6f7d306f"}]},"branch":"refs/heads/master"},"5f0f14a7f8b99ac008be4604bcaeec2dbc149d06":{"kind":"REWORK","_number":9,"created":"2023-02-08 16:07:26.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/86/868486/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/86/868486/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/86/868486/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/86/868486/9"}}},"commit":{"parents":[{"commit":"1d4beed95b1384ae0615ab735706e989c46ebddd","subject":"db: Migrate \"backend info\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1d4beed95b1384ae0615ab735706e989c46ebddd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-12-23 12:41:37.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-02-08 16:05:28.000000000","tz":0},"subject":"Add quiesce_wait_time option to share replica promote API.","message":"Add quiesce_wait_time option to share replica promote API.\n\nCurrently netapp_snapmirror_quiesce_timeout is an option of replica\npromote that can only be set by the operator. There are scenarios where\nhigh timeout value set by operator does not fit well e.g. disaster\nstroke and user want to do an unplanned failover fast. Added new\noption \u0027quiesce_wait_time\u0027 to share replica promote API which allows\nto use specified wait_time(in seconds) instead of config option.\n\nCloses-bug: #2000171\nChange-Id: Ib02063ee8b82f7374cd89f90e7f24a845c6c7cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f0f14a7f8b99ac008be4604bcaeec2dbc149d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f0f14a7f8b99ac008be4604bcaeec2dbc149d06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
