)]}'
{"id":"openstack%2Fmanila~873087","triplet_id":"openstack%2Fmanila~master~Ib68f71e19f2ec1190cc431f029d996193008d557","project":"openstack/manila","branch":"master","topic":"bug/2007060","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-02-22 14:45:25.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-02-23 06:59:10.000000000","reason":"Change was submitted"},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-02-23 06:59:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib68f71e19f2ec1190cc431f029d996193008d557","subject":"Fix Manila API error message","status":"MERGED","created":"2023-02-13 02:52:07.000000000","updated":"2023-02-23 23:36:21.000000000","submitted":"2023-02-23 06:59:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873087-bug/2007060","meta_rev_id":"79e9f5424bbf1225e5a8e94defe9d4ff42069d1e","_number":873087,"virtual_id_number":873087,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-23 06:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"recommended":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"all":[{"value":1,"date":"2023-02-21 07:55:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2023-02-22 14:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-02-20 22:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2023-02-22 14:45:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 04:25:14.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-02-13 05:47:17.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 12:40:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-13 13:32:03.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-02-13 14:13:49.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-20 22:21:57.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-20 22:22:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-02-20 22:22:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2023-02-21 07:55:50.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"}],"messages":[{"id":"c389d061b97a6c608ede7a9488e34fae05e04f59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-13 02:52:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b756aff95bb74cfcc176024355d41c2ec55029a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-02-13 04:25:14.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/87/873087/1/thirdparty-check/pure-devstack-manila-tempest-aio/0652bbf/ : SUCCESS in 1h 12m 25s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/87/873087/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/2c81046/ : SUCCESS in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3778a37e49f0b57f9a1275b593d93f078451493","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-02-13 05:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-87-873087-1/ns5_manila : SUCCESS in 2h 13m 56s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"f05994c63822985531e00a40ff8d64326ac7ade9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 12:40:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4abd0ab8c454fd698f103d22d99dd08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11997e6cfd504429994409819beef94d : SUCCESS in 10m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d30cb3f904014c11a936b639666e15ee : SUCCESS in 20m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9a24a1cee44ab7a6c7580fcbefb9aa : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8200ec5fb7d14028a08a308c38e3d797 : SUCCESS in 9m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fcb3ef2be804878a73935931552341b : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1aec07ba014a8cb491437582546179 : SUCCESS in 5m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/530836d4ba064e27964aff2cf684f185 : SUCCESS in 1h 27m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aca06759091e4406acfd6dd395da4816 : TIMED_OUT in 3h 12m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c53af092002a4dafa1e5d16feab1f27b : SUCCESS in 1h 11m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e9fbedec99a04c3cb593df4514e0994c : SUCCESS in 1h 37m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/037196c234474cdd886e5692e9e84cb3 : RETRY_LIMIT in 3h 01m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/74a3370155d54eee959e40a7c9c28cc2 : FAILURE in 1h 18m 45s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0b2639d0618a4f65b65ebbc56cdff2e1 : SUCCESS in 4m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a5499757af2e49ed91b0dbf7be263855 : SUCCESS in 4m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/763a6de7059548dc92f98175bd5402c6 : SUCCESS in 39m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbdc41499e8f4ab2b19ec8851bddaecb : SUCCESS in 33m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/37ceb3754e744407bfd24c756e559404 : SUCCESS in 43m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/94e1de7f1f684ca4b575bf081ae61fcb : SUCCESS in 34m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/151e51379e644248a032407a043d615b : SUCCESS in 27m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/231e38c77c594e6aaa428ea52dd4c906 : FAILURE in 1h 03m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ec3f8bf3e7a94b03b05b07692cb4b4a6 : SUCCESS in 51m 41s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9928a313217442ebeae9934be7ab9d1 : SUCCESS in 1h 02m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/646ad17e5144470497e96ad15f7f99a9 : SUCCESS in 36m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/27c6822da67a43119e0e7c157aca58b5 : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfe16c2d734c76b598d3a451c3e6a82c07d2409e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-13 13:32:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/87/873087/1/check/DellEMC_Unity_Manila_dhss_true/aa24f56/DellEMC_Unity_Manila_dhss_true/1220 : FAILURE in 50m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/87/873087/1/check/DellEMC_Unity_Manila_dhss_false/4807cb7/DellEMC_Unity_Manila_dhss_false/1319 : SUCCESS in 50m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ef38c917d4621a23cfe1d6cc26ca1fc7f7ba12d6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-13 14:13:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/873087/1/upstream-check/manila-tempest-plugin-ontap-dhss/de239df/ : SUCCESS in 1h 31m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/87/873087/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/c00849b/ : SUCCESS in 1h 14m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"82714f12b709967012d07ab852e2294e567a0ebc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-20 22:21:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e648c4a7562217901431827b9d460ad4ce61053","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-02-21 07:55:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8cda10772c9474b03bee385e319bed8cb991f436","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-22 14:45:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d43d2b6ab5d230500f8bc88c1731efafd3de6f06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 14:45:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"910c178396b84e0ac0f074549b5fe478bfcebf18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 15:56:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46cb10a38e943259f07bb61c35f6322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6fa90bed774509bda94c280da6ab77 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8663ae9552e64421b3409387e5c9292d : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60fe2e1eba72479b9b511c2be3fab564 : SUCCESS in 11m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dceefe8ea2ae41fe99d329022def1d57 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7498790b61fc4a91a5bd56280a977079 : SUCCESS in 6m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3f15df9648934171b7f2e8628b7701fd : SUCCESS in 1h 07m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a96974f36ac44553b436bd33a5638e84 : SUCCESS in 33m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad854bb9b8a84548a848efb765ab34c3 : FAILURE in 47m 15s","accounts_in_message":[],"_revision_number":1},{"id":"50ba6c5201b855989d400647b65926cc37f12d9a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-22 15:58:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32632b70843836e4e38318801aa6fcb121b4cd18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 18:36:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcde4e3f19ce4f6e87940f750b05f22a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab393ea32aaf43a684380d941ca80dda : SUCCESS in 8m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/644358e96aaa4305ae17aa82e26adde6 : SUCCESS in 19m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4657f891e9054b59a9340b337b47d5a6 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/167c1fb083c64cf5ace4b93495380cc2 : SUCCESS in 9m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46dcabe8376a4e6bac4839b1eed4b5f0 : FAILURE in 18m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6471b0cbe73458aa178cf2331126953 : SUCCESS in 5m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/94a569e1c3814cd495a092a730fee811 : SUCCESS in 1h 24m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/900ad57b17404451bfb9465830b71f1b : SUCCESS in 2h 22m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/340c1762101642199bb5edd6ff6ab02b : SUCCESS in 55m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/09a7e7b9d4d3483da3684fd80da7c970 : SUCCESS in 1h 34m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9aa66900f2ec415a9695104380bd29eb : FAILURE in 2h 32m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/347308c518474439acbe1fe4b7f7c2ff : SUCCESS in 1h 08m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/daef0aa2cb9d4d27aee1a8b3586303d8 : SUCCESS in 5m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c00be4cc692748b49c9f80034010a98c : SUCCESS in 5m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9091d9740e34422497cc1e820cbef5fc : SUCCESS in 40m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f13298ba2cbf4f2096185b39f885e57b : SUCCESS in 44m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dc6b3bf572424f848329a89af4b568c3 : SUCCESS in 47m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/cf7b50bf5a1a461b97029e2195a77e8b : SUCCESS in 28m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/23394377be434c2d96bd825606590e7d : SUCCESS in 29m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d826079ec5da43d2a01c5b1fb3d7846d : FAILURE in 1h 52m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0084634857414a0f8974316e5a916631 : SUCCESS in 1h 01m 03s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5f42b631ffdd43edb28c37ff6f4e811d : SUCCESS in 59m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f4174615040147c9970e5cc5600058ae : SUCCESS in 36m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a6c4078c419347559e4474a73883cc7d : SUCCESS in 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c7453faeda433025242f8b724dcdc892720e8f4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-22 19:44:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45b684c15e1a0231cdace0bd963d30a36996a08a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 22:56:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af9deef1f40c4f0082b16e4eef175a43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe5cac022f04619832bb06177a37828 : SUCCESS in 8m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa2121681ed4bdfb86d17180323f221 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d90d50eaf7404688fc7f86ba77238e : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00962d6f4dd44579a61b1fddcecf7e24 : SUCCESS in 10m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e77422bcc22b406a892d3df95b874e10 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/698a82e3763244bdb53c0d7fce520ac3 : SUCCESS in 7m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dcb3b3b224b940eaae5aaea20d2e08a5 : SUCCESS in 56m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e6c48bf29f1b4ef2ae0e1297ea8e7d49 : TIMED_OUT in 3h 02m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/73ade13b968b481fb7d682074dff6e1a : SUCCESS in 59m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bace9d639e174c0691ddece8be7bbdce : SUCCESS in 1h 13m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/250e119057194c478bb791f6c96c002c : SUCCESS in 1h 23m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/af8fe2a22bed4f588231f169aafe89cd : SUCCESS in 58m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9834be0c1792449ba450c78dc61cfae8 : SUCCESS in 4m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f2911f224cec443f9514d94b9e2ea1e6 : SUCCESS in 5m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6d1cf011c6ef4f6bafa45532e07d6ab4 : SUCCESS in 23m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2052bee60b6344268ddbb64d84ee5c4c : FAILURE in 39m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0f599c4da52f42c18406b5821708bbcc : SUCCESS in 46m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6fa2fe7aab3143fb81a41e84ebde7996 : SUCCESS in 34m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7cf8b6ac39664920a0a15dc5e05d3f8a : SUCCESS in 20m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/cd4400f637324a54862258daf6dd2eb5 : FAILURE in 1h 18m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1cea0731aff64e0ca9c7df2bf80ed0be : SUCCESS in 48m 46s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e33c9c55d6124b94b010c4626ef5a2d4 : SUCCESS in 55m 17s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/81cd0bb99a474816b057e29cc8a18a42 : SUCCESS in 39m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f4a5a52d87734e7093c94e69008dfe44 : SUCCESS in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8291c99f165241bb2419626a545ebfa0863b081c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-22 23:48:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36a18c80c706bc22fa93abd4e68ea37dcd46b432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b648629ba142465d951e50abf9744e72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0036981b5c7d478f99503f16ba5cd779 : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/336786a891bb40de817224ba8c52d8a2 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bd297b41e14d068a7e0c8be577df1b : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c341f07dc8c64c3ebc6afb655f6cb029 : SUCCESS in 13m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11aad5fc32d141e1b2bc024fa795cbc5 : SUCCESS in 15m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/417d9aedd53b4def9aca57106d092c3d : SUCCESS in 6m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7b10e884af5346c5a505bd11d8f6687f : SUCCESS in 1h 18m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2042c989efcf4abdb210fffd880aafab : SUCCESS in 2h 02m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b3a3647ed8ea487a93d0d8ffbd069126 : FAILURE in 2h 39m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ee1dfa668e6a4122bc84ee42cb26ea5d : SUCCESS in 1h 26m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ba4ac8884d1642f4b624fc99a0054005 : FAILURE in 2h 56m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2de77f746643489bb6bc491fdb17b255 : SUCCESS in 1h 09m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3c6565c820ec431281b83c2add6840a6 : SUCCESS in 5m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2d50de35f3cd4e31a36d4638ce5b4409 : SUCCESS in 3m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e562a55ef6754e6a95580f0d011ab73c : SUCCESS in 33m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a4d76dbbf5df4b40b4b2fdd7e974d40c : SUCCESS in 49m 37s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1227a36f2c5242e7b0f56c0c0ccbd5b6 : SUCCESS in 1h 02m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/acd53672f8cf4856a1935e24f5f10a48 : SUCCESS in 30m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/11dbadf6c2b3477c989859e1a0f52c9e : SUCCESS in 33m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6b2ee608ac504916b933da7129f63582 : FAILURE in 1h 11m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/18e57a5d400c42e1b5a0dcbc21a5cb2c : SUCCESS in 59m 06s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d8ff5a7e26294f00a160e2cc6101bf3d : SUCCESS in 57m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f84dd9a2cdee4b5182a1110a26e2da60 : SUCCESS in 32m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c442df8439994c68904f02f761f6babf : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"468457ab260bbae11a12f5d322e6412af7755554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:01:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"654c6908285e5861190cf9776d5edc86177c9b6c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-23 06:52:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/87/873087/1/check/DellEMC_Unity_Manila_dhss_true/d20b364/DellEMC_Unity_Manila_dhss_true/1304 : FAILURE in 51m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/87/873087/1/check/DellEMC_Unity_Manila_dhss_false/e350456/DellEMC_Unity_Manila_dhss_false/1404 : FAILURE in 40m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a5affe593eb0b092f100143582bf568684f79030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:59:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00341df4bd3a4bbe8dc0416014c41ef4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/297d95b4fb6846dcaf07be65984ab129 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647371e5cfb44c0f9b9de612179465f6 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1af08b6d91e48928e8abd1be5a1e6d1 : SUCCESS in 10m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03817cd666424dcebd44d4b3f3cc4fce : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04d2415800e84f4bb6192f286252e5b2 : SUCCESS in 7m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cf6b80f4bc2c4e5a98614184f7d450b1 : SUCCESS in 56m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e2df02a25c0e4c8ca8e859e620ba7f47 : SUCCESS in 37m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d7570c337624435d8f88dbbbbba95800 : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":1},{"id":"94e2e527d473c39d843b133c6b273a7384c8bbf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 06:59:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb30885a6a02216b2c1307d4501d154ae12d4a72","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 07:00:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f67c8c6b3149b5894358522830abfb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05be43244f954d299725356214c6e2de : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4351e3c9aa4d4e989f119292f249440a : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1},{"id":"79e9f5424bbf1225e5a8e94defe9d4ff42069d1e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-23 23:36:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a65dffc0c7c6699d2e8ebbee7521913ff0c02917","revisions":{"a65dffc0c7c6699d2e8ebbee7521913ff0c02917":{"kind":"REWORK","_number":1,"created":"2023-02-13 02:52:07.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/87/873087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/87/873087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/873087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/873087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/87/873087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/87/873087/1"}}},"commit":{"parents":[{"commit":"f6f93ff9516f0ae24801d7b422c0e504bfebd0f5","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f6f93ff9516f0ae24801d7b422c0e504bfebd0f5"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-02-13 02:49:00.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-02-13 02:49:00.000000000","tz":-180},"subject":"Fix Manila API error message","message":"Fix Manila API error message\n\nSome Manila error messages were wrongly returning the string marker\n\"%s\" instead of the string content. This patch fixes them.\n\nCloses-Bug: #2007060\nChange-Id: Ib68f71e19f2ec1190cc431f029d996193008d557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a65dffc0c7c6699d2e8ebbee7521913ff0c02917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a65dffc0c7c6699d2e8ebbee7521913ff0c02917"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
